)]}'
{"id":"openstack%2Fneutron~968232","triplet_id":"openstack%2Fneutron~stable%2F2025.2~I662b3eb7fb8c17d0ad30b10e8537d4a142003256","project":"openstack/neutron","branch":"stable/2025.2","attention_set":{},"removed_from_attention_set":{"6577":{"account":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"last_update":"2025-11-25 11:18:37.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I662b3eb7fb8c17d0ad30b10e8537d4a142003256","subject":"Replace response body in after hook for 404 error","status":"MERGED","created":"2025-11-24 17:29:30.000000000","updated":"2025-11-25 11:19:44.000000000","submitted":"2025-11-25 11:18:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"968232","cherry_pick_of_change":966498,"cherry_pick_of_patch_set":6,"meta_rev_id":"e3df0f490cfc986e7d83ead558a9cddcfe2c36cc","_number":968232,"virtual_id_number":968232,"owner":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-25 11:18:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2025-11-25 09:17:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2025-11-24 21:05:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2025-11-25 09:17:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-24 18:14:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-11-24 20:07:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-24 21:05:00.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2025-11-25 09:17:39.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"ec545c2f7a4e3fc67c84dd11d3098493f27d9980","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2025-11-24 17:29:30.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"a2ba1f8f97cc8bd5211699224bbeb65983adca5c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-24 18:14:09.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0257d29b92c84d0dabc708aba8f0cc22\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/394f041a1aa04aeda5fa9f5d2eddf46e : SUCCESS in 36m 24s\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/41232748acc6452589244ac1fe8a87a2 : NODE_FAILURE Node(set) request 60c981e74d0643c29b00baa6c79fd3ab failed in 0s\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/54f6fffc338f4ababb6d8852294dc46e : NODE_FAILURE Node(set) request a1f4a4ff9cb14d2eaadde94ca313c8cd failed in 0s","accounts_in_message":[],"_revision_number":1},{"id":"71cbf189906bc443bac81113e91ac505baa6f706","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-24 20:07:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/44b92f85e8b94fe9a30eedf4ca374500\n\n- neutron-tempest-plugin-openvswitch-2025-2 https://zuul.opendev.org/t/openstack/build/69478d973b0a4d639bde10180db775ca : SUCCESS in 1h 35m 00s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-2 https://zuul.opendev.org/t/openstack/build/98549f381d894f89b9307982373daca1 : SUCCESS in 1h 57m 23s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/3e226d2aa2374a368c3a66d6ffb35986 : SUCCESS in 2h 34m 05s\n- neutron-tempest-plugin-designate-scenario-2025-2 https://zuul.opendev.org/t/openstack/build/ebc4545d8a72442a95b7c50885d78c3a : SUCCESS in 31m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/453973ab2a4f4219896a1548a6d5c60e : SUCCESS in 32m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e170ddfbb68f42329d652eda09f5e8af : SUCCESS in 13m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/07247883689d477eb4bc7d14bb47536e : SUCCESS in 21m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/84331c499ec846cf86a97929919c9a50 : SUCCESS in 21m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/53733eba764f4a7eaeaf9c104e026b6e : SUCCESS in 23m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/359b89509e5f46739f5c9fd673d545fc : SUCCESS in 11m 20s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d3331db43f5e4a1f999c3ada6f72cfd8 : SUCCESS in 1h 45m 34s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/34f1eb9aabd44f5f9f1e29521715f991 : SUCCESS in 32m 51s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/6404838e16aa4b04b56f0c669368442b : SUCCESS in 1h 19m 39s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/4998a8bfbca84922a2625b264b6074ce : SUCCESS in 1h 54m 56s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4911ff934875422f9b83ea21d778837d : SUCCESS in 1h 05m 26s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/53a964a036d14dcdbe289c558f8d9d3b : SUCCESS in 1h 00m 12s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/a4ae1632d60343afafe966034fd1d12a : SUCCESS in 27m 14s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/74e564ee03b8493886a2a480401ae6de : FAILURE in 42m 31s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3a9a5a67653d4a2ea8a6f2c4395c25c9 : SUCCESS in 1h 36m 07s","accounts_in_message":[],"_revision_number":1},{"id":"76bab6497c0bb18cb644fdc1b30869d97d06b0f5","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-11-24 21:05:00.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"55c135715035c528545fa2896f0323e3f19b270b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-24 23:10:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/367326638e594d8a99c144a8d75ad4e7\n\n- neutron-tempest-plugin-openvswitch-2025-2 https://zuul.opendev.org/t/openstack/build/61fe90941fa94657a7786839839e691e : SUCCESS in 1h 28m 46s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-2 https://zuul.opendev.org/t/openstack/build/5d2d216d9a9f4284a53564c79d2c1cc4 : SUCCESS in 1h 39m 13s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/a8b44dc30f61471c99b5875b0ab3373a : SUCCESS in 2h 03m 43s\n- neutron-tempest-plugin-designate-scenario-2025-2 https://zuul.opendev.org/t/openstack/build/c3e7269ed3504492a777a2b1ed300295 : SUCCESS in 34m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b93c4db8bdc9464e929c5ab0b351387b : SUCCESS in 44m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e295185dbec94a0b85e033da07a7e4f3 : SUCCESS in 12m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f83fd436ba5947bdb320bc82a47d6f4a : SUCCESS in 23m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cf206dc513a74359abf3012b14197fc4 : SUCCESS in 21m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9c68a77257534de78b78abfa8211bbbb : SUCCESS in 28m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8cb20b4016aa4e048cb5aff1324ee204 : SUCCESS in 7m 48s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/6b1cd94d0fd647af95a076a824bcf848 : SUCCESS in 1h 36m 01s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c56b2c97b4444884b45d1b8d411d7506 : SUCCESS in 1h 06m 02s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/d6e702542279426c8f5b9d07592142cc : SUCCESS in 1h 09m 22s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/f93c44169aef45fdb583db02a4f9e972 : SUCCESS in 1h 22m 02s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/056d06f74b52455ab534fc08a8996a57 : SUCCESS in 1h 34m 43s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/0588d50410054db1b04c587d2d736a03 : SUCCESS in 1h 03m 45s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/0a5cac748ffa40fbae7af52c0ca8f80b : SUCCESS in 59m 24s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/12862bb230d248cbaadd2f61d08b861a : SUCCESS in 50m 50s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/2df17a92e7da4a718297e70d0723d05f : SUCCESS in 1h 17m 19s","accounts_in_message":[],"_revision_number":1},{"id":"ba44a1f98c6b60f71d99849c839fb416f17b53ec","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-24 23:14:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c791fe75073484b92ee1a49710fdec4\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/69155e0aca2442d49194441dc231c601 : SUCCESS in 46m 56s\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1edfa1215bd642519888ee69cccc200a : SUCCESS in 41m 59s\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/67d9cc99f7f9469895587bcf7db92e3e : SUCCESS in 27m 18s","accounts_in_message":[],"_revision_number":1},{"id":"471ae0afc46177acedfc69fbce14010bf23a0a4f","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-11-25 09:17:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d8b16cc6c49c98a58debf98285072cef53da4d04","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 09:18:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"aaf6a95a03efa09f4060b3092e48649e44cad5ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 11:18:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f05a94e36a914d62ab7ae7ed9f03592f\n\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/0c67355eca1a4eb8bbd4ef0c45844477 : SUCCESS in 2h 00m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbc092d5f89e43c6aba0687d1ab02d2e : SUCCESS in 7m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/02b8a802d1014e9ca0451746361445dd : SUCCESS in 21m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4a840b1e2e4f4f5e907394dadd59f83d : SUCCESS in 19m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72145d1d37df4daea8cd85231f5e64db : SUCCESS in 12m 14s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a547fa8c60a5441a98c16549f7489468 : SUCCESS in 1h 50m 22s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/5fa1334e580b4baf8ba4672793355f40 : SUCCESS in 1h 52m 06s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/cda6d6a9574f4cc381d4088414140383 : SUCCESS in 1h 12m 34s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/47e7ae46be0949ada197c8042e108f63 : SUCCESS in 58m 57s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/16c24471bff743858f3e7144f2749f3a : SUCCESS in 48m 28s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e4c1904eafd045a4a5cc3c74af5eb0c8 : SUCCESS in 1h 06m 29s","accounts_in_message":[],"_revision_number":1},{"id":"6ba90f59498bae86985bd39c2ebb9f5e83cc089b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 11:18:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e3df0f490cfc986e7d83ead558a9cddcfe2c36cc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 11:19:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01aa525ddd4349b3b1a216f7572cee09\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85e9758ff4244cfb99fe90204628369f : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"274f97f54b8cafd57b817a35b7dc0eceb058b64a","revisions":{"274f97f54b8cafd57b817a35b7dc0eceb058b64a":{"kind":"REWORK","_number":1,"created":"2025-11-24 17:29:30.000000000","uploader":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"ref":"refs/changes/32/968232/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/32/968232/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/32/968232/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/32/968232/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/32/968232/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/32/968232/1"}}},"commit":{"parents":[{"commit":"483aea6b41d9706d6e95c0d5bedb6abe55875dad","subject":"[FT] Wait for the manager to be created","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/483aea6b41d9706d6e95c0d5bedb6abe55875dad"}]}],"author":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2025-11-08 17:19:17.000000000","tz":240},"committer":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2025-11-24 17:29:30.000000000","tz":0},"subject":"Replace response body in after hook for 404 error","message":"Replace response body in after hook for 404 error\n\nCloses-Bug: 2130972\n\nChange-Id: I662b3eb7fb8c17d0ad30b10e8537d4a142003256\nSigned-off-by: Sergey Kraynev \u003csergejyit@gmail.com\u003e\n(cherry picked from commit d95e3e5eaf24ee788d65255d669671e175f25208)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/274f97f54b8cafd57b817a35b7dc0eceb058b64a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/274f97f54b8cafd57b817a35b7dc0eceb058b64a"}]},"branch":"refs/heads/stable/2025.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
