)]}'
{"id":"openstack%2Fneutron~968233","triplet_id":"openstack%2Fneutron~stable%2F2025.1~I662b3eb7fb8c17d0ad30b10e8537d4a142003256","project":"openstack/neutron","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"6577":{"account":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"last_update":"2025-11-25 11:20:47.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I662b3eb7fb8c17d0ad30b10e8537d4a142003256","subject":"Replace response body in after hook for 404 error","status":"MERGED","created":"2025-11-24 17:29:56.000000000","updated":"2025-11-25 11:21:42.000000000","submitted":"2025-11-25 11:20:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"968233","cherry_pick_of_change":968232,"cherry_pick_of_patch_set":1,"meta_rev_id":"2e4ee5222a0cef516bdb0eff58fafe35a72aa1d0","_number":968233,"virtual_id_number":968233,"owner":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-25 11:20:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2025-11-25 09:17:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2025-11-25 04:17:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2025-11-25 09:17:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-24 17:55:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-11-24 19:22:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-25 04:17:26.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2025-11-25 09:17:51.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"c298a11dfbe899ad369e1e67998b27b0d6f8af40","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2025-11-24 17:29:56.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2025.2.","accounts_in_message":[],"_revision_number":1},{"id":"ff7f4ba545387b71f9aaa3bee1b5bc0dbfd6dd31","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-24 17:55:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43ca02640e77429e81effc2417972432\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/434caef6afa44c0ab662bdd0f1d04f8f : NODE_FAILURE Node(set) request dafa70b84564402ba0a536c8d3998d74 failed in 0s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ca38e26266644d82bc49d27abcff3cd6 : NODE_FAILURE Node(set) request e89a8e4d60fc4bcbab183f74a65381ba failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7a5e72755f3a34c2ca3a1876990482729726e88c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-24 19:22:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/941048431ad74f8793e7550666efa768\n\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/4e3ea950faae47268564b42f5d68ee8e : SUCCESS in 1h 35m 42s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/24cd5a4b0c7a4145a4aae1ecfea6f69c : SUCCESS in 1h 44m 30s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/d866db7b943a45baa6389850ee107d7a : SUCCESS in 1h 42m 03s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/2c6fbf66c621490fa3f54120125dcdcd : SUCCESS in 31m 45s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/c5ad63e92af6488cafd465353bd87928 : SUCCESS in 1h 13m 39s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/3fcb8f129fc3434084f2db824cc09c1b : SUCCESS in 1h 15m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/669ec8ab92854b5c9a1ffa60a4fdd7f9 : SUCCESS in 32m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8179665493c4897b147d0fa34d9d4c6 : SUCCESS in 11m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/82dec04297e64698a7cadc2515400e7e : SUCCESS in 19m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1a4f2881b0a94ba9b097e7438e441084 : SUCCESS in 19m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5525208a25048708ed6a55e084115ff : SUCCESS in 12m 29s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/f408136bbae240968b7938d80ae1ee31 : SUCCESS in 1h 04m 19s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/fedabc662ec447d997798e12319ee03b : SUCCESS in 1h 10m 38s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/9bf05568b7694c65ae471c1b5aec19ce : SUCCESS in 1h 29m 28s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/03150dc9774345bd80b44f508fe2db46 : SUCCESS in 1h 25m 26s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/3144210cb4e347f68b4091ff15d1c6d5 : SUCCESS in 1h 05m 05s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/46c9d46fdf4c48568d697ba4c17c63d7 : SUCCESS in 1h 08m 10s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/b239ae5e80934321a9f574423279ac87 : SUCCESS in 1h 38m 30s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/abe76eeddd804a22abc570ee8414b3fb : SUCCESS in 1h 28m 31s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/ecaf3062f3f3447eb7584fad289b10a9 : SUCCESS in 58m 21s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/c3aa216335284e4ab1ed93c3e1bbe5f4 : SUCCESS in 1h 04m 51s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a8a06b289d3741efbad20f0ed1071ccb : SUCCESS in 1h 15m 12s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/fc8e45adc4d8458793d3e48ac465f79f : SUCCESS in 1h 19m 31s","accounts_in_message":[],"_revision_number":1},{"id":"dd0fd710878930f41b25bd0ca8afef94dcb20214","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-11-25 04:17:26.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"214fa9cddb50756c59db72b7cf48ea18b5c3a788","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-11-25 09:17:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7337d7594fa11a9d26972f8302fce22cd456215c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 09:18:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1272bc1a1615db60d16059c0e16e51b43b9c170c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 11:20:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20f6f57ccec6423a88442ef3e74c4bdd\n\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/f05503be43b3409c837c40aef9bcc97e : SUCCESS in 1h 47m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46c38c851fbb48ca82feef7c0d696a63 : SUCCESS in 9m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a9576b2821bb4687a4cfcbc99270a528 : SUCCESS in 23m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/97ff741ddf86499dabbb3095dbeeed3e : SUCCESS in 19m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73b0d45f4ec546be930f17334d5617c8 : SUCCESS in 11m 19s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a4d4ab770f3a469886c8132ac4a9a381 : SUCCESS in 1h 29m 01s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/42be638667df44498e4f7f9f62674d92 : SUCCESS in 1h 56m 09s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/060ee961546f4c89b968582ccc057a40 : SUCCESS in 1h 09m 52s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/095c2538bf20445488e04a2d6cb86181 : SUCCESS in 1h 20m 13s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/8da16b19cbce4a0e9ceb3248bda1b0e7 : SUCCESS in 1h 11m 22s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/86b61623da544067b6526ca7cba8c7de : SUCCESS in 1h 16m 39s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/0b179abbd2e84e6ebd1464eba8d3e935 : SUCCESS in 1h 16m 31s","accounts_in_message":[],"_revision_number":1},{"id":"e47af048def2b777f5f718ef562a80afc49a36b5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 11:20:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2e4ee5222a0cef516bdb0eff58fafe35a72aa1d0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 11:21:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f7e4cd4fa1545c593a32d512c7e5362\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ecec2f05b7b427ba5df8c835f9d2425 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"72165d09aa203cdb0b204a2c33ebbee3358d1f63","revisions":{"72165d09aa203cdb0b204a2c33ebbee3358d1f63":{"kind":"REWORK","_number":1,"created":"2025-11-24 17:29:56.000000000","uploader":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"ref":"refs/changes/33/968233/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/33/968233/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/968233/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/968233/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/968233/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/33/968233/1"}}},"commit":{"parents":[{"commit":"5c6576bae99ca1067b97293cf6dde9f9c6b5bd6a","subject":"Merge \"[OVN] Initialize OVN agent in ``start`` method\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5c6576bae99ca1067b97293cf6dde9f9c6b5bd6a"}]}],"author":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2025-11-08 17:19:17.000000000","tz":240},"committer":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2025-11-24 17:29:56.000000000","tz":0},"subject":"Replace response body in after hook for 404 error","message":"Replace response body in after hook for 404 error\n\nCloses-Bug: 2130972\n\nChange-Id: I662b3eb7fb8c17d0ad30b10e8537d4a142003256\nSigned-off-by: Sergey Kraynev \u003csergejyit@gmail.com\u003e\n(cherry picked from commit d95e3e5eaf24ee788d65255d669671e175f25208)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/72165d09aa203cdb0b204a2c33ebbee3358d1f63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/72165d09aa203cdb0b204a2c33ebbee3358d1f63"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
