)]}'
{"id":"openstack%2Fneutron~968336","triplet_id":"openstack%2Fneutron~unmaintained%2F2024.1~I662b3eb7fb8c17d0ad30b10e8537d4a142003256","project":"openstack/neutron","branch":"unmaintained/2024.1","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2025-12-01 19:54:23.000000000","reason":"\u003cGERRIT_ACCOUNT_17685\u003e replied on the change","reason_account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2025-12-01 21:51:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I662b3eb7fb8c17d0ad30b10e8537d4a142003256","subject":"Replace response body in after hook for 404 error","status":"MERGED","created":"2025-11-25 15:40:15.000000000","updated":"2025-12-01 21:53:10.000000000","submitted":"2025-12-01 21:51:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"968336","cherry_pick_of_change":966498,"cherry_pick_of_patch_set":6,"meta_rev_id":"0bb9f290a250d7b936545ade06a5eeefdb6d4f9e","_number":968336,"virtual_id_number":968336,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-01 21:51:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"recommended":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"all":[{"value":1,"date":"2025-11-25 16:05:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2025-11-26 03:29:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"value":2,"date":"2025-12-01 19:54:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2025-11-26 09:04:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"value":1,"date":"2025-12-01 19:54:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-25 15:40:15.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"state":"CC"},{"updated":"2025-11-25 16:05:29.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2025-11-25 17:31:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-26 03:29:00.000000000","updated_by":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"reviewer":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"state":"REVIEWER"},{"updated":"2025-11-26 09:04:01.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2025-12-01 17:56:38.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"5f4d5f6b7faee13ce839d194756d315873ca4704","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-11-25 15:40:15.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"b4e91688cc525a924878e61480a304ca9e122b1b","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-11-25 15:41:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ce55de1565046656e0fb11a61805e748a6c8b282","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-11-25 16:05:29.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"9d5cb090e7f34181d3e2e633d90260f9e0807754","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 17:31:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/043fcb3577d74d36822b0aadf59ace26\n\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/7a6bcdd3f0154229be16f98f59d5a590 : SUCCESS in 1h 03m 01s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/0cc9ba3849554030b7bd19c2e9568d65 : SUCCESS in 1h 10m 58s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/ceaf951069824a61a3c07bc3b7c6ae8a : SUCCESS in 1h 34m 07s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/1258a8941b954ce8a743f95f8a7d09bd : SUCCESS in 28m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e6cf1fc2c3df43419b3b25eadac2339d : SUCCESS in 40m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1cd05260c44442789b93e72c29e2324e : SUCCESS in 7m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f7c45fd24750482b9748036eec46cec5 : SUCCESS in 26m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c7a56dddf82b466cae85ef841806939b : SUCCESS in 34m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b82b519ef8364bbdb60390378638d206 : SUCCESS in 11m 44s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/87d41cba382f486c928db0add61b24eb : SUCCESS in 1h 29m 06s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8778164cc535452892591f486bb9f7d9 : SUCCESS in 1h 17m 50s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/7fe5194a83284d7f893bda024701d4fe : SUCCESS in 35m 02s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/e8379fd1bcf04cee9b51e98b04b40c72 : SUCCESS in 51m 17s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8ae782183572462eb9e3be5025505969 : SUCCESS in 1h 36m 57s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/10714580755d4988af3f5dc9c327a009 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":1},{"id":"f89fb5e0641dbda2e50168955452662f04babc02","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2025-11-26 03:29:00.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"345a1dd5d7ecb5a659564c25a9c9fea70b4dbac9","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-11-26 09:04:01.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0f4d4c05422e45815df1bf469109c893ab6bbd09","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-12-01 19:54:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7f707709adbd59e07c3167fbd4104e4277bece34","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-01 19:54:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"414b1d6cae9893a6f7f4d6995d68e9772d047a75","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-01 21:51:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/499ff5b669674a979559117fa2dc61ae\n\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/f81d1e5a2b2443b58ac00d890c1c309e : SUCCESS in 1h 43m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ccd1632bbf4c4c8889395983c50c06aa : SUCCESS in 6m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/733506b43d29492886d61c39d11c8a28 : SUCCESS in 28m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/989ca919536e44569114be1a5b411004 : SUCCESS in 23m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41aaf343261b4d3995dba3cbd38734ca : SUCCESS in 12m 01s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6efdf1b58f58441eba3d1dbc39e12e44 : SUCCESS in 1h 52m 07s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/40fe9cb3b4224fa886ad20427481964c : SUCCESS in 1h 39m 58s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/6b6a9beea7e8408dbd68c09a12369af9 : SUCCESS in 52m 22s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/7630b453e64d42528c171ab24ecd63e1 : SUCCESS in 50m 03s","accounts_in_message":[],"_revision_number":1},{"id":"7d4684e3d1bfc816531dc2835749918da3d0da2f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-01 21:51:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0bb9f290a250d7b936545ade06a5eeefdb6d4f9e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-01 21:53:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/243fc1a2a0bc4aa9bf771fe66356d968\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3681ebe2978b4f638bdf718d6950a221 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"453d87c43a356cf66121b8935e12f880057d4ad2","revisions":{"453d87c43a356cf66121b8935e12f880057d4ad2":{"kind":"REWORK","_number":1,"created":"2025-11-25 15:40:15.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/36/968336/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/968336/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/968336/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/968336/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/968336/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/968336/1"}}},"commit":{"parents":[{"commit":"6593cad135f87f3ac2c9fb24b467d1d9cd35f6df","subject":"[CI][9-stream][stable only] Use last python3.9 supported tempest tag","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6593cad135f87f3ac2c9fb24b467d1d9cd35f6df"}]}],"author":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2025-11-08 17:19:17.000000000","tz":240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2025-11-25 15:40:15.000000000","tz":0},"subject":"Replace response body in after hook for 404 error","message":"Replace response body in after hook for 404 error\n\nCloses-Bug: 2130972\n\nChange-Id: I662b3eb7fb8c17d0ad30b10e8537d4a142003256\nSigned-off-by: Sergey Kraynev \u003csergejyit@gmail.com\u003e\n(cherry picked from commit d95e3e5eaf24ee788d65255d669671e175f25208)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/453d87c43a356cf66121b8935e12f880057d4ad2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/453d87c43a356cf66121b8935e12f880057d4ad2"}]},"branch":"refs/heads/unmaintained/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
