)]}'
{"id":"openstack%2Fneutron~968450","triplet_id":"openstack%2Fneutron~master~I86ed1aa473a41ece4478e19b1e0f37aadaf93238","project":"openstack/neutron","branch":"master","topic":"bug/2132982","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2025-11-26 16:45:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I86ed1aa473a41ece4478e19b1e0f37aadaf93238","subject":"[OVN] Add \u0027qos-bw-minimum-ingress\u0027 ML2 extension","status":"MERGED","created":"2025-11-26 11:41:25.000000000","updated":"2025-11-26 16:46:42.000000000","submitted":"2025-11-26 16:45:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"968450-bug/2132982","meta_rev_id":"66b40fc04836a00addfff2ccb4e095d283875a95","_number":968450,"virtual_id_number":968450,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-26 16:45:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2025-11-26 14:29:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2025-11-26 11:48:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-11-26 11:44:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":1,"date":"2025-11-26 14:29:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-26 11:44:55.000000000","updated_by":{"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"},"reviewer":{"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"},"state":"REVIEWER"},{"updated":"2025-11-26 11:48:18.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2025-11-26 12:42:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-11-26 13:55:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-26 14:29:40.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"26c96c359cac92c6741e59eb3c3965b4f1484608","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-11-26 11:41:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"21da7b6973ee172a507f7ee3e19ad70218e32e99","author":{"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"},"date":"2025-11-26 11:44:55.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"91bfe834227f7273cfae847b89a707169464f02e","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-11-26 11:48:18.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7e7a4cf82cd9e16ac2f5f3cf3615bbb9e838127d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-26 12:42:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0becaed618d84c5eb57992bb915b67a0\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/256f8a3689944c38a6c8b226f19d5f3b : SUCCESS in 45m 36s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fca2c367e13e487f881bf0262180d7c2 : SUCCESS in 22m 37s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/087879f6c62e4eeb9e030437c284aaba : SUCCESS in 28m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e3d2e02c8c169db9fd2a28ae93272ec97c7fd0d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-26 13:55:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f9abaa2558d41d0ab9d5e4ac18e85dd\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/8be5f3b7af094de0b7a13baee09baadb : SUCCESS in 2h 11m 59s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/91fadde7b06643f89e0b1adc7a6e5eb6 : SUCCESS in 31m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/18d8ffab7d4344eab34da3ae7db9f1ed : SUCCESS in 1h 02m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/510b5c76eb6847deab07bb91cdc63161 : SUCCESS in 11m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6162d7fbf01c48b9b27f2c128ecd6b2e : SUCCESS in 35m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ed4fa794bdba42c6a952f0893fdeea00 : SUCCESS in 40m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60738592a8e441a6bfa542071bfc6139 : SUCCESS in 5m 58s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/878bda1d8a5c4492b94aeac281ddf88d : SUCCESS in 58m 42s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a99a24d402f04b309ddb33ed7fc495d8 : SUCCESS in 1h 52m 07s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/e9167e38e7024a6daaa4dfaa9d33ff7c : SUCCESS in 1h 12m 44s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/0fc5aee4813d44b4a81cf7e92841b29a : SUCCESS in 1h 10m 09s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6935398ab3194914877bc42c65e909d4 : SUCCESS in 1h 01m 43s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/23112239456b4e5b966669fe9a23e420 : SUCCESS in 1h 29m 00s","accounts_in_message":[],"_revision_number":1},{"id":"f6426bb5b32ced63cb99f9beb5b5fecc34db9297","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-11-26 14:29:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"01b9f33c579db0d480c12691fd9c4b13313628bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-26 14:30:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a8e656dbc0eda3c79292d9def800792528d720c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-26 16:45:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dfab3c367c4e490891560adcb017d747\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/bb1447077a714e3aa57d336ad3959de9 : SUCCESS in 2h 01m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b413f8a207c4e25854f2daa9d780e41 : SUCCESS in 13m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1e3faf630cc641e486a3c359394ba6bc : SUCCESS in 35m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/03d0546906a0451e8a6434d7f5831838 : SUCCESS in 33m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a61bcb441584ff6b8bdc34488e777e8 : SUCCESS in 6m 43s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/20312629f5f84cb0b3e08429dd3da674 : SUCCESS in 2h 09m 48s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/484ed2ca3a104cd5adc63b336182ef78 : SUCCESS in 41m 07s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/c35955c632eb4e2996ec2a7b1e93c62b : SUCCESS in 1h 12m 06s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/8dcfd12ad0154bccac26b08c832e9b91 : SUCCESS in 1h 32m 18s","accounts_in_message":[],"_revision_number":1},{"id":"cc2e7f027f6c2b8a71fdd85da91f07e7d913bb5c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-26 16:45:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"66b40fc04836a00addfff2ccb4e095d283875a95","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-26 16:46:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00a8e3831d94410092c33604c32eb832\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f0a94cc8a1a464bb6ff9760eb8061d5 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"45b8dfe6a8d06ba5dc39ab28d2cb5adc4e7a8f5a","revisions":{"45b8dfe6a8d06ba5dc39ab28d2cb5adc4e7a8f5a":{"kind":"REWORK","_number":1,"created":"2025-11-26 11:41:25.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/50/968450/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/50/968450/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/50/968450/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/50/968450/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/50/968450/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/50/968450/1"}}},"commit":{"parents":[{"commit":"65b7790e8ad9b445a28e76e8e41d0cbd7cf3900e","subject":"Merge \"Add ovn-agent to metadata caching doc\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/65b7790e8ad9b445a28e76e8e41d0cbd7cf3900e"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-11-26 11:40:42.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-11-26 11:40:42.000000000","tz":60},"subject":"[OVN] Add \u0027qos-bw-minimum-ingress\u0027 ML2 extension","message":"[OVN] Add \u0027qos-bw-minimum-ingress\u0027 ML2 extension\n\nCloses-Bug: #2132982\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I86ed1aa473a41ece4478e19b1e0f37aadaf93238\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/45b8dfe6a8d06ba5dc39ab28d2cb5adc4e7a8f5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/45b8dfe6a8d06ba5dc39ab28d2cb5adc4e7a8f5a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
