)]}'
{"id":"openstack%2Fneutron~968455","triplet_id":"openstack%2Fneutron~stable%2F2025.1~I36d3c036c9795603d93b01bfd94d2ae80dc2ec7b","project":"openstack/neutron","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2025-11-27 15:05:05.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I36d3c036c9795603d93b01bfd94d2ae80dc2ec7b","subject":"Fix incorrect log message in ovn_client","status":"MERGED","created":"2025-11-26 11:46:44.000000000","updated":"2025-11-27 15:06:15.000000000","submitted":"2025-11-27 15:05:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"968455","cherry_pick_of_change":968454,"cherry_pick_of_patch_set":1,"meta_rev_id":"c3f06d23a532688fd8275f305227d5e02da4246f","_number":968455,"virtual_id_number":968455,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-27 15:05:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2025-11-26 11:48:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2025-11-27 09:30:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2025-11-27 09:30:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-26 11:46:44.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2025-11-26 11:48:47.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2025-11-26 13:27:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-11-26 13:48:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-27 09:30:14.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"7c79178614e32c44cb289f13a585bfb8e55d417f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-11-26 11:46:44.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2025.2.","accounts_in_message":[],"_revision_number":1},{"id":"df11394198cc97f85d1a765cef9eac8a456db804","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-11-26 11:48:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"69932475f47fbe42448a572a58542497de8d6341","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-26 13:27:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/04cb076f6d96403d8196932048b78f50\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9ff7e0612fe14a6d9c8711b12b82da23 : SUCCESS in 24m 28s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0625a3fe50504a4e932ca2d6ade759fa : SUCCESS in 23m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c30eb17f037f64f42d1159c0df579d81b41dc908","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-26 13:48:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b7994ba4fc864342940344f36c502af7\n\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/c4da1173fc8b4bdfb483613a6c5a414f : SUCCESS in 2h 00m 25s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/55a4d1e3f97f4128a56af84f4c12df22 : SUCCESS in 1h 16m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/16c5f04e1916451c8cb441b0ba432273 : SUCCESS in 27m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc3ca4ac3d1948e2bf43c4ec2ca3bcfc : SUCCESS in 5m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/16be741bfdd04ea4803602a5e07b232c : SUCCESS in 14m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0a82509c91b04a9985ca79542f7afd51 : SUCCESS in 14m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfdb4899a8c34250973eb4e534e955bb : SUCCESS in 11m 28s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/49c8e8e5444e46b385738c396b7f9a5f : SUCCESS in 1h 02m 19s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/2675a5f7947b4fe1b142f16ee349e7a3 : SUCCESS in 1h 19m 43s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/8095bca776b141abb24daaa95745dfa0 : SUCCESS in 1h 03m 47s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/da8c0083d43e433ba1946af4c7170694 : SUCCESS in 55m 09s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e3a849e51af8467b8a2cad517d06d49b : SUCCESS in 51m 03s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/395915f582684d5e86bb53d3f161ad41 : SUCCESS in 1h 16m 55s","accounts_in_message":[],"_revision_number":1},{"id":"0c5d323aedaf3df4e66d074d95c7b7805d2975c2","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-11-27 09:30:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"16a388a52723181b430e58020a45526bc6e273b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-27 09:30:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d6d9e818715976d3830b318bbd5b90d273fdbe15","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-27 15:05:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d55e5b480f64ba1a31e6bf760b3af05\n\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/4efe2bec676440f8a20518925bab1883 : SUCCESS in 2h 28m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f761aa973b64f05b24311f41508ad08 : SUCCESS in 5m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/076e249f7129480a92b2f6e0038c1f77 : SUCCESS in 21m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7ee1ef2035564a2287c27173a657551b : SUCCESS in 17m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0ccdb6d9caf41a8befa888cd8d15c8c : SUCCESS in 12m 30s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/133c4c6943274febb4274ee43f1280e8 : SUCCESS in 1h 26m 21s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/c333770e084e4b5ca4ccf74013f10099 : SUCCESS in 1h 16m 10s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a47e6b579d034da5a34cee25a9e400dd : SUCCESS in 59m 34s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/2c4905fabd704a49a26b6ae417c7b042 : SUCCESS in 1h 17m 37s","accounts_in_message":[],"_revision_number":1},{"id":"9cf4533ede7af38cc87596acdd3213a602722eea","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-27 15:05:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c3f06d23a532688fd8275f305227d5e02da4246f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-27 15:06:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b23b9bdb497e4acea40642a025a5f712\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8ebd660b98a4141b891763940fd3fbe : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c1219812bf18d6babbac13f2affbf2e4eb3df826","revisions":{"c1219812bf18d6babbac13f2affbf2e4eb3df826":{"kind":"REWORK","_number":1,"created":"2025-11-26 11:46:44.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/55/968455/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/55/968455/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/968455/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/968455/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/968455/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/55/968455/1"}}},"commit":{"parents":[{"commit":"d8775ea7e45ee3a7326bfdf547a93289e4899d58","subject":"Merge \"Remove false-positive ACLs in OVN DB sync\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d8775ea7e45ee3a7326bfdf547a93289e4899d58"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2025-11-20 23:19:33.000000000","tz":-300},"committer":{"name":"yatin","email":"ykarel@redhat.com","date":"2025-11-26 11:46:44.000000000","tz":0},"subject":"Fix incorrect log message in ovn_client","message":"Fix incorrect log message in ovn_client\n\nString contained %s but had no argument, added port id.\n\nCloses-bug: #2132088\nChange-Id: I36d3c036c9795603d93b01bfd94d2ae80dc2ec7b\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n(cherry picked from commit 807e6a5cf633cf7e51534bec655b98a1281861bb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c1219812bf18d6babbac13f2affbf2e4eb3df826"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c1219812bf18d6babbac13f2affbf2e4eb3df826"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
