)]}'
{"id":"openstack%2Fneutron~968769","triplet_id":"openstack%2Fneutron~master~I9ebf3048cb14fb79e9de612952752c9c56b14420","project":"openstack/neutron","branch":"master","topic":"bug/2125553","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2025-12-01 19:34:59.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9ebf3048cb14fb79e9de612952752c9c56b14420","subject":"[OVN] Add \u0027provider:physical_network\u0027 field if there is physnet","status":"MERGED","created":"2025-11-28 08:52:44.000000000","updated":"2025-12-01 19:36:03.000000000","submitted":"2025-12-01 19:34:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"968769-bug/2125553","meta_rev_id":"ff4858acbc8bacff4721c1d79bebcb821090a211","_number":968769,"virtual_id_number":968769,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-01 19:34:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2025-11-28 09:26:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2025-12-01 17:39:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2025-12-01 17:39:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-28 08:53:01.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"},{"updated":"2025-11-28 09:26:59.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2025-11-28 09:55:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-11-28 10:47:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-28 11:23:42.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2025-11-28 11:23:42.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2025-12-01 17:39:59.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"a2a8989fb284aaf3cc7e946b6e99a9bb882398c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-11-28 08:52:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3678a1679f21a569a9d970e66e8d107e494ec60d","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-11-28 09:26:59.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6b3ed2c122a3806504fa70c3d27bb1489994551e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 09:55:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4696b75908a4d7787dcee343a9dc3ea\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/475fb96227c54022840265c135896323 : SUCCESS in 1h 00m 46s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f92340e578784d6f99dab7859af02289 : SUCCESS in 25m 36s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1c01bffa3a00425aa2f60f5c98823be9 : SUCCESS in 31m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"35375d74ce51477f31cc16a81fdf66ae30f941ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 10:47:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/42c0e31bd7e84d72871721842aa8f0cf\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/63c70091658b4a09b1a3cac50cc792d8 : SUCCESS in 1h 53m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/04c11d3ce24c4706b6afc46c800fe24d : SUCCESS in 58m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c45b998c0a1549fea4943b5c6a0f931b : SUCCESS in 23m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dba97b924b7f41e3845c2bd4d01577f1 : SUCCESS in 32m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c8b05fefb42c469ca1cb13d20bd37469 : SUCCESS in 21m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11a444669549480fab89cf97cb3dfcbf : SUCCESS in 10m 47s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/fd535daef1d646bfb011d4de2ad35180 : SUCCESS in 1h 12m 39s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/3987e3add77b4964b11f8131419e5e0e : SUCCESS in 1h 46m 59s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/c8feac64564e4c10b859af1600a71c14 : SUCCESS in 1h 12m 32s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a6445891831a4dab9e7d70135b56f711 : SUCCESS in 1h 03m 23s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/374ef887509b4b3fb4e339221962a688 : SUCCESS in 1h 00m 51s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a584aa9decff4d348ae51c5f1b33a113 : SUCCESS in 1h 31m 17s","accounts_in_message":[],"_revision_number":1},{"id":"7e77863476963eb22ddab90a177ca0c71d344448","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-12-01 17:39:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a21c5f4f50e5fef54711a764e430dabd4e346d03","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-01 17:40:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9836fef1e03c468343f18fc4fcca2af38696a98a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-01 19:34:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f745104bb3944a509a50c8d5a382dd48\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d188376697414c9591c7e16b4d9fc8c8 : SUCCESS in 1h 49m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7eeb0dace5349ff852c1d6917cd74c2 : SUCCESS in 24m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/49db68bc134a4798b84262b494cfbebd : SUCCESS in 25m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ecff8cfaa1454622a34ce403bb787489 : SUCCESS in 35m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99a94a08afdf4ede977f511d8a698a30 : SUCCESS in 12m 46s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a19772c17ac84404bcec0e28961c7dfb : SUCCESS in 1h 49m 38s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/354a31fe07374251a72628fddbb5f6e0 : SUCCESS in 1h 04m 05s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/51dafb35071b4c2280ea58d64e1385b8 : SUCCESS in 50m 25s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/46d595378ae642429bf1b552d040d2d4 : SUCCESS in 1h 07m 44s","accounts_in_message":[],"_revision_number":1},{"id":"421085d16c43da095a1ec3c8d19ed8e0f3307096","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-01 19:34:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ff4858acbc8bacff4721c1d79bebcb821090a211","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-01 19:36:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/24a537b97c1f4eb0988db1ccce322936\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85b22ecdb9c94ed69f9958846bb8f81e : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a924eb9faadfbf7683af4ed5435b29e45166ee91","revisions":{"a924eb9faadfbf7683af4ed5435b29e45166ee91":{"kind":"REWORK","_number":1,"created":"2025-11-28 08:52:44.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/69/968769/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/69/968769/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/69/968769/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/69/968769/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/69/968769/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/69/968769/1"}}},"commit":{"parents":[{"commit":"1c9bfe52614ba50813dc492bab9f4a24031d7951","subject":"Merge \"[OVN] Add \u0027qos-bw-minimum-ingress\u0027 ML2 extension\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1c9bfe52614ba50813dc492bab9f4a24031d7951"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-11-28 08:48:00.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-11-28 08:48:00.000000000","tz":60},"subject":"[OVN] Add \u0027provider:physical_network\u0027 field if there is physnet","message":"[OVN] Add \u0027provider:physical_network\u0027 field if there is physnet\n\nIn [1], the commit message stated that the field\n\u0027provider:physical_network\u0027 would be present in the\n``Logical_Switch`` if the network has an associated physical network.\nIf not, the field won\u0027t be present.\n\nThis field is used in [2], but the code uses ``get`` to retrieve this\nfield from the dictionary, expecting that this field could be not\npresent.\n\n[1]https://review.opendev.org/c/openstack/neutron/+/962568\n[2]https://github.com/openstack/neutron/blob/1c9bfe52614ba50813dc492bab9f4a24031d7951/neutron/common/ovn/utils.py#L1312\n\nRelated-Bug: #2125553\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I9ebf3048cb14fb79e9de612952752c9c56b14420\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a924eb9faadfbf7683af4ed5435b29e45166ee91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a924eb9faadfbf7683af4ed5435b29e45166ee91"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
