)]}'
{"id":"openstack%2Fneutron~968770","triplet_id":"openstack%2Fneutron~master~I1df5d7984201498493026dc06c3e7750b8780a5e","project":"openstack/neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-01-09 16:10:54.000000000","reason":"Change was submitted"},"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2026-01-09 16:10:54.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2025-12-15 22:24:10.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I1df5d7984201498493026dc06c3e7750b8780a5e","subject":"Cleanup stale DHCP ports not bound to host","status":"MERGED","created":"2025-11-28 09:49:24.000000000","updated":"2026-01-09 16:12:58.000000000","submitted":"2026-01-09 16:10:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":22,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"968770","meta_rev_id":"d37f1e112768cd9e96ad0d65d45e0f996a066b22","_number":968770,"virtual_id_number":968770,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-09 16:10:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":2,"date":"2026-01-09 08:02:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2025-12-15 22:24:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":1,"date":"2026-01-09 08:02:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-28 11:52:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-11-28 12:05:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-01 20:59:36.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2025-12-23 20:14:34.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2025-12-23 20:14:34.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2026-01-09 08:02:26.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"9ec6a2411b506c6d6f6c4e9e3bfae346b423d482","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-11-28 09:49:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a3c6c352e3eded7cf88e4cc22c6121196b5efd9d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 11:52:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0970b295a4954a38923f11979f020380\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/23ce17b105884bed8578fa4211a0b602 : SUCCESS in 27m 02s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9148fc963a4d45ee9dc34df67874a908 : SUCCESS in 28m 57s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f16fc44622a04c18bdc468a46b1f2259 : SUCCESS in 1h 03m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ad604b7a5f834a2a6c4b2a58ce805a1c3ae3b928","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 12:05:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81440de5c81e4367acfa1d100e9fca39\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/ffc67ffd3bc34906ac6476834d97f663 : SUCCESS in 1h 26m 10s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d0b15b2c204d46ddabfeda0cd57eaa96 : SUCCESS in 2h 14m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2ee1aadfda8c427b8223eb2d5585b382 : SUCCESS in 1h 01m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/383ce2765ab0429784fb6533c9cff7f1 : SUCCESS in 10m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/71897f5e457e41e397cb624b80d200fd : SUCCESS in 34m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7cd03e9193cc484888e40619de06d956 : SUCCESS in 30m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b743ec266bb46e1b87ef924fdaf6bb7 : SUCCESS in 12m 01s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/3e736d1376a244bbbc512ac574fe051c : SUCCESS in 58m 46s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/365f99bb8b8841de883864320965a262 : SUCCESS in 1h 08m 10s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/7ed5249201144cb88dd4aa5ea2572b79 : SUCCESS in 1h 44m 17s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/698d350eae704bc1a7d9b234731ff901 : SUCCESS in 57m 35s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/d5c7626ee6d843d19f22f9356564c2f7 : SUCCESS in 42m 25s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/a3cfdf08119044cd984cc1191b9d8da1 : SUCCESS in 1h 43m 20s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b6bb04e265904a28939b8c9ae7c74663 : SUCCESS in 1h 40m 16s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d0230f581cad4ac9814f52e0f1b168b2 : SUCCESS in 46m 01s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/ffa5970383fa4ddeaba9d7b536ce5143 : SUCCESS in 1h 07m 58s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/cfebb80714d34a828f2e7d39469e6709 : SUCCESS in 1h 17m 45s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e3184ba06692437fadfdedafd1550ed2 : SUCCESS in 1h 09m 02s","accounts_in_message":[],"_revision_number":1},{"id":"0c17ab9faa7bf5107f2a19502793eb8b5ff1fb7f","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-12-01 20:59:36.000000000","message":"Patch Set 1: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"55d5730de14267f5083acc872d8f99007d38c6ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-12-02 08:02:37.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"00cc62178bafbc9fa0152e787c5dbd3de2192674","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-12-02 08:03:20.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"b615480c8862dbc23852c56ecb10c4a586d61883","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-02 09:14:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b87c7225fa044e0bbb9fe1c74520fc4\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f4e43e96e6244fb995c55964a714eafe : SUCCESS in 1h 09m 20s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2a45bcec28f34650ab100568b4933de8 : SUCCESS in 54m 50s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d0e83890f6874fd99d32183c56848348 : SUCCESS in 1h 08m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2f0c8f3d1343da9ac59506de8391b925a9fc3c89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-02 10:00:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b5550aee6e6a4e8ba8e3bd4f851cee13\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f7c3b40664cf476fa338414e3ef2f5c3 : SUCCESS in 1h 28m 16s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c95c8540dd5c4fbc86b606f6cccc1fd5 : SUCCESS in 1h 56m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2dd8fbb3bb554e8d9a49d605e0d4ff3f : SUCCESS in 36m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af16271ca42b4cc19743867025311b1f : SUCCESS in 23m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ca53b86e462f4171ab34b7d7722a24ae : SUCCESS in 25m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/309ca143342241ce8a0b9a865d307bf2 : SUCCESS in 21m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1576192955ac4f329008335d432133b6 : SUCCESS in 6m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/958ca603defd40ceb0feebd6a7a14f42 : SUCCESS in 26m 39s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/67a8e1d8f49747d8b17d11915b7c26d6 : SUCCESS in 58m 02s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/0c4ecb6057c94a4cb2220764757cd2a8 : SUCCESS in 1h 03m 53s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/bcb1ca83a481486eb70302b4d5c0df2d : SUCCESS in 1h 10m 26s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e6457fe12caf41bd9f9394ff4273a692 : SUCCESS in 1h 08m 08s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/cfe0f16645734df4b236ef67cc973671 : SUCCESS in 1h 09m 26s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/8b476ec6185f4190baa51fe94622df33 : SUCCESS in 1h 36m 16s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8a93e571bb3444d880edbf108e83fb6f : SUCCESS in 1h 21m 44s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a5e4627ddc97444aa96449ae33b8d53b : SUCCESS in 1h 04m 11s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/0d2a95f4c08a4117800160bfafa78cc7 : SUCCESS in 34m 26s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/902dcdcd436d4555812e5ab42d243e57 : SUCCESS in 1h 35m 25s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a8d74967db4f4b58906e91a75c8fcc44 : SUCCESS in 1h 14m 43s","accounts_in_message":[],"_revision_number":2},{"id":"2a6c5701c2735be5fa130145e0b97d424f2362af","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-12-02 11:04:21.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"cabfd7575bb27a3d460c954b35405a2b838be58d","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-12-02 19:08:39.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"fe164c4f1fc8424d5ec6e62753b1e02b4fd8a977","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-12-02 20:40:28.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"6c8e74fd423eb01a1ac2372c2655631914eb4d14","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-12-02 20:40:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7609a59dccd2c8ee3037f27bfe013c9917b43c26","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-12-03 08:16:11.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"0227b971e51353c6dcb6406d14b8065af901a5a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-03 10:17:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76c186254c55452384bef09c8f2a3df6\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/13b3485b92564b2da87d473f76c6cbfd : SUCCESS in 1h 27m 53s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/cdf09cfa41554813b8b312afaf8f04b8 : SUCCESS in 1h 42m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6adfed5a8f8e4c46a14aeb99f112fa71 : SUCCESS in 36m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0622a8b35f7247ee884edc28b0c01026 : SUCCESS in 11m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/52481bba3ed44db392ced5ae73fa870f : SUCCESS in 36m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1b82da90ed364bc5b4e82ab9b81184ec : SUCCESS in 35m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba7de4d8609e480cb15420e10d482166 : SUCCESS in 6m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a6e09c95413465f98050a0fc31fc9ae : SUCCESS in 12m 20s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/0d5a3c04b1ec425c99aa3140c86ba6fe : SUCCESS in 1h 05m 18s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/3a7fdda1fa284434a616a698c71534b3 : SUCCESS in 1h 09m 54s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/074fc49be1d640e495ca72b162691921 : SUCCESS in 1h 52m 31s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/fae717e220b64f5c974dea991ff3a9c9 : SUCCESS in 42m 17s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/931cd564fc714715b4bbac5d6cd1d1c6 : SUCCESS in 38m 34s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/b93a330ba4c8423685dbe1d665b70d50 : SUCCESS in 1h 06m 22s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d35d95d3a28d408d8059171fe14d24ab : SUCCESS in 1h 47m 17s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/9ddfec98ffda4fd8b0e8c99c9eea00a6 : SUCCESS in 1h 08m 55s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/7357c36cf29340bfa63b75249f2df162 : SUCCESS in 1h 10m 53s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5e8c0dd5b73e46a989ab4b33993b7b5c : SUCCESS in 1h 01m 17s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3a2bfdc9bd7241f79427939d8fe3e34f : SUCCESS in 1h 13m 26s","accounts_in_message":[],"_revision_number":3},{"id":"efdac49ec2263b95ac7a95b183b8aa289c0995df","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-03 12:41:56.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8dd02090495f4792a00f8c2f1e1cbbbe\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/dc23e5e313bc4322a29a54d46d629031 : SUCCESS in 30m 42s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6bae26c0579f4a7894d0bea305acb614 : SUCCESS in 21m 40s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/fb80d28994754c53b4f7e6dfe1f6d64e : SUCCESS in 1h 05m 14s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"84acd03516f958ba80ee06ae2220b475a483248c","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-12-03 21:56:56.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"f1912a4fcf1eceb128b9e765eba080335fd62d4d","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-12-04 19:09:42.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"18b23afae1c818e18a2c0641a54bfde7419953ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-12-05 08:13:28.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"5e9a7f1a1d405fbee7f6e11526a83b9f5a0aaf37","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-12-05 08:13:40.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d5f08f0043f72802c10057f65a09cb3424250bfa","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-12-05 08:17:28.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"98b0d636a1b8536d04ebfb50251f372d7164bcbb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 09:37:18.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/14f8492246924bf69ec4af01868f04c7\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c76db40770d9438db73994f12ed416f9 : SUCCESS in 1h 11m 00s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7275cabbfa104c028725e7447ec464e7 : SUCCESS in 22m 41s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c1c0572104f74bc5910645839ae90fa6 : SUCCESS in 1h 08m 48s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"bc278c364a24c882f87d47f9109212050f669a68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 10:14:03.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/88f97d26fa0c4651bb179289364fa010\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/da7af33e3703458d862d7a7685ef8e43 : SUCCESS in 1h 28m 55s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/3500e850433b4459b27280fba7b9cb9e : SUCCESS in 1h 43m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b86cd726622545da92a468ab484a681a : SUCCESS in 1h 12m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a798b4c498694a328028ea5be89657d1 : SUCCESS in 10m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/acebc3f0c65a4aefb5fc78ceae74b99d : SUCCESS in 22m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/149072e431d24adc93fc262d5d865c33 : SUCCESS in 24m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8037c84990bc4244898388aa595a8c18 : SUCCESS in 14m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a71adda3eaed4d8e8689da671d0033a9 : POST_FAILURE in 17m 11s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/ea6b3bf457b5493a8bf568b219849204 : SUCCESS in 57m 33s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/52e405ab256c46b5a116f263afadb718 : SUCCESS in 1h 08m 41s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/1aac59e06f9c44f787652b6c0fc2d68c : SUCCESS in 1h 48m 47s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1b300b47fe3448aa964555743524eec1 : SUCCESS in 1h 08m 03s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/133587f4509a4a1f8c59a91523da20bb : SUCCESS in 1h 01m 36s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/1abe2ea5bbe74a4d9d34e180dd10f671 : SUCCESS in 1h 09m 04s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c86941913aa4462488d913528c51b34d : SUCCESS in 1h 39m 23s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/312c51a3c1924a50b2543b3345904342 : SUCCESS in 59m 04s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/22461481f20842d6b87f33a0144213ab : SUCCESS in 1h 01m 15s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/bbdec5ae92994b8799ebdd07bd776d62 : SUCCESS in 40m 14s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/51624ec4de0f465ab57bbaeadbbf039b : SUCCESS in 1h 34m 57s","accounts_in_message":[],"_revision_number":4},{"id":"25c2ad1e3b01e29d4473e262d6758bc26148969f","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-12-05 13:37:24.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c12334405170d36d50bdabdc821caa48a62c2d98","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 14:46:25.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/67b58cdd6cf241d98f7e6043b8183fe2\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ed73fec06d124374aa8a643b95ade0fe : SUCCESS in 1h 06m 46s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/513e4a0ed0dc4dcfbc24f20003f52512 : SUCCESS in 52m 15s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8da79de93f7d4123bfcca0e2338f3c75 : SUCCESS in 1h 06m 14s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"3c9ef7a8f3b433e18cee9450623216dfc8642794","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 15:48:21.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/713e7acc4a884403990076faddf580df\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/4b29cab849ba485088dda3d193f9f27c : SUCCESS in 1h 32m 41s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/3637822f89784820bf760387ad18adaf : SUCCESS in 2h 08m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4f4158c12e51404f9cb25928f2c80cc1 : SUCCESS in 41m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b983dc661c6f45d3ae25441899204736 : SUCCESS in 23m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8410e3cd79bc4d60b28a0e5c186cbdd7 : SUCCESS in 33m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fbc3d328a6dc4bb98076632c3a1e5d1c : SUCCESS in 24m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b7aa92f56e94675b2f320cd2c384bcf : SUCCESS in 9m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/82bbb9f4f6304ea7a7e5f3c29e82ab22 : SUCCESS in 24m 12s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/7748186cdf3641a5abe09bcdcc23fe6c : SUCCESS in 56m 45s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/d3ebceabe82a4b1ea215607937824136 : SUCCESS in 1h 24m 06s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a3e7d84b558d48908d71f1ddd852d0da : SUCCESS in 1h 01m 42s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/bf94682040454841bd073c6785cf51b3 : SUCCESS in 1h 08m 42s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/117170a52141493ab136997f89fda572 : SUCCESS in 39m 44s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/f8ee4e5fd6dd4ba1a36c75a8c41fe239 : SUCCESS in 1h 57m 24s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/744f7b8d4d4444ddaa2507f0c41e0c5b : SUCCESS in 1h 41m 15s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/bcdd812b3a2d4a0a8a36c45b0ada3d82 : SUCCESS in 1h 08m 06s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/a61446771a844457a49d76fdfd37ce6d : SUCCESS in 1h 10m 14s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/477102d1e26244af9385011072754223 : SUCCESS in 37m 46s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/274ecedc1a5d4b34a1ca6ea4734fa066 : SUCCESS in 1h 13m 34s","accounts_in_message":[],"_revision_number":4},{"id":"74cedcbf38ec6b15045a573f8bd8b0ea319c3f40","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-12-15 22:24:10.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"21a9f3ae274c78b7c0b4c02483be66866d0102b2","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-01-09 08:02:26.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"bc9ebaeed55134c2e17e7fa5eb5376ce82277a40","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-09 08:03:18.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"e147c2e1d5d91c179b89749c3d92a7a0db8f9286","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-09 09:27:44.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/46685ac97da044e58807dd82c4f7ab6b\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/cbd4edd1554542d19e9255d6b83751c1 : RETRY_LIMIT in 44s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/367affd12f10418980292a253234bd09 : RETRY_LIMIT in 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55e5888697e142ba9f3551d4be3e5f1f : SUCCESS in 26m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/220f9a9d65954c10bd1c654ae9cf53be : SUCCESS in 38m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/78e586793ed84a52b1927f00edf61658 : SUCCESS in 34m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b757b452388480b95ee1379fe169bac : SUCCESS in 12m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/acc4651f52ca46bcbb7b5c5ed459ed21 : SUCCESS in 13m 27s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/1bd4ab8da8b343bda4f3f8fca10ed5c9 : RETRY_LIMIT in 10s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/308d78c7510f4c4eb953d684b3d42bb5 : RETRY_LIMIT in 43s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/ecca6309f9b6408bac211fd203e8a82f : SUCCESS in 1h 12m 57s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/33c49a69d56643049f00dc803f3887ab : SUCCESS in 1h 09m 58s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a56e68768eb442babbc2c2ee0a3ec1b5 : RETRY_LIMIT in 40s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/86d4abb2bcce4d85ab9dfb9bede14bdd : RETRY_LIMIT in 23s","accounts_in_message":[],"_revision_number":4},{"id":"426d64d969512280449f4ddbf54b1409ee277c5d","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-09 11:18:08.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4bf51a983d68b090ce6a08ec7d383fcf85675852","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-09 12:51:35.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee307b8316ec46028732fa5d607464ac\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0f526d8f0caa47449987f4dd72f4b50f : SUCCESS in 26m 11s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/25866d0531134815a3585f106649b82e : SUCCESS in 24m 21s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2c56e5590d0a4b6c8bbacf5e1e82b674 : SUCCESS in 25m 54s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d9f4a125d2d3647048ca86a4be838eba6d7015cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-09 13:27:58.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e15f3b36cd543dc805423b458eeffab\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/59b9584275dc4f848c12a05de83e25d4 : SUCCESS in 1h 24m 23s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/eb53bbab43424772ab44369a236fcf9e : SUCCESS in 2h 07m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b8a99b6fc2134fc8bf70f3060542ff31 : SUCCESS in 59m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05a0fe9dd6964f8fb2618da313bf28e1 : SUCCESS in 24m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b51a1548c60e41a494a815fe62e3f0bf : SUCCESS in 26m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6c67848320144263af1d2b48ca0a67e3 : SUCCESS in 23m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/134b5f5ffbc04a10bc2069de11b2fe32 : SUCCESS in 11m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5d1e796c7fa04c7cbf3e5794356ee411 : SUCCESS in 24m 36s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/18bf49636ed64c9884b77094cacd1621 : SUCCESS in 52m 01s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/fb91d9c524e64b9092662982155c4195 : SUCCESS in 1h 14m 05s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c216e42da2ac4cc8b7d1172b929768e0 : SUCCESS in 1h 05m 07s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/0ddf59ab7c78435bbeefc9d49ad3de19 : SUCCESS in 1h 09m 41s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/541a6468b4a84b4697141401ae3c5b42 : SUCCESS in 36m 56s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/2c416679f8d44a4ca273b79b86fb6ffa : SUCCESS in 1h 35m 43s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6768e1fe93cd4a09b173a2d20d902a48 : SUCCESS in 1h 12m 31s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/fd128a8e81ee4aeba35ab4f7ec499ae2 : SUCCESS in 1h 00m 52s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/718415a68e2648f6b22e8746eca2e24c : SUCCESS in 1h 16m 48s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/49964cb6e06247de83ba7245bcdd8e90 : SUCCESS in 1h 42m 55s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f1b370b22b7b47bb9cd3deb903139e97 : SUCCESS in 1h 26m 54s","accounts_in_message":[],"_revision_number":4},{"id":"2589ac817aa364480eaca2b1f44e0d20cafc2297","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-09 13:28:28.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"1bb1bcd46a91f3df114629994fef815ad063a1d5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-09 16:10:54.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f61910190d9241b2979dac2227fd01cf\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/5f2175cbac834c9fad92c8cb7d9407f4 : SUCCESS in 1h 38m 35s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e4f5b3f109fe4a3389f858fc53f375f4 : SUCCESS in 2h 41m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e70864a51db341208885dd5ce72074f9 : SUCCESS in 30m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1cf700f9f50344e6a6d3dcb6ccb753be : SUCCESS in 38m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/94491c858fed4b2e9c5cfe9308269ad9 : SUCCESS in 37m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49e669e627504320a0133fe927f49233 : SUCCESS in 11m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c875ddf0187e4eae8a5c6c0ab64271b2 : SUCCESS in 25m 21s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/5c99e210c2ab45719b915a417ff47bb4 : SUCCESS in 1h 39m 15s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/1856582cebe644aeb1d0b62c24de8739 : SUCCESS in 1h 37m 06s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/99066a8e06f1440eaa706c0288a9b7ef : SUCCESS in 1h 15m 57s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/eb886022e3a14d4da63c712e955d9188 : SUCCESS in 1h 08m 34s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/5e13c84360eb480fbee57b22249d5097 : SUCCESS in 47m 36s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/990f8f18bd644d24a2ff4a3d913b8626 : SUCCESS in 1h 16m 20s","accounts_in_message":[],"_revision_number":4},{"id":"07296295ba7fc2b0dd9b5a5fc6794ffa16e8f617","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-09 16:10:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"d37f1e112768cd9e96ad0d65d45e0f996a066b22","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-09 16:12:58.000000000","message":"Patch Set 4:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/63de13ae55d74efb9d59837ea0d2caf5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/248023fc66354ff596866a6b717d558e : FAILURE in 39s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/67d4089444854584996027987e31fec9 : FAILURE in 34s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"479718d362c9077a57eeb9263e93b274ea10f677","revisions":{"d7a40a597d1d188df42185c074c375b9e9f27fa8":{"kind":"REWORK","_number":1,"created":"2025-11-28 09:49:24.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/70/968770/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/70/968770/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/70/968770/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/70/968770/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/70/968770/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/70/968770/1"}}},"commit":{"parents":[{"commit":"1c9bfe52614ba50813dc492bab9f4a24031d7951","subject":"Merge \"[OVN] Add \u0027qos-bw-minimum-ingress\u0027 ML2 extension\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1c9bfe52614ba50813dc492bab9f4a24031d7951"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-11-28 09:45:07.000000000","tz":60},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-11-28 09:45:07.000000000","tz":60},"subject":"Cleanup stale DHCP ports not bound to host","message":"Cleanup stale DHCP ports not bound to host\n\nThis loops through the network ports when checking\nfor stale devices that should not be there anymore\nand if the tap port exist and the binding:host_id\nsays it\u0027s bound to another host than what the DHCP\nagent is on the port is considered stale and is\nunplugged.\n\nThis makes sure that Neutron has self heal itself\nduring this error condition where a DHCP port\nthat is not bound to itself is still there causes\nconnectivity issues due to for example IP conflict\nwith the other DHCP agent.\n\nPartial-Bug: #2130885\nChange-Id: I1df5d7984201498493026dc06c3e7750b8780a5e\nSigned-off-by: Tobias Urdin \u003ctobias.urdin@binero.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d7a40a597d1d188df42185c074c375b9e9f27fa8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d7a40a597d1d188df42185c074c375b9e9f27fa8"}]},"branch":"refs/heads/master"},"a14ce2524b61ff94dd0e32ffcf6f39e732f6efd5":{"kind":"REWORK","_number":2,"created":"2025-12-02 08:02:37.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/70/968770/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/70/968770/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/70/968770/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/70/968770/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/70/968770/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/70/968770/2"}}},"commit":{"parents":[{"commit":"1c9bfe52614ba50813dc492bab9f4a24031d7951","subject":"Merge \"[OVN] Add \u0027qos-bw-minimum-ingress\u0027 ML2 extension\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1c9bfe52614ba50813dc492bab9f4a24031d7951"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-11-28 09:45:07.000000000","tz":60},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-12-02 08:02:28.000000000","tz":60},"subject":"Cleanup stale DHCP ports not bound to host","message":"Cleanup stale DHCP ports not bound to host\n\nThis loops through the network ports when checking\nfor stale devices that should not be there anymore\nand if the tap port exist and the binding:host_id\nsays it\u0027s bound to another host than what the DHCP\nagent is on the port is considered stale and is\nunplugged.\n\nThis makes sure that Neutron has self heal itself\nduring this error condition where a DHCP port\nthat is not bound to itself is still there causes\nconnectivity issues due to for example IP conflict\nwith the other DHCP agent.\n\nPartial-Bug: #2130885\nChange-Id: I1df5d7984201498493026dc06c3e7750b8780a5e\nSigned-off-by: Tobias Urdin \u003ctobias.urdin@binero.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a14ce2524b61ff94dd0e32ffcf6f39e732f6efd5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a14ce2524b61ff94dd0e32ffcf6f39e732f6efd5"}]},"branch":"refs/heads/master"},"16594c4e246fc61e422211bc6a327f57d755417d":{"kind":"REWORK","_number":3,"created":"2025-12-03 08:16:11.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/70/968770/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/70/968770/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/70/968770/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/70/968770/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/70/968770/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/70/968770/3"}}},"commit":{"parents":[{"commit":"1c9bfe52614ba50813dc492bab9f4a24031d7951","subject":"Merge \"[OVN] Add \u0027qos-bw-minimum-ingress\u0027 ML2 extension\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1c9bfe52614ba50813dc492bab9f4a24031d7951"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-11-28 09:45:07.000000000","tz":60},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-12-03 08:16:03.000000000","tz":60},"subject":"Cleanup stale DHCP ports not bound to host","message":"Cleanup stale DHCP ports not bound to host\n\nThis loops through the network ports when checking\nfor stale devices that should not be there anymore\nand if the tap port exist and the binding:host_id\nsays it\u0027s bound to another host than what the DHCP\nagent is on the port is considered stale and is\nunplugged.\n\nThis makes sure that Neutron has self heal itself\nduring this error condition where a DHCP port\nthat is not bound to itself is still there causes\nconnectivity issues due to for example IP conflict\nwith the other DHCP agent.\n\nPartial-Bug: #2130885\nChange-Id: I1df5d7984201498493026dc06c3e7750b8780a5e\nSigned-off-by: Tobias Urdin \u003ctobias.urdin@binero.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/16594c4e246fc61e422211bc6a327f57d755417d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/16594c4e246fc61e422211bc6a327f57d755417d"}]},"branch":"refs/heads/master"},"479718d362c9077a57eeb9263e93b274ea10f677":{"kind":"REWORK","_number":4,"created":"2025-12-05 08:13:28.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/70/968770/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/70/968770/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/70/968770/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/70/968770/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/70/968770/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/70/968770/4"}}},"commit":{"parents":[{"commit":"1c9bfe52614ba50813dc492bab9f4a24031d7951","subject":"Merge \"[OVN] Add \u0027qos-bw-minimum-ingress\u0027 ML2 extension\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1c9bfe52614ba50813dc492bab9f4a24031d7951"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-11-28 09:45:07.000000000","tz":60},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-12-05 08:13:27.000000000","tz":0},"subject":"Cleanup stale DHCP ports not bound to host","message":"Cleanup stale DHCP ports not bound to host\n\nThis loops through the network ports when checking\nfor stale devices that should not be there anymore\nand if the tap port exist and the binding:host_id\nsays it\u0027s bound to another host than what the DHCP\nagent is on the port is considered stale and is\nunplugged.\n\nThis makes sure that Neutron has self heal itself\nduring this error condition where a DHCP port\nthat is not bound to itself is still there causes\nconnectivity issues due to for example IP conflict\nwith the other DHCP agent.\n\nPartial-Bug: #2130885\nChange-Id: I1df5d7984201498493026dc06c3e7750b8780a5e\nSigned-off-by: Tobias Urdin \u003ctobias.urdin@binero.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/479718d362c9077a57eeb9263e93b274ea10f677"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/479718d362c9077a57eeb9263e93b274ea10f677"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
