)]}'
{"id":"openstack%2Fneutron~970204","triplet_id":"openstack%2Fneutron~master~I0f95712be44426fb5184702c1969d7fe4bf4147e","project":"openstack/neutron","branch":"master","topic":"bug/2134387","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2025-12-15 13:42:33.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2025-12-16 10:28:05.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2025-12-16 10:28:05.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0f95712be44426fb5184702c1969d7fe4bf4147e","subject":"Check if network is external in auto-allocate code","status":"MERGED","created":"2025-12-08 23:58:07.000000000","updated":"2025-12-16 10:30:53.000000000","submitted":"2025-12-16 10:28:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"970204-bug/2134387","meta_rev_id":"1f5b6ece7bf41df6bfe426c1442ce0a2b4b7c530","_number":970204,"virtual_id_number":970204,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-16 10:28:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2025-12-15 13:42:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2025-12-16 08:30:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2025-12-16 08:30:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2025-12-15 13:42:33.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}],"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-09 00:57:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-12-09 02:46:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-10 17:05:12.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2025-12-12 10:35:12.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2025-12-15 13:43:43.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"},{"updated":"2025-12-15 13:43:43.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2025-12-16 08:30:42.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"f1b3146172815431191a913cd69b072cbb3c4e08","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-12-08 23:58:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8cba3e1e5d110d4ddbd5e6d941834f2bb3559eb6","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-12-09 00:23:39.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d6ca2922c642ff81509502ce8ad4b1dba61a2622","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-09 00:57:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9ac4d92b5a2418b96213b75a15f520d\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fb10ac4b84304216a3ce7a31104fd475 : SUCCESS in 27m 40s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/916ec9ba28da481a9ed2d7325b03ae7a : SUCCESS in 42m 41s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/17fd2bb9c2664e629cd5f8b9c562cd82 : SUCCESS in 57m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"15e63ffb047fa5c3ac09b62e981ca9c18d65a56b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-12-09 01:11:47.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"957394e5090c8f29ce506c43aed323a36845f052","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-12-09 01:11:58.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":2},{"id":"1d134a768634e9174da4003c9965548a31f94aab","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-09 02:12:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a60a779b6ff04fb091a8dd755554d95f\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/762863578e044ec0b480ea8558b19a05 : SUCCESS in 59m 00s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ec42b11e9768478887962fcf0e7d418c : SUCCESS in 44m 38s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/91df9d31af984e45bbc83fbc1b62999c : SUCCESS in 57m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"eec7f8398cb4b083ca4d0dd5a645f66f96a65e02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-09 02:46:01.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/743d8ca795e9474a90c1dd448ba3fbb0\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/851e1c289ddc42449fae84bb69432f81 : FAILURE in 13m 35s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/103075750fac4e8d96bcc3af7e3f4e48 : FAILURE in 14m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9dac2f75d19645efa5b55324ebd60940 : SUCCESS in 1h 04m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c0b4dbd196f484c82fcdcbc54023366 : FAILURE in 24m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/172c5eb680054a6785bf7ef10bf13dee : SUCCESS in 22m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/972a3f91c4464c4ab40d91c7fd8609a0 : SUCCESS in 37m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30a311807b534484a83e01dbaf871f44 : SUCCESS in 9m 29s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/08b1fdde14f845d9ac22ba74a61363be : FAILURE in 39m 41s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/d4c648e77f0c4923bd2ef8b02038c53a : FAILURE in 1h 06m 04s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/e56812fd5ad34db69b70281a4083d56f : SUCCESS in 1h 32m 26s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/20bb391876664ff7a3343b64a4f39a02 : FAILURE in 39m 30s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/538c5a6565fa4085b207a7cd44c9b641 : FAILURE in 34m 16s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/a214135cf773426f966b8aa343d968a0 : FAILURE in 23m 46s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7b25ce1c8dad45e78f7ced0aa2adb85a : FAILURE in 24m 52s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/af52b89eedf14faba9d0463c53201a54 : FAILURE in 22m 24s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/382559916f904dad92efa2bf8385614d : FAILURE in 22m 26s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/55ca50ac7f5b48f281f4e3c1b1bd40de : FAILURE in 25m 50s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/295d6c4206dd4a4d8ba1b262d51549e5 : FAILURE in 22m 57s","accounts_in_message":[],"_revision_number":2},{"id":"63b717fd6443a8e2d632eacc98a86ec4ae652323","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-12-09 22:03:53.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2a2b0db85f0605550d24f9b01335371c00c97ecd","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-12-09 22:04:01.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"f86ac5a8525435dfe8bd8fab3ee7ff62a91b22bf","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-12-09 22:30:51.000000000","message":"Patch Set 3:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"5b2130aa2033e14e24baceafa0ecc08022066ec8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-09 23:17:58.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d800f03300824c258dfe7b9da7d8630f\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/208d0a95729a4a5985237379692228f1 : SUCCESS in 1h 11m 48s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5c9c4b66d18f4a5a9f352a83f66dfa05 : SUCCESS in 55m 34s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/06fb0a4154e247bdb3ed100be5cfd892 : SUCCESS in 1h 11m 39s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7da83cc955f575120662ef025221f4af80bdf65b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-10 00:18:06.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/19397251a9284efa9db9afc6f8f30192\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/23d37ebf45bf4f838af5ca671175cf69 : FAILURE in 1h 32m 16s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/1fde767551e74db4838f74b8da66e833 : FAILURE in 2h 12m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cabb6743c3b542318734fceb65acea16 : SUCCESS in 36m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a1ea04d45704be2ae92f493a31447c4 : SUCCESS in 28m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0768957f179e484483c79f01e75a377f : SUCCESS in 36m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8ad2b08686b243a9968696ff53314148 : SUCCESS in 36m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9ff5438942e45efad7165e6c50ead6f : SUCCESS in 13m 28s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/bb57378919b1407f9b6de79412565cfb : SUCCESS in 30m 12s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/f4f5d05ada2a4a71b6609e7125083740 : SUCCESS in 35m 16s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/9182a38d32914eb6a4a813ff8c62195d : SUCCESS in 1h 14m 03s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a61e86272e76462eae7dbcd5c4c6b3a0 : SUCCESS in 54m 34s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/07eace47722646faa8eb9c2a561f7419 : SUCCESS in 1h 13m 19s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/c018a0bde96a4b9985532fba4653eece : SUCCESS in 1h 47m 24s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0c8c8d5cd79d469196c1908d2e91dac7 : SUCCESS in 1h 41m 57s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/6d9ad5f48a5b4455936226cfb26f6fe6 : SUCCESS in 54m 34s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/2053c30c407f474ab73731e72945f823 : SUCCESS in 34m 50s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8eaba298bfa344de824130078e56c874 : SUCCESS in 35m 40s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/afb282a47c87445e961724646fc236c6 : SUCCESS in 1h 20m 59s","accounts_in_message":[],"_revision_number":3},{"id":"3300c35bf08d223bc823730bab14f3b4f3e1ed2e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-12-10 16:04:03.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"d35bfeec7201232c2880773cb55d2cb4a22aa540","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-12-10 17:05:12.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c232739bf97d42235ea103f555c498617b96e051","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-12-10 17:15:23.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"96741803e51fd561035896ec28ef934d51917fbe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-10 17:17:55.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0a6d56e1c864e0bbd7a26f9a7d408e1\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/333b3de65e8d42ca876c3c6f5fbf43b0 : SUCCESS in 26m 17s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2513a1ecf69545fe839f9a167e02456e : SUCCESS in 25m 43s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d6d91641c44746748e24d4d92db9cb2d : SUCCESS in 45m 16s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"dd91a031972f1a385ac64756b87e8aa165f0aa0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-10 18:03:02.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3b69ed170f4d4f3296640422c367d02d\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/504ad03182b34623bd220c15fbb6da45 : SUCCESS in 1h 38m 01s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/146164e6492c41bf9631478c04f4e272 : SUCCESS in 1h 58m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ac2f4ad4aa274e868d1a2f484d800039 : TIMED_OUT in 1h 20m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb5ea0406ec34ec388582f9ab2a098ca : SUCCESS in 17m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/40baa7f92db74f8182e982b8425c750f : SUCCESS in 28m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/27b4bded852e471bb3bb58770cb9cba7 : SUCCESS in 34m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0eea1ea302a0466dbebf84c935792d5f : SUCCESS in 12m 51s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/8db1ff94a0974ae98c8a4832380b1da3 : SUCCESS in 1h 00m 50s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/5f3e45d5223146a2bbc4d9bc5e882589 : SUCCESS in 1h 13m 19s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/666870d28d744b9b93295d45bbe06c3f : SUCCESS in 1h 14m 06s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/5aad9c2a960e416a8eb7e4da650fb2a4 : SUCCESS in 1h 13m 51s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/d17054e16e5545f5b74a610a325c863f : SUCCESS in 1h 12m 31s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/1f62f527ff574db7ab5c9e0ab3139a5d : SUCCESS in 1h 37m 15s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/cbd14db48964445cb023556a0d16177f : SUCCESS in 1h 45m 58s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/5663043dad2a47589b9194d8763f9898 : SUCCESS in 1h 07m 33s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/90252c15d9c34e548f3a3ddaa628b369 : SUCCESS in 34m 55s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d0670e3f158c4ac4902d285b5886d5c5 : SUCCESS in 1h 03m 30s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/ec38c86422444ab294e19dda3a2eafbe : SUCCESS in 1h 03m 29s","accounts_in_message":[],"_revision_number":4},{"id":"4f74f3feb20df4d0e1b65c10dd63f3f74dca5bc0","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-12-10 18:08:12.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8aa6093ba47aca702bcae76964a61b89eb328094","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-10 19:15:18.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/71a7ade7ed5644ef86bbf46adaba831c\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/72497a7fc5e94823b6b88b3e03198965 : SUCCESS in 29m 58s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a8ca5976ef8a4217b56ffd14c7a375a2 : SUCCESS in 21m 09s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/01450ce2b24c47a1888d334faa55de99 : SUCCESS in 1h 01m 11s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f1d8560c8533c73da2248c28cec556507328180c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-10 20:09:26.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7096f5f910ed48d4b486f3a523f4f7cc\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c5c85ba3e1ba41a18cc2a260e6bcf8b1 : SUCCESS in 1h 21m 36s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/830b1bae5d2c4c1b9759a3207a929296 : SUCCESS in 1h 55m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8b5f9f932fc045a981ea70c63a31326e : SUCCESS in 59m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f161e94fdd645ccbc98a15d36e2482a : SUCCESS in 30m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3d88c24258c14e56bc0a228f3ca9332f : SUCCESS in 23m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4c35104c1ef343f0bc2089eae5fa09e9 : SUCCESS in 22m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4341817be00d429a9c9d562a1fea1b6c : SUCCESS in 12m 15s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/51441e4495f047cd8f8ec0c6acf94259 : SUCCESS in 1h 00m 20s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/12fbd77b05464d04888e3ef9f8a58a3f : SUCCESS in 1h 13m 25s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/13b287f4104d4d988f85279ce26e652e : FAILURE in 26m 40s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/9a0a7189cdfc4cb29ceed4683b07b0fb : SUCCESS in 1h 10m 00s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/002f215583af459abc65159e68944d63 : SUCCESS in 1h 10m 28s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/dc6be3e2137146299215b8c842e3e7c5 : SUCCESS in 1h 48m 52s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7f78be54faf24760a111ea3c3bb54409 : SUCCESS in 1h 37m 04s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/ae771ddb375b47ae9ebe1f5a10236309 : SUCCESS in 1h 13m 48s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/62e1f7d79ac544b7904254201187dde9 : SUCCESS in 31m 23s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5b7d33782b2e4c70b4ac9abc9ac83abd : SUCCESS in 1h 36m 21s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d000a58dbfb241558e899b2ffa075eb7 : SUCCESS in 1h 24m 36s","accounts_in_message":[],"_revision_number":4},{"id":"7938cc900f933631dd9feaad28040121fa75c291","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-12-10 20:36:21.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"85f499e4a2727d77b8f87666c71d216dcb194a3c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-10 21:43:35.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0cae391062e14d10b439d08be55fb985\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/149ada8265454627bcc5df79e2e58309 : SUCCESS in 57m 59s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5da20a60951840d99f539702e5a8ab43 : SUCCESS in 44m 17s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/020abfa1e8794ab69e4e6c6d97335ebd : SUCCESS in 58m 11s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a7bb31005a2e22394fff611bac8934983afc0070","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-10 22:36:28.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2536c2a8a54d4a12bb6807fd2ea1d11b\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/230c046797de4a218ef1abb46d5b2f54 : SUCCESS in 1h 49m 28s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/40a9d26529c74b45a5f3ffc1fcbb2e3c : SUCCESS in 1h 52m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ccb5904ee9914d9785f2c9926db3ab1d : SUCCESS in 1h 14m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2adf03a9155c4463aed1c5bf22fc4eb6 : SUCCESS in 10m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/53d63104c46d42c7b2c33f5b3c717f7e : FAILURE in 22m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/905acadf3cbf478e9be23494b263a42b : SUCCESS in 35m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/136f8b552aa2435e8bcb07505a6264b0 : SUCCESS in 12m 41s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/73a638322eb444eab0214634320d5506 : SUCCESS in 58m 19s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/2cee49b1644f472888c51f251752451c : SUCCESS in 48m 50s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/ff9283a9b6974addab2d67f7286ab7eb : SUCCESS in 1h 48m 12s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a6ec2dd752b640b3b9b2c150e3da965f : SUCCESS in 53m 06s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/2eb39d3edd804a9abab3aeaeedf2771c : SUCCESS in 36m 09s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/b88c2717632144d092745d0fe91318b6 : SUCCESS in 1h 47m 04s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2e0e52e4f1894e90a4f6ae85e0913ea6 : SUCCESS in 1h 43m 09s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/db80cf6192b243ceb860630025e4456d : SUCCESS in 58m 31s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/f39b262ce6ba4c1b9efc274f33a4e679 : SUCCESS in 1h 12m 02s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/fa62334add7940889fc2251a44b44f30 : SUCCESS in 39m 26s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3ae9ee4de3f14a4c8786df46c17f1742 : SUCCESS in 1h 12m 23s","accounts_in_message":[],"_revision_number":4},{"id":"9a925b02705ac5b1c19882a4b229cc02f2cd10cf","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-12-11 00:01:39.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a0668eb88ae0441e26a833442f758e8779b77c62","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-11 01:14:45.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92a2f1a3e491427cb03f4140a73e3dc4\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fd5f6c85699d40129d65341357ff46b3 : SUCCESS in 37m 14s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6215a44f872a4d358df088e115447290 : SUCCESS in 20m 37s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1e01566697a346f0b225f4c96728472c : SUCCESS in 1h 01m 17s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"6cbb3b9155e050625fe09c496a27702c1ddd3758","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-11 02:09:46.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f87d025f83ec429d8b7ac29978d9cbe1\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/66546189b81f4e988276911f53582b2c : SUCCESS in 1h 15m 28s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/fa9871ade83c4999b392698b9bde5315 : SUCCESS in 2h 04m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/06ebccd901a24fa589b6c0a99c78935d : SUCCESS in 1h 01m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f1d3cb8ca6e4c3f98344fc814768229 : SUCCESS in 10m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0e0fda172247456fb6798b327c4649e2 : SUCCESS in 20m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4a54763974b144f18d967339ed6d9daa : SUCCESS in 24m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2376eefa0c54a5fa1d4505e38f31bbf : SUCCESS in 12m 53s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/2c191cafff47442192d34bdce3c7e358 : SUCCESS in 57m 03s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/20bd7c319b884c4cbb51c1ddce64556b : SUCCESS in 1h 18m 15s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/ffe7f6837eb749c7ace75b1df4fe34cf : SUCCESS in 1h 08m 59s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b36b948bd2014b23ab0f39aad7978998 : SUCCESS in 43m 20s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/7c0cf001147b4a108c421db9cf8df276 : SUCCESS in 1h 09m 05s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/31bedcd612144457ba4689f590cade28 : SUCCESS in 1h 00m 41s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ed9d7044b6744499aded6b0ba9001ada : SUCCESS in 1h 40m 28s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/2a49078dea274c3a9175bea59b31d1f6 : SUCCESS in 53m 14s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/a43f5c9082ae463091f24664ab63638f : SUCCESS in 34m 30s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7539e91d4f97443ba978ac35fc16a7ca : SUCCESS in 59m 43s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f096d6c99d1145718b95cdf54a775e46 : SUCCESS in 1h 07m 31s","accounts_in_message":[],"_revision_number":4},{"id":"8ed544ba1db404c0c2039dd98a9c8eb7fcbacbaf","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-12-12 10:24:05.000000000","message":"Topic set to bug/2134387","accounts_in_message":[],"_revision_number":4},{"id":"15b4b452b9c9ab46c16315c7e26da8cb32cb23ba","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-12-12 10:35:12.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7f3436789b9b98a15df93a8274c40fab44ba29d2","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-12-12 10:40:41.000000000","message":"Patch Set 4: Review-Priority+1","accounts_in_message":[],"_revision_number":4},{"id":"d5a7b29af76fcbf61164834b4a3978ce05d8fa10","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-12-12 12:27:27.000000000","message":"Patch Set 4: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7a6c4e27b807f4b87031850a65d8560ec9fae20f","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-12-12 15:39:09.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d314b60c4dfc7421bb5bd77c6a9323e125f37e62","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-12-12 15:39:16.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"6ae5a5880f3ead38132b790a8af4290f72d1253e","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-12-12 15:39:58.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"fd2af398d4ed85d397742310232d83094020a777","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-12-12 17:05:02.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"105d49ac62c2cab6ac6e55ca223d1179b173575d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 17:15:25.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2df5e0c447e9447f9dbe495c34e70d9a\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3f433a1ad640457c8c8b992603b38620 : SUCCESS in 35m 40s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2cd0179192394c388a2b846a94f09aa9 : SUCCESS in 45m 04s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5a9401242ef9486da634bf6e739edef7 : SUCCESS in 48m 37s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"1620ed23a3eb3ea90d9397a33eb927e46f7b30b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 17:49:38.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a21d34b1bcff472aae76510655e53995\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/cd3f0889c92a446d98839350dd071033 : SUCCESS in 1h 29m 36s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/26b7bc23b1ad46528c567aac855642a8 : SUCCESS in 2h 09m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/171bf548d1d84429b36203da27a4b2f7 : SUCCESS in 1h 08m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a9dd4575e25412cb5b5f25073e7471e : SUCCESS in 11m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/212c99e0114544e2a848afe6cded0ce1 : SUCCESS in 32m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/94cb3bb943b842b191d16811e9a300e6 : SUCCESS in 34m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90d356d250f14fa3a991e8bd28d6da44 : SUCCESS in 13m 45s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/8ac85ef07d75446fbaa557ca6f8d1533 : SUCCESS in 44m 51s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/0077cf73680c4b4b87b360221118e033 : SUCCESS in 39m 03s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/271444deca4f4aa79df44259e2bc0ce6 : SUCCESS in 1h 29m 57s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/948b06aeea94402299105ba8ccc2623e : SUCCESS in 43m 23s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/6ba1183e4d5b48d78a672ae92cb15923 : SUCCESS in 1h 13m 26s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/9d45038f5ff04545b8df5f5e35df70d8 : SUCCESS in 1h 30m 31s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c47ffe363fac483891efcb7017ed4902 : SUCCESS in 1h 36m 50s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/fcdf739ea3be4edaa99aa89ed5631ba3 : SUCCESS in 1h 20m 27s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/b89c124415584ce6bfdfd57d58a48f45 : SUCCESS in 31m 26s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f5140f608a2b48b2bacb7c4a88230b41 : SUCCESS in 1h 26m 11s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e406b510492243f28ca3cbf23defe8a1 : SUCCESS in 1h 05m 28s","accounts_in_message":[],"_revision_number":5},{"id":"aacf6fd2eb4b2bb2fdd54509ab6da1cbc928b577","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-12-15 13:42:33.000000000","message":"Patch Set 5: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":5},{"id":"c9261fdbb062fefbe5da9af8838b14a7b6d67b31","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-12-16 08:30:42.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"7d88ef74b4f4fc795330ab496e28747057347f4e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-16 08:31:22.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"5e190c26bcb7ca68b16694cb60d61518359c08f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-16 10:28:05.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d649d859409247c1aef891046f0b6ead\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/588f5d2103de4bb0a9b69f17b1c53de0 : SUCCESS in 1h 21m 10s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/674ba143203d4ad28135df2ced5cd250 : SUCCESS in 1h 39m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6aa165e832145bcb8b071ce331b31a9 : SUCCESS in 11m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/650bef6cd01446bc90ee3473dfe52688 : SUCCESS in 34m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e3fffdf3da294c9eb8cbe9564a650cde : SUCCESS in 25m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/921629b910c54be2a56a4935ec1656cb : SUCCESS in 12m 25s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/abbf717356d14b7a825ce3fc4d5c1717 : SUCCESS in 1h 08m 38s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/6fc2903967e9407fa2538af20a059c54 : SUCCESS in 1h 09m 33s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/96cc3f8cda42470da76b9b99f065caa9 : SUCCESS in 1h 12m 59s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e010d1764974481787ae8f0a192b3034 : SUCCESS in 1h 05m 12s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/414ae5967ee04946bc17ca1c8915b377 : SUCCESS in 58m 04s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a6764af1c38d4ffe83d88a9b6c3bcef0 : SUCCESS in 1h 02m 55s","accounts_in_message":[],"_revision_number":5},{"id":"a889286c9a4c7eb2fea51f6d2f456fe0af105f0a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-16 10:28:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"1f5b6ece7bf41df6bfe426c1442ce0a2b4b7c530","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-16 10:30:53.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ac6f9f6e14a4acab541ec0c6eb6a966\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71172a02f74c4d04a26d8aaf9afd2430 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"4d196d1cd12182082e51f574626fd3002ac4c6b8","revisions":{"f79737a2b6f3184600739c83cedbdce3d264e732":{"kind":"REWORK","_number":1,"created":"2025-12-08 23:58:07.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/04/970204/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/04/970204/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/970204/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/970204/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/970204/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/04/970204/1"}}},"commit":{"parents":[{"commit":"a991dd1d6b0d38dd34a2f2a01cc191e496f32bb5","subject":"Fix maintenance LRP prefix adding fails over \u0027name\u0027 attribute missing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a991dd1d6b0d38dd34a2f2a01cc191e496f32bb5"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2025-12-08 23:50:12.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2025-12-08 23:50:12.000000000","tz":-300},"subject":"Check if original network is external in auto-allocate code","message":"Check if original network is external in auto-allocate code\n\nWhen using OSC to change the auto-allocate default value of\na network (is_default), the \u0027router:external\u0027 value is not\nsupplied. For example the request will just have:\n\n  {\u0027is_default\u0027: True}\nor\n  {\u0027is_default\u0027: False}\n\nThis causes the call to fast-exit, even when the change to\nthe network would/should have succeeded.\n\nWe should instead also check if there is an existing network,\nand verify the \u0027router:external\u0027 value of that before\nfast-exiting.\n\nBroken in [0] in 2025.1.\n\n[0] https://review.opendev.org/c/openstack/neutron/+/933132\n\nCloses-bug: #2134387\n\nChange-Id: I0f95712be44426fb5184702c1969d7fe4bf4147e\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f79737a2b6f3184600739c83cedbdce3d264e732"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f79737a2b6f3184600739c83cedbdce3d264e732"}]},"branch":"refs/heads/master"},"2ee305cf4cbeb57013d4a746f440d0045004f43b":{"kind":"REWORK","_number":2,"created":"2025-12-09 01:11:47.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/04/970204/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/04/970204/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/970204/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/970204/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/970204/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/04/970204/2"}}},"commit":{"parents":[{"commit":"a991dd1d6b0d38dd34a2f2a01cc191e496f32bb5","subject":"Fix maintenance LRP prefix adding fails over \u0027name\u0027 attribute missing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a991dd1d6b0d38dd34a2f2a01cc191e496f32bb5"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2025-12-08 23:50:12.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2025-12-09 01:11:30.000000000","tz":-300},"subject":"Check if original network is external in auto-allocate code","message":"Check if original network is external in auto-allocate code\n\nWhen using OSC to change the auto-allocate default value of\na network (is_default), the \u0027router:external\u0027 value is not\nsupplied. For example the request will just have:\n\n  {\u0027is_default\u0027: True}\nor\n  {\u0027is_default\u0027: False}\n\nThis causes the call to fast-exit, even when the change to\nthe network would/should have succeeded.\n\nWe should instead also check if there is an existing network,\nand verify the \u0027router:external\u0027 value of that before\nfast-exiting.\n\nBroken in [0] in 2025.1.\n\n[0] https://review.opendev.org/c/openstack/neutron/+/933132\n\nCloses-bug: #2134387\n\nChange-Id: I0f95712be44426fb5184702c1969d7fe4bf4147e\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2ee305cf4cbeb57013d4a746f440d0045004f43b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2ee305cf4cbeb57013d4a746f440d0045004f43b"}]},"branch":"refs/heads/master"},"6bda50745bdb9db4de23097003dcdf4d3178bad4":{"kind":"REWORK","_number":3,"created":"2025-12-09 22:04:01.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/04/970204/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/04/970204/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/970204/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/970204/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/970204/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/04/970204/3"}}},"commit":{"parents":[{"commit":"a991dd1d6b0d38dd34a2f2a01cc191e496f32bb5","subject":"Fix maintenance LRP prefix adding fails over \u0027name\u0027 attribute missing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a991dd1d6b0d38dd34a2f2a01cc191e496f32bb5"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2025-12-08 23:50:12.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2025-12-09 22:03:20.000000000","tz":-300},"subject":"Check if original network is external in auto-allocate code","message":"Check if original network is external in auto-allocate code\n\nWhen using OSC to change the auto-allocate default value of\na network (is_default), the \u0027router:external\u0027 value is not\nsupplied. For example the request will just have:\n\n  {\u0027is_default\u0027: True}\nor\n  {\u0027is_default\u0027: False}\n\nThis causes the call to fast-exit, even when the change to\nthe network would/should have succeeded.\n\nWe should instead also check if there is an existing network,\nand verify the \u0027router:external\u0027 value of that before\nfast-exiting.\n\nBroken in [0] in 2025.1.\n\n[0] https://review.opendev.org/c/openstack/neutron/+/933132\n\nCloses-bug: #2134387\n\nChange-Id: I0f95712be44426fb5184702c1969d7fe4bf4147e\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6bda50745bdb9db4de23097003dcdf4d3178bad4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6bda50745bdb9db4de23097003dcdf4d3178bad4"}]},"branch":"refs/heads/master"},"004b4c743b50236d13a39762f122a8751f1d2dd9":{"kind":"REWORK","_number":4,"created":"2025-12-10 16:04:03.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/04/970204/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/04/970204/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/970204/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/970204/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/970204/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/04/970204/4"}}},"commit":{"parents":[{"commit":"a991dd1d6b0d38dd34a2f2a01cc191e496f32bb5","subject":"Fix maintenance LRP prefix adding fails over \u0027name\u0027 attribute missing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a991dd1d6b0d38dd34a2f2a01cc191e496f32bb5"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2025-12-08 23:50:12.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2025-12-10 16:03:24.000000000","tz":-300},"subject":"Check if network is external in auto-allocate code","message":"Check if network is external in auto-allocate code\n\nWhen using OSC to change the auto-allocate default value of\na network (is_default), the \u0027router:external\u0027 value is not\nsupplied. For example the request will just have:\n\n  {\u0027is_default\u0027: True}\nor\n  {\u0027is_default\u0027: False}\n\nThis causes the call to fast-exit, even when the change to\nthe network would/should have succeeded.\n\nWe should instead also check if there is an existing network,\nand verify the \u0027router:external\u0027 value of that before\nfast-exiting.\n\nBroken in [0] in 2025.1.\n\n[0] https://review.opendev.org/c/openstack/neutron/+/933132\n\nCloses-bug: #2134387\n\nChange-Id: I0f95712be44426fb5184702c1969d7fe4bf4147e\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/004b4c743b50236d13a39762f122a8751f1d2dd9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/004b4c743b50236d13a39762f122a8751f1d2dd9"}]},"branch":"refs/heads/master"},"4d196d1cd12182082e51f574626fd3002ac4c6b8":{"kind":"REWORK","_number":5,"created":"2025-12-12 15:39:16.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/04/970204/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/04/970204/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/970204/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/970204/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/970204/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/04/970204/5"}}},"commit":{"parents":[{"commit":"a991dd1d6b0d38dd34a2f2a01cc191e496f32bb5","subject":"Fix maintenance LRP prefix adding fails over \u0027name\u0027 attribute missing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a991dd1d6b0d38dd34a2f2a01cc191e496f32bb5"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2025-12-08 23:50:12.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2025-12-12 15:36:33.000000000","tz":-300},"subject":"Check if network is external in auto-allocate code","message":"Check if network is external in auto-allocate code\n\nWhen using OSC to change the auto-allocate default value of\na network (is_default), the \u0027router:external\u0027 value is not\nsupplied. For example the request will just have:\n\n  {\u0027is_default\u0027: True}\nor\n  {\u0027is_default\u0027: False}\n\nThis causes the call to fast-exit, even when the change to\nthe network would/should have succeeded.\n\nWe should instead also check if there is an existing network,\nand verify the \u0027router:external\u0027 value of that before\nfast-exiting.\n\nBroken in [0] in 2025.1.\n\n[0] https://review.opendev.org/c/openstack/neutron/+/933132\n\nCloses-bug: #2134387\n\nChange-Id: I0f95712be44426fb5184702c1969d7fe4bf4147e\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4d196d1cd12182082e51f574626fd3002ac4c6b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4d196d1cd12182082e51f574626fd3002ac4c6b8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
