)]}'
{"id":"openstack%2Fneutron~970269","triplet_id":"openstack%2Fneutron~master~I2e096d2c85ecef0eb9cba920fb1af2d2fb0c0eb9","project":"openstack/neutron","branch":"master","topic":"bug/1886798","attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-12-10 16:02:19.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}}},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2025-12-10 12:42:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2e096d2c85ecef0eb9cba920fb1af2d2fb0c0eb9","subject":"Use NumaAffinityPoliciesEnumField from neutron-lib","status":"MERGED","created":"2025-12-09 15:33:21.000000000","updated":"2025-12-10 16:02:19.000000000","submitted":"2025-12-10 12:42:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"970269-bug/1886798","meta_rev_id":"9821e66a2f18ef37e5ad003b62130d24376e6899","_number":970269,"virtual_id_number":970269,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-10 12:42:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2025-12-10 10:33:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2025-12-09 17:47:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2025-12-10 10:33:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-09 15:34:03.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2025-12-09 17:09:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-12-09 17:47:03.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2025-12-09 17:56:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-10 10:33:10.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"7524114471353d6441637d02723f56b0e9c00001","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-12-09 15:33:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"309419677367377bfad54a6f0e38ad47fa2752d6","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-12-09 15:34:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"df2fc57956bcab97064239de25f98faf3ad126e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-12-09 16:05:02.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":2},{"id":"4711b60c86ae1490af1015b29b50daa1ab4c3a97","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-09 17:09:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7905db41f93342658084bcc046f0f659\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e49b2c266ea847d5bd4f3a5ad3595f62 : SUCCESS in 1h 01m 49s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2b81a2d1f7234c77874d80218a2c9b12 : SUCCESS in 50m 01s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/52821fde32e34e2b9b8806d5f1c43c04 : SUCCESS in 1h 01m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ea4e2a5e7bd866dce962536a34d1bdc62b2ccd79","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-12-09 17:47:03.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"8c6460c8b45d91650a92273b864116d920ec438d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-09 17:56:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9fff9fb139464884b5224d3f0919d349\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/3fdad924e00e47ad8d0dfae89347b16a : SUCCESS in 1h 28m 16s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b5291731520f4151a70448eb7258643c : SUCCESS in 1h 49m 39s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a9648b43c39949579df6e620da2605a1 : SUCCESS in 31m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e5ed5fdc05fb44a0a33b1cb598ee01e6 : SUCCESS in 1h 03m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6388d5929b340d0b4eb45eb2e4fd147 : SUCCESS in 27m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1064bedde764479fb7d979057bfe5bc7 : SUCCESS in 20m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/65e9d42cbfe7495987119aa7071e5154 : SUCCESS in 24m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f81c7b3800134b29af2083976f374a89 : SUCCESS in 7m 01s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/87dcbf0dfb0d4cb69b196e423346b8a6 : SUCCESS in 1h 01m 38s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/7e74599c97894b3e9ff542b027c81aa0 : SUCCESS in 37m 44s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/918e6b3f809c40548a50103c2355e7e6 : SUCCESS in 1h 32m 49s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/cadbe500f8064f61abfcd0290174b283 : SUCCESS in 1h 04m 40s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/88fa436f8c5740708069812538ccbff8 : SUCCESS in 39m 44s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/0cb92c9e88634718ad9b58a3823b75f3 : SUCCESS in 1h 25m 57s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/61adf3a315a94b99966e4e8ceb9a44cc : SUCCESS in 1h 39m 13s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/8da2fb9792934d579971ff89290965fe : SUCCESS in 1h 10m 30s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/f92ab0b05d6a4feb86634ef13dcbe8a6 : SUCCESS in 53m 20s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9112828ef3cf455a971527234013fbdc : SUCCESS in 39m 49s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/bc089d2d1e7d4e148d9c8898e97614f1 : SUCCESS in 1h 24m 33s","accounts_in_message":[],"_revision_number":2},{"id":"bc2373c7609be1bafd4a1243d97f7eba8e3f19a3","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-12-10 10:33:10.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"44fc3c2cf7aee9daf1f93bab4824bca5c4bc8716","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-10 10:33:31.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c88d442d96f371648648a709d54d4c9cca905e59","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-10 12:42:24.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb30db3581194925a9697224fb0ae3d0\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/fa23a0dccfa0404987060fbb40e4040a : SUCCESS in 1h 25m 32s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/99e6a8817eb34a02976c5d6ed7bd6742 : SUCCESS in 2h 08m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c86e3f8828a1439eb8925919b2992321 : SUCCESS in 25m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cfd583f4f0914dd78c50cbe5468cd854 : SUCCESS in 37m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/77221fec5a9c4ea5841d44f80c691eaa : SUCCESS in 37m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8084590da00454a9cdd827fb05e5436 : SUCCESS in 13m 45s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/5b8f5ae8e3a34e6abde8e02b67bf51a3 : SUCCESS in 1h 51m 52s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e30f0639fefd40ea911816a936be9a85 : SUCCESS in 1h 44m 40s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/e2713214d0e54ad3b09edec9ce3182d5 : SUCCESS in 1h 13m 22s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/479b5dedba3a4e22863fcc62a195b6af : SUCCESS in 1h 10m 14s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/9634f8a7f21143c496a6147287e20bef : SUCCESS in 1h 02m 55s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/97268546f9ba440db430331c6885745d : SUCCESS in 1h 26m 23s","accounts_in_message":[],"_revision_number":2},{"id":"6581b8152b31404d3817876ce2684fab477f6da5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-10 12:42:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"d3628fdff77eead9e312b6f4bd4b10bae7e31264","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-10 12:43:22.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/916e4e3fb260428fa660be459cf20a99\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7933e58085d0409f9bbaaa1cbe7cb27c : SUCCESS in 42s","accounts_in_message":[],"_revision_number":2},{"id":"9821e66a2f18ef37e5ad003b62130d24376e6899","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-12-10 16:02:19.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d251834cc6b883e5ca98d783318a11a429c80263","revisions":{"e0d99042ef58f1f4213fe2564858db7a0bfd71f5":{"kind":"REWORK","_number":1,"created":"2025-12-09 15:33:21.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/69/970269/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/69/970269/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/69/970269/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/69/970269/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/69/970269/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/69/970269/1"}}},"commit":{"parents":[{"commit":"7fa7737947092c77cb2c210fd18c5adbb81e5223","subject":"Merge \"Don\u0027t prevent setting CIDRs as allowed_address_pair for port\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7fa7737947092c77cb2c210fd18c5adbb81e5223"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-09 15:32:05.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-09 15:32:49.000000000","tz":540},"subject":"Use NumaAffinityPoliciesEnumField from neutron-lib","message":"Use NumaAffinityPoliciesEnumField from neutron-lib\n\nIt was added in neutron-lib 2.21.0 [1].\n\n[1] 096fb92766007a6cdb01d94d26217c54c1039b30\n\nRelated-Bug: #1886798\nChange-Id: I2e096d2c85ecef0eb9cba920fb1af2d2fb0c0eb9\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e0d99042ef58f1f4213fe2564858db7a0bfd71f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e0d99042ef58f1f4213fe2564858db7a0bfd71f5"}]},"branch":"refs/heads/master"},"d251834cc6b883e5ca98d783318a11a429c80263":{"kind":"REWORK","_number":2,"created":"2025-12-09 16:05:02.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/69/970269/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/69/970269/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/69/970269/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/69/970269/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/69/970269/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/69/970269/2"}}},"commit":{"parents":[{"commit":"7fa7737947092c77cb2c210fd18c5adbb81e5223","subject":"Merge \"Don\u0027t prevent setting CIDRs as allowed_address_pair for port\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7fa7737947092c77cb2c210fd18c5adbb81e5223"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-09 15:32:05.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-09 15:54:59.000000000","tz":540},"subject":"Use NumaAffinityPoliciesEnumField from neutron-lib","message":"Use NumaAffinityPoliciesEnumField from neutron-lib\n\nIt was added in neutron-lib 2.21.0 [1].\n\n[1] 096fb92766007a6cdb01d94d26217c54c1039b30\n\nRelated-Bug: #1886798\nChange-Id: I2e096d2c85ecef0eb9cba920fb1af2d2fb0c0eb9\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d251834cc6b883e5ca98d783318a11a429c80263"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d251834cc6b883e5ca98d783318a11a429c80263"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
