)]}'
{"id":"openstack%2Fneutron~970407","triplet_id":"openstack%2Fneutron~master~I179abea54d81b31a1627e3e5054f814018ac774f","project":"openstack/neutron","branch":"master","topic":"review/970407","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2025-12-17 08:00:33.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2025-12-18 11:26:03.000000000","reason":"\u003cGERRIT_ACCOUNT_13861\u003e replied on the change","reason_account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2025-12-18 14:29:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I179abea54d81b31a1627e3e5054f814018ac774f","subject":"Refactor ovn-db-sync script","status":"MERGED","created":"2025-12-10 15:22:04.000000000","updated":"2025-12-18 14:30:41.000000000","submitted":"2025-12-18 14:29:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"970407-review/970407","meta_rev_id":"d5eea924f435c2ef7d94d9b9bfcad28b15c5b82b","_number":970407,"virtual_id_number":970407,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-18 14:29:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2025-12-17 08:00:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2025-12-18 11:29:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2025-12-18 11:29:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}],"REVIEWER":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-10 16:41:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-12-10 17:30:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-10 18:07:33.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2025-12-11 10:53:24.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2025-12-12 10:13:43.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2025-12-18 09:40:33.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"CC"},{"updated":"2025-12-18 11:29:01.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"}],"messages":[{"id":"f5aeeeb881e86dde46a741fa5cf2d29e311ce8d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-12-10 15:22:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ef02ba3d38bc10d935dfe16080988efe704f9f03","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-10 16:41:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee5693bebdd4468f909ea90b1e80c415\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/97fc73187f2041efb021281ce8b3695a : SUCCESS in 1h 16m 48s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c69f2f418dc142219758c44f94b5e9a0 : SUCCESS in 59m 51s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d2d83e9775af4c64bac357deadf54d0d : SUCCESS in 1h 15m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5d1a1f305a6c317ef3b8f2304c39cd8f5233c286","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-10 17:30:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35a6cf9eefb4426488e550e0873879d2\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/55f6c57089c54d8ca6966d0ad8827365 : SUCCESS in 1m 34s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b4f9727e02ed4f698cbb46ac1fe5fbe4 : SUCCESS in 1h 21m 39s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b9624eeedeb64efe9bd5c6f0d5c1e35e : SUCCESS in 2h 07m 57s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c8295432fc4b4446b1d39c6170577030 : SUCCESS in 31m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aef3d24458ed423890bce05e455a9403 : SUCCESS in 1h 17m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2abb8d2ddd34a6e80be5e3429e328b5 : SUCCESS in 10m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f1deaf11982f44f7bd42145379effd01 : SUCCESS in 26m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/da627cfe84134c9eb204b29cbdcd0b2f : SUCCESS in 24m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4627d72cb8cc49898d90c3ba6d205b1d : SUCCESS in 5m 57s\n- requirements-check https://zuul.opendev.org/t/openstack/build/25ed5f1423ac426db36345b4a4c56025 : SUCCESS in 1m 55s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/0dda358aa5264cdca1643e2bbdbf1277 : SUCCESS in 1h 01m 29s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/ebea13c6010c4c9db64d322bcf9b7f34 : SUCCESS in 1h 01m 28s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/ec186f6a846f432c8379a18e5cf931cf : SUCCESS in 1h 00m 14s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/4c4d0d479d6f4fe3a69a026654fbe5fe : SUCCESS in 1h 06m 41s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/655abb4dae40484caee7a92665865fe7 : SUCCESS in 40m 41s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/1a947fe578fc4fd5abe74ca7e8cfb28c : SUCCESS in 1h 06m 14s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6f95ef653b4443fc877edebeda976526 : SUCCESS in 1h 20m 36s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/b9fbeadacf5c46cdb6fbf4ec55360784 : SUCCESS in 46m 06s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/40a98b401e6f4ae480c9b57387a3f1e0 : SUCCESS in 55m 50s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/13058c3e4d744c79bb7fdefe101a8c53 : SUCCESS in 1h 24m 20s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/45a4599a98c74847a0ab7fd14c084f63 : SUCCESS in 1h 08m 23s","accounts_in_message":[],"_revision_number":1},{"id":"89cdc441afcd6a578d9b05d9d96398dd8975c59e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-12-11 09:50:34.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"863570cfc2749761c0e55e68f421f89a77c30f3e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-12-11 10:23:36.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"49e3f0b2f8496caf3fac9a353d594067c4b7e9fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-12-11 11:12:30.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"c8f93e318033e034c7be3e7d74501d8184781754","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-11 12:19:16.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/315f6277608748e79abc29a57b6bd6b7\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8020bf0d13e24358b732a4b130e73836 : FAILURE in 1h 04m 16s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1ebbdd8fe9034dcbbb9ba973a3d69493 : FAILURE in 38m 36s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b58e34e7f7534419abe8a271b45bc354 : FAILURE in 1h 02m 50s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f9ae2987d2af562031298128acb6441f6a81fd85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-11 13:00:26.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d7fd52bc478d418ba6994f3cc4723837\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a191f8798ba34ff59e864a7ae7e7a65b : SUCCESS in 2m 37s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/3371b8fa70754f9998ccbdb14193291e : SUCCESS in 1h 34m 12s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f6eee1b179264dd886b5c8f968d94274 : SUCCESS in 1h 46m 54s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7d0dfad47f164398adf9ee72ef92d404 : SUCCESS in 39m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/157da93f15c949baba6e6053aaee5db7 : FAILURE in 1h 01m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9022130282f349ed82e78d8c1d57fd5a : FAILURE in 3m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3f3b4417ba8f4672800ce3ba283fea51 : FAILURE in 39m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1f4f282835ec431c866ccc1318c06688 : FAILURE in 28m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3783a92352e84fafbd944cce123ba687 : SUCCESS in 13m 44s\n- requirements-check https://zuul.opendev.org/t/openstack/build/9907bcc11fe7440599f23b1ebb67741b : SUCCESS in 2m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c586f6363cce47fe823c931ae9f59bbc : SUCCESS in 21m 58s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/54be3c38dda5409c8dde6db228a899bf : SUCCESS in 1h 05m 34s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/1c96cb85747f40baa28aa6e7d942d3dd : SUCCESS in 1h 19m 53s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/6d2466bb95ac4d67a29134da46b7f283 : SUCCESS in 1h 26m 23s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/103ee4e9a8ae4532b476ac04314e1804 : SUCCESS in 1h 09m 55s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/2138062d4e7346909daac49ec60821dc : SUCCESS in 57m 46s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/7c5ba40428c54bc290716c3081b3c6b1 : SUCCESS in 1h 36m 00s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d6a25e2c725842908bf6f7933921e8bc : SUCCESS in 52m 16s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/2d4d926355524defa053ae6de6f9dff3 : SUCCESS in 1h 01m 08s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/39daae150410487ebb66d5ac409ac284 : SUCCESS in 57m 10s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0f1b1e2391dd4204ab0bbdbe717d7651 : SUCCESS in 36m 27s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/059b519275cc49439b629ae88d46513c : SUCCESS in 1h 14m 10s","accounts_in_message":[],"_revision_number":4},{"id":"20db3be2bb3c1c913045a37a7880847c5f7433cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-12-11 14:14:45.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"13a2824e1d364cd90aac2da27298db794632b8db","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-11 15:18:44.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85c8bf473a9847a98d764c3eefb5627b\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/74d5bb3b780445adaa87361341001f6a : SUCCESS in 46m 27s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d50ddca3860045b3b2e261dedf6f1202 : SUCCESS in 22m 47s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2275262847b8421a938a4c6d638ab42d : SUCCESS in 25m 15s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"bba6e74d7b00a160595eead1cd57887307f613c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-11 16:25:38.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba6a20e25b5b42398d73039a5e9d11ce\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/35f4ab323ae94a2ab4628317ade7ada7 : SUCCESS in 3m 01s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/42825667fa6843a58af3ed369a0cbdbc : SUCCESS in 1h 26m 12s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/5bd2174ce7e2453ea40cb8d6acb6f6e3 : SUCCESS in 2h 03m 36s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5aa96164a0c740938c706cb97542d14f : SUCCESS in 33m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f41ad38ca06548428716f465a42bd129 : SUCCESS in 36m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45a15bbcfae9440ea887117dde97a50d : SUCCESS in 26m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0b26a75d3ae547a6bd1790b95799e62e : SUCCESS in 20m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/11728d3ebb9b494bac9b24f392f4321d : SUCCESS in 41m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12d7fb0d57784deb91456c3d2ab9dac7 : SUCCESS in 12m 40s\n- requirements-check https://zuul.opendev.org/t/openstack/build/38d3e891ae344ce9bf65ad976e4ba08f : SUCCESS in 4m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/820d2940bb594a4e8c744fcc64cfe2ab : SUCCESS in 20m 52s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/a2f2d206452b474abf43db699dfaa417 : SUCCESS in 1h 06m 27s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/2511d85bc53e4f3795071a2c0f6cee65 : SUCCESS in 58m 51s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/84bf534b439d4cdcb9eeee369819fd6f : SUCCESS in 59m 27s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c460c079dd2340d3888455947196181c : SUCCESS in 1h 12m 38s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/3e9011239ab44328af2211f8b16addb2 : SUCCESS in 1h 06m 55s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/537118b523be49af8de8fe862a9282a3 : SUCCESS in 2h 02m 47s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8ac34814bf9a408c9afb01aa5ffc475f : SUCCESS in 55m 47s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/81524d54887243b0a2269049eff63815 : SUCCESS in 1h 00m 32s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/c14e6c1de4ca4ca691ec66b78f6b2e80 : SUCCESS in 1h 19m 01s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9cbf5cb5501344e3bac164d362afca4f : SUCCESS in 1h 26m 00s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/b7dd97ab48624aeb8b2a527b5a57d7bb : SUCCESS in 1h 25m 17s","accounts_in_message":[],"_revision_number":5},{"id":"85ee2221759d388b76dd533940459a9ad5a70f34","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-12-12 10:13:43.000000000","message":"Patch Set 5: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"d4126e9e376b8d3bb2aee84a299db909a69718c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-12-15 15:17:13.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"4552bd33840b5518575790e37442c1d5a6865573","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-12-15 16:04:48.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a883e2b0d4214a1db956cf6262eb4e79e325b129","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-15 16:47:44.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f48705b4c7ff47bf86c4a9fa1c90b569\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fa8643a3a9d6472ba6c9e5aca2646166 : SUCCESS in 48m 12s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d16074cbdcf14efaa8be097b5d1477d3 : SUCCESS in 38m 49s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b5aa13a1d8ba47229b94fd88539a667f : SUCCESS in 1h 13m 38s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"46e7428332e827d47f9f82ed1a47c38380c5325e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-15 17:42:10.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5520ca995ac41f9b4651392e137cb84\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8a2afe1ebfe34e57bf3e815254b124e0 : SUCCESS in 2m 46s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c42910c9b29f4e79b065d7a192e52519 : SUCCESS in 1h 49m 26s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a8b6ab84d51b4e9b97f932b59fa83028 : SUCCESS in 2h 23m 39s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f85d942df96c4bb3baf7421ec4370de9 : SUCCESS in 33m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f01fca66813c452a9456c23da2044bda : SUCCESS in 57m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81c5c345cdd84ff2bb5fbf98e48d1776 : SUCCESS in 26m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/654fb4a4fd9048059c26351a48c344bd : SUCCESS in 44m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/24bf03d1b0934371a03593285b1e5be6 : SUCCESS in 22m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/864889dbadde4161b19d924884eb48b5 : SUCCESS in 12m 10s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a023cdbfcc2a4ef6b7d083c3c9b36d89 : SUCCESS in 2m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c917077c745949d4bbe05f3f1ecc1534 : SUCCESS in 23m 05s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/ccf1245daec84d33bcce182009999ac0 : SUCCESS in 35m 11s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/355a53e941d5429e996a33b28dc4ecd2 : SUCCESS in 51m 15s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c19c598e7ce545839047869ea11eaaba : SUCCESS in 1h 22m 54s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/4a7a60457d414d57a3b4cac54813b00a : SUCCESS in 1h 09m 05s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/5f12f7b6fced4c44ab6cdf582b4b23cb : SUCCESS in 1h 04m 43s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/8757823417b04434815e3054e54b4f46 : SUCCESS in 1h 03m 35s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/982b159f0b834017911e47b699c13ce9 : SUCCESS in 1h 33m 35s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/83d8d4a39b0e4b63804366f5a697f42b : SUCCESS in 53m 03s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/53b170c31dac4ed9b428af6a364e53eb : SUCCESS in 1h 11m 27s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6a87b106976248f7a7d5641c4f9b005a : SUCCESS in 42m 55s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/37504251406844d6ac4a39db74f3ba50 : SUCCESS in 1h 09m 09s","accounts_in_message":[],"_revision_number":6},{"id":"0efa1d33c096ee623cabc7ad64bc703ea0044057","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-12-17 08:00:33.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"03014a367c32612aaeb4d591973a50857a46d0a3","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-12-18 09:40:33.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"737fe1c0ef357a749a7d1643c0fd15c364f2c6b7","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-12-18 09:47:22.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"7c7cf7a171fcdfcdbc4245ba36044a84e74f8496","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-12-18 11:26:03.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"1608d711466482cd26ce7c9b4abbdae8002f7cbf","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-12-18 11:29:01.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"53fae7c81d32672c3a3410a01d3c5bca8b97dc06","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-18 11:29:22.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"0d1e062e13d3e11530a9bde107f21960c622e8d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-18 14:29:01.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee9fae17fd7f4d36baef32f11361e1b0\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0d71a30d8528442391c448c012e98677 : SUCCESS in 3m 05s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/2d473bcc534a41989363491ed855533a : SUCCESS in 1h 38m 53s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ee6583c2a8854e8cbb73f9c8645dfa82 : SUCCESS in 2h 58m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d85e1cff9f447019757298fcbd14b46 : SUCCESS in 11m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/581f9f16e35f4c0ab37741f975deccfd : SUCCESS in 32m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f024747975e24ed8a324160ec04ebbcf : SUCCESS in 42m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae0301584a584a12b1f326a367a692af : SUCCESS in 7m 20s\n- requirements-check https://zuul.opendev.org/t/openstack/build/5e07319208d848d99c24c756d46b65cd : SUCCESS in 2m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b351b3bf7d1544e1b51485d2e2825265 : SUCCESS in 29m 26s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/26428d9c0699474a85ae1bccb4e966d0 : SUCCESS in 1h 46m 07s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/32dfa092cede47c29fc3256517469aa1 : SUCCESS in 1h 49m 23s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/3d2e19de4e1a4108a572287697837459 : SUCCESS in 1h 10m 12s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/328639f4c9344b01ab6f43a3c4d73bd7 : SUCCESS in 50m 49s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/1d3f15354fa64e4fb49f864a42ad7c4d : SUCCESS in 1h 04m 33s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/ba54aad9b202493d83b3e88eda1c9b1d : SUCCESS in 1h 18m 37s","accounts_in_message":[],"_revision_number":6},{"id":"f3c21e816ffb396d5739cb7c903dcf8945f6111b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-18 14:29:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"d5eea924f435c2ef7d94d9b9bfcad28b15c5b82b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-18 14:30:41.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa9838a3b09f436aba71879227fd1910\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd37104faba34d9586d2bec6901f100e : SUCCESS in 51s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/748fab81a72945d6b739fd9db3d9c4c1 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"2faba4ede3291d49c1f74afb66fe93a73e3007ef","revisions":{"fb7acce0054ba4d87bea0f050354c87bd6c4f16a":{"kind":"REWORK","_number":1,"created":"2025-12-10 15:22:04.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/07/970407/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/07/970407/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/07/970407/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/07/970407/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/07/970407/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/07/970407/1"}}},"commit":{"parents":[{"commit":"4d2c562d2250a02b88524fcddae2066b1b17abe2","subject":"Merge \"Fix ignored metadata options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4d2c562d2250a02b88524fcddae2066b1b17abe2"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2025-12-10 14:38:20.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2025-12-10 15:22:01.000000000","tz":60},"subject":"[WIP] Refactor ovn-db-sync script","message":"[WIP] Refactor ovn-db-sync script\n\nThis patch refactors the ovn-db-sync CLI tool provided by Neutron in the\nway that actual sync tasks are now made as \"plugins\" which can be loaded\nthrough the entry_points. With that change stadium or 3rd party projects\ncan add their own ovn-db-sync tasks and Neutron will load them together\nwith tasks implemented in Neutron directly.\n\nTODO:\n- migration mode\n- release note\n\nCloses-bug: #2130570\n\nChange-Id: I179abea54d81b31a1627e3e5054f814018ac774f\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fb7acce0054ba4d87bea0f050354c87bd6c4f16a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fb7acce0054ba4d87bea0f050354c87bd6c4f16a"}]},"branch":"refs/heads/master"},"35a954b642e30484d8543793bfb702371b3208b7":{"kind":"REWORK","_number":2,"created":"2025-12-11 09:50:34.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/07/970407/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/07/970407/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/07/970407/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/07/970407/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/07/970407/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/07/970407/2"}}},"commit":{"parents":[{"commit":"4d2c562d2250a02b88524fcddae2066b1b17abe2","subject":"Merge \"Fix ignored metadata options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4d2c562d2250a02b88524fcddae2066b1b17abe2"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2025-12-10 14:38:20.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2025-12-11 09:47:29.000000000","tz":60},"subject":"[WIP] Refactor ovn-db-sync script","message":"[WIP] Refactor ovn-db-sync script\n\nThis patch refactors the ovn-db-sync CLI tool provided by Neutron in the\nway that actual sync tasks are now made as \"plugins\" which can be loaded\nthrough the entry_points. With that change stadium or 3rd party projects\ncan add their own ovn-db-sync tasks and Neutron will load them together\nwith tasks implemented in Neutron directly.\n\nTODO:\n- migration mode\n- release note\n\nCloses-bug: #2130570\n\nChange-Id: I179abea54d81b31a1627e3e5054f814018ac774f\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/35a954b642e30484d8543793bfb702371b3208b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/35a954b642e30484d8543793bfb702371b3208b7"}]},"branch":"refs/heads/master"},"93cc223eb0619c9830eec2869ceb56e0d6b8de44":{"kind":"REWORK","_number":3,"created":"2025-12-11 10:23:36.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/07/970407/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/07/970407/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/07/970407/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/07/970407/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/07/970407/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/07/970407/3"}}},"commit":{"parents":[{"commit":"4d2c562d2250a02b88524fcddae2066b1b17abe2","subject":"Merge \"Fix ignored metadata options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4d2c562d2250a02b88524fcddae2066b1b17abe2"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2025-12-10 14:38:20.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2025-12-11 10:23:11.000000000","tz":60},"subject":"[WIP] Refactor ovn-db-sync script","message":"[WIP] Refactor ovn-db-sync script\n\nThis patch refactors the ovn-db-sync CLI tool provided by Neutron in the\nway that actual sync tasks are now made as \"plugins\" which can be loaded\nthrough the entry_points. With that change stadium or 3rd party projects\ncan add their own ovn-db-sync tasks and Neutron will load them together\nwith tasks implemented in Neutron directly.\n\nTODO:\n- release note\n- docs\n\nCloses-bug: #2130570\n\nChange-Id: I179abea54d81b31a1627e3e5054f814018ac774f\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/93cc223eb0619c9830eec2869ceb56e0d6b8de44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/93cc223eb0619c9830eec2869ceb56e0d6b8de44"}]},"branch":"refs/heads/master"},"38a4bd93d50234dd05adf68dd9cf5182158ea0c5":{"kind":"REWORK","_number":4,"created":"2025-12-11 11:12:30.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/07/970407/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/07/970407/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/07/970407/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/07/970407/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/07/970407/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/07/970407/4"}}},"commit":{"parents":[{"commit":"4d2c562d2250a02b88524fcddae2066b1b17abe2","subject":"Merge \"Fix ignored metadata options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4d2c562d2250a02b88524fcddae2066b1b17abe2"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2025-12-10 14:38:20.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2025-12-11 11:12:05.000000000","tz":60},"subject":"Refactor ovn-db-sync script","message":"Refactor ovn-db-sync script\n\nThis patch refactors the ovn-db-sync CLI tool provided by Neutron in the\nway that actual sync tasks are now made as \"plugins\" which can be loaded\nthrough the entry_points. With that change stadium or 3rd party projects\ncan add their own ovn-db-sync tasks and Neutron will load them together\nwith tasks implemented in Neutron directly.\n\nAssisted-by: Claude Code\n\nCloses-bug: #2130570\n\nChange-Id: I179abea54d81b31a1627e3e5054f814018ac774f\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/38a4bd93d50234dd05adf68dd9cf5182158ea0c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/38a4bd93d50234dd05adf68dd9cf5182158ea0c5"}]},"branch":"refs/heads/master"},"816d076515054a73aa63ee646c0910ed242db5e7":{"kind":"REWORK","_number":5,"created":"2025-12-11 14:14:45.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/07/970407/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/07/970407/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/07/970407/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/07/970407/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/07/970407/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/07/970407/5"}}},"commit":{"parents":[{"commit":"4d2c562d2250a02b88524fcddae2066b1b17abe2","subject":"Merge \"Fix ignored metadata options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4d2c562d2250a02b88524fcddae2066b1b17abe2"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2025-12-10 14:38:20.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2025-12-11 14:14:33.000000000","tz":60},"subject":"Refactor ovn-db-sync script","message":"Refactor ovn-db-sync script\n\nThis patch refactors the ovn-db-sync CLI tool provided by Neutron in the\nway that actual sync tasks are now made as \"plugins\" which can be loaded\nthrough the entry_points. With that change stadium or 3rd party projects\ncan add their own ovn-db-sync tasks and Neutron will load them together\nwith tasks implemented in Neutron directly.\n\nAssisted-by: Claude Code\n\nCloses-bug: #2130570\n\nChange-Id: I179abea54d81b31a1627e3e5054f814018ac774f\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/816d076515054a73aa63ee646c0910ed242db5e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/816d076515054a73aa63ee646c0910ed242db5e7"}]},"branch":"refs/heads/master"},"2faba4ede3291d49c1f74afb66fe93a73e3007ef":{"kind":"REWORK","_number":6,"created":"2025-12-15 15:17:13.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/07/970407/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/07/970407/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/07/970407/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/07/970407/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/07/970407/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/07/970407/6"}}},"commit":{"parents":[{"commit":"4d2c562d2250a02b88524fcddae2066b1b17abe2","subject":"Merge \"Fix ignored metadata options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4d2c562d2250a02b88524fcddae2066b1b17abe2"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2025-12-10 14:38:20.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2025-12-15 15:05:03.000000000","tz":60},"subject":"Refactor ovn-db-sync script","message":"Refactor ovn-db-sync script\n\nThis patch refactors the ovn-db-sync CLI tool provided by Neutron in the\nway that actual sync tasks are now made as \"plugins\" which can be loaded\nthrough the entry_points. With that change stadium or 3rd party projects\ncan add their own ovn-db-sync tasks and Neutron will load them together\nwith tasks implemented in Neutron directly.\n\nAssisted-by: Claude Code\n\nCloses-bug: #2130570\n\nChange-Id: I179abea54d81b31a1627e3e5054f814018ac774f\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2faba4ede3291d49c1f74afb66fe93a73e3007ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2faba4ede3291d49c1f74afb66fe93a73e3007ef"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
