)]}'
{"id":"openstack%2Fneutron~970646","triplet_id":"openstack%2Fneutron~stable%2F2025.2~I2bc635e80f23d38683ae635ac15f4fc3e7a79b73","project":"openstack/neutron","branch":"stable/2025.2","topic":"bug/2134703","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2025-12-16 16:07:09.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2bc635e80f23d38683ae635ac15f4fc3e7a79b73","subject":"[OVN] Fix the `_gen_router_port_ext_ids` call","status":"MERGED","created":"2025-12-11 16:00:14.000000000","updated":"2025-12-16 16:08:37.000000000","submitted":"2025-12-16 16:07:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"970646-bug/2134703","meta_rev_id":"3c9f3ff17eb5c98334db4580e097085601dfeb82","_number":970646,"virtual_id_number":970646,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-16 16:07:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2025-12-16 14:05:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2025-12-15 06:35:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2025-12-16 14:05:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-11 17:02:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-12-11 18:02:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-13 01:41:15.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2025-12-16 14:05:47.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"b4b1485b45d35d033364cf59740306aecbae903e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-12-11 16:00:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"70aec17c8f57fc0d2abb7d9731f00daa2d96fc32","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-12-11 16:00:22.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"17af318f6988b9b0d6e515478460c92d253c493f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-11 17:02:33.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/83509b0a116a433482cba0caafddc560\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4cb4d91a7e7b47beb1b741cc4118f7fd : FAILURE in 24m 18s\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/780e35482cf547039187b623e75e20e4 : FAILURE in 36m 02s\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1a8ad8ab183d420e8d769313fb0b53ce : FAILURE in 18m 20s","accounts_in_message":[],"_revision_number":1},{"id":"8030a2466736e49de5fcbdcddc4dce538a25ffa1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-11 18:02:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5661eb9a2d5f44f3a83e7f263393496b\n\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/6195d6f6a2e742c08c72e98cdb281d95 : SUCCESS in 2h 01m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8ff111b2070b4d6f9c3b6abba9dd76b7 : FAILURE in 26m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d761ccee7c34c41b9577fac8d5b825e : FAILURE in 5m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/13d0f26c851f4f5eb149ce646f7876f2 : FAILURE in 16m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a57006a124ac4dc5a5265238ec1bf261 : FAILURE in 19m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b84c8b4fc49e4c0bb25969474fbfa5e3 : FAILURE in 22m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dff13f2cc67d4677b9d77efdc7f8fb05 : SUCCESS in 9m 35s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/be5f08f6741b443e86e1990a6802dfa3 : SUCCESS in 1h 41m 39s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/b0630fa297394ea8bdf8f28cbe40b7a5 : FAILURE in 18m 42s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a585db373cb5483ab4494bc3315f3639 : SUCCESS in 1h 02m 01s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ffa7be39fc7f4e03929de81be53489e3 : SUCCESS in 32m 06s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/6b1e3dc8eabf453b9a9833ce9c3fad01 : SUCCESS in 1h 15m 39s","accounts_in_message":[],"_revision_number":1},{"id":"8c9012de26d49014c2f325c23fa5349c86b67feb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-12-12 08:43:21.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"b31e8a4ad9c90868a563a2aeda9f84718380e2e8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 10:27:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30d6f18b17134bcca91b95549af2f154\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/82a8ce54bc744aefa6ffd17a5bfd7345 : SUCCESS in 23m 09s\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/67025c4492db469eae31c9df297d0be5 : SUCCESS in 35m 10s\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4bc327c988d64993961106301336621b : SUCCESS in 41m 46s","accounts_in_message":[],"_revision_number":2},{"id":"7e749e626ee40b8f2a294fc62b2ae8e56bcf85fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 11:07:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/88fa8f295bcb4b79841a7679bc3ab3bd\n\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/5338297c08f64f0196791c9eac520f02 : SUCCESS in 2h 17m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/448f25a9a3c242e6b2b599d344a31fab : SUCCESS in 34m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9b9145a07c84a1d8aeb7f0a34d7c811 : SUCCESS in 14m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a061b7b5a9ea42b98dba17e7f983dce6 : SUCCESS in 21m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fb9196fe6ff44b4fbc18a8d563308031 : SUCCESS in 18m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/32b9392910234a3dbe7b69180c94ab57 : SUCCESS in 18m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1baa5fcd85a642f28a75f45910a4982e : SUCCESS in 6m 26s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/aff4596651a74eb486a89ba61e58b5f9 : SUCCESS in 1h 33m 33s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/0381fc34b70a4117bd731e03c13e47b3 : SUCCESS in 36m 51s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/f25d2659528048c783b822729e158240 : SUCCESS in 50m 04s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b05dfc74822c4bcdb9f3ebc8690afaf2 : SUCCESS in 1h 21m 31s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/90bceab46a1a403c929de25d282a2397 : SUCCESS in 1h 15m 41s","accounts_in_message":[],"_revision_number":2},{"id":"b424a8ef4a39b8215558d44507206f7b2d645347","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-12-12 16:41:36.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"035b4dee3452305e1dca3f04f33ea468b4197e08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 18:28:34.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73535f0fe61e4bf782c3839ad00b55b7\n\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/84942b11bcf4409cab703ee61753bfbe : SUCCESS in 1h 45m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e3ea3c264a82402fabca2fb277405da2 : SUCCESS in 47m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d23b5ace2b94225852dc7a1c14c1734 : SUCCESS in 5m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8969475bdcd54e9589cd8f1c1f1a8341 : SUCCESS in 22m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/478ed86066d94add8a53283507d33d24 : SUCCESS in 20m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2ebac90283624baab2a52b9fda90a769 : SUCCESS in 24m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a65dc4f5e4a54738bc6aa13734a0c633 : SUCCESS in 12m 06s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/ba40a543c2984392b3d408b6e082c146 : SUCCESS in 1h 32m 16s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/305a9130b37a46068634fb17ec740a5a : SUCCESS in 1h 06m 35s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/3380c7eecf094401ae7816fcb0c1b65e : SUCCESS in 1h 08m 03s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/75872e02ba554c0392afb93b7d3bd0be : SUCCESS in 1h 18m 11s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/eea298ef594549bc8b8375bbc3d2ae70 : SUCCESS in 1h 24m 17s","accounts_in_message":[],"_revision_number":3},{"id":"85ec04f36804f27fb128e5b0fdd0d55ea4e69fa6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 19:12:38.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9bd8e3feab9e4e988979f5277720c1a0\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5116fef489c5419ab4e008f0f643990b : SUCCESS in 38m 27s\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2f1ecfd48a724228b1ed0034b18736fa : NODE_FAILURE Node(set) request 9da9d252e23542d09224d54c95ee2386 failed in 0s\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/989db7dc32784e9591c5d09af66fabf0 : SUCCESS in 31m 01s","accounts_in_message":[],"_revision_number":3},{"id":"9531921f60c57dda6187b0fcdac9f47f710459ec","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-12-13 01:41:15.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"4312e94314c7cf5d67d5763ab6d71dde6414ae46","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-12-15 06:35:38.000000000","message":"Uploaded patch set 4: New patch set was added with same tree, parent tree, and commit message as Patch Set 3.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"0e9e71caf1c3699bc8fbad4c9d4b35a122ee93e1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-15 07:18:46.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f53822b0f244645be17dd97af88fd72\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/847bbca10bb640268470301cc767f6a1 : SUCCESS in 41m 31s\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f2913a16624742e68fcd8d77a934cc7f : SUCCESS in 36m 52s\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/69fc3f6339104bfc9e014e8a8fa2ecee : SUCCESS in 24m 56s","accounts_in_message":[],"_revision_number":4},{"id":"172d304876070cbc7ea483911ee3a2232acdad92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-15 08:45:29.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/17a387806b914cf1a8fb8856bbccd5d2\n\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/812b2c15860147c3940cd7f402444b0e : SUCCESS in 2h 01m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0af6973c1e58479a899f3f12aafa7e90 : SUCCESS in 45m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/115959c6189c4146ac5f747548c002e9 : SUCCESS in 13m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eb5744baefd14a80b81ae96dc38e33c1 : SUCCESS in 21m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/572f07dd0be843099063f93bdc052eb5 : SUCCESS in 22m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/021538a547684ad4b5f785e9984097bd : SUCCESS in 24m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9da69c92932415f9078a8fe9d25ba38 : SUCCESS in 13m 21s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/5d4c272c43854a65818589d26f945931 : SUCCESS in 2h 03m 05s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/6ca845611b8d47b694cfb00ba046eb38 : SUCCESS in 1h 10m 55s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/6aa3c9adc980491da8ba78d3a5127ae2 : SUCCESS in 1h 04m 42s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/bac32d8f486d4af7b4309b0c24b9e73b : SUCCESS in 1h 16m 16s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/53f7415083a44ef09c6481ee252b5ae6 : SUCCESS in 1h 13m 17s","accounts_in_message":[],"_revision_number":4},{"id":"e6cea895fd3bdc783df8226212a8912128e5d332","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2025-12-16 14:05:47.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8b7af97094f7603c8757d4e7fcdafe2e2bcb47ed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-16 14:06:21.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"317eba36d5973718d2d96521250624915d18aab2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-16 16:07:09.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f12c0c65e94740f19ea03126d30230b0\n\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/63aad4396ab6425fab1a6ad3f54dd897 : SUCCESS in 1h 40m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8cd980e8b6f147b68a7a0474c21035a1 : SUCCESS in 13m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1cf04b01857c48ffa9bf8e9a84447ba6 : SUCCESS in 16m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7a9a6f90faa240499e7b06468278a88b : SUCCESS in 15m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14ba47989c6e4e2ba22738b3a381949d : SUCCESS in 6m 45s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a460e45c01c4490eb248b20d6107ccb3 : SUCCESS in 1h 54m 44s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/a22da613d8ef4963a44e753fb8102768 : SUCCESS in 38m 40s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/220e13c0b5734e9fa140f8e6640c5d4d : SUCCESS in 52m 26s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/81169474fcc04727826ca10510e63388 : SUCCESS in 1h 03m 01s","accounts_in_message":[],"_revision_number":4},{"id":"499837d22bfca8ba392d6f2275657db45261f807","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-16 16:07:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"3c9f3ff17eb5c98334db4580e097085601dfeb82","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-16 16:08:37.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7764d71530764e0c8413d2c789a049d2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab24cb75e8414220b6fd9daf79c3ab23 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"2c87bfe4f02f1c5c601788f2d299c8c1476f59fd","revisions":{"b1b954b67bf69932b1c94f2794efab86eeb8f534":{"kind":"REWORK","_number":1,"created":"2025-12-11 16:00:14.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/46/970646/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/46/970646/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/970646/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/970646/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/970646/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/46/970646/1"}}},"commit":{"parents":[{"commit":"6d3592b9495e40acb34108aa1a666d3bd4317d29","subject":"Fix maintenance LRP prefix adding fails over \u0027name\u0027 attribute missing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6d3592b9495e40acb34108aa1a666d3bd4317d29"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-12-11 15:27:36.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-12-11 16:00:02.000000000","tz":60},"subject":"[OVN] Fix the `_gen_router_port_ext_ids` call","message":"[OVN] Fix the `_gen_router_port_ext_ids` call\n\nIn the maintenance method `update_lrouter_ports_ext_ids_name_prefix`,\nthe call to `_gen_router_port_ext_ids` should be done with a Neutron\nDB port, not the Logical_Switch_Port.\n\nCloses-Bug: #2134703\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I2bc635e80f23d38683ae635ac15f4fc3e7a79b73\n(cherry picked from commit 9d5735812c070979c65b0dce061b97454405f8cc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b1b954b67bf69932b1c94f2794efab86eeb8f534"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b1b954b67bf69932b1c94f2794efab86eeb8f534"}]},"branch":"refs/heads/stable/2025.2"},"9558787770496bbc9b7d69b819e36fec461f168c":{"kind":"REWORK","_number":2,"created":"2025-12-12 08:43:21.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/46/970646/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/46/970646/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/970646/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/970646/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/970646/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/46/970646/2"}}},"commit":{"parents":[{"commit":"6d3592b9495e40acb34108aa1a666d3bd4317d29","subject":"Fix maintenance LRP prefix adding fails over \u0027name\u0027 attribute missing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6d3592b9495e40acb34108aa1a666d3bd4317d29"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-12-11 15:27:36.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-12-12 08:43:05.000000000","tz":60},"subject":"[OVN] Fix the `_gen_router_port_ext_ids` call","message":"[OVN] Fix the `_gen_router_port_ext_ids` call\n\nIn the maintenance method `update_lrouter_ports_ext_ids_name_prefix`,\nthe call to `_gen_router_port_ext_ids` should be done with a Neutron\nDB port, not the Logical_Switch_Port.\n\nCloses-Bug: #2134703\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I2bc635e80f23d38683ae635ac15f4fc3e7a79b73\n(cherry picked from commit 0758518b128fe379efc2d554ecd73d164339e7a3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9558787770496bbc9b7d69b819e36fec461f168c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9558787770496bbc9b7d69b819e36fec461f168c"}]},"branch":"refs/heads/stable/2025.2"},"e9e6656d613e7e35e7c782743edc64ffff4571b4":{"kind":"REWORK","_number":3,"created":"2025-12-12 16:41:36.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/46/970646/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/46/970646/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/970646/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/970646/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/970646/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/46/970646/3"}}},"commit":{"parents":[{"commit":"6d3592b9495e40acb34108aa1a666d3bd4317d29","subject":"Fix maintenance LRP prefix adding fails over \u0027name\u0027 attribute missing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6d3592b9495e40acb34108aa1a666d3bd4317d29"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-12-11 15:27:36.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-12-12 16:41:22.000000000","tz":60},"subject":"[OVN] Fix the `_gen_router_port_ext_ids` call","message":"[OVN] Fix the `_gen_router_port_ext_ids` call\n\nIn the maintenance method `update_lrouter_ports_ext_ids_name_prefix`,\nthe call to `_gen_router_port_ext_ids` should be done with a Neutron\nDB port, not the Logical_Switch_Port.\n\nCloses-Bug: #2134703\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I2bc635e80f23d38683ae635ac15f4fc3e7a79b73\n(cherry picked from commit 9edf82b24d5180f18fa08b45900d45cf9ab9f2dc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e9e6656d613e7e35e7c782743edc64ffff4571b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e9e6656d613e7e35e7c782743edc64ffff4571b4"}]},"branch":"refs/heads/stable/2025.2"},"2c87bfe4f02f1c5c601788f2d299c8c1476f59fd":{"kind":"NO_CHANGE","_number":4,"created":"2025-12-15 06:35:38.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/46/970646/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/46/970646/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/970646/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/970646/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/970646/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/46/970646/4"}}},"commit":{"parents":[{"commit":"6d3592b9495e40acb34108aa1a666d3bd4317d29","subject":"Fix maintenance LRP prefix adding fails over \u0027name\u0027 attribute missing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6d3592b9495e40acb34108aa1a666d3bd4317d29"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-12-11 15:27:36.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-12-15 06:35:25.000000000","tz":60},"subject":"[OVN] Fix the `_gen_router_port_ext_ids` call","message":"[OVN] Fix the `_gen_router_port_ext_ids` call\n\nIn the maintenance method `update_lrouter_ports_ext_ids_name_prefix`,\nthe call to `_gen_router_port_ext_ids` should be done with a Neutron\nDB port, not the Logical_Switch_Port.\n\nCloses-Bug: #2134703\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I2bc635e80f23d38683ae635ac15f4fc3e7a79b73\n(cherry picked from commit 9edf82b24d5180f18fa08b45900d45cf9ab9f2dc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2c87bfe4f02f1c5c601788f2d299c8c1476f59fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2c87bfe4f02f1c5c601788f2d299c8c1476f59fd"}]},"branch":"refs/heads/stable/2025.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
