)]}'
{"id":"openstack%2Fneutron~970665","triplet_id":"openstack%2Fneutron~master~I02ba844e24b8a8d21119a955b90623105e0345f2","project":"openstack/neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-12-12 12:54:54.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I02ba844e24b8a8d21119a955b90623105e0345f2","subject":"Drop redundant target-version option","status":"MERGED","created":"2025-12-11 18:17:13.000000000","updated":"2025-12-12 12:56:00.000000000","submitted":"2025-12-12 12:54:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"970665","meta_rev_id":"f3e428c3a88d430e87dae4582f14057bfc23e88f","_number":970665,"virtual_id_number":970665,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-12 12:54:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2025-12-12 10:42:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2025-12-12 00:28:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2025-12-12 10:42:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-11 20:21:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-12 00:28:16.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2025-12-12 10:42:50.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"4b02a78a52c5072e325886fd25c31762cbf41708","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-12-11 18:17:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9bdc7f67e5444c2d598c51f60c3ddf25cf25ad37","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-12-11 18:21:24.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"4542486da6b5dde8e901226416847c1588f44d73","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-12-11 18:21:57.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"120aebcc8ace91d9b9db5d34909217879dece9b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-11 20:21:01.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5335f399f711436799c7b1a9d1503654\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/71098e54ad834295a60c8f3c37cf1785 : SUCCESS in 3m 15s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/7c419033d75242579fb8bf784b2e230b : SUCCESS in 1h 24m 46s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/51c01ac9d1d2460d8e92eda356950e5b : SUCCESS in 1h 57m 42s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8f37b30ff2bf458b986a33dd7a1c64b2 : SUCCESS in 31m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e056cbed928f40be802778913336d787 : SUCCESS in 11m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9558368a4a63413fbc7dbcb563e18bcd : SUCCESS in 6m 23s\n- requirements-check https://zuul.opendev.org/t/openstack/build/142197bfcced4735af93ad9b0abc9cef : SUCCESS in 1m 52s","accounts_in_message":[],"_revision_number":3},{"id":"472c1a860f93979ba2a5cd2f388d0dff93ffd07e","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-12-12 00:28:16.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"43e4cd33baaf4ee1830e219b9953e466b7db418b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-12-12 10:42:50.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"9eff1cf325280be882bfa338289e9b299a13cbe9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 10:43:10.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a8ddd164466eca5a2046b43859dce7314f3ba34b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 12:54:54.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c24126c4c12e4c82adfb43881c0f7337\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4a6549c31e0649029d0a06fbd8829bad : SUCCESS in 1m 50s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/1829ea99270348878ac7275536d81a9f : SUCCESS in 1h 24m 30s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d13a76c94b2e49b099641b0a460bcf9c : SUCCESS in 2h 10m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f45dc90c50e450183f0249c24e05782 : SUCCESS in 19m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dedeb4bddf104ddbbcb2988f59f9f11c : SUCCESS in 6m 15s\n- requirements-check https://zuul.opendev.org/t/openstack/build/da78b325702d4e4d8384d2c4eacf6d6b : SUCCESS in 1m 42s","accounts_in_message":[],"_revision_number":3},{"id":"d656157f72f309dc2635572d9305e50a437ff426","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 12:54:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"f3e428c3a88d430e87dae4582f14057bfc23e88f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 12:56:00.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/75ec01d872d947bca72e9598069c5af0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce907433b55d4ca5b77ffcf9539919f3 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"0e9f221de189e29dff9df87df71a61a2a7a9d1dc","revisions":{"736b6a2f5a1bdaec7acc67a8b4cbbf7bc63cb584":{"kind":"REWORK","_number":1,"created":"2025-12-11 18:17:13.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/65/970665/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/970665/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/970665/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/970665/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/970665/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/970665/1"}}},"commit":{"parents":[{"commit":"7fa7737947092c77cb2c210fd18c5adbb81e5223","subject":"Merge \"Don\u0027t prevent setting CIDRs as allowed_address_pair for port\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7fa7737947092c77cb2c210fd18c5adbb81e5223"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-11 18:16:06.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-11 18:16:56.000000000","tz":540},"subject":"Drop redundant target-version option","message":"Drop redundant target-version option\n\nruff uses [python] requires-python by default.\n\nChange-Id: I02ba844e24b8a8d21119a955b90623105e0345f2\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/736b6a2f5a1bdaec7acc67a8b4cbbf7bc63cb584"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/736b6a2f5a1bdaec7acc67a8b4cbbf7bc63cb584"}]},"branch":"refs/heads/master"},"63c7330a7c395fc85d269bf8e5cf950947ebdcd4":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-12-11 18:21:24.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/65/970665/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/970665/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/970665/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/970665/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/970665/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/970665/2"}}},"commit":{"parents":[{"commit":"7fa7737947092c77cb2c210fd18c5adbb81e5223","subject":"Merge \"Don\u0027t prevent setting CIDRs as allowed_address_pair for port\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7fa7737947092c77cb2c210fd18c5adbb81e5223"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-11 18:16:06.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-11 18:20:43.000000000","tz":540},"subject":"Drop redundant target-version option","message":"Drop redundant target-version option\n\nruff uses [python] requires-python by default[1].\n\n[1] https://docs.astral.sh/ruff/settings/#target-version\n\nChange-Id: I02ba844e24b8a8d21119a955b90623105e0345f2\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/63c7330a7c395fc85d269bf8e5cf950947ebdcd4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/63c7330a7c395fc85d269bf8e5cf950947ebdcd4"}]},"branch":"refs/heads/master"},"0e9f221de189e29dff9df87df71a61a2a7a9d1dc":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-12-11 18:21:57.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/65/970665/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/970665/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/970665/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/970665/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/970665/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/970665/3"}}},"commit":{"parents":[{"commit":"7fa7737947092c77cb2c210fd18c5adbb81e5223","subject":"Merge \"Don\u0027t prevent setting CIDRs as allowed_address_pair for port\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7fa7737947092c77cb2c210fd18c5adbb81e5223"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-11 18:16:06.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-11 18:21:44.000000000","tz":540},"subject":"Drop redundant target-version option","message":"Drop redundant target-version option\n\nruff uses [project] requires-python by default[1].\n\n[1] https://docs.astral.sh/ruff/settings/#target-version\n\nChange-Id: I02ba844e24b8a8d21119a955b90623105e0345f2\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0e9f221de189e29dff9df87df71a61a2a7a9d1dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0e9f221de189e29dff9df87df71a61a2a7a9d1dc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
