)]}'
{"id":"openstack%2Fneutron~971049","triplet_id":"openstack%2Fneutron~master~I102d027f4d29c6f46a50b9ab1c902458e8ebc791","project":"openstack/neutron","branch":"master","topic":"bug/2136142","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2025-12-16 08:19:26.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}}},"hashtags":[],"change_id":"I102d027f4d29c6f46a50b9ab1c902458e8ebc791","subject":"Remove the \"nftables\" explicit CI jobs","status":"MERGED","created":"2025-12-15 14:30:45.000000000","updated":"2025-12-16 10:30:02.000000000","submitted":"2025-12-16 10:27:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"971049-bug/2136142","meta_rev_id":"8d6e5cc7052d3c3bb2deba8c3f7cec5d65483fff","_number":971049,"virtual_id_number":971049,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-16 10:27:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2025-12-16 08:18:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2025-12-15 20:23:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":1,"date":"2025-12-16 08:18:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-15 14:31:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-12-15 14:31:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-15 16:19:12.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2025-12-15 20:23:56.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2025-12-16 08:18:59.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"15a38203825f5e91177c2523b1844983730058a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-12-15 14:30:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ec515d85fe044e7cb0cc509db326d665e94c6a40","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-15 14:31:05.000000000","message":"Patch Set 1:\n\n(10 comments)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/neutron on branch master.  The\nproblem was:\n\n  Job neutron-ovs-tempest-plugin-iptables_hybrid not defined\n\nThe problem appears in the the \"neutron-periodic-jobs\" project-template stanza:\n\n  project-template:\n      name: neutron-periodic-jobs\n      periodic:\n        jobs: \u0026neutron-periodic-jobs\n          - neutron-functional\n          - neutron-functional-fips\n          - neutron-functional-with-pyroute2-master\n          - neutron-functional-with-sqlalchemy-master\n          - neutron-functional-with-ovsdbapp-master\n  ...\n\n  in \"openstack/neutron/zuul.d/job-templates.yaml@master\", line 91","accounts_in_message":[],"_revision_number":1},{"id":"dc41e9c7838bc1350f026a63c9db0090696a33e3","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-12-15 14:31:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2aa9bc4686aa808aca730a70b35c80fa5d514196","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-15 14:31:24.000000000","message":"Patch Set 1: Verified-1\n\n(10 comments)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/neutron on branch master.  The\nproblem was:\n\n  Job neutron-ovs-tempest-plugin-iptables_hybrid not defined\n\nThe problem appears in the the \"neutron-periodic-jobs\" project-template stanza:\n\n  project-template:\n      name: neutron-periodic-jobs\n      periodic:\n        jobs: \u0026neutron-periodic-jobs\n          - neutron-functional\n          - neutron-functional-fips\n          - neutron-functional-with-pyroute2-master\n          - neutron-functional-with-sqlalchemy-master\n          - neutron-functional-with-ovsdbapp-master\n  ...\n\n  in \"openstack/neutron/zuul.d/job-templates.yaml@master\", line 91","accounts_in_message":[],"_revision_number":1},{"id":"af0ebaf325f75fb59d9311c26e9bf822ed7fc305","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-15 14:31:39.000000000","message":"Patch Set 1:\n\n(10 comments)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/neutron on branch master.  The\nproblem was:\n\n  Job neutron-ovs-tempest-plugin-iptables_hybrid not defined\n\nThe problem appears in the the \"neutron-periodic-jobs\" project-template stanza:\n\n  project-template:\n      name: neutron-periodic-jobs\n      periodic:\n        jobs: \u0026neutron-periodic-jobs\n          - neutron-functional\n          - neutron-functional-fips\n          - neutron-functional-with-pyroute2-master\n          - neutron-functional-with-sqlalchemy-master\n          - neutron-functional-with-ovsdbapp-master\n  ...\n\n  in \"openstack/neutron/zuul.d/job-templates.yaml@master\", line 91","accounts_in_message":[],"_revision_number":1},{"id":"3bb9ef138a8ef6590dfa44f214ee1c94785a2087","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-12-15 15:35:15.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"e62a41b90d0d234e32c9b38b3133c59221e91b40","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-12-15 15:35:27.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"79595464de24d3ec99cbfc88eb67e456d282b89d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-15 17:32:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5f642100b8a454d8ad152d046ab1915\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/ef7db432438844b98de7dd8f317c2efe : SUCCESS in 1h 25m 58s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/707b5ec2cfdf47528de0e192d6b6d2f2 : SUCCESS in 1h 55m 39s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/42a3e9cda2b14ef698747db498001534 : SUCCESS in 33m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fd0994635b0e40f2b12dda00c0ad7151 : SUCCESS in 50m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2e7a353676b457dbed4d1f5638c9807 : SUCCESS in 21m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e774e6db903d40af882b20705c1286b9 : SUCCESS in 34m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d06505ec2e76498788ab80583e77f69d : SUCCESS in 24m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4da0d814cd849bca3b2b65cac924be5 : SUCCESS in 12m 28s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/b452069f025243418c79786c8a5c8e48 : SUCCESS in 58m 44s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/ebd4cc2434bc4407af40601bee99cb8b : SUCCESS in 1h 15m 19s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/755b7f8043004b7a8b06f87cfb971711 : SUCCESS in 58m 11s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/286ef3a9258744d68607f6e94c3a0d4d : SUCCESS in 39m 37s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/21f46045d7bb40faa593d624e844d068 : SUCCESS in 1h 43m 52s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ea61151b23f2428f9584f80483265711 : SUCCESS in 1h 25m 00s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/58c26a8c33cf4291b7e896e2e657405a : SUCCESS in 1h 05m 30s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/e1fcd6baadfe444a8476541043098a4b : SUCCESS in 1h 01m 38s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/74e3388aa23c4b7c9987ca64954b48a5 : SUCCESS in 1h 10m 20s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/9f0ba0d387c3477fb492721006ddaa3f : SUCCESS in 1h 29m 01s","accounts_in_message":[],"_revision_number":2},{"id":"86b22213e6723830a30abc46002b75fb1e7db4e4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-15 17:38:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3fba547020274f5ba8d6fd36ceb10356\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f1ca4e5526a7471698b6f5c01ac24a1e : SUCCESS in 53m 58s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8897383463ad41be868398b0a63676b0 : SUCCESS in 20m 38s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e3faec1318a64591a041f98e4c3a8427 : SUCCESS in 1h 06m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d5ae4d57cd1ff5393a06f3a01e1dfa3b2982150c","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-15 17:50:32.000000000","message":"Patch Set 2:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4980f9a657848a59e3de9f7251b004c\n\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/a82c4ca5b50346caad47bd2d63eb9fd3 : SUCCESS in 1h 19m 35s (non-voting)\n- neutron-tempest-plugin-openvswitch-distributed-dhcp https://zuul.opendev.org/t/openstack/build/b747719a91304196826ec3297a865b33 : SUCCESS in 1h 50m 58s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-distributed-dhcp https://zuul.opendev.org/t/openstack/build/9117d4b72c834dfca6cefc2e387d95d1 : FAILURE in 1h 46m 50s\n- neutron-tempest-plugin-ovn-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/36bab16afeeb422499b954078e8493fc : SUCCESS in 2h 10m 46s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7e416f3f4e7a40aea999aa1b5c5ae95e : SUCCESS in 2h 06m 15s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/6f596e52b054445dbd9b4770148f6a3e : SUCCESS in 44m 39s\n- neutron-ovs-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/488ed0a8f84b4b639dce7df7e7c14dbe : SUCCESS in 38m 40s\n- neutron-ovn-tempest-loki https://zuul.opendev.org/t/openstack/build/cd29307a2c454cadbbcb621e8d376e4e : SUCCESS in 1h 17m 19s\n- openstack-tox-py312-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/93cb546fa1a9489aa9333d211eb472db : SUCCESS in 28m 29s\n- openstack-tox-py312-with-oslo-master https://zuul.opendev.org/t/openstack/build/fffe315e32ba4abc864052a8739ea9fc : SUCCESS in 21m 11s\n- openstack-tox-py312-with-ovsdbapp-master https://zuul.opendev.org/t/openstack/build/c026f188ab08498a92c445e266f766b5 : SUCCESS in 27m 48s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/6b7ffc58e63449db9a776c44c5d397ca : SUCCESS in 1h 07m 52s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/0cb2cdbe4d224f20832ddf06df6f852c : SUCCESS in 1h 10m 47s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/a75341282c904746bda9803d8a40451b : SUCCESS in 2h 08m 42s\n- neutron-ovs-tempest-with-os-ken-master https://zuul.opendev.org/t/openstack/build/d39903e17e1840d39ec78fea5927b185 : SUCCESS in 1h 09m 36s\n- neutron-ovn-tempest-mariadb-full https://zuul.opendev.org/t/openstack/build/1b03090e574e4fb8af648315660a92a3 : SUCCESS in 1h 06m 11s\n- neutron-ovn-tempest-ipv6-only-ovs-master https://zuul.opendev.org/t/openstack/build/ac96329393c944d9a8728ffd80620a56 : SUCCESS in 1h 18m 00s\n- neutron-ovn-tempest-ovs-master-centos-9-stream https://zuul.opendev.org/t/openstack/build/01d23933da2a4f8f925c6032027cec70 : SUCCESS in 1h 18m 08s\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/78a8938641ec4375b022cb847e71c080 : SUCCESS in 1h 12m 45s\n- neutron-ovn-tempest-full-multinode-ovs-lts https://zuul.opendev.org/t/openstack/build/10a6b721af9541be89c2cdeb28b00533 : SUCCESS in 1h 44m 48s\n- neutron-ovn-tempest-ovs-master https://zuul.opendev.org/t/openstack/build/4f19f8f6b7774b339217bec153292b3e : SUCCESS in 44m 07s\n- neutron-ovn-tempest-ovs-lts https://zuul.opendev.org/t/openstack/build/28964451ea6a477e8358a76df590b4e3 : SUCCESS in 49m 13s\n- neutron-ovn-tempest-plugin-ovs-master https://zuul.opendev.org/t/openstack/build/26b2329141cd4a388e439ea78fad48e0 : SUCCESS in 1h 45m 26s\n- neutron-ovn-tempest-plugin-ovs-lts https://zuul.opendev.org/t/openstack/build/9cd772b6c86444659d6b04a79b68f4e2 : SUCCESS in 1h 55m 15s\n- neutron-ovn-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/ac75358193f646c4970ba9a7636ac5c9 : SUCCESS in 1h 11m 44s\n- neutron-ovn-tempest-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/22cdff533c2544d29aded7a2b775d5d4 : SUCCESS in 1h 24m 52s\n- neutron-ovs-tempest-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/ef574716a5684151befa72e2f05f1765 : SUCCESS in 45m 00s\n- neutron-ovs-tempest-fips https://zuul.opendev.org/t/openstack/build/2e30c46465db4aee8815740731dfc616 : SUCCESS in 1h 24m 51s\n- neutron-ovn-tempest-ovs-release-fips https://zuul.opendev.org/t/openstack/build/e2dacd760a284050aa0d5bdf0e068ae8 : SUCCESS in 45m 22s\n- devstack-tobiko-neutron https://zuul.opendev.org/t/openstack/build/7b074a8b898c48f2be2c1831037c4555 : SUCCESS in 46m 31s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/51a858be05904818815fd713ef0db65a : SUCCESS in 1h 08m 04s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/6dd10542eadd4af5a31343a6fe291d77 : SUCCESS in 57m 06s\n- devstack-enforce-scope https://zuul.opendev.org/t/openstack/build/6d8cca29670241f8aace6093c6acdb64 : SUCCESS in 25m 37s\n- openstack-tox-py312-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/9e6247a651714c0a9642051efbb18a8a : SUCCESS in 27m 56s\n- neutron-ovs-tempest-with-oslo-master https://zuul.opendev.org/t/openstack/build/e119fe9d937a42459e8cf5341e2311f0 : SUCCESS in 1h 24m 41s\n- neutron-ovn-tempest-ovs-release-with-oslo-master https://zuul.opendev.org/t/openstack/build/07d5da3ca9f940ad866f1e4636a3e60a : SUCCESS in 1h 21m 26s\n- neutron-tempest-plugin-ovn-with-ovn-metadata-agent https://zuul.opendev.org/t/openstack/build/30cd6b86480247ceaa0fe927ade21d40 : SUCCESS in 1h 43m 47s\n- neutron-ovn-grenade-multinode-ovn-metadata-agent https://zuul.opendev.org/t/openstack/build/2b9b3fef74c44b1396d4678120eeb5b1 : SUCCESS in 50m 22s\n- ironic-tempest-ovn-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/03bb713fc33247a8af372e5679d96384 : SUCCESS in 24m 46s","accounts_in_message":[],"_revision_number":2},{"id":"8624dccb28449b8bbad520651b7863ce487a81cb","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-12-15 20:23:56.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e2f1f460d3491d4a949ca4aac861cfd409745302","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-12-16 08:18:59.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7d2751df09baaca36d3382bc535842f1250b3e71","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-12-16 08:19:26.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4332a16dbb97b30b87bacb537d790d9fd6ca4fdb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-16 08:19:42.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e3d4f946cbc9d55cc7c61bbc15e9d53190f8205a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-16 10:27:56.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f450ef2b947c4b00901500f1cc3a33fb\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b054b62d12514e7eb7b0d93b5be2794c : SUCCESS in 1h 19m 23s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/60f6d2333884449ea6f04f9dbeff94e0 : SUCCESS in 2h 07m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9907555a210b45a788f460e10c7889c3 : SUCCESS in 18m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d2c8f0f91f2e4a0c92d3e74381d84a87 : SUCCESS in 26m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e5c1e27404d345cd81c153c844d4b73a : SUCCESS in 32m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d3ccca978dc4dc39ed5b371be88398b : SUCCESS in 8m 44s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/7ddcdf3bd05b45f1a6d540e7a08261d6 : SUCCESS in 1h 51m 51s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/2265492b906f48ada014c5660b53731a : SUCCESS in 1h 13m 03s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/2471db0d6810413cad610356314dbc4c : SUCCESS in 1h 08m 41s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/6c1a4380a5e34adba2298d6eedccca1a : SUCCESS in 56m 14s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/add0c7cae06940f296ea38dbb3dfd8fe : SUCCESS in 1h 18m 59s","accounts_in_message":[],"_revision_number":2},{"id":"66b2a6ab2fe54555c84e03cd2d425fba5a23ef61","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-16 10:27:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"8d6e5cc7052d3c3bb2deba8c3f7cec5d65483fff","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-16 10:30:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f3d63a39a304062bbe470f44fd79ccc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98a1f46093fa4e6aba59bacb8e1eab5c : SUCCESS in 42s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4808ad7901721585e1b1374848a7958b1dc49d76","revisions":{"a83599d00101d745a067308cf5cc8d0a3fbc09a3":{"kind":"REWORK","_number":1,"created":"2025-12-15 14:30:45.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/49/971049/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/49/971049/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/971049/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/971049/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/971049/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/49/971049/1"}}},"commit":{"parents":[{"commit":"467b2827f2a17d223032db3b33ad2a3e9781d668","subject":"Merge \"Remove redundant mock statement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/467b2827f2a17d223032db3b33ad2a3e9781d668"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-12-15 14:28:25.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-12-15 14:28:25.000000000","tz":60},"subject":"Remove the \"nftables\" explicit CI jobs","message":"Remove the \"nftables\" explicit CI jobs\n\nThe latest supported OS use the \"iptables\" compatible binary provided\nby tje \"nftables\" framework. The \"iptables\" commands are actually using\nthe \"nftables\" backend. This patch removes the installation of the\ncompatible binaries because this is now redundant.\n\nCloses-Bug: #2136142\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I102d027f4d29c6f46a50b9ab1c902458e8ebc791\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a83599d00101d745a067308cf5cc8d0a3fbc09a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a83599d00101d745a067308cf5cc8d0a3fbc09a3"}]},"branch":"refs/heads/master"},"4808ad7901721585e1b1374848a7958b1dc49d76":{"kind":"REWORK","_number":2,"created":"2025-12-15 15:35:15.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/49/971049/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/49/971049/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/971049/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/971049/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/971049/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/49/971049/2"}}},"commit":{"parents":[{"commit":"467b2827f2a17d223032db3b33ad2a3e9781d668","subject":"Merge \"Remove redundant mock statement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/467b2827f2a17d223032db3b33ad2a3e9781d668"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-12-15 14:28:25.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-12-15 15:22:21.000000000","tz":60},"subject":"Remove the \"nftables\" explicit CI jobs","message":"Remove the \"nftables\" explicit CI jobs\n\nThe latest supported OS use the \"iptables\" compatible binary provided\nby tje \"nftables\" framework. The \"iptables\" commands are actually using\nthe \"nftables\" backend. This patch removes the installation of the\ncompatible binaries because this is now redundant.\n\nCloses-Bug: #2136142\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I102d027f4d29c6f46a50b9ab1c902458e8ebc791\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4808ad7901721585e1b1374848a7958b1dc49d76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4808ad7901721585e1b1374848a7958b1dc49d76"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
