)]}'
{"id":"openstack%2Fneutron~971239","triplet_id":"openstack%2Fneutron~stable%2F2024.2~Ia07610a17afb55b802290c8910a17dace7794c4f","project":"openstack/neutron","branch":"stable/2024.2","attention_set":{},"removed_from_attention_set":{"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2025-12-19 18:23:35.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia07610a17afb55b802290c8910a17dace7794c4f","subject":"Search for active port binding when updating LSP host info","status":"MERGED","created":"2025-12-16 18:44:48.000000000","updated":"2025-12-19 18:25:30.000000000","submitted":"2025-12-19 18:23:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"971239","cherry_pick_of_change":970564,"cherry_pick_of_patch_set":1,"meta_rev_id":"826d65958f117b01220269994b195cde6032fe1d","_number":971239,"virtual_id_number":971239,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":7156,"name":"Mathieu Gagné","email":"mgagne@calavera.ca","username":"mgagne"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-19 18:23:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2025-12-16 22:30:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2025-12-16 18:58:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7156,"name":"Mathieu Gagné","email":"mgagne@calavera.ca","username":"mgagne"},{"value":2,"date":"2025-12-19 16:39:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":7156,"name":"Mathieu Gagné","email":"mgagne@calavera.ca","username":"mgagne"},{"value":1,"date":"2025-12-19 16:39:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":7156,"name":"Mathieu Gagné","email":"mgagne@calavera.ca","username":"mgagne"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7156,"name":"Mathieu Gagné","email":"mgagne@calavera.ca","username":"mgagne"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-16 18:58:17.000000000","updated_by":{"_account_id":7156,"name":"Mathieu Gagné","email":"mgagne@calavera.ca","username":"mgagne"},"reviewer":{"_account_id":7156,"name":"Mathieu Gagné","email":"mgagne@calavera.ca","username":"mgagne"},"state":"REVIEWER"},{"updated":"2025-12-16 20:48:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-16 22:30:27.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2025-12-19 16:39:51.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"9de7733ea709967caf90e4f242ad55da49d8555f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-12-16 18:44:48.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"7c705ca39031403730512f02b5b652bcf6021cff","author":{"_account_id":7156,"name":"Mathieu Gagné","email":"mgagne@calavera.ca","username":"mgagne"},"date":"2025-12-16 18:58:17.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"6a34a9436bf710c6af950685627a6c27c56b46b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-16 20:48:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a76494a5c40487cac12142760894df3\n\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/2f9437929c594e0ca3be8590c7f38d52 : SUCCESS in 2h 02m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4947bf99e5084c41b396773c94aa25a2 : SUCCESS in 40m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5332bae55d994f3fa644dadeb7d0aefe : SUCCESS in 10m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/670e181830294564add70842fce4636e : SUCCESS in 15m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/48cd124b159f4992bc9d7bcb719d58f4 : SUCCESS in 17m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5d17152d2e3041ff8fffd4f8947eeedc : SUCCESS in 21m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a999b1584a740fa9ebf6e4a1a347936 : SUCCESS in 6m 02s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ff881f143330416da264b8f528ff87c4 : SUCCESS in 1h 09m 52s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/89f5be18dbdb4507a97fc053d98f9c12 : SUCCESS in 1h 05m 15s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/c8c7d2cb94bf4057b69808f379b3ad96 : SUCCESS in 43m 15s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5fa54d873f00406f9066a601143513df : SUCCESS in 29m 06s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/cb4228dde44d4a38b4293bf19ba3bd18 : SUCCESS in 59m 50s","accounts_in_message":[],"_revision_number":1},{"id":"f79033bcd0dd3ec77ba28a95baad6ea2d1e5cbbf","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-12-16 22:30:27.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"785cfcee9a298e68d73a5dfe9b9af6a274019cbc","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2025-12-19 16:39:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"06781f5115fc1ddb96e89fe2593ed42c92963405","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-19 16:40:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a71111ab07b8f3ec735d8cce11b5dfd5a56a0bf9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-19 18:23:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9aa011fc12b5449f8f216de0ca878bc1\n\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/b771e6827cc0462d8d540b87bd1388f4 : SUCCESS in 1h 37m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49152a3c36294b48a0a8734d62e1e755 : SUCCESS in 10m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0db691609a04439d98bdedd3f0ed54aa : SUCCESS in 23m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2345c12667d74f17b8d189ffe0dbd07e : SUCCESS in 15m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e768d7f8437042b985a641a3e7b569f9 : SUCCESS in 6m 08s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/63bf8b5e6c02497285afec3f5578d1ab : SUCCESS in 1h 07m 14s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/cf8261a230fc4b4e987caf90a35c0fa6 : SUCCESS in 35m 33s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d48e40134ad848bf9d7e1694cc0a01f8 : SUCCESS in 51m 57s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/b0d41e1a3d004bf794bdbfb69d99d991 : SUCCESS in 1h 07m 48s","accounts_in_message":[],"_revision_number":1},{"id":"6825f041588892ee5ab3b63388e2e30d0146079b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-19 18:23:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"826d65958f117b01220269994b195cde6032fe1d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-19 18:25:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e97fc08c58c64b239dd86080775cfe06\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab50c775c69848ea8d2ca0f20768d14d : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f3d32497ac43b8e4734a12378e3f4c1bd4720e56","revisions":{"f3d32497ac43b8e4734a12378e3f4c1bd4720e56":{"kind":"REWORK","_number":1,"created":"2025-12-16 18:44:48.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/39/971239/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/971239/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/971239/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/971239/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/971239/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/971239/1"}}},"commit":{"parents":[{"commit":"e81a569e6b3d54aa46c498e11e443dd869281ff3","subject":"Update OVN_AGENT_METADATA_SB_CFG_KEY when OVN metadata agent starts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e81a569e6b3d54aa46c498e11e443dd869281ff3"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2025-12-10 23:27:25.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2025-12-16 18:44:48.000000000","tz":0},"subject":"Search for active port binding when updating LSP host info","message":"Search for active port binding when updating LSP host info\n\nWhen updating LSP host info for an UP port, the code\nalways uses the first port binding in the DB entry\n(db_port.port_bindings[0].host).\n\nSince there could be multiple port bindings present,\nsearch for the active one when determining which host\nvalue to use.\n\nCloses-bug: #2134504\nChange-Id: Ia07610a17afb55b802290c8910a17dace7794c4f\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n(cherry picked from commit 137949ee0781e138de8955c74cefb9bea612ce13)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f3d32497ac43b8e4734a12378e3f4c1bd4720e56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f3d32497ac43b8e4734a12378e3f4c1bd4720e56"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
