)]}'
{"id":"openstack%2Fneutron~971561","triplet_id":"openstack%2Fneutron~master~Id9579052b1963b925f7f655586f0995524359917","project":"openstack/neutron","branch":"master","topic":"bug/2111276","attention_set":{},"removed_from_attention_set":{"5756":{"account":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"last_update":"2025-12-19 18:02:13.000000000","reason":"\u003cGERRIT_ACCOUNT_5756\u003e replied on the change","reason_account":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}},"8655":{"account":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"last_update":"2026-01-05 14:37:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id9579052b1963b925f7f655586f0995524359917","subject":"bgp: Improve existing server Commands","status":"MERGED","created":"2025-12-19 16:51:34.000000000","updated":"2026-01-05 14:40:38.000000000","submitted":"2026-01-05 14:37:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"971561-bug/2111276","meta_rev_id":"bcbd4892960ac1d660d67951128416f49d1804bd","_number":971561,"virtual_id_number":971561,"owner":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-05 14:37:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":2,"date":"2026-01-05 12:16:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2025-12-19 23:15:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":1,"date":"2026-01-05 12:16:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-19 17:30:53.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"CC"},{"updated":"2025-12-19 20:03:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-12-19 20:56:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-19 23:15:04.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2026-01-05 12:16:44.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"5696ec6b15e7e69176fed768ffc0f4dc489c3cdf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2025-12-19 16:51:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b60a70522b0beaa2fd85da18914720c8e61c5e2c","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2025-12-19 17:30:53.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"4dc9852503a37587dd7e08522fada278dde2b118","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2025-12-19 17:33:08.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"1a616c6c34b5e554a89d06e518675d605d0296da","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2025-12-19 18:02:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d542a5704b107b12a5d7e3279fc957c6a558801c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2025-12-19 18:52:45.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d7a018d44cdf3798df0853175f09170ad3ecde6e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-19 20:03:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d35a3fb4c80c45d5884e031f1bf83a0d\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/822b0da0d8cd405485f7b98fbe8b7be8 : SUCCESS in 1h 03m 13s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6e0835c12f984c5f9b4c9a3b72af3794 : SUCCESS in 20m 22s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/02a81adab474497aa173c4a816d315e8 : SUCCESS in 24m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d1634485c9486ba348802c811263c603f85ffe6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-19 20:56:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8230bea9f0054574a338689825653fda\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/cb42d85d536741a9a50770594654af87 : SUCCESS in 1h 16m 44s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/706e8a465a5e49e18331014bb07202ef : SUCCESS in 1h 58m 21s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9d4be352b7a1463d87c086bf2cd686af : SUCCESS in 31m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0cd09c0045de4d85b0170d479805f2fe : SUCCESS in 1h 03m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/130013c2ddb24258856b2ede60842551 : SUCCESS in 22m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/25ebf1b8bab54056917097c8bce1b6bd : SUCCESS in 24m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0b10e25d2807480591d714be9041373d : SUCCESS in 29m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/158252fb50f84cc0b3354801febe2b05 : SUCCESS in 13m 30s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/b9f1f643e1684a68a4dfce0c2aae072e : SUCCESS in 59m 26s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/64b662d1d2524edd85d7e810c6c36301 : SUCCESS in 39m 23s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/6f7d2b3d238844adbf7bd8a314a75022 : SUCCESS in 1h 52m 29s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c0fb2db938f24f8b868d8a29a4020bb5 : SUCCESS in 1h 15m 23s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/3f498148505a4b92bfc0b5433819d133 : SUCCESS in 54m 10s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/754036706a8f4b888c7ca3d72edccc7d : SUCCESS in 1h 51m 40s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c3848146fb934dbe9b0a77e740fc17ab : SUCCESS in 1h 36m 38s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/281f804a388243049e3a8ad0a3e8aa46 : SUCCESS in 49m 52s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/c939a8758e85492d951c5525b1bb2103 : SUCCESS in 35m 20s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4ce6756897d448208fa72d8a4ed42ff3 : SUCCESS in 1h 27m 31s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/91f6861cce4548d395122eb6f1dced6d : SUCCESS in 1h 22m 32s","accounts_in_message":[],"_revision_number":2},{"id":"88fd492d206b2b6caa6add18952487a38c8f2079","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2025-12-19 23:15:04.000000000","message":"Patch Set 2: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"b7fd2ad29982d6e133213cd92a0b0e8d98c7ae36","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-01-05 12:16:44.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"488ca52b3a0a83ebc66e775987392c47e39bee7d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-05 12:17:17.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4aa0ea00d9cfc866113eda43cc1ab870875d7c41","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-05 14:37:38.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c0e80965bc441c190d2e1f802e7fa0d\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/42c2d0dedf644152ad3862bbd75baac7 : SUCCESS in 1h 32m 40s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e34778f6b58f4e70acc4767ce299b53c : SUCCESS in 1h 48m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a0b0f87e82348228d5f3e507a843fb3 : SUCCESS in 29m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/264ed9eb37684e1dbc711342620a3f11 : SUCCESS in 35m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0faa47fb10fd46848e86406859c017a6 : SUCCESS in 21m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57c797fcc5a545879618921743f6aab4 : SUCCESS in 14m 06s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/0c0de03c62084abbb4d669682f87dc3d : SUCCESS in 1h 20m 06s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ab1214ec1ea34b63abc8e7a5022ddb70 : SUCCESS in 1h 55m 52s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/636b6a1e13d14efbacb5b0949d0a8027 : SUCCESS in 41m 02s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/37e98b99a2e34ab3bbe8ef3d53e2bd7d : SUCCESS in 1h 07m 54s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/c49098ce055c443db48b14cfd6c1731f : SUCCESS in 1h 11m 36s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/9a0d0b6f89be4d4aafc5b4ac2c26a5ad : SUCCESS in 1h 12m 16s","accounts_in_message":[],"_revision_number":2},{"id":"51858a41769ab2af0880160a5a26df6f5c9f2af1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-05 14:37:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"bcbd4892960ac1d660d67951128416f49d1804bd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-05 14:40:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a96e9c76fada4bf69c67ee7fb71c3d15\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e18e9a30f3cb49888b5ea20a989dbcc1 : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ef78d8e978800b207952799fc27ea253aaa522af","revisions":{"f0f274d53a18b17957c55971b4dc47d0398fb7f9":{"kind":"REWORK","_number":1,"created":"2025-12-19 16:51:34.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/61/971561/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/61/971561/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/61/971561/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/61/971561/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/61/971561/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/61/971561/1"}}},"commit":{"parents":[{"commit":"e288617ba6abdb40f3707f1aa44f64383989fbdc","subject":"Merge \"bgp: Remove chassis indexing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e288617ba6abdb40f3707f1aa44f64383989fbdc"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2025-12-19 16:46:07.000000000","tz":0},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2025-12-19 16:51:09.000000000","tz":0},"subject":"bgp: Improve existing server Commands","message":"bgp: Improve existing server Commands\n\nThere were some cases where result is used in post_commit and requires a\nuuid and the existing commands may introduce an issue. This patch makes\nsure the `result` object is as expected while storing the Row in a\ndifferent instance attribute.\n\nRelated-Bug: #2111276\n\nChange-Id: Id9579052b1963b925f7f655586f0995524359917\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f0f274d53a18b17957c55971b4dc47d0398fb7f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f0f274d53a18b17957c55971b4dc47d0398fb7f9"}]},"branch":"refs/heads/master"},"ef78d8e978800b207952799fc27ea253aaa522af":{"kind":"REWORK","_number":2,"created":"2025-12-19 18:52:45.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/61/971561/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/61/971561/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/61/971561/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/61/971561/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/61/971561/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/61/971561/2"}}},"commit":{"parents":[{"commit":"e288617ba6abdb40f3707f1aa44f64383989fbdc","subject":"Merge \"bgp: Remove chassis indexing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e288617ba6abdb40f3707f1aa44f64383989fbdc"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2025-12-19 16:46:07.000000000","tz":0},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2025-12-19 18:50:16.000000000","tz":0},"subject":"bgp: Improve existing server Commands","message":"bgp: Improve existing server Commands\n\nThere were some cases where result is used in post_commit and requires a\nuuid and the existing commands may introduce an issue. This patch makes\nsure the `result` object is as expected while storing the Row in a\ndifferent instance attribute.\n\nRelated-Bug: #2111276\n\nChange-Id: Id9579052b1963b925f7f655586f0995524359917\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ef78d8e978800b207952799fc27ea253aaa522af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ef78d8e978800b207952799fc27ea253aaa522af"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
