)]}'
{"id":"openstack%2Fneutron~971593","triplet_id":"openstack%2Fneutron~master~I86c741e8eba5413875787e636f8d60f79238875d","project":"openstack/neutron","branch":"master","topic":"bug/2106776","attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-10-06 04:31:12.000000000","reason":"Reviewer was added"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2026-10-05 19:40:46.000000000","reason":"Vote got outdated and was removed: Code-Review-1"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2026-01-25 17:01:33.000000000","reason":"Added by \u003cGERRIT_ACCOUNT_38070\u003e using the hovercard menu","reason_account":{"_account_id":38070,"name":"James Bhattarai","display_name":"james","email":"Jamesbhattarai14@gmail.com","username":"jamesii-b"}}},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-10-06 04:31:17.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"38070":{"account":{"_account_id":38070,"name":"James Bhattarai","display_name":"james","email":"Jamesbhattarai14@gmail.com","username":"jamesii-b"},"last_update":"2026-10-06 04:31:16.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-10-02 15:54:18.000000000","reason":"Only change owner, uploader, reviewers, and cc can be in the attention set"}},"hashtags":[],"change_id":"I86c741e8eba5413875787e636f8d60f79238875d","subject":"Fix: per-gateway SNAT settings not persisted in multi-gateway routers","status":"NEW","created":"2025-12-20 22:07:52.000000000","updated":"2026-10-06 04:31:17.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9e55d87abefaed514353b4c8f5e20c60019f2115","_number":971593,"virtual_id_number":971593,"owner":{"_account_id":38070,"name":"James Bhattarai","display_name":"james","email":"Jamesbhattarai14@gmail.com","username":"jamesii-b"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":38070,"name":"James Bhattarai","display_name":"james","email":"Jamesbhattarai14@gmail.com","username":"jamesii-b"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-10-05 23:57:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":38070,"name":"James Bhattarai","display_name":"james","email":"Jamesbhattarai14@gmail.com","username":"jamesii-b"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":38070,"name":"James Bhattarai","display_name":"james","email":"Jamesbhattarai14@gmail.com","username":"jamesii-b"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":38070,"name":"James Bhattarai","display_name":"james","email":"Jamesbhattarai14@gmail.com","username":"jamesii-b"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":38070,"name":"James Bhattarai","display_name":"james","email":"Jamesbhattarai14@gmail.com","username":"jamesii-b"}],"CC":[{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-20 23:08:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-12-21 00:05:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-21 21:00:38.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2025-12-23 07:06:15.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2026-10-02 16:49:18.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-10-05 08:45:37.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2026-10-05 19:41:06.000000000","updated_by":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"CC"},{"updated":"2026-10-06 04:31:12.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2026-10-06 04:31:12.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"cac32eb32ed23555f87481dc1f858343ed57608e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38070,"name":"James Bhattarai","display_name":"james","email":"Jamesbhattarai14@gmail.com","username":"jamesii-b"},"date":"2025-12-20 22:07:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f5b79de472af117e341fbeddada5d6240a8dbb2e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-20 23:08:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4152411ffed4a08ae887c067add7e03\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6f18839ba7164cb4b720e77a0e675b65 : NODE_FAILURE Node(set) request 78f3f5adb9d549b1826c230f4c05936b failed in 0s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/283ad7adf3f1421bac8514b80757028c : FAILURE in 44m 26s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/19d850fd8434466cae052b80418497c4 : FAILURE in 58m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"48991eeaa2c10247202c596540d539c2c0151b0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-21 00:05:49.000000000","message":"Patch Set 1: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0f238298660b4a8b81a908be4b799411\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c2faf58ebbb24646af1be4e106055e6e : FAILURE in 1h 01m 04s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/7d3c720559ee47bea3b33dfab9a67f84 : FAILURE in 13m 21s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/bc68c798d9e9439391884e514992016b : FAILURE in 11m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a0cf8bd88a73424b8b4fd9f270b15005 : FAILURE in 41m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f912594ce4cc4f3fbb48ab0d00c83492 : FAILURE in 7m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cf0e2d8270254cbaa4e95e9e22fe1b22 : FAILURE in 29m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0b6a8e2144d648b8b58f0424c9764eae : FAILURE in 25m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2df6170c20d64c778f786eac4dda695c : SUCCESS in 5m 43s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/0d1fcb003828421fbf6ed30a579bc2e9 : SUCCESS in 46m 10s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/41c1277200874eafa3963a665177b486 : FAILURE in 35m 56s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f8bec30b62ef43fa85ba093452505f4f : FAILURE in 42m 37s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/471a784080554d998eaabda642262fec : SUCCESS in 1h 11m 49s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/92f8a2c271484e50aa2578ace8b059da : FAILURE in 1h 09m 17s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/660f850526864d469de69958eae8661a : SUCCESS in 1h 52m 16s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f47c73a279014d58b2d079163612bf7f : SUCCESS in 54m 49s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/67cc83474c6e4be6a8a0b213c19b1963 : FAILURE in 22m 16s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/b7b3e69b835e4ab1a8ec91a52ec96597 : SUCCESS in 1h 15m 33s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9fe076404a4b4beb9cca111527430321 : FAILURE in 26m 47s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/082462a5a3854c74a20e526fe64fab62 : FAILURE in 15m 59s","accounts_in_message":[],"_revision_number":1},{"id":"7e775b416a7662c39b50032a90b725835a34cd45","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38070,"name":"James Bhattarai","display_name":"james","email":"Jamesbhattarai14@gmail.com","username":"jamesii-b"},"date":"2025-12-21 07:49:39.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"e4526653e1bd2fd97c61a0b816a840fc4954f9b2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-21 08:41:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8cdb4c0d07cd4eea98a6c95bd64bd268\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f756bc20067f4908b457b39539b9f9e3 : FAILURE in 50m 01s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1ccc7b8615f245e0b96f2fb2bbf40343 : FAILURE in 21m 19s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c2ad0313418548f79f93afb91197c584 : FAILURE in 24m 07s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"369c6fcd2fbf1203bf5c3051dfe6eb31ba59c625","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-21 10:06:31.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7efcdc2ab7d645ae9c6b438ac1ab88fe\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/9a744ff2f10044ef9f44f7e86b76d52d : SUCCESS in 1h 37m 34s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/8ece5b1f78d9406f9d52efbc3bd6edca : SUCCESS in 2h 14m 46s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2802ae24b80144f9b1976028efad6e58 : SUCCESS in 30m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e29c317e29d8471098772a78eee8817f : FAILURE in 39m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/317e3fe4902f44c0bc19065d933d4196 : FAILURE in 7m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3319f348be334f4d9d9c27623f001181 : FAILURE in 37m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b6d87426d6044d82bf76ef0f4158e53a : FAILURE in 24m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/667aba50ebe1427c8e05ce17758601f0 : SUCCESS in 12m 33s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/eaa20c44be604d32948a47626b9952f6 : SUCCESS in 1h 02m 05s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/1e3b397283814d76902c056a4e9bba2f : SUCCESS in 1h 12m 17s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/5271d99284d54baf9f5772f3fe5050e0 : SUCCESS in 1h 11m 04s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/cae9119c45b447bfa8d02783226518b2 : SUCCESS in 57m 16s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/e2281cc80272428bad21f8c3c3d654d9 : SUCCESS in 1h 10m 47s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/a6abaf0e5fcb48d380f99f4b7f03d180 : SUCCESS in 1h 55m 50s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/eb400d066aee416688f32b6aa188774c : SUCCESS in 1h 43m 06s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/bc95cddf50294373ae70d83d07b0658e : SUCCESS in 1h 03m 55s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/1a71ce920d7047ddba3c6b1e3b89d604 : SUCCESS in 1h 08m 11s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8d2aedf3ea504664afd1e2529da61ba9 : SUCCESS in 35m 19s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c71563bc0dc541d4bfcd384efa37cc84 : SUCCESS in 1h 32m 48s","accounts_in_message":[],"_revision_number":2},{"id":"52fa91826ac16b7ebe3d6107f75a50ea8a04138a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38070,"name":"James Bhattarai","display_name":"james","email":"Jamesbhattarai14@gmail.com","username":"jamesii-b"},"date":"2025-12-21 11:13:55.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"59a9b38448c3be6216d0d7903ea5e24e9ebed39f","author":{"_account_id":38070,"name":"James Bhattarai","display_name":"james","email":"Jamesbhattarai14@gmail.com","username":"jamesii-b"},"date":"2025-12-21 11:19:49.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"ef0e9e97c167655d5bf1ed769b1c3f3cad218229","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-21 12:04:29.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce1197a0b56b4bf283e510df161fd650\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/93c93af8deeb4b9a8788ad5d39bb7fb5 : SUCCESS in 48m 57s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/393685a8e9ca4421b358f5327294866a : SUCCESS in 29m 31s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/674a3463df39415f8aa8b759dc2a2147 : SUCCESS in 25m 00s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b3773e5c0192914dc62104213ba301e8888b9208","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-21 13:43:17.000000000","message":"Patch Set 3: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6b7640f5b56d44c1b5f910936a9f5e68\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/a40638cb1edd42eaa4917f46f7d0efc7 : FAILURE in 1h 28m 52s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/21fdd85aec9c4bf89ded3f1ea4a61c08 : SUCCESS in 2h 27m 31s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1657650dc4b740cc9a2c037a6bf38498 : SUCCESS in 19m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2a80fb57c5e84e2ca0ce9abd1ec25cc3 : SUCCESS in 35m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61c8832ca036417e9ca59d4f0f27adbb : FAILURE in 21m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6aa42a4a80864959b0ce7930fb244270 : SUCCESS in 27m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0c8c3b6bf4c444a58aacc3e2de0c345f : SUCCESS in 22m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a2a0694f7b644f9bbf7775175a4783d : SUCCESS in 8m 01s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/936b8cb65f494f159d9ce4a2409bb342 : SUCCESS in 40m 55s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/4a93e203a9c64a88a6953459e3e931c1 : SUCCESS in 40m 43s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d60e23f6d55847c8a4e05f19dbd567bb : SUCCESS in 1h 13m 42s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/7d42844326df4b6cbecafa1ca91088fd : FAILURE in 1h 04m 19s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/fdc34b395bd24f5fb6a0993918dc2966 : FAILURE in 1h 03m 00s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e47b3139cb1a43b59992c4a87bee9c57 : SUCCESS in 1h 34m 14s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f7dbe71fdfad4997b8dc64cac483035e : SUCCESS in 1h 33m 10s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/87e408e504ba4882ab0774651ce85cfd : SUCCESS in 46m 46s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/1279be8163d34c26bdf710871e7a3e30 : SUCCESS in 31m 18s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d6ca855815ef4dc09e61178ec89128d5 : SUCCESS in 1h 30m 34s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/192586b136b14ca794ec7acb1244c9c6 : SUCCESS in 1h 02m 46s","accounts_in_message":[],"_revision_number":3},{"id":"67b3e821a1a7e0b6086036ea49a4d959dc10e232","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-12-21 21:00:38.000000000","message":"Patch Set 3: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"7029c957ddbd117b0485e9685ca7321a1bf56493","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38070,"name":"James Bhattarai","display_name":"james","email":"Jamesbhattarai14@gmail.com","username":"jamesii-b"},"date":"2025-12-25 10:00:21.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"08e9bcabad197b2df400eb86fd160059ee09de00","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-25 10:28:23.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c1de93476854ce1990121d9fc143cfd\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/332c378524f345638c1cc40f2695bb21 : FAILURE in 24m 54s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0cd4928c0c5c46279d3d194296fa0cc6 : FAILURE in 22m 27s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2f7d462a3f75409798197ad7f9711944 : FAILURE in 26m 08s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"00e77b08c6ed579ead53ff8f7fc66e0032b32da8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-25 11:31:59.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f217f2953f9f46c7bb668ee0c308ffdc\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/5436a2a081984ae5b856194aefbcb2bc : FAILURE in 46m 38s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/5d7c1e04898c43af8e20efd79561be2b : FAILURE in 13m 49s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9a6d56a3fd0a4ad79a376540f5a5cbe8 : FAILURE in 9m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f36b5a174d90418894f809d1419c7510 : FAILURE in 1h 01m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2ab1e0f38de4550b5eebb891011791f : SUCCESS in 27m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9cc5ab2176c343e1b8eef65f44597daf : FAILURE in 26m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8e92c9e3c0434623a7e2e945ea5e8672 : FAILURE in 24m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8f114a091a64e88b0b302420f602e0b : SUCCESS in 12m 18s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/655ec16fcdb9411898d2bf2e43e56bd5 : SUCCESS in 58m 28s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/9dc6f1f900cd4f179cbc192c16665103 : FAILURE in 1h 03m 23s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/380e4d9b164a49ccbbcb350830e72486 : FAILURE in 26m 18s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/0161b7acc121445f93b62480d0997288 : SUCCESS in 1h 09m 16s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/6d1f9e1c2cc84672a35d94dd12148fdb : FAILURE in 31m 02s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/3ea8996c384744fca0e22efbb125e5c1 : SUCCESS in 1h 29m 42s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/bab57da4ca1d43c59444d6d98102355a : SUCCESS in 1h 13m 11s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/3c1148359b674456ba905ff27f5fcd24 : FAILURE in 11m 09s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/6f91ecb1bcee4323a2efb021e9c2d9f5 : SUCCESS in 34m 40s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/930def746c77444d9b12664d8be0ce7a : FAILURE in 26m 15s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3382539425344fbea9b517b68db9650d : FAILURE in 23m 28s","accounts_in_message":[],"_revision_number":4},{"id":"29b6e06057061fd4568dd539633cddff2650642a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38070,"name":"James Bhattarai","display_name":"james","email":"Jamesbhattarai14@gmail.com","username":"jamesii-b"},"date":"2026-01-05 12:48:32.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"4c3ce9afb219810f1af2760331dbace3cf12f80d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-05 13:39:18.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e3934ee180784c0c83ded50ae9045d17\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/104709b4bfa14a08b8752c87880552d1 : SUCCESS in 49m 02s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6b426dedc9d94c9799d727fa039ec1d4 : SUCCESS in 20m 39s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a935dec0efce4a588aa5923b493071d9 : SUCCESS in 27m 36s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c76c8872916d62803fef795579ce108e7af06dfb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-05 14:42:54.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5a39753c1bbe4d2684ec289e74106a9f\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c7c9fdc1709e41dd9b7a07216e258431 : SUCCESS in 1h 21m 49s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/168f6b0e9a80415789a03a12f2a55877 : SUCCESS in 1h 52m 54s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/be10c9c76ecf4ee9be9fcc9f06eff755 : SUCCESS in 32m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d22b9ab7f3444dafa251c7124653ec3b : SUCCESS in 1h 05m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70369d59d278465e8b86387942194804 : SUCCESS in 26m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8fe0fa2cabcc43fab0051efa699b7410 : SUCCESS in 36m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6ec43ed102a24ffb9f0fe34b04f243fb : SUCCESS in 23m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4846e1204fc4624adfca58bf2683f24 : SUCCESS in 14m 02s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/b06b251c92f347d3a6783a6bfb171546 : SUCCESS in 31m 22s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/01280ea4ae2b44f8a8502b8882d89556 : SUCCESS in 1h 02m 27s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/b9a2a9916cd4404dac61ba58fdfa0334 : FAILURE in 43m 08s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/7440525166cc47b8ac693d7d9bf77a67 : SUCCESS in 1h 03m 04s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/e7a529961fc54889ae3c8e4d38de978c : SUCCESS in 1h 00m 15s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/8f1013c93a394189a45ab63a9d360f65 : SUCCESS in 1h 36m 10s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/37f50f6cba2f4c9dbf3815878a07688e : SUCCESS in 1h 37m 25s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/2bb18705016342749fc9edcb0db0624f : SUCCESS in 1h 11m 19s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/351ce2a22cf44824a7980608b464baa6 : SUCCESS in 1h 14m 50s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b3a3fb3d2ffd4d4fbe3d8c63d5eb0b48 : SUCCESS in 1h 44m 28s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/28fef354f13946e6b0d9ddc0a7894221 : SUCCESS in 1h 34m 30s","accounts_in_message":[],"_revision_number":5},{"id":"af0b70f0af5f67930a496194a9fb37e7938502fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38070,"name":"James Bhattarai","display_name":"james","email":"Jamesbhattarai14@gmail.com","username":"jamesii-b"},"date":"2026-01-05 15:02:51.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"6483bf3f289cc981a04f3cb6fe1275b3c7457ad9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-05 17:19:52.000000000","message":"Patch Set 6: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ab9011bacea54ee6a67e8f07551a023d\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/03f53e16f00644d383068f330c2a9850 : SUCCESS in 1h 33m 56s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b664daae15b64da792c7572dfdfd6bc2 : SUCCESS in 1h 49m 46s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/420c1a461cf04bbeb55422b40adc5677 : SUCCESS in 30m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c0dd4e7cddee473e958926c3fb63ed20 : SUCCESS in 1h 05m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d88362faed564ccea3c69880f1003598 : FAILURE in 7m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b5553429e960487d849c20613447a8e7 : SUCCESS in 35m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1873309bc0d94863ac9a100491224bb2 : SUCCESS in 31m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f45f720ac1f84b7cad8d6efee71fc2bb : SUCCESS in 11m 28s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/ab0857fe481742afb7cff2be942ee81e : SUCCESS in 51m 40s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/e2dbcc89506f4e6bb10a85758f0efdc8 : SUCCESS in 41m 35s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/0a4ca5b045c941d08a1eb2175a65fc2a : SUCCESS in 2h 03m 11s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/0702b1fe42e6450b932e50cf9cdc4620 : SUCCESS in 1h 01m 24s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/72e61cae09d44ada8a644cea373318d8 : SUCCESS in 55m 35s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/5c3e7e8688ff4732b8ecfa9b2d7f29ca : SUCCESS in 1h 35m 26s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/32f55ffaa1d444e094807f273f703d36 : SUCCESS in 1h 41m 03s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d5185a7cfc6343c2af0d934b08ae338f : SUCCESS in 1h 10m 20s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/70d497956bff46a199497f7cd801e30c : SUCCESS in 30m 57s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1e3b4bfc46354d6c844e83915b40c759 : SUCCESS in 1h 07m 33s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/543ac2cf3301478cb3199f28523dd751 : SUCCESS in 1h 13m 32s","accounts_in_message":[],"_revision_number":6},{"id":"a51115d0a2a84cde79b9e0959aab7b0df93d6141","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38070,"name":"James Bhattarai","display_name":"james","email":"Jamesbhattarai14@gmail.com","username":"jamesii-b"},"date":"2026-01-05 17:43:14.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"9680388f0d8f287e9fe2f728cc286c6793dfc02e","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":38070,"name":"James Bhattarai","display_name":"james","email":"Jamesbhattarai14@gmail.com","username":"jamesii-b"},"date":"2026-01-05 19:07:22.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":7},{"id":"532943bc97d1d59ffee322a7ca040384a578a329","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":38070,"name":"James Bhattarai","display_name":"james","email":"Jamesbhattarai14@gmail.com","username":"jamesii-b"},"date":"2026-01-05 19:07:31.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":7},{"id":"2ab426a4001bae669066f028f133032cf59bb816","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-05 19:20:52.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c4d328773e54f35aa9956344736bbca\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a8d3408032c2458cb1d65f7f1752b628 : SUCCESS in 52m 25s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6e3bc55331b743c79691509282040039 : SUCCESS in 23m 00s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4012e56473b946cf8d3056701829e23b : SUCCESS in 25m 56s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"e0ab1806eba95fc418483c2fe83fd512dc50c38e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-05 19:38:20.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe90c96db6b74382b5c36cb1f0ec8e17\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/3e03b7b200e24b449921555b6c4c6a38 : SUCCESS in 1h 32m 35s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/4d8ce80441704a36b60c323f11a707cb : SUCCESS in 1h 48m 51s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fd6e1b530aeb426c8bb7e7a195e75c0c : SUCCESS in 32m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4f4df61dcbcf4385b10626d49de79bd5 : SUCCESS in 1h 03m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ed8d4388cb348039ed2b7370c5aca93 : SUCCESS in 10m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d527041235a6450faec9ddf6ffc07d94 : SUCCESS in 35m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7510633b84c841d998133e81b5c76b65 : SUCCESS in 21m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4afe39d870f9454a9b0b8ec469639aae : SUCCESS in 12m 57s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/14819a37d10b4548973891a153ce504d : SUCCESS in 1h 01m 51s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/3662c3f2018d42868782fc4922f1f625 : SUCCESS in 57m 55s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c14792dc5bd54cdfaff815f57568c98d : SUCCESS in 1h 52m 31s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/65106f7fdb8c47c482a78c64e61141ca : SUCCESS in 1h 01m 31s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/25e87611011b467c80bc83fdc53ed7bb : SUCCESS in 1h 10m 15s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/0c0e58c9667a4f9f8068fe9d6e8f5678 : SUCCESS in 1h 25m 41s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6e908c7abb714fd285f83a47e30a9ad9 : SUCCESS in 1h 37m 09s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/68349bddab1245669064a662f4a73f36 : SUCCESS in 1h 02m 27s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/8af44a7da04c4ffb8cfc6e8d79baab86 : SUCCESS in 1h 08m 25s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a4f8c849219f4b14ab83b88977f4637a : SUCCESS in 1h 30m 33s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c7dabb1243f84671a20a6ea2d7f9a65f : SUCCESS in 1h 12m 24s","accounts_in_message":[],"_revision_number":7},{"id":"ae09bc9cae76d05f380740fa4bfbaa0ee7e3ea18","author":{"_account_id":38070,"name":"James Bhattarai","display_name":"james","email":"Jamesbhattarai14@gmail.com","username":"jamesii-b"},"date":"2026-01-05 20:04:31.000000000","message":"Patch Set 7:\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"f3dd1253e9ac79ee5e8e70dcd868e1b7bb33ae53","author":{"_account_id":38070,"name":"James Bhattarai","display_name":"james","email":"Jamesbhattarai14@gmail.com","username":"jamesii-b"},"date":"2026-01-05 20:05:30.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"642180e4c14ca719e6ab6f4cf886a192d71c4725","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":38070,"name":"James Bhattarai","display_name":"james","email":"Jamesbhattarai14@gmail.com","username":"jamesii-b"},"date":"2026-01-05 20:05:40.000000000","message":"Removed Code-Review+1 by \u003cGERRIT_ACCOUNT_38070\u003e\n","accounts_in_message":[{"_account_id":38070,"name":"James Bhattarai","display_name":"james","email":"Jamesbhattarai14@gmail.com","username":"jamesii-b"}],"_revision_number":7},{"id":"f29a549797ce43067a4593285ee2e6f34fe3c94a","author":{"_account_id":38070,"name":"James Bhattarai","display_name":"james","email":"Jamesbhattarai14@gmail.com","username":"jamesii-b"},"date":"2026-01-10 08:54:48.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"9d32f97fef0562683c60d7dc94aa34b88017dede","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-10-02 07:39:07.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"94df36db13f38e47274bde604730cf2f1ad4ef3e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-02 08:35:18.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/882068d2b7214f32ad442d9c022b30a2\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b390dc637fb54ef78ef095d0a1e6d437 : FAILURE in 46m 14s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/f482af443b7c41b895af0060218282ea : FAILURE in 54m 37s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"939b77b6a7dda71cff8195ba3bdaf20181370fee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-02 09:18:12.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fdf25552b7be4ba99799d6452cc37bf1\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/cc5a0ac5aaeb45c188756f223b319249 : SUCCESS in 42m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6a9a52dbdd9e4a1aaf22c553bcadbf10 : SUCCESS in 32m 56s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/3c80931e5d044eb4925f660e6020ecbe : SUCCESS in 1h 02m 42s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/316d561a614a45fb835f0b2fb4059b05 : SUCCESS in 1h 33m 31s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1a5fbb71bdb249a1b495e60f8425e859 : SUCCESS in 28m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e7548b4ca1da41d68f831d73e87508bd : FAILURE in 49m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ffa3d63009a4bec9147f1f5004dc875 : FAILURE in 9m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/af8d997756a3425291042a288a312d40 : FAILURE in 23m 12s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3172820688ac4259aaa955761a496929 : FAILURE in 37m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6987b4dc250e4dca979a59ec83a5148a : SUCCESS in 13m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/58da314f2b0a489f849fec973120eef4 : SUCCESS in 7m 49s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/74ec9aee3cca4ee18fd5d52973243f34 : SUCCESS in 1h 33m 52s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/ef0662c9303a4780985293d12833921e : SUCCESS in 34m 19s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/de03e4aa0eb24d5395530ad66d7c0289 : SUCCESS in 38m 17s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/693f026e99ae49dd8f755993c46e2f14 : SUCCESS in 1h 25m 23s (non-voting)\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e0298f6585404cba9214c48b5bef0db2 : SUCCESS in 42m 47s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/59931c2feca949a4900412ef158d76aa : SUCCESS in 1h 03m 12s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/de491b3d011b42b896ffb95c4bf7457e : SUCCESS in 1h 17m 54s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/984999fcc20b41b281137f7c56270f2e : SUCCESS in 1h 33m 17s","accounts_in_message":[],"_revision_number":8},{"id":"50f11c72c406459ef4c9fd63c5b4b6d8fb4736f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-10-02 15:14:13.000000000","message":"Patch Set 9: Published edit on patch set 8.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":9},{"id":"9b67c4dd40c56053462f33ac6f0f354ff4f94aac","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-02 15:54:18.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0fc7675aeacd4ccda3d327744408a476\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a123871286de4734b8504128a215d462 : FAILURE in 38m 08s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/551628942cef44a5848d37b336ad0f3e : SUCCESS in 29m 19s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"db08b13806c43fe1de26aa191dca0df5e41986ac","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-10-02 16:49:18.000000000","message":"Patch Set 8: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2026/Arista_Tempest_Testrefs/changes/93/971593/8 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":8},{"id":"93f9bd4692bce8d2de802643adb5809b40a26a27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-02 17:16:51.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e170c95622d41d4895d924d723d4a27\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/0255063c40a042f1ab6081aa196fa13d : SUCCESS in 1h 17m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1a7a4d4218f540cabf06c09735a59ded : SUCCESS in 48m 25s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/558a18ea36d244ffbe6dffbe6d20fe96 : SUCCESS in 1h 05m 36s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/9cf4ee7aadd94904bda9df187b506c11 : SUCCESS in 1h 59m 51s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/03ca340d012c4f7996a54a07d6f64375 : SUCCESS in 30m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a10a7303a10d4a728049e36ce6e40ae1 : SUCCESS in 1h 03m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/206ff7c5891f45669290c57b8618d493 : SUCCESS in 12m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/554f6d9b39c4410f98e51429b4394047 : SUCCESS in 29m 45s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/632549882cbd443a8b311badb03b729b : SUCCESS in 34m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81e90a25622446b88bd63e4dd7d7d95c : SUCCESS in 14m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/af7980323a034afd85d9745af5547497 : SUCCESS in 26m 02s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a53d63a7d8b44b888f28615b51bd8309 : SUCCESS in 1h 39m 38s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/2e7a43e4c9d0445b958b38cd681a25ba : SUCCESS in 1h 06m 21s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/edcfb742e2c748d1b3f950983c43fd89 : SUCCESS in 1h 18m 13s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e9bae51501714d2d979b9f84240eb008 : SUCCESS in 1h 20m 17s (non-voting)\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d4ac648d623e4dcf8f8f84796f504e00 : SUCCESS in 1h 24m 15s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/31952da0a5544be88bccb1935642a3f1 : SUCCESS in 36m 03s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/0c533638c3334548b5bc111747fa4f49 : SUCCESS in 1h 07m 37s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1515758129a54697b5007b16fed8552a : SUCCESS in 1h 16m 57s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":9},{"id":"f5730e59f5abfb5842cb64ace3681dbdd412c10f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-10-02 20:57:58.000000000","message":"Patch Set 9:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/October-2026/Arista_Tempest_Testrefs/changes/93/971593/9 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":9},{"id":"77899afd7d53e4a77e5efdbe3931fd80255e14b9","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-10-05 08:45:37.000000000","message":"Patch Set 9: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":9},{"id":"7f98cf173e08aa2c4c2703518165149d4697f5df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-10-05 19:40:46.000000000","message":"Patch Set 10: Published edit on patch set 9.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"da60805e59eaa6a3502b1982e9eeac888ead6714","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-10-05 19:41:06.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"6cac3e2a83d1e75a77b0af1c9a6b04990f8118f8","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-10-05 19:56:46.000000000","message":"Patch Set 10:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/October-2026/Arista_Tempest_Testrefs/changes/93/971593/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"34e011a54a5dacdbd81bd0ff2a543443653ff781","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-05 20:24:44.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b086750ed67f4863a85c190c19f133ac\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/425b16dbc6a84ecc96b4c8c0b6a40508 : FAILURE in 29m 07s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/73a27b7f61574756ada16af20d2b10dd : FAILURE in 41m 51s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"82f3f57400794e37f38186aeec15b9462e74924c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-05 21:45:19.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0b8f2cffb5454bf98d2b2c265b202013\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/7477c59f60a6472789baa7a17c915161 : SUCCESS in 1h 03m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c6644f800e9347f4980daece3bac1039 : SUCCESS in 1h 02m 46s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/99574eda21c54b999010f56816fc2eec : SUCCESS in 1h 03m 05s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/428a8b1ee46b40be99862156f001ff65 : FAILURE in 1h 41m 52s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8e2f171d67894b4480ccc9bbbc1eb8db : SUCCESS in 29m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6985dd60f50b462894bc427f48eb5927 : FAILURE in 53m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/feaf7a1b9d664f3da7cf94e014b4a42e : SUCCESS in 11m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e7114aef0f954a8f97a2ee0d3cbf6999 : FAILURE in 21m 21s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d1861bddad884653b87c59b6c2ba828c : FAILURE in 34m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/367389a895544cbfa2207a0d7ccefc0d : SUCCESS in 10m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fcbbc3b2aa4748c3992460ca20d15c71 : SUCCESS in 23m 02s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/377e624fff3e47019fa90fe9e521ade6 : SUCCESS in 1h 50m 24s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c67324bca20845d189ad7327fed74c9c : SUCCESS in 43m 52s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/1fe4dfd66d0e49989501f275a79f5f0d : SUCCESS in 49m 37s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/fb43ee57d52f4982a6e4bf7985840bf5 : SUCCESS in 1h 26m 13s (non-voting)\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/73640096e24f46b6b063098a4aa7cc80 : SUCCESS in 1h 05m 01s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a11b62786eff4b01b2fcc94d5c30678a : SUCCESS in 52m 23s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/16f50b1baf384c10925058252dc01c7f : SUCCESS in 39m 17s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3f19220fd72e4801888d04eb88fd6d9c : SUCCESS in 1h 17m 54s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":10},{"id":"1039ca93f8f3dea0c8e7b7eb5eeb8f0cebbcf6a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-10-05 22:14:14.000000000","message":"Patch Set 11: Published edit on patch set 10.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":11},{"id":"28abfa8dfba5c8238ac605536a9512fd319c4f83","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-10-05 22:23:22.000000000","message":"Patch Set 11:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/October-2026/Arista_Tempest_Testrefs/changes/93/971593/11 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":11},{"id":"929e74645b4bd71b3d82a02e84adff0977cc82ac","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-05 23:16:46.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/120d9c62491d4bb7bc42e498871814ac\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/884e1313643f40d4b2456508c7362df7 : SUCCESS in 48m 23s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/eecdaa512cb94356b984e423da553b5e : SUCCESS in 1h 00m 38s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"094fc9ac41d659b7079f939fbbc7cfc0238e1cd9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-05 23:57:24.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/716c88748f854872a5616ba277262319\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/7159cd82835e4d3cbc62e1f306465554 : SUCCESS in 1h 21m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f09a8d2111ec427bae52083b381cfb45 : SUCCESS in 32m 03s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/09c6472b7a1d4f3cbd3891c7cb0edb40 : SUCCESS in 1h 03m 11s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d399acd2d805455dbbe1d20948550289 : SUCCESS in 1h 42m 17s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/609a2119d8164724a389a6df1bf3ff04 : SUCCESS in 27m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/434874ed775147f6a6c5d9c3dadb1e21 : SUCCESS in 44m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aca965d762f34ff8890727806d4b5b65 : SUCCESS in 11m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a709300218444548968f0739fd0e2fc7 : SUCCESS in 31m 58s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8d3be0df7e6c4c9fa1b114323aff0985 : SUCCESS in 21m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0266efd7929b4ad69f56c8a479634681 : SUCCESS in 12m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/262931a4229240e78e790150eda24c0f : SUCCESS in 20m 14s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/1dcad1751c3e4084b899ec5049a95297 : SUCCESS in 1h 24m 12s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/59b0dfd64b0f4beeb0b9554f6a0352a3 : SUCCESS in 38m 28s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/47d4bed6cde943ceb61a5a055af49b88 : SUCCESS in 41m 13s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/7a3f037c7299495db43e3a8145c96877 : SUCCESS in 1h 36m 23s (non-voting)\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1afe3b29958a4c0cb2fd54e4da42ffb1 : SUCCESS in 1h 19m 36s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/1ce3f0b670fd4d6e9779941697d995c0 : SUCCESS in 1h 04m 35s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/67e7a611be99482f9ee6cb71e043564e : SUCCESS in 1h 00m 53s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ff3f4d81956e4caebf09c5d810732d01 : SUCCESS in 1h 20m 06s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"eef65821ee21b22145388aa7f4b3ca2f79a70766","revisions":{"1ed3074ba7c3fa8fb11da9d63642bde1bb8fe9e9":{"kind":"REWORK","_number":1,"created":"2025-12-20 22:07:52.000000000","uploader":{"_account_id":38070,"name":"James Bhattarai","display_name":"james","email":"Jamesbhattarai14@gmail.com","username":"jamesii-b"},"ref":"refs/changes/93/971593/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/93/971593/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/971593/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/971593/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/971593/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/93/971593/1"}}},"commit":{"parents":[{"commit":"1a0e4115995497ac0df9ed5353fabca483a7afb1","subject":"Merge \"Refactor ovn-db-sync script\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1a0e4115995497ac0df9ed5353fabca483a7afb1"}]}],"author":{"name":"James Bhattarai","email":"jamesbhattarai14@gmail.com","date":"2025-12-20 21:57:22.000000000","tz":345},"committer":{"name":"James Bhattarai","email":"jamesbhattarai14@gmail.com","date":"2025-12-20 21:57:22.000000000","tz":345},"subject":"Fix: per-gateway SNAT settings not persisted in multi-gateway routers","message":"Fix: per-gateway SNAT settings not persisted in multi-gateway routers\n\nProblem:\nMulti-gateway routers were ignoring per-gateway enable_snat settings.\nAll gateways would receive SNAT rules regardless of their individual enable_snat value.\n\nSolution:\n1. Add RouterPort.enable_snat column to store per-gateway SNAT setting\n2. Update API layer to store and retrieve per-gateway enable_snat\n3. Update OVN client to read per-gateway settings and apply rules individually\n4. Implement Relevant Migration Files\n\nChanges:\n- neutron/db/models/l3.py: Add enable_snat column to RouterPort\n- neutron/db/l3_extra_gws_db.py: Store per-gateway enable_snat in RouterPort\n- neutron/plugins/ml2/drivers/ovn/mech_driver/ovsdb/ovn_client.py:\n  * Add _iter_ipv4_gw_addrs_with_snat() to read per-gateway settings\n  * Update update_nat_rules() to apply rules per-gateway\n\nhttps://bugs.launchpad.net/neutron/+bug/2106776\n\nCloses-Bug: #2106776\nChange-Id: I86c741e8eba5413875787e636f8d60f79238875d\nSigned-off-by: James Bhatarai \u003cjamesbhattarai14@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1ed3074ba7c3fa8fb11da9d63642bde1bb8fe9e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1ed3074ba7c3fa8fb11da9d63642bde1bb8fe9e9"}]},"branch":"refs/heads/master"},"2184b8632284b6019bb20506a141e7bfc1cc06df":{"kind":"REWORK","_number":2,"created":"2025-12-21 07:49:39.000000000","uploader":{"_account_id":38070,"name":"James Bhattarai","display_name":"james","email":"Jamesbhattarai14@gmail.com","username":"jamesii-b"},"ref":"refs/changes/93/971593/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/93/971593/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/971593/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/971593/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/971593/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/93/971593/2"}}},"commit":{"parents":[{"commit":"1a0e4115995497ac0df9ed5353fabca483a7afb1","subject":"Merge \"Refactor ovn-db-sync script\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1a0e4115995497ac0df9ed5353fabca483a7afb1"}]}],"author":{"name":"James Bhattarai","email":"jamesbhattarai14@gmail.com","date":"2025-12-20 21:57:22.000000000","tz":345},"committer":{"name":"James Bhattarai","email":"jamesbhattarai14@gmail.com","date":"2025-12-21 07:48:41.000000000","tz":345},"subject":"Fix: per-gateway SNAT settings not persisted in multi-gateway routers","message":"Fix: per-gateway SNAT settings not persisted in multi-gateway routers\n\nProblem:\nMulti-gateway routers were ignoring per-gateway enable_snat settings.\nAll gateways would receive SNAT rules regardless of their individual enable_snat value.\n\nSolution:\n1. Add RouterPort.enable_snat column to store per-gateway SNAT setting\n2. Update API layer to store and retrieve per-gateway enable_snat\n3. Update OVN client to read per-gateway settings and apply rules individually\n4. Implement Relevant Migration Files\n\nChanges:\n- neutron/db/models/l3.py: Add enable_snat column to RouterPort\n- neutron/db/l3_extra_gws_db.py: Store per-gateway enable_snat in RouterPort\n- neutron/plugins/ml2/drivers/ovn/mech_driver/ovsdb/ovn_client.py:\n  * Add _iter_ipv4_gw_addrs_with_snat() to read per-gateway settings\n  * Update update_nat_rules() to apply rules per-gateway\n\nPS: Fixed the test failed due to PEP8 E501, DetachedInstanceError, Migration Branch Conflicts\n\nhttps://bugs.launchpad.net/neutron/+bug/2106776\n\nCloses-Bug: #2106776\nChange-Id: I86c741e8eba5413875787e636f8d60f79238875d\nSigned-off-by: James Bhatarai \u003cjamesbhattarai14@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2184b8632284b6019bb20506a141e7bfc1cc06df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2184b8632284b6019bb20506a141e7bfc1cc06df"}]},"branch":"refs/heads/master"},"6b585416ed53818566770a38c485f7d2a1d4f282":{"kind":"REWORK","_number":3,"created":"2025-12-21 11:13:55.000000000","uploader":{"_account_id":38070,"name":"James Bhattarai","display_name":"james","email":"Jamesbhattarai14@gmail.com","username":"jamesii-b"},"ref":"refs/changes/93/971593/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/93/971593/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/971593/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/971593/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/971593/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/93/971593/3"}}},"commit":{"parents":[{"commit":"1a0e4115995497ac0df9ed5353fabca483a7afb1","subject":"Merge \"Refactor ovn-db-sync script\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1a0e4115995497ac0df9ed5353fabca483a7afb1"}]}],"author":{"name":"James Bhattarai","email":"jamesbhattarai14@gmail.com","date":"2025-12-21 11:11:40.000000000","tz":345},"committer":{"name":"James Bhattarai","email":"jamesbhattarai14@gmail.com","date":"2025-12-21 11:11:40.000000000","tz":345},"subject":"Fix: per-gateway SNAT settings not persisted in multi-gateway routers","message":"Fix: per-gateway SNAT settings not persisted in multi-gateway routers\n\nProblem:\nMulti-gateway routers were ignoring per-gateway enable_snat settings.\nAll gateways would receive SNAT rules regardless of their individual enable_snat value.\n\nSolution:\n1. Add RouterPort.enable_snat column to store per-gateway SNAT setting\n2. Update API layer to store and retrieve per-gateway enable_snat\n3. Update OVN client to read per-gateway settings and apply rules individually\n4. Implement Relevant Migration Files\n\nChanges:\n- neutron/db/models/l3.py: Add enable_snat column to RouterPort\n- neutron/db/l3_extra_gws_db.py: Store per-gateway enable_snat in RouterPort\n- neutron/plugins/ml2/drivers/ovn/mech_driver/ovsdb/ovn_client.py:\n  * Add _iter_ipv4_gw_addrs_with_snat() to read per-gateway settings\n  * Update update_nat_rules() to apply rules per-gateway\n\nPS: Fixed the test failed due to PEP8 E501, DetachedInstanceError, Migration Branch Conflicts + Migration revision pointing\n\nhttps://bugs.launchpad.net/neutron/+bug/2106776\n\nCloses-Bug: #2106776\nChange-Id: I86c741e8eba5413875787e636f8d60f79238875d\nSigned-off-by: James Bhatarai \u003cjamesbhattarai14@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6b585416ed53818566770a38c485f7d2a1d4f282"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6b585416ed53818566770a38c485f7d2a1d4f282"}]},"branch":"refs/heads/master"},"4534f0da89aa8400374910e7d945a32ce1cb81be":{"kind":"REWORK","_number":4,"created":"2025-12-25 10:00:21.000000000","uploader":{"_account_id":38070,"name":"James Bhattarai","display_name":"james","email":"Jamesbhattarai14@gmail.com","username":"jamesii-b"},"ref":"refs/changes/93/971593/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/93/971593/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/971593/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/971593/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/971593/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/93/971593/4"}}},"commit":{"parents":[{"commit":"0ba83a9c07758f8fc36219422a669ef109e97733","subject":"Merge \"Change some functional test code to use project_id\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0ba83a9c07758f8fc36219422a669ef109e97733"}]}],"author":{"name":"James Bhattarai","email":"jamesbhattarai14@gmail.com","date":"2025-12-25 09:57:55.000000000","tz":345},"committer":{"name":"James Bhattarai","email":"jamesbhattarai14@gmail.com","date":"2025-12-25 09:57:55.000000000","tz":345},"subject":"Fix: per-gateway SNAT settings not persisted in multi-gateway routers","message":"Fix: per-gateway SNAT settings not persisted in multi-gateway routers\n\nProblem:\nMulti-gateway routers were ignoring per-gateway enable_snat settings.\nAll gateways would receive SNAT rules regardless of their individual enable_snat value.\n\nSolution:\n1. Add RouterPort.enable_snat column to store per-gateway SNAT setting\n2. Update API layer to store and retrieve per-gateway enable_snat\n3. Update OVN client to read per-gateway settings and apply rules individually\n4. Implement Relevant Migration Files\n\nPS: Fixed the test failed due to PEP8 E501 + Migration revision pointing\n\nhttps://bugs.launchpad.net/neutron/+bug/2106776\n\nCloses-Bug: #2106776\nChange-Id: I86c741e8eba5413875787e636f8d60f79238875d\nSigned-off-by: James Bhatarai \u003cjamesbhattarai14@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4534f0da89aa8400374910e7d945a32ce1cb81be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4534f0da89aa8400374910e7d945a32ce1cb81be"}]},"branch":"refs/heads/master"},"5785aa7e189c6dd71d07ff97ed29d369020a1137":{"kind":"REWORK","_number":5,"created":"2026-01-05 12:48:32.000000000","uploader":{"_account_id":38070,"name":"James Bhattarai","display_name":"james","email":"Jamesbhattarai14@gmail.com","username":"jamesii-b"},"ref":"refs/changes/93/971593/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/93/971593/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/971593/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/971593/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/971593/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/93/971593/5"}}},"commit":{"parents":[{"commit":"185b5ffe2b8f91dbbac38667e03c07969dc99e7a","subject":"Doc: add warnig for fullstack","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/185b5ffe2b8f91dbbac38667e03c07969dc99e7a"}]}],"author":{"name":"James Bhattarai","email":"jamesbhattarai14@gmail.com","date":"2026-01-05 12:42:58.000000000","tz":345},"committer":{"name":"James Bhattarai","email":"jamesbhattarai14@gmail.com","date":"2026-01-05 12:42:58.000000000","tz":345},"subject":"Fix: per-gateway SNAT settings not persisted in multi-gateway routers","message":"Fix: per-gateway SNAT settings not persisted in multi-gateway routers\n\nProblem:\nMulti-gateway routers were ignoring per-gateway enable_snat settings.\nAll gateways would receive SNAT rules regardless of their individual enable_snat value.\n\nSolution:\n1. Add RouterPort.enable_snat column to store per-gateway SNAT setting\n2. Update API layer to store and retrieve per-gateway enable_snat\n3. Update OVN client to read per-gateway settings and apply rules individually\n4. Implement Relevant Migration Files\n\nPS: Fixed the test failed+ Migration revision pointing\n\nhttps://bugs.launchpad.net/neutron/+bug/2106776\n\nCloses-Bug: #2106776\nChange-Id: I86c741e8eba5413875787e636f8d60f79238875d\nSigned-off-by: James Bhatarai \u003cjamesbhattarai14@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5785aa7e189c6dd71d07ff97ed29d369020a1137"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5785aa7e189c6dd71d07ff97ed29d369020a1137"}]},"branch":"refs/heads/master"},"291b0d516fb70efd55000055ee15add8935ce530":{"kind":"REWORK","_number":6,"created":"2026-01-05 15:02:51.000000000","uploader":{"_account_id":38070,"name":"James Bhattarai","display_name":"james","email":"Jamesbhattarai14@gmail.com","username":"jamesii-b"},"ref":"refs/changes/93/971593/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/93/971593/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/971593/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/971593/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/971593/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/93/971593/6"}}},"commit":{"parents":[{"commit":"43f17c2bf54c2ff6c7ffe0fdafe3b076588fd4fb","subject":"Merge \"bgp: Improve existing server Commands\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/43f17c2bf54c2ff6c7ffe0fdafe3b076588fd4fb"}]}],"author":{"name":"James Bhattarai","email":"jamesbhattarai14@gmail.com","date":"2026-01-05 14:56:13.000000000","tz":345},"committer":{"name":"James Bhattarai","email":"jamesbhattarai14@gmail.com","date":"2026-01-05 14:59:01.000000000","tz":345},"subject":"Fix: per-gateway SNAT settings not persisted in multi-gateway routers","message":"Fix: per-gateway SNAT settings not persisted in multi-gateway routers\n\nProblem:\nMulti-gateway routers were ignoring per-gateway enable_snat settings.\nAll gateways would receive SNAT rules regardless of their individual enable_snat value.\n\nSolution:\n1. Add RouterPort.enable_snat column to store per-gateway SNAT setting\n2. Update API layer to store and retrieve per-gateway enable_snat\n3. Update OVN client to read per-gateway settings and apply rules individually\n4. Implement Relevant Migration Files\n\nPS: Fixed the test failed+ Migration revision points\n\nhttps://bugs.launchpad.net/neutron/+bug/2106776\n\nCloses-Bug: #2106776\nChange-Id: I86c741e8eba5413875787e636f8d60f79238875d\nSigned-off-by: James Bhatarai \u003cjamesbhattarai14@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/291b0d516fb70efd55000055ee15add8935ce530"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/291b0d516fb70efd55000055ee15add8935ce530"}]},"branch":"refs/heads/master"},"70a2cc0474f47769b14f9b08facc46975a2b6cf2":{"kind":"REWORK","_number":7,"created":"2026-01-05 17:43:14.000000000","uploader":{"_account_id":38070,"name":"James Bhattarai","display_name":"james","email":"Jamesbhattarai14@gmail.com","username":"jamesii-b"},"ref":"refs/changes/93/971593/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/93/971593/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/971593/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/971593/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/971593/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/93/971593/7"}}},"commit":{"parents":[{"commit":"43f17c2bf54c2ff6c7ffe0fdafe3b076588fd4fb","subject":"Merge \"bgp: Improve existing server Commands\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/43f17c2bf54c2ff6c7ffe0fdafe3b076588fd4fb"}]}],"author":{"name":"James Bhattarai","email":"jamesbhattarai14@gmail.com","date":"2026-01-05 17:29:23.000000000","tz":345},"committer":{"name":"James Bhattarai","email":"jamesbhattarai14@gmail.com","date":"2026-01-05 17:29:23.000000000","tz":345},"subject":"Fix: per-gateway SNAT settings not persisted in multi-gateway routers","message":"Fix: per-gateway SNAT settings not persisted in multi-gateway routers\n\nProblem:\nMulti-gateway routers were ignoring per-gateway enable_snat settings.\nAll gateways would receive SNAT rules regardless of their individual enable_snat value.\n\nSolution:\n1. Add RouterPort.enable_snat column to store per-gateway SNAT setting\n2. Update API layer to store and retrieve per-gateway enable_snat\n3. Update OVN client to read per-gateway settings and apply rules individually\n4. Implement Relevant Migration Files\n\nPS: Fixed the test failed+ Migration revision points\n\nhttps://bugs.launchpad.net/neutron/+bug/2106776\n\nCloses-Bug: #2106776\nChange-Id: I86c741e8eba5413875787e636f8d60f79238875d\nSigned-off-by: James Bhatarai \u003cjamesbhattarai14@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/70a2cc0474f47769b14f9b08facc46975a2b6cf2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/70a2cc0474f47769b14f9b08facc46975a2b6cf2"}]},"branch":"refs/heads/master"},"0c04223e8e3ad8d3e3fc73b9a3e4ffb5b9127e32":{"kind":"REWORK","_number":8,"created":"2026-10-02 07:39:07.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/93/971593/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/93/971593/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/971593/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/971593/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/971593/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/93/971593/8"}}},"commit":{"parents":[{"commit":"59fb77d697c8caea813ee0f287ecb68304c88795","subject":"Merge \"Minor refactor of the EVPN FSM\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/59fb77d697c8caea813ee0f287ecb68304c88795"}]}],"author":{"name":"James Bhattarai","email":"jamesbhattarai14@gmail.com","date":"2026-01-05 17:29:23.000000000","tz":345},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-10-02 07:38:57.000000000","tz":120},"subject":"Fix: per-gateway SNAT settings not persisted in multi-gateway routers","message":"Fix: per-gateway SNAT settings not persisted in multi-gateway routers\n\nProblem:\nMulti-gateway routers were ignoring per-gateway enable_snat settings.\nAll gateways would receive SNAT rules regardless of their individual enable_snat value.\n\nSolution:\n1. Add RouterPort.enable_snat column to store per-gateway SNAT setting\n2. Update API layer to store and retrieve per-gateway enable_snat\n3. Update OVN client to read per-gateway settings and apply rules individually\n4. Implement Relevant Migration Files\n\nPS: Fixed the test failed+ Migration revision points\n\nhttps://bugs.launchpad.net/neutron/+bug/2106776\n\nCloses-Bug: #2106776\nChange-Id: I86c741e8eba5413875787e636f8d60f79238875d\nSigned-off-by: James Bhatarai \u003cjamesbhattarai14@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0c04223e8e3ad8d3e3fc73b9a3e4ffb5b9127e32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0c04223e8e3ad8d3e3fc73b9a3e4ffb5b9127e32"}]},"branch":"refs/heads/master"},"3cecad2bcaab7537f274458f6ce953216ad9ed93":{"kind":"REWORK","_number":9,"created":"2026-10-02 15:14:13.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/93/971593/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/93/971593/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/971593/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/971593/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/971593/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/93/971593/9"}}},"commit":{"parents":[{"commit":"59fb77d697c8caea813ee0f287ecb68304c88795","subject":"Merge \"Minor refactor of the EVPN FSM\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/59fb77d697c8caea813ee0f287ecb68304c88795"}]}],"author":{"name":"James Bhattarai","email":"jamesbhattarai14@gmail.com","date":"2026-01-05 17:29:23.000000000","tz":345},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-10-02 15:14:05.000000000","tz":0},"subject":"Fix: per-gateway SNAT settings not persisted in multi-gateway routers","message":"Fix: per-gateway SNAT settings not persisted in multi-gateway routers\n\nProblem:\nMulti-gateway routers were ignoring per-gateway enable_snat settings.\nAll gateways would receive SNAT rules regardless of their individual enable_snat value.\n\nSolution:\n1. Add RouterPort.enable_snat column to store per-gateway SNAT setting\n2. Update API layer to store and retrieve per-gateway enable_snat\n3. Update OVN client to read per-gateway settings and apply rules individually\n4. Implement Relevant Migration Files\n\nPS: Fixed the test failed+ Migration revision points\n\nhttps://bugs.launchpad.net/neutron/+bug/2106776\n\nCloses-Bug: #2106776\nChange-Id: I86c741e8eba5413875787e636f8d60f79238875d\nSigned-off-by: James Bhatarai \u003cjamesbhattarai14@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3cecad2bcaab7537f274458f6ce953216ad9ed93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3cecad2bcaab7537f274458f6ce953216ad9ed93"}]},"branch":"refs/heads/master"},"34ef448ee4e87fbba68cdf79e03f625702ab3c94":{"kind":"REWORK","_number":10,"created":"2026-10-05 19:40:46.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/93/971593/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/93/971593/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/971593/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/971593/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/971593/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/93/971593/10"}}},"commit":{"parents":[{"commit":"59fb77d697c8caea813ee0f287ecb68304c88795","subject":"Merge \"Minor refactor of the EVPN FSM\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/59fb77d697c8caea813ee0f287ecb68304c88795"}]}],"author":{"name":"James Bhattarai","email":"jamesbhattarai14@gmail.com","date":"2026-01-05 17:29:23.000000000","tz":345},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-10-05 19:40:14.000000000","tz":0},"subject":"Fix: per-gateway SNAT settings not persisted in multi-gateway routers","message":"Fix: per-gateway SNAT settings not persisted in multi-gateway routers\n\nProblem:\nMulti-gateway routers were ignoring per-gateway enable_snat settings.\nAll gateways would receive SNAT rules regardless of their individual enable_snat value.\n\nSolution:\n1. Add RouterPort.enable_snat column to store per-gateway SNAT setting\n2. Update API layer to store and retrieve per-gateway enable_snat\n3. Update OVN client to read per-gateway settings and apply rules individually\n4. Implement Relevant Migration Files\n\nPS: Fixed the test failed+ Migration revision points\n\nhttps://bugs.launchpad.net/neutron/+bug/2106776\n\nCloses-Bug: #2106776\nChange-Id: I86c741e8eba5413875787e636f8d60f79238875d\nSigned-off-by: James Bhatarai \u003cjamesbhattarai14@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/34ef448ee4e87fbba68cdf79e03f625702ab3c94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/34ef448ee4e87fbba68cdf79e03f625702ab3c94"}]},"branch":"refs/heads/master"},"eef65821ee21b22145388aa7f4b3ca2f79a70766":{"kind":"REWORK","_number":11,"created":"2026-10-05 22:14:14.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/93/971593/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/93/971593/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/971593/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/971593/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/971593/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/93/971593/11"}}},"commit":{"parents":[{"commit":"59fb77d697c8caea813ee0f287ecb68304c88795","subject":"Merge \"Minor refactor of the EVPN FSM\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/59fb77d697c8caea813ee0f287ecb68304c88795"}]}],"author":{"name":"James Bhattarai","email":"jamesbhattarai14@gmail.com","date":"2026-01-05 17:29:23.000000000","tz":345},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-10-05 22:12:41.000000000","tz":0},"subject":"Fix: per-gateway SNAT settings not persisted in multi-gateway routers","message":"Fix: per-gateway SNAT settings not persisted in multi-gateway routers\n\nProblem:\nMulti-gateway routers were ignoring per-gateway enable_snat settings.\nAll gateways would receive SNAT rules regardless of their individual enable_snat value.\n\nSolution:\n1. Add RouterPort.enable_snat column to store per-gateway SNAT setting\n2. Update API layer to store and retrieve per-gateway enable_snat\n3. Update OVN client to read per-gateway settings and apply rules individually\n4. Implement Relevant Migration Files\n\nPS: Fixed the test failed+ Migration revision points\n\nhttps://bugs.launchpad.net/neutron/+bug/2106776\n\nCloses-Bug: #2106776\nChange-Id: I86c741e8eba5413875787e636f8d60f79238875d\nSigned-off-by: James Bhatarai \u003cjamesbhattarai14@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/eef65821ee21b22145388aa7f4b3ca2f79a70766"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/eef65821ee21b22145388aa7f4b3ca2f79a70766"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
