)]}'
{"id":"openstack%2Fneutron~971876","triplet_id":"openstack%2Fneutron~master~Id8c0dbfd38c5d81fd089e9ebe0d521a03b367079","project":"openstack/neutron","branch":"master","topic":"lp2136733-gw-port-binding-fail","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-03-13 13:08:56.000000000","reason":"Change was abandoned"},"30534":{"account":{"_account_id":30534,"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","username":"wilkmar"},"last_update":"2026-03-13 13:05:44.000000000","reason":"\u003cGERRIT_ACCOUNT_30534\u003e replied on the change","reason_account":{"_account_id":30534,"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","username":"wilkmar"}},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2026-03-13 13:08:56.000000000","reason":"Change was abandoned"},"5756":{"account":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"last_update":"2026-03-13 13:08:56.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Id8c0dbfd38c5d81fd089e9ebe0d521a03b367079","subject":"Use SB DB for checking if an OVN agent is alive","status":"ABANDONED","created":"2025-12-28 13:42:09.000000000","updated":"2026-03-13 13:08:56.000000000","total_comment_count":18,"unresolved_comment_count":4,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"59bd02be1dcc1ca6dd0f32157370c08caae1e0d6","_number":971876,"virtual_id_number":971876,"owner":{"_account_id":30534,"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","username":"wilkmar"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-03-02 22:55:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2025-12-28 14:43:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-12-28 15:57:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-29 16:48:53.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2026-01-05 09:59:51.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"CC"},{"updated":"2026-01-05 09:59:51.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2026-01-26 08:09:03.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2026-01-29 08:57:38.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2026-02-06 12:25:12.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"CC"},{"updated":"2026-02-09 17:39:09.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"}],"messages":[{"id":"b15736b828b7028942f4a99ee015d853f2d2011f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30534,"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","username":"wilkmar"},"date":"2025-12-28 13:42:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5ac793376d682240583f82ccc188350de1321983","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":30534,"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","username":"wilkmar"},"date":"2025-12-28 13:42:53.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"b311af56efa483cd925812e854992c0cbc7be8d0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-28 14:43:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e489a12539294874942046425925dd70\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/da7b0a8a913d42eebe1f3f66fbac4184 : SUCCESS in 59m 24s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/db10dfe264674cdba8a3acd59bfdd586 : SUCCESS in 30m 20s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/bf1f8233685243ac9992fb8ee87fdbde : SUCCESS in 34m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6351370c304bb97d91bc6a85ea0a94d330599269","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-28 15:57:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6b83881f858b49d4be95eb73776790d8\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/ea14dc3c695844279a798986463f51f0 : FAILURE in 44m 38s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a5343615c19b44fb9cefc66585c08c31 : SUCCESS in 2h 13m 15s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c75e1f075b44478a969dfe2648364b4f : SUCCESS in 34m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/df9712c8461a466fa8ffdde931410604 : SUCCESS in 1h 12m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/702a4478fac742c4ae89a0452dd694cc : SUCCESS in 25m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a1c33985adde4518a3910d593a6e8461 : SUCCESS in 34m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d54800b4cf304a1f9bf53e7d93bf4670 : SUCCESS in 33m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/981658fd5aa84a72b4ab77179df56b58 : SUCCESS in 10m 16s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/354ccbc1a34c4d238449946a3c09d875 : SUCCESS in 33m 09s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/8efaaf5b4ac04f9883d769525d52bb1e : SUCCESS in 1h 08m 16s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/017c26b18f964caa932762bf7259e533 : FAILURE in 26m 19s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/97f541a194014cff9c8fe786dd16ba3b : SUCCESS in 54m 58s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/adaf58714a6a4d4597a9329a09dda245 : SUCCESS in 52m 25s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/183a7e2f38e049ca8265660fa431340b : SUCCESS in 1h 49m 42s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e5e8b46a8f0443c08fde8cc4f318c45e : SUCCESS in 1h 34m 00s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/3abc85bbaea44ef8a65ec45bb1b77997 : SUCCESS in 1h 00m 45s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/cbc1bca3933d499dbdb39641d686765f : SUCCESS in 1h 07m 57s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/11a0e4a4432e44adaaaf97f311532f75 : SUCCESS in 40m 40s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f9901da0767b451ba1af5d51725fa218 : SUCCESS in 1h 16m 17s","accounts_in_message":[],"_revision_number":1},{"id":"81538f285fad7e36493cb10921418654ba419648","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30534,"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","username":"wilkmar"},"date":"2025-12-29 07:21:32.000000000","message":"Uploaded patch set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"a45a54e4bff6eb123ad43028545e448e04e81249","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-29 07:44:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b388e7434964a4cb342c2bcfa1b8285\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e11083d556084aa08e5c7319f6142d24 : NODE_FAILURE Node(set) request 4e2fca63e9df43bd8f1e9d0a46149d55 failed in 0s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c5102b50e7534bbfbe30ed9e4abb01ac : NODE_FAILURE Node(set) request 7c06e6b6a59b4c98937a7c7835fe358a failed in 0s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1600f6f2367a4947b960234869d74605 : NODE_FAILURE Node(set) request 9d6145938b6846059e97abf299a9bc34 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f2ce7e0660b31b55de8e50e9c4e9296d372c99ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-29 09:28:56.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8e6fd848861f45168d8f2b8271a8476f\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/83243cd00e714ce7b2d89fd6ed019c7a : FAILURE in 1h 26m 01s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/5f0c4c29433748d1b678b7e12b19997a : SUCCESS in 2h 05m 32s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c34a12b487d744048203463c7eef49a5 : SUCCESS in 32m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8c72f42019194cf3a4278f56c705e2c8 : SUCCESS in 42m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55c83e2273404fb5ba7ad4d017dbbbaa : SUCCESS in 11m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8320d70c144a49ba9a3eac4165fe0049 : SUCCESS in 42m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e13b1c21f3f840f082d06bb5a5c11c15 : SUCCESS in 31m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77473ff318e747f19b28de3ee88de25a : SUCCESS in 7m 04s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/0bb400f84c2b4fc79562e08eb6ddd5cc : SUCCESS in 45m 26s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/8007f0585d9d4da4849f35226b415a2c : SUCCESS in 48m 49s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/fa889b170fc348f4b17d114b30dc4296 : FAILURE in 46m 12s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/5b04ce70cb724f99b6c9f5a62000c153 : SUCCESS in 56m 06s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/c6047f63943a4eb8bedd2fe3e3cc5e06 : SUCCESS in 39m 25s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/2df84594c5b84902b8f6d4b688bed9d0 : SUCCESS in 1h 51m 04s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/da6effb64f5b4f64a37f5f451ee030bf : SUCCESS in 1h 32m 12s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/5eb0c0a10d1545f989d084bd901c0f9d : SUCCESS in 1h 04m 31s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/38952ce7b5bf469ebe6a1e88a1178d58 : SUCCESS in 55m 51s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1a654af35ccf4da29090aaa8a7dfd2da : SUCCESS in 35m 14s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/965d27a7e8954ee6a50f117d55d6280a : SUCCESS in 1h 27m 28s","accounts_in_message":[],"_revision_number":2},{"id":"fd1ec52a707021673671c04d2f38508b5a5aee8a","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-01-05 15:25:01.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3a7d7ce16d8e3f85d888856ad95e9424205e21a1","author":{"_account_id":30534,"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","username":"wilkmar"},"date":"2026-01-06 14:49:46.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"aab9b1e2206cfe6fbf08773d13d25f2be390a037","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30534,"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","username":"wilkmar"},"date":"2026-01-07 09:02:22.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"338441bb0f0095860f075510af7cee6d5c5fc4ae","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 10:09:21.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb4b2bd2aa9f4da888afb9cbf4f45ebe\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b91cbe13533d4d04a531cf3608694bc2 : SUCCESS in 31m 47s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0fd931f5d47b4dfa90f35ba7f2c38d7b : SUCCESS in 53m 01s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5622451e88d24b5a8d1efb71d4f68707 : SUCCESS in 1h 04m 53s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"182a875cd610c3e098ee274d9b0227363f1c0c85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 11:05:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ebe20b7a5644705aa8edc000f6cf494\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/5d9738179840408e95f4e83c7e0b46b9 : SUCCESS in 1h 17m 52s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/bd3c99d36b4943d98055e6d0a9e90944 : SUCCESS in 1h 51m 32s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b5f68c9553a9440fb5bc6019450cff87 : SUCCESS in 29m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ee8e97aa3e934945b9f1daef9408d0e2 : SUCCESS in 56m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4490e68d67f4e28bdac6939afb54c5e : SUCCESS in 24m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9c4c68c7eebe427cba542221b9675a1d : SUCCESS in 42m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/11301dc48ec149f8b3cf08476b2031da : SUCCESS in 40m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f8bf224637747db872f9a0b0404f4fb : SUCCESS in 14m 47s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/128980214aea4b97b3d01ba39559afca : SUCCESS in 1h 00m 19s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/a6679b6ce79a415e9ecf63ecbc41bf5b : SUCCESS in 39m 06s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/6b23f557f9424abea1f9db644afe70df : SUCCESS in 1h 51m 27s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e5999acc039d495b8c39af12d69be3d8 : SUCCESS in 1h 08m 46s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/731f342a95bb4598b1d04e016008c306 : SUCCESS in 1h 10m 46s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/f842a001c4734c359e52eee80ff64b84 : SUCCESS in 1h 05m 39s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/5500c25961f24a59b6e0e27d17a3fcc6 : SUCCESS in 58m 50s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/233d6035dc2543f289d45ef610aca988 : SUCCESS in 1h 06m 55s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/bc29899044554077b2f9c21cabe769f6 : SUCCESS in 52m 41s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0f3b778b7ff14431b6b85d075e7dc97f : SUCCESS in 1h 00m 03s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/1b624c9fedb74697ba51af8e25770d71 : SUCCESS in 1h 00m 17s","accounts_in_message":[],"_revision_number":3},{"id":"5b0ba1447ce71b98a0119e9524480b893fef8ec1","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-01-26 08:09:03.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"804395de1969821a00e1a400bd3a9217a889d61c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30534,"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","username":"wilkmar"},"date":"2026-02-04 17:16:21.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"73bc3a82894a9ecfc94cf35746e6659e18be0e38","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-04 18:06:13.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f37febf9af9c49ffbb2052cd883b7f8c\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/445cc882ff0e4a67bd8a459b94c66a12 : SUCCESS in 35m 44s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/eacc747719974dedb3d1deddb0de91a6 : SUCCESS in 45m 38s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/214a2eb4eed24812b3bda79c6f72663a : RETRY_LIMIT in 4m 11s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e5ae69ee3d75033e73fbcff04d841874c1ca25a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-04 19:29:59.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be634f1716dd4477876fbb3f999e5774\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/35ab4f9267074628b3d257a3922d275a : SUCCESS in 1h 25m 23s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/fe8e554a582e4b70b86f9e9376f1c035 : SUCCESS in 2h 08m 58s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/af43fdeaa37c4708b5eddcf936a344f8 : SUCCESS in 32m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a680339400634f79a5abee2b363dbb42 : SUCCESS in 41m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53b04125ca254451aec83a163fda75ae : SUCCESS in 28m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fd42e7af0891430ab08d8e8986daeb8e : SUCCESS in 39m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2d8175b6585b45ada12d85d434892627 : SUCCESS in 34m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0550f9a99f8d4e16873699ce0fc416d0 : SUCCESS in 11m 41s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/0b8c0456152c44fd8dbd78e8e35b5e18 : SUCCESS in 53m 51s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/97255b54b4e345d586a46d88b4e068d2 : SUCCESS in 1h 10m 24s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/89d813383d8a468193f8acb80a888599 : SUCCESS in 45m 17s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c0f4680a2aa34ef5810e3cd3151b642b : SUCCESS in 57m 49s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/fe3ad09f721541bb86a88c0afe95a010 : SUCCESS in 40m 55s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/8d9a38da5ff84831bcb8c66821243308 : SUCCESS in 2h 00m 16s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9aa395e58d9048b785eac3b2dd380e1a : SUCCESS in 1h 06m 32s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/7990757b3c894753b1cb7fa4c2520d43 : SUCCESS in 1h 11m 12s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/9f9a6f05ea82474aa072bf0635beb182 : SUCCESS in 52m 34s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e6cf55f4860648aeb8a05346381ea3ff : SUCCESS in 1h 46m 01s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c8665b2950504ad2badbf6fc59bec05e : SUCCESS in 1h 06m 07s","accounts_in_message":[],"_revision_number":4},{"id":"fd069ac383591cf851f8185edc226adac4033c37","author":{"_account_id":30534,"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","username":"wilkmar"},"date":"2026-02-05 07:54:10.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"c39fa204a64194a540c7f8b5371b520284751438","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-02-06 12:24:51.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e7a058d7efa337c6e31696bdbbe6338fdeb1cead","author":{"_account_id":30534,"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","username":"wilkmar"},"date":"2026-02-06 15:24:31.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b5799d880e1aef2f9506916c48d0ddb68fde0f34","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-02-09 08:29:16.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a04ac8697cc8c1fcde15a55eaa63f8f88d4dc206","author":{"_account_id":30534,"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","username":"wilkmar"},"date":"2026-02-09 14:27:47.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5df22ca0579ab3d30707cf300a9afa476e2c7e47","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2026-02-09 17:39:09.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4589162fee131077ab43d9a894919176d4834b7f","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":30534,"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","username":"wilkmar"},"date":"2026-02-11 14:03:47.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":4},{"id":"e2c041fec54b40b538dace3108f9c26e42ee6b0b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30534,"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","username":"wilkmar"},"date":"2026-02-11 16:58:57.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"a812e2bfbbc44ba8232fc1ad73731dd4fc1e25db","author":{"_account_id":30534,"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","username":"wilkmar"},"date":"2026-02-11 17:05:58.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"1f2605099ae7b942683df6e5f2542f85461c5c60","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 18:07:56.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e942bde39fa745a1ab7dde35fc2b2270\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/41e6132dc2fc47e2b602b7594a293a7c : SUCCESS in 1h 02m 41s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3a0a66203895443ea0889e3ed674cbe0 : SUCCESS in 42m 41s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1f908ee4c44b41eab3ca844d3b636502 : RETRY_LIMIT in 3m 57s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"3e84d24a97e9da0de565bbc9e1b555291593dfc7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 19:13:57.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2aae8f14456d4d019d15501fd691344b\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/bc1268b36b234e65a1059fb7841f8210 : SUCCESS in 2h 10m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e768479326ff4df094011e4c4bbc9ab1 : SUCCESS in 52m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c3c1dd6f5b2475caa340208e8d81b14 : SUCCESS in 25m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/845ded618f0148e28fdf7c27d8f2f728 : SUCCESS in 32m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8bb55a88d5274c19a167a294adba7cba : SUCCESS in 23m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bec29ca34e6434fbfc917a34085e8d4 : SUCCESS in 10m 19s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/30d820c223a24d43bf95f124a0294542 : SUCCESS in 53m 42s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/050cbaba3dfe4e708ef286ebe7c01d70 : SUCCESS in 42m 26s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/dde97146f27d47ad95baffe57039876b : SUCCESS in 1h 33m 35s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/b5f66f29e70146b1bbc2c45fd5229021 : SUCCESS in 1h 16m 21s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8408de249d0e4ed8a6c101ded695e2dc : SUCCESS in 1h 22m 06s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/75198e2f0b0b46b18e48420c24812c9d : SUCCESS in 1h 17m 43s","accounts_in_message":[],"_revision_number":5},{"id":"6a5f0349bfd81eeca45c08f9626477d520d2d2a3","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2026-02-12 14:55:07.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"8d43e7c120b00ceaa91ca429706da4dba58857b9","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30534,"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","username":"wilkmar"},"date":"2026-02-23 08:42:22.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"fd1b8cfddb5abe697c77fa2e7ddf37af6342b6bf","author":{"_account_id":30534,"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","username":"wilkmar"},"date":"2026-02-23 09:05:06.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"3dfefea5cf7939612c50b9c0086533a230d54507","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-23 09:15:02.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1473351ff6764bbda4e89ddfd73051d9\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1f5a6ac27dee4beaa3e93f02fe2c9b41 : SUCCESS in 25m 44s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8cde0f3f38954a29b31a746e41d543bf : SUCCESS in 29m 33s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/40565c9df7d94ee2b417504dbe98a649 : RETRY_LIMIT in 3m 16s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"f6e00c0a040d9106d36cc8e4815ba40a61d77eeb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-23 10:44:06.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/088f8cdf88394b46802ac0ef4a90d4bd\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/97ebadc29cf44f1a902b9e2054e6989e : SUCCESS in 2h 00m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f9a3e72b946447d599c8b036bd3b3737 : SUCCESS in 1h 17m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2a60eef4d7541a2bf35cf4a9811edee : SUCCESS in 23m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/53dba522a76d4a84a5f4461288f17071 : SUCCESS in 25m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/81de3dd5a7464e54aba57b96939b9b64 : SUCCESS in 17m 49s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ad0a633de4d24eeba291ce8a1255e2ce : SUCCESS in 29m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1545504914f74b90ae89b32568403b80 : SUCCESS in 9m 04s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/d0b8d4cd36b64131a4eb66f62dbaeeab : SUCCESS in 51m 35s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/9595b696818946a78eae1fe2e04c45a9 : FAILURE in 8m 52s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/2e72e80ee3fd44afa1137aed4f659783 : SUCCESS in 1h 18m 00s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/4464fbf7c996499fb23b689d33cbfe06 : SUCCESS in 1h 09m 52s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d5e155e372e94a2e8c7579d3f33a800b : SUCCESS in 1h 25m 09s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a02bde881c9a4c6f8aa18c453c9fed73 : SUCCESS in 1h 03m 39s","accounts_in_message":[],"_revision_number":6},{"id":"4b57a06254eb77e4ee1c5aceb81ea0953c5a9bba","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-02-25 17:51:05.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"1aa6367ef44097daa9dda9a5f092e88067467937","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-25 19:35:25.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/62df489a58ad4e5cb785667ffa21257b\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c2a77001a3ee40b393d8ea530e15bae7 : FAILURE in 28m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6b4a6b56053f4cdaa1783605dbe519d9 : SUCCESS in 46m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca2e8156cf204037a842b1e9a907e511 : SUCCESS in 23m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b1f7b6f1e1db4164a276825619383485 : SUCCESS in 35m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/371aab91fe8e4dc3b6c6439dc39cb5f4 : SUCCESS in 25m 18s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0e83ef76d03b4140b8384a6233cbdee6 : SUCCESS in 32m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1f81586d68d4f0197b6e3f1826651b3 : SUCCESS in 11m 53s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/00299a4c7a124714be8ce8c71c3258e9 : SUCCESS in 41m 00s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/cd6b30b93f474a92a458bd7c34229c2e : SUCCESS in 1h 27m 38s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/6dc39275632b40a8aece68d8623280dc : SUCCESS in 40m 11s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/7b2bd856ded94e38988f93572c3597a7 : SUCCESS in 1h 07m 49s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1b9efdbcbdd446888974a9ee70acf01c : SUCCESS in 1h 38m 36s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3f1cd00ca13a45a28244804b9d5e6ca1 : SUCCESS in 1h 27m 12s","accounts_in_message":[],"_revision_number":6},{"id":"5780b9405f33c967d16edc2553f801d68feabf2a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-25 20:07:49.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f15160fa7a9647eda26ed50b2108b9df\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ae51d2f8b1b5488ba49e4bd5690a2488 : SUCCESS in 31m 40s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/aae22a7d964a4dab9cf6ce638453e9d4 : RETRY_LIMIT in 3m 45s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/1cddba1bfb19499188c28ab681dae86f : SUCCESS in 24m 23s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/756ba645eda14082b1703035047e9a49 : SUCCESS in 25m 21s","accounts_in_message":[],"_revision_number":6},{"id":"3a6478df937426fe96f059819c3032e9e53f0ed1","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-03-02 21:00:43.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"e0329bfab53bbf381c5ca82dab87a6c6cf8ab24c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-02 22:55:43.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e952151949194fda9a53d7566cba9209\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a347185f4eed4a419735dd0bf21931a5 : SUCCESS in 1h 12m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3a51e3ddee6940f6a2ddb26ef2966d9b : SUCCESS in 1h 03m 52s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c90658d97d9842b7b89fa755bab409f9 : SUCCESS in 1h 54m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/78cf94377cf4405e8a2204f48bd08168 : SUCCESS in 37m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3a4b3233a654db998f1ee15596a63bb : SUCCESS in 18m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/02c0821be86f45d1b78428ff936bd97d : SUCCESS in 41m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a9c220f7fdf2474c9aed5e3bcf4ed461 : SUCCESS in 15m 57s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/364a627fbf4c41c4b330c71e3db17ccc : SUCCESS in 38m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1bc215ccf7d2432c9aa63f72364276d7 : SUCCESS in 6m 44s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/0ef310d7965e4f668012e7c30a08706e : SUCCESS in 1h 02m 15s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/82a077ca9b104ee29b489ca98b5957d4 : SUCCESS in 1h 27m 05s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/5f936daa428b4eb49b0bd5866c5d90d3 : SUCCESS in 41m 53s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/397f727c19bc47e2b6f3ac2d797d12c8 : SUCCESS in 1h 08m 34s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1e41b97903b044b08ce4ea828c7d2216 : SUCCESS in 1h 31m 25s","accounts_in_message":[],"_revision_number":6},{"id":"93f613e5d048a98595ed9c022efa31eeee141011","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-02 23:23:56.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9048f7c5a2a8437a8df7f5371480dd37\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a5e2d3e2ce1d462d841d6e44ed69fd0b : SUCCESS in 28m 14s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e70ad22d150f419aa3399081f659aa57 : RETRY_LIMIT in 3m 27s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/f52dad64d35447338c47a2c9114ca217 : SUCCESS in 25m 13s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c108c2b13b5841ea88a9e4c2da054aa3 : SUCCESS in 22m 57s","accounts_in_message":[],"_revision_number":6},{"id":"5e9f078a9676cbc1d45555bc42d36954bb05368e","author":{"_account_id":30534,"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","username":"wilkmar"},"date":"2026-03-13 13:08:13.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"59bd02be1dcc1ca6dd0f32157370c08caae1e0d6","tag":"autogenerated:gerrit:abandon","author":{"_account_id":30534,"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","username":"wilkmar"},"date":"2026-03-13 13:08:56.000000000","message":"Abandoned\n\nI think it doesn\u0027t make sense to continue working on this patch as the fail back of the highest priority chassis, is only a side effect of some other issue. Please check my LP bug comment for details [1]. Thank you All for sharing your comments so far.\n\n[1] https://bugs.launchpad.net/neutron/+bug/2136733/comments/6","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"f3269deb622fa46a34f21c5a42bfdc523d663989","revisions":{"07384a8ebbe638ff2b4f227fb8943870f7cd9681":{"kind":"REWORK","_number":1,"created":"2025-12-28 13:42:09.000000000","uploader":{"_account_id":30534,"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","username":"wilkmar"},"ref":"refs/changes/76/971876/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/76/971876/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/971876/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/971876/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/971876/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/76/971876/1"}}},"commit":{"parents":[{"commit":"0ba83a9c07758f8fc36219422a669ef109e97733","subject":"Merge \"Change some functional test code to use project_id\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0ba83a9c07758f8fc36219422a669ef109e97733"}]}],"author":{"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","date":"2025-12-28 13:14:57.000000000","tz":60},"committer":{"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","date":"2025-12-28 13:41:35.000000000","tz":60},"subject":"Add option to skip cache when checking if an OVN agent is alive","message":"Add option to skip cache when checking if an OVN agent is alive\n\nThis patch allows for checking if an OVN agent is alive based on the state in\nthe Southbound DB rather than based on the cached values in the ``AgentCache``.\nIn some situations different RPC or API workers have the ``AgentCache`` out of\nsync with the Southbound DB. As a result they are not aware of the actual state\nof an agent, which may lead to problems, such as a GW port binding failure\non the highest priority OVN Chassis.\nThis patch doesn\u0027t solve the problem of the ``AgentCache`` being out of sync with\nthe Southbound DB, but instead, it adds the DB check to see if an OVN agent\nis alive during port binding.\n\nCloses-Bug: #2136733\nChange-Id: Id8c0dbfd38c5d81fd089e9ebe0d521a03b367079\nSigned-off-by: Marcin Wilk \u003cmarcin.wilk@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/07384a8ebbe638ff2b4f227fb8943870f7cd9681"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/07384a8ebbe638ff2b4f227fb8943870f7cd9681"}]},"branch":"refs/heads/master"},"f7088fd05362bb22eee36ba98a036b1c94963d5d":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-12-29 07:21:32.000000000","uploader":{"_account_id":30534,"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","username":"wilkmar"},"ref":"refs/changes/76/971876/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/76/971876/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/971876/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/971876/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/971876/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/76/971876/2"}}},"commit":{"parents":[{"commit":"0ba83a9c07758f8fc36219422a669ef109e97733","subject":"Merge \"Change some functional test code to use project_id\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0ba83a9c07758f8fc36219422a669ef109e97733"}]}],"author":{"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","date":"2025-12-28 13:14:57.000000000","tz":60},"committer":{"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","date":"2025-12-29 07:12:57.000000000","tz":60},"subject":"Use SB DB for checking if an OVN agent is alive","message":"Use SB DB for checking if an OVN agent is alive\n\nThis patch allows for checking if an OVN agent is alive based on the\nstate in the Southbound DB rather than based on the cached values in\nthe ``AgentCache``. In some situations different RPC or API workers\nhave the ``AgentCache`` out of sync with the Southbound DB. As a\nresult they are not aware of the actual state of an agent, which may\nlead to problems, such as a GW port binding failure on the highest\npriority OVN Chassis.\nThis patch doesn\u0027t solve the problem of the ``AgentCache`` being out\nof sync with the Southbound DB, but instead, it adds the DB check to\nsee if an OVN agent is alive during port binding.\n\nCloses-Bug: #2136733\nChange-Id: Id8c0dbfd38c5d81fd089e9ebe0d521a03b367079\nSigned-off-by: Marcin Wilk \u003cmarcin.wilk@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f7088fd05362bb22eee36ba98a036b1c94963d5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f7088fd05362bb22eee36ba98a036b1c94963d5d"}]},"branch":"refs/heads/master"},"db81a7ccfb2738612bba0db22a73587d3f2020a3":{"kind":"REWORK","_number":3,"created":"2026-01-07 09:02:22.000000000","uploader":{"_account_id":30534,"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","username":"wilkmar"},"ref":"refs/changes/76/971876/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/76/971876/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/971876/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/971876/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/971876/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/76/971876/3"}}},"commit":{"parents":[{"commit":"0ba83a9c07758f8fc36219422a669ef109e97733","subject":"Merge \"Change some functional test code to use project_id\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0ba83a9c07758f8fc36219422a669ef109e97733"}]}],"author":{"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","date":"2025-12-28 13:14:57.000000000","tz":60},"committer":{"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","date":"2026-01-07 09:00:15.000000000","tz":60},"subject":"Use SB DB for checking if an OVN agent is alive","message":"Use SB DB for checking if an OVN agent is alive\n\nThis patch allows for checking if an OVN agent is alive based on the\nstate in the Southbound DB rather than based on the cached values in\nthe ``AgentCache``. In some situations different RPC or API workers\nhave the ``AgentCache`` out of sync with the Southbound DB. As a\nresult they are not aware of the actual state of an agent, which may\nlead to problems, such as a GW port binding failure on the highest\npriority OVN Chassis.\nThis patch doesn\u0027t solve the problem of the ``AgentCache`` being out\nof sync with the Southbound DB, but instead, it adds the DB check to\nsee if an OVN agent is alive during port binding.\n\nCloses-Bug: #2136733\nChange-Id: Id8c0dbfd38c5d81fd089e9ebe0d521a03b367079\nSigned-off-by: Marcin Wilk \u003cmarcin.wilk@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/db81a7ccfb2738612bba0db22a73587d3f2020a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/db81a7ccfb2738612bba0db22a73587d3f2020a3"}]},"branch":"refs/heads/master"},"11f5a2d08f91b044084d1f3ceafd9beada9e68d9":{"kind":"REWORK","_number":4,"created":"2026-02-04 17:16:21.000000000","uploader":{"_account_id":30534,"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","username":"wilkmar"},"ref":"refs/changes/76/971876/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/76/971876/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/971876/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/971876/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/971876/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/76/971876/4"}}},"commit":{"parents":[{"commit":"6e0e6a5bc91ed3c91c2cbec75d4ea2e06ca89dfe","subject":"Merge \"Add information to network ip availabilities\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6e0e6a5bc91ed3c91c2cbec75d4ea2e06ca89dfe"}]}],"author":{"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","date":"2026-02-04 17:13:30.000000000","tz":60},"committer":{"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","date":"2026-02-04 17:14:57.000000000","tz":60},"subject":"Use SB DB for checking if an OVN agent is alive","message":"Use SB DB for checking if an OVN agent is alive\n\nThis patch allows for checking if an OVN agent is alive based on the\nstate in the Southbound DB rather than based on the cached values in\nthe ``AgentCache``. In some situations different RPC or API workers\nhave the ``AgentCache`` out of sync with the Southbound DB. As a\nresult they are not aware of the actual state of an agent, which may\nlead to problems, such as a GW port binding failure on the highest\npriority OVN Chassis.\nThis patch doesn\u0027t solve the problem of the ``AgentCache`` being out\nof sync with the Southbound DB, but instead, it adds the DB check to\nsee if an OVN agent is alive during port binding.\n\nCloses-Bug: #2136733\nChange-Id: Id8c0dbfd38c5d81fd089e9ebe0d521a03b367079\nSigned-off-by: Marcin Wilk \u003cmarcin.wilk@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/11f5a2d08f91b044084d1f3ceafd9beada9e68d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/11f5a2d08f91b044084d1f3ceafd9beada9e68d9"}]},"branch":"refs/heads/master"},"6e571d6a6365b19f9d5709004e19514db5896e6b":{"kind":"REWORK","_number":5,"created":"2026-02-11 16:58:57.000000000","uploader":{"_account_id":30534,"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","username":"wilkmar"},"ref":"refs/changes/76/971876/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/76/971876/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/971876/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/971876/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/971876/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/76/971876/5"}}},"commit":{"parents":[{"commit":"025ea04a3919cdc1ef4111574ea46d6f4d1dce6b","subject":"Merge \"Remove configuration variable ``enable_signals``.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/025ea04a3919cdc1ef4111574ea46d6f4d1dce6b"}]}],"author":{"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","date":"2026-02-11 16:48:40.000000000","tz":60},"committer":{"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","date":"2026-02-11 16:48:40.000000000","tz":60},"subject":"Use SB DB for checking if an OVN agent is alive","message":"Use SB DB for checking if an OVN agent is alive\n\nThis patch allows for checking if an OVN agent is alive based on the\nstate in the Southbound DB rather than based on the cached values in\nthe ``AgentCache``. In some situations different RPC or API workers\nhave the ``AgentCache`` out of sync with the Southbound DB. As a\nresult they are not aware of the actual state of an agent, which may\nlead to problems, such as a GW port binding failure on the highest\npriority OVN Chassis.\nThis patch doesn\u0027t solve the problem of the ``AgentCache`` being out\nof sync with the Southbound DB, but instead, it adds the DB check to\nsee if an OVN agent is alive during port binding.\n\nCloses-Bug: #2136733\nChange-Id: Id8c0dbfd38c5d81fd089e9ebe0d521a03b367079\nSigned-off-by: Marcin Wilk \u003cmarcin.wilk@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6e571d6a6365b19f9d5709004e19514db5896e6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6e571d6a6365b19f9d5709004e19514db5896e6b"}]},"branch":"refs/heads/master"},"f3269deb622fa46a34f21c5a42bfdc523d663989":{"kind":"REWORK","_number":6,"created":"2026-02-23 08:42:22.000000000","uploader":{"_account_id":30534,"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","username":"wilkmar"},"ref":"refs/changes/76/971876/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/76/971876/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/971876/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/971876/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/971876/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/76/971876/6"}}},"commit":{"parents":[{"commit":"33f1c11bed4342211fdbe817885d28f911295f1d","subject":"Merge \"[OVN] Add a maintenace method to create the AS for each AG\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/33f1c11bed4342211fdbe817885d28f911295f1d"}]}],"author":{"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","date":"2026-02-23 08:33:18.000000000","tz":60},"committer":{"name":"Marcin Wilk","email":"marcin.wilk@canonical.com","date":"2026-02-23 08:41:04.000000000","tz":60},"subject":"Use SB DB for checking if an OVN agent is alive","message":"Use SB DB for checking if an OVN agent is alive\n\nThis patch allows for checking if an OVN agent is alive based on the\nstate in the Southbound DB rather than based on the cached values in\nthe ``AgentCache``. In some situations different RPC or API workers\nhave the ``AgentCache`` out of sync with the Southbound DB. As a\nresult they are not aware of the actual state of an agent, which may\nlead to problems, such as a GW port binding failure on the highest\npriority OVN Chassis.\nThis patch doesn\u0027t solve the problem of the ``AgentCache`` being out\nof sync with the Southbound DB, but instead, it adds the DB check to\nsee if an OVN agent is alive during port binding.\n\nCloses-Bug: #2136733\nChange-Id: Id8c0dbfd38c5d81fd089e9ebe0d521a03b367079\nSigned-off-by: Marcin Wilk \u003cmarcin.wilk@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f3269deb622fa46a34f21c5a42bfdc523d663989"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f3269deb622fa46a34f21c5a42bfdc523d663989"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
