)]}'
{"id":"openstack%2Fneutron~972039","triplet_id":"openstack%2Fneutron~master~I47c73e95f9a48b29ce26d31ba921f3fb673f78f6","project":"openstack/neutron","branch":"master","topic":"bug/2069718","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-01-16 11:22:10.000000000","reason":"Change was submitted"},"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2026-01-16 11:22:10.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I47c73e95f9a48b29ce26d31ba921f3fb673f78f6","subject":"Add flag \"ovs_create_tap\" in the port VIF details","status":"MERGED","created":"2026-01-02 10:52:35.000000000","updated":"2026-01-16 11:23:16.000000000","submitted":"2026-01-16 11:22:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"972039-bug/2069718","meta_rev_id":"6e89a32fe30e32f6aea4d4107a6ffe73c9a0e9ba","_number":972039,"virtual_id_number":972039,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-16 11:22:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2026-01-16 09:36:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2026-01-16 08:48:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2026-01-16 09:29:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2026-01-16 09:29:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}],"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-02 11:43:58.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"},{"updated":"2026-01-02 11:43:58.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2026-01-02 11:43:58.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"CC"},{"updated":"2026-01-02 11:54:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-01-02 12:54:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-16 08:48:54.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2026-01-16 09:29:06.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2026-01-16 09:36:34.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"}],"messages":[{"id":"2ae80b86eac32e1733997f3c167ece6300cbe4f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-01-02 10:52:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e747ba301c0971f37b290e9d46441c9dabcdca6c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-02 11:54:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/609ab25f68964f53a1b6df368006244c\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/731ede086f814ffbbbe8c246d567aed6 : FAILURE in 58m 56s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/886c886641454cbcb0852d197887e3c9 : FAILURE in 41m 57s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3fc15271aa8c45ac8c0eb1b953fe01d8 : FAILURE in 53m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"43b1de3487c7806dab0b88338e7ac5f02e351d81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-02 12:54:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ac58ce8225b049f58850cdc77acc8a0a\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/80bfbfa40335410a913cdd736bb2619a : SUCCESS in 1h 20m 42s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/dd7b75f6eaec43f6b5f527de1b99d31f : SUCCESS in 1h 50m 05s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/139a74d8cc2f44e689fd40c05b2dfe1b : SUCCESS in 31m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/40c0a5043e3e44998376cb1406e7d05d : FAILURE in 1h 00m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f949a7c81374f1e9faa27a2671f003d : SUCCESS in 10m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5c9a6f042d7e466ca153892ccfffc1fd : FAILURE in 22m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dae676a4793440a2b535cddffaef264d : FAILURE in 22m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a4a72a33e3d47268d062f505cfcf2fd : SUCCESS in 12m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/67ebbf7f0fba4081bb2ebb7727f4c33e : SUCCESS in 16m 12s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/30cfd2e01f1f4035b419ea748319578e : SUCCESS in 1h 00m 58s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/fdb70117963447cf8d6c32d743c431cd : SUCCESS in 47m 53s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/4dbae4da0bfb4680923ce4bd3a9bdcfb : FAILURE in 47m 09s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/937d04f2165940a6bbb82da9e448d470 : SUCCESS in 1h 12m 38s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/0fb0745519294d898429751dd6ed1e8d : SUCCESS in 1h 10m 21s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/19d9cd07fbdd49989f975afbeeccb137 : SUCCESS in 1h 56m 00s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/fe49b179421a449790110f11c5049d98 : SUCCESS in 1h 37m 31s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/2e6e6522db75497c9dc8df4da002e0bb : SUCCESS in 1h 07m 24s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/b95540cf81b043f7a3946a05f1184009 : SUCCESS in 1h 15m 37s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2b3bfb89a26c4de2927715d2021c9fc9 : SUCCESS in 1h 21m 19s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/1f35458d752b461781fb6e8937375756 : SUCCESS in 1h 30m 44s","accounts_in_message":[],"_revision_number":1},{"id":"af06b65d32cd8e2639dfb5f4953be4f67a409614","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-01-05 12:45:14.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"ee828ff79e1aee252422d6c76e3377e6051ef8ea","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-05 13:44:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c032a1875290471ca6dd265f3f14100c\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e18b5c494a9f412693962140b59bbf2a : SUCCESS in 56m 40s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2edc582e0ab14a1bac54a88bbcc1dc4a : SUCCESS in 45m 50s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/47fb2302fd8745aaa13e645bbd065590 : SUCCESS in 28m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1d56b3dbdf05ba0cdfcfd9cf75a2d989e400486a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-05 14:56:10.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f3e6cd2c19aa408f8da58108d35cd041\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/136161d006c94e9384e623fb951e5eba : SUCCESS in 1h 21m 16s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/8002e1cf70b04aaab1833c39fbdc2665 : SUCCESS in 2h 09m 31s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4224373ecac743a284afa1e3984fcfb0 : SUCCESS in 32m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/57cf101cb9c340d884f967ede4d7953e : SUCCESS in 43m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69b27a5940304bccab9317a1920a5257 : SUCCESS in 10m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1f1b184b4ff143598db27db39186ce1f : SUCCESS in 26m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1851e6f5eb7049cab2aebf29e04aea82 : SUCCESS in 21m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0af07d0e84194190b9c10a071d36ab35 : SUCCESS in 11m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d914f158d4e4495c815010cb540672a3 : SUCCESS in 26m 09s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/b1a255d3719749b7884587a8771cf3f5 : SUCCESS in 46m 35s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/8af8571031a943adb5118f5b3dcfb6d4 : SUCCESS in 1h 12m 53s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/09f19a027d1e442bb3f548b6071720a0 : SUCCESS in 1h 14m 29s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/7eda9d5012d44b13b0e44a6d8ffc7db8 : SUCCESS in 1h 10m 15s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/040d99962f3948c48a6d5d23087c4a4b : SUCCESS in 1h 14m 35s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/8fc89932d22a4d48b861bc81d7882241 : SUCCESS in 2h 01m 24s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/090fcda6b8fc47989815fd6b1e748a1f : FAILURE in 1h 16m 30s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/5c733a83eb454ad2be346af94a88546e : SUCCESS in 1h 06m 41s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/1507d602836d45c0b7f8c292b2db96e1 : SUCCESS in 1h 17m 04s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b177b0d10e38451c8c9863540578924f : SUCCESS in 1h 33m 34s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/0e7d1b12c7f9470285dc1ed2fbee671b : SUCCESS in 1h 49m 02s","accounts_in_message":[],"_revision_number":2},{"id":"af88806302c8e64f837ff2d396bde0d6f9b30367","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-01-07 07:58:49.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7ed670e6c3357c85408273e394f24f90f7d2696b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 09:01:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4ffaabb00284758aee1601c1b70b914\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/112579b19e2545fdbea16dd517677476 : SUCCESS in 1h 00m 48s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f39b4ad122e94879b7fbbf048ae57001 : SUCCESS in 48m 40s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6c01589968b341148a544531dd69a995 : SUCCESS in 1h 01m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"93ea3a46c132f97f7440a7918e5f90812a98cf55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 10:10:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a03affd8dbc46f3a46c733fba51480d\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/5409afa12049442f860eea61a418ee7f : SUCCESS in 1h 29m 30s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/3c5e212fe5f44c44b2e2112f76890021 : SUCCESS in 2h 10m 30s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8f95bf4294ab4efb99279ea098a68d17 : SUCCESS in 30m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/942c8db68e7d42e9a434687315260e25 : SUCCESS in 43m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba50b73ab39a4ce2bbca8fa08461488a : SUCCESS in 25m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/94abef0a33694f1cbfe323e6d295aac8 : SUCCESS in 27m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6358c4fa748844ad9da530cd1e8e3a9f : SUCCESS in 23m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4137c8fc7fd647d08653649e970fddb8 : SUCCESS in 11m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/08d846db46904e1c8078347d1ee91915 : SUCCESS in 13m 17s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/4e26c17f56134ab38396489712f7f4ee : SUCCESS in 34m 12s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/ff8d9abf02384d00aaabb5c9056caa7a : SUCCESS in 49m 57s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f8710f3124054dad8ac93e0b514244e3 : SUCCESS in 1h 41m 08s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1e2163b04a344b768b68c565fcfd7975 : SUCCESS in 1h 04m 17s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/404ff02ec90e4014917b80bbf7310edf : SUCCESS in 39m 27s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/d0117ad7d03544c0a2f5191e267e6281 : SUCCESS in 1h 09m 10s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/59819e60548049ff9157cd101baca688 : SUCCESS in 1h 25m 57s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/ae4b582debd24b298e561c2af1b01fd9 : SUCCESS in 1h 03m 03s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/5560902a2668439798ca3aa66a3a9ada : SUCCESS in 1h 11m 22s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5b79e9077eb84a0fac0bd0f556e550f3 : SUCCESS in 1h 21m 27s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/6c506d99a6ab415aa3e0f080ed7a3baa : SUCCESS in 1h 07m 23s","accounts_in_message":[],"_revision_number":2},{"id":"50d9442acee8084e14e9131f985322f05b6105d4","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-01-09 16:58:46.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"cf3646b2bf0ca066ef749c3cf8c15b184401ee2b","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-01-12 14:03:16.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ad345e78556268aa433a0adb235960891b878b0f","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-01-13 09:29:47.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e760d7585ea680c001abb8060d159f7a083e1ab8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-01-13 09:39:19.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"32cc95557dba54a0de77387c11022be63fd98671","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-01-13 09:39:41.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"dbc0a4417b7f82d5e86e24aec714287b54ad3640","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 10:39:09.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/735b9cd0b5f94a009ce7855f9edac38a\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/97e3ccd57b884376b468ffd37bb8094d : SUCCESS in 33m 51s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/da5b6d5b2b694935b8da828d42416443 : SUCCESS in 44m 04s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c3e5d4243c464988a840a1f5c1f27f06 : SUCCESS in 54m 32s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8ef72f794543610aaa2362925818667e7a6bd297","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 11:30:09.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5c55c04bdc9d4e0985769ddbddcc48e3\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/077f22193e0340a3ae3bc6041adedcc7 : SUCCESS in 1h 32m 26s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/0281ce7298eb43b5844cacac66aeb56b : SUCCESS in 1h 42m 20s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7b963d7777c84418b33fb38592b34815 : SUCCESS in 33m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/97dc62e0a22946bdb5e14bb93888e8e1 : SUCCESS in 1h 07m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff5184e768c54a73b4bf69f3bc6dc1b2 : SUCCESS in 21m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/587876bdb1bb40889d2034ba6b273122 : SUCCESS in 20m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1f1c01a695fa4cf69edf23338aa44c19 : SUCCESS in 37m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c635816673254c23b4a01c6ce44a64dd : SUCCESS in 9m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dda8e7a0fa834b1eb9b7fac72511da8c : SUCCESS in 25m 17s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/c0a6e9303b4549aab773b01d6727883d : SUCCESS in 1h 03m 37s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/5bd707025b2c4c9682245b77baebd3db : SUCCESS in 1h 02m 33s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c59c605805bf4fa780408a01cbe94739 : FAILURE in 26m 28s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/186a27379b4c4ce0bb23927d72330384 : SUCCESS in 1h 11m 59s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/6c29276a064e4345a09749f16fdde9a9 : SUCCESS in 37m 00s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/cd7ac69293684501879ac243b36361e9 : SUCCESS in 1h 47m 44s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/49d863dd530e458c9eb0e8b0890b89e4 : SUCCESS in 1h 11m 25s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/5ec78c587bc345119e4a6106893352fc : SUCCESS in 1h 06m 44s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/69d87b0b2d5941cc9e58e6e5b719098c : SUCCESS in 1h 11m 55s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/86159e72592d44fbae7e70fe9caa6614 : SUCCESS in 41m 13s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/fff1127faafa43b09d26d725b2eb2a04 : SUCCESS in 1h 04m 43s","accounts_in_message":[],"_revision_number":3},{"id":"6fa4363d532ca0dfbc018b32e69e1923f3838132","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-01-13 11:41:06.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"e770f7904b6d89484817c25a16c19e1be270b31c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 12:17:16.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/11c5a0e9c5e94566a9a5bb38d7b3dfb1\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0d67d28635a74b6789ed3b0f36f62635 : SUCCESS in 33m 29s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4386366b98ce402991e574eab50aeb98 : SUCCESS in 25m 23s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/770743535fd34a1a899b79379529cf36 : SUCCESS in 28m 09s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"ddea889bd4485942c0b1ea2c9ed14484a446b08c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 13:59:42.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c62407318cc45099121be73db924aa4\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/fa4118fcc37a4d8cab0dff174db622c3 : SUCCESS in 1h 19m 08s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/7a2e2ac510a34981914466bd35c552a9 : SUCCESS in 2h 16m 16s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c2305da483c24bbe90eb229426622c00 : SUCCESS in 32m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/921febad7f0f421f95c1db4f322a0114 : SUCCESS in 37m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3dfce3d5bf104389a09690379933d083 : SUCCESS in 24m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cb1784e4189845aca724446ca5d1302b : SUCCESS in 26m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/11f42b68e01c4168b70dbe19129b8216 : SUCCESS in 21m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbcc4831b49a4ae7a982bc9e5be04ff9 : SUCCESS in 13m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/83ac7df1c5544e7f81d51d64c0b83ae6 : SUCCESS in 23m 50s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/7ffca79173ea4c0d9ae0994fcc290d33 : SUCCESS in 58m 02s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/9b5e4b072fb64881bcc5b738cc29e433 : SUCCESS in 1h 08m 08s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/6eb93d96970d4b4187cfb2b2aa1797f3 : SUCCESS in 1h 49m 23s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/5e1d25265ef944c5abb42064d91141ae : SUCCESS in 59m 07s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/d3eaacf0120b4d879d2ebc390fd43a22 : SUCCESS in 41m 27s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/b06a88e476e44e4581e2ee6dae2cdb37 : SUCCESS in 1h 45m 25s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/de87051630ad40ce8449043345ea10f6 : SUCCESS in 1h 43m 16s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/7424cb773d304698bd37bd933f57d199 : SUCCESS in 55m 05s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/d0e0b9994a394ec58ee575d96968d1d7 : SUCCESS in 53m 59s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/74c275c059da4ef89c513abca15e3fac : SUCCESS in 1h 06m 41s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/1203854cd48f429181f8f4c279e2ede2 : SUCCESS in 1h 06m 26s","accounts_in_message":[],"_revision_number":4},{"id":"f5ae0a1b25b8513ac3ef5885f597145aec8db96d","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-01-16 08:48:54.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"6bf27f61531c93106343f7640f3cb0d05ca61825","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-01-16 09:29:06.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"61055eeb0806f7ab4cc86f127931e2659efed034","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 09:29:28.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"d69e4d11448b546ba0afaff80e50239df8420024","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2026-01-16 09:36:34.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"9f917d6371ad6b0d213a44f4ee4005a06f0bb128","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 11:22:09.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/54bf77b2b69346dbb3144536ee738d9e\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/e5c11aa7d63a4dcda53d019ad79e84fd : SUCCESS in 1h 29m 58s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c2deb6fd57fb47f0a31624be734e61e0 : SUCCESS in 1h 39m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e0ce164bfe44e33acad470c7f41fc75 : SUCCESS in 21m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0206b031af2543a39648e882ab1d2802 : SUCCESS in 26m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/eb9e82dd9d394955864a8f1ca861f6e2 : SUCCESS in 28m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/810393dbe4c74ab6822418412c7a350d : SUCCESS in 13m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fc3f4a9fc24d4b9a95fcd206ac2bfd3a : SUCCESS in 27m 23s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a7e7393babef41fb970eff625e163a2d : SUCCESS in 1h 06m 44s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/450084b4f2894ae3a0ed264b1b63469b : SUCCESS in 1h 48m 01s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/f1d7cb2386b54862aa9d86484774fb14 : SUCCESS in 1h 07m 28s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/10e457daafb449c7961e49a2daa6bdd9 : SUCCESS in 39m 46s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/4377376640f24008a36ba383ec3c5863 : SUCCESS in 45m 46s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/632f3dc796e645779dbb954ee5168848 : SUCCESS in 59m 42s","accounts_in_message":[],"_revision_number":4},{"id":"a5bf1e69fedc295bb20f3e1e369f586e5cb566a4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 11:22:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"6e89a32fe30e32f6aea4d4107a6ffe73c9a0e9ba","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 11:23:16.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08a594a19b5b430fa095f16c20584c87\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1bd1d9fc0b9454a981d2ee367476e41 : SUCCESS in 49s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2f3dd38008c34b9592869e462c4f87fc : SUCCESS in 43s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"1072ffeb34ac1e8112553c308553cde4fca33c7a","revisions":{"e6cf28be7b7f2d41da5b0b8a6f2a284893856b09":{"kind":"REWORK","_number":1,"created":"2026-01-02 10:52:35.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/39/972039/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/972039/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/972039/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/972039/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/972039/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/972039/1"}}},"commit":{"parents":[{"commit":"0ba83a9c07758f8fc36219422a669ef109e97733","subject":"Merge \"Change some functional test code to use project_id\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0ba83a9c07758f8fc36219422a669ef109e97733"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-01-02 10:18:10.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-01-02 10:52:21.000000000","tz":60},"subject":"Add flag \"os_vif_tap_creation\" in the port VIF details","message":"Add flag \"os_vif_tap_creation\" in the port VIF details\n\nThis flag determines if os-vif pre-creates the TAP device or not (in the\nsecond case, it will be libvirt). This flag applies only to ML2/OVN\nmechanism driver.\n\nThe flag is added to the ``vif_details`` dictionary passed to Nova in\nthe port binding operation.\n\nThis flag is statically configured using a configuration variable (False\nby default, keeping the current behaviour):\n  [ovn]\n  os_vif_tap_creation\u003dTrue\n\nRelated-Bug: #2069718\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I47c73e95f9a48b29ce26d31ba921f3fb673f78f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e6cf28be7b7f2d41da5b0b8a6f2a284893856b09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e6cf28be7b7f2d41da5b0b8a6f2a284893856b09"}]},"branch":"refs/heads/master"},"c369368a895a29a945a14abbc10f4bf46d61f21b":{"kind":"REWORK","_number":2,"created":"2026-01-05 12:45:14.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/39/972039/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/972039/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/972039/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/972039/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/972039/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/972039/2"}}},"commit":{"parents":[{"commit":"185b5ffe2b8f91dbbac38667e03c07969dc99e7a","subject":"Doc: add warnig for fullstack","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/185b5ffe2b8f91dbbac38667e03c07969dc99e7a"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-01-02 10:18:10.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-01-05 12:44:58.000000000","tz":60},"subject":"Add flag \"os_vif_tap_creation\" in the port VIF details","message":"Add flag \"os_vif_tap_creation\" in the port VIF details\n\nThis flag determines if os-vif pre-creates the TAP device or not (in the\nsecond case, it will be libvirt). This flag applies only to ML2/OVN\nmechanism driver.\n\nThe flag is added to the ``vif_details`` dictionary passed to Nova in\nthe port binding operation.\n\nThis flag is statically configured using a configuration variable (False\nby default, keeping the current behaviour):\n  [ovn]\n  os_vif_tap_creation\u003dTrue\n\nRelated-Bug: #2069718\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I47c73e95f9a48b29ce26d31ba921f3fb673f78f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c369368a895a29a945a14abbc10f4bf46d61f21b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c369368a895a29a945a14abbc10f4bf46d61f21b"}]},"branch":"refs/heads/master"},"90505f00fa11fcf4bf54a7cf5de53942e4ea6885":{"kind":"REWORK","_number":3,"created":"2026-01-13 09:39:19.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/39/972039/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/972039/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/972039/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/972039/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/972039/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/972039/3"}}},"commit":{"parents":[{"commit":"9656a5fcf2c706d2f159655f9fa40c9768bbf5c1","subject":"Merge \"bgp: Introduce BGP chassis commands\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9656a5fcf2c706d2f159655f9fa40c9768bbf5c1"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-01-02 10:18:10.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-01-13 09:39:02.000000000","tz":60},"subject":"Add flag \"ovs_create_tap\" in the port VIF details","message":"Add flag \"ovs_create_tap\" in the port VIF details\n\nThis flag determines if os-vif pre-creates the TAP device or not (in the\nsecond case, it will be libvirt). This flag applies only to ML2/OVN\nmechanism driver.\n\nThe flag is added to the ``vif_details`` dictionary passed to Nova in\nthe port binding operation.\n\nThis flag is statically configured using a configuration variable (False\nby default, keeping the current behaviour):\n  [ovn]\n  ovs_create_tap\u003dTrue\n\nThis flag is marked as deprecated and will be removed in 2027.1 (2\nreleases later).\n\nRelated-Bug: #2069718\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I47c73e95f9a48b29ce26d31ba921f3fb673f78f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/90505f00fa11fcf4bf54a7cf5de53942e4ea6885"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/90505f00fa11fcf4bf54a7cf5de53942e4ea6885"}]},"branch":"refs/heads/master"},"1072ffeb34ac1e8112553c308553cde4fca33c7a":{"kind":"REWORK","_number":4,"created":"2026-01-13 11:41:06.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/39/972039/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/972039/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/972039/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/972039/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/972039/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/972039/4"}}},"commit":{"parents":[{"commit":"9656a5fcf2c706d2f159655f9fa40c9768bbf5c1","subject":"Merge \"bgp: Introduce BGP chassis commands\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9656a5fcf2c706d2f159655f9fa40c9768bbf5c1"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-01-02 10:18:10.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-01-13 11:40:49.000000000","tz":60},"subject":"Add flag \"ovs_create_tap\" in the port VIF details","message":"Add flag \"ovs_create_tap\" in the port VIF details\n\nThis flag determines if os-vif pre-creates the TAP device or not (in the\nsecond case, it will be libvirt). This flag applies only to ML2/OVN\nmechanism driver.\n\nThe flag is added to the ``vif_details`` dictionary passed to Nova in\nthe port binding operation.\n\nThis flag is statically configured using a configuration variable (False\nby default, keeping the current behaviour):\n  [ovn]\n  ovs_create_tap\u003dTrue\n\nThis flag is marked as deprecated and will be removed in 2027.1 (2\nreleases later).\n\nRelated-Bug: #2069718\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I47c73e95f9a48b29ce26d31ba921f3fb673f78f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1072ffeb34ac1e8112553c308553cde4fca33c7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1072ffeb34ac1e8112553c308553cde4fca33c7a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
