)]}'
{"id":"openstack%2Fneutron~972453","triplet_id":"openstack%2Fneutron~master~I2de68420284e31afcb749e07e0103a36bb5b0099","project":"openstack/neutron","branch":"master","topic":"bug/2137631","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-01-07 19:58:34.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2de68420284e31afcb749e07e0103a36bb5b0099","subject":"Add ``path_separator`` to alembic configuration","status":"MERGED","created":"2026-01-07 11:06:35.000000000","updated":"2026-01-07 20:00:06.000000000","submitted":"2026-01-07 19:58:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"972453-bug/2137631","meta_rev_id":"cac3df34275b1b5d43f3b88d6bd5359bb3a6d28c","_number":972453,"virtual_id_number":972453,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-07 19:58:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2026-01-07 15:31:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2026-01-07 16:38:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2026-01-07 16:38:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-07 12:21:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-07 15:31:55.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2026-01-07 15:57:55.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2026-01-07 16:38:27.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"8aa7361fe9a833316004bcea773261a664aabfec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-01-07 11:06:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eecfe455bfaf391031b2301265fb5a3aecf79888","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 12:21:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ba970d9b41084053ae9deccb739d7d5d\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/02dc000ce4714890bb2cb4a85773f027 : FAILURE in 15m 25s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a27622dc85314fb98d5e612e3e65d1a0 : FAILURE in 15m 18s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/cbd2b739257e49b590609b48cd8a954e : FAILURE in 14m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/478770034e7b47e18074acc84db9d7ee : FAILURE in 34m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83d2505fcfa54e71b84c9c728c39f38f : FAILURE in 25m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fd7e4b109c734fdba24e909a8f53f5bf : FAILURE in 23m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/32d5bcdba31d417b804bc859904836e3 : FAILURE in 33m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83bc52e165554c92b6964b19968efced : SUCCESS in 13m 23s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/e0b5b9c7e2394014aa93b125bfd05313 : FAILURE in 50m 04s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/1b35c4b8f10f463fa3837a74a8ad7016 : FAILURE in 52m 35s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/0c496266c8a1452d978e72ca899db50a : FAILURE in 30m 34s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/546f9ac4fccb443ca0f36bc3eb7adcaf : FAILURE in 36m 43s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/e362b8e305854daa8ce0345fd86ba570 : FAILURE in 1h 02m 57s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/3a60aaa315d243f6888f1bbf8c386679 : FAILURE in 23m 31s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/5e828f1bb2d046aaa9ae81a7cb151908 : FAILURE in 24m 27s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a1ef377ac23249ce8c4da51117d639c5 : FAILURE in 21m 13s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/069c20f52f1841e1b1cf17541dd61004 : FAILURE in 22m 48s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e9af3ca819df42ac8f420c87aa1e6a00 : FAILURE in 16m 57s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/371f54a757904448b059a0cf5785a508 : FAILURE in 27m 12s","accounts_in_message":[],"_revision_number":1},{"id":"df2069cdb3113c54e407ee56c9b0aec30e1ef688","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 12:34:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74a6d6c406674f81a14d91f80d5b5d6f\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/523aef9e33f643e2b9c093e8b3855dec : FAILURE in 25m 23s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e5667ba3c6774705b5bfccc0eca0347b : FAILURE in 43m 52s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/77e6e04bdd3444dc8ebfe3f6550e5cd4 : FAILURE in 55m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ac994d991502a76e9202b51ce4069dd6ad0bc6f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-01-07 13:47:42.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"9187790b6430cff6b9e5cddb5b69923301e32caa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 15:07:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06b7c2baa58648e58a8796cbd2420fdc\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0ba23d99350544e0a819f5169cfe532f : SUCCESS in 26m 43s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/98e34b82f8b2406ab15687134878852a : SUCCESS in 42m 22s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/eda5ac4df45848d1a82a12ea040e2273 : SUCCESS in 52m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0ebd9531a589d7832f0282d52d51ed5931c9f150","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-01-07 15:31:55.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"47f524062cf987be94fd17c4b65d1da880900ec0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 15:58:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81b7bd081043455787d105ea6122945a\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/5f9a3664060c4d6aad60623f00534514 : SUCCESS in 1h 31m 06s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/3f06ebe6ff9c4112a8e457af85389808 : SUCCESS in 1h 55m 12s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a8ab7595c8f34588a7f1c3da4493be9a : SUCCESS in 30m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4845a47ceda24995a932253c71cbf9e4 : SUCCESS in 43m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/435089a7aaa649f79b03af8b2530c48c : SUCCESS in 12m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c785b58ad78c4bd383c948c2a3b875be : SUCCESS in 35m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e0a4f8c45085467db502447fce30bcae : SUCCESS in 29m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd51650c3a024afbbaf88d15795c12c9 : SUCCESS in 12m 20s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/fd0a79d5d6ab499a92bdcaacd3ced89a : SUCCESS in 58m 40s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/a96dfbcc6660465a915d69fd41cf98dc : SUCCESS in 39m 22s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/39d7bf2b3f0341e3b0d84f42c6c808ad : SUCCESS in 2h 03m 51s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/ff558e239b2541d09da5b19b45bb7323 : SUCCESS in 1h 07m 19s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/dfcb27347fc74adb81b22a8ebc68f6df : SUCCESS in 1h 12m 09s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/a1bb57c0900e4d41a3c86d2ce8bc1509 : SUCCESS in 1h 35m 51s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8a0273e7241f4cefaadf171f43d09e0a : SUCCESS in 1h 47m 38s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/b39a4f6ea31a4de5adbd89b628f3cb1e : SUCCESS in 1h 00m 54s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/054ae9b88d734ec3b5de62e3eac1964f : SUCCESS in 1h 28m 26s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e8aaea05af3f4950b4a0f597ba2bc78d : SUCCESS in 37m 44s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/052b101bd8a54b269afc7b18d4ce46e4 : SUCCESS in 1h 04m 54s","accounts_in_message":[],"_revision_number":2},{"id":"e6405dc7beff589b328925bce5f01a61efa7d650","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-01-07 16:38:27.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5e5eb2675a6232992aec14ec960711e6cd326d5d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 16:38:50.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e1d342901d04c3272a54b21f8521001ef559d02f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 19:58:34.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/789fc237fc5f41de9b521a70bea51fa9\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/ecf8c6704bd44580a2d5ed9a32db0aba : SUCCESS in 1h 17m 59s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e453b84fb70043d6b4aee6969048c096 : SUCCESS in 2h 20m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8197c92107b24480845b067c92c50b13 : SUCCESS in 22m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a6729008a02b4e6b8dea60be8cada30c : SUCCESS in 35m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a1fc76fd04254454be152e39e8521a5a : SUCCESS in 36m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2e6b53b2b2b44b6b8d5a012c57faada : SUCCESS in 6m 00s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f6e39628097b4fe3b0157249a7cc71e9 : SUCCESS in 59m 41s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/5636d923e19f485b85f80ca0977ff648 : SUCCESS in 1h 06m 50s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/507f87ef21064c249711d7816e0d1558 : SUCCESS in 1h 08m 51s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c8625d3751a94188b894e677c9b27640 : SUCCESS in 40m 55s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/3de3416bc20548edb240bc5dc146672d : SUCCESS in 3h 01m 48s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/2cf8fb19d70249b48a75c6ce66e13a25 : SUCCESS in 1h 16m 21s","accounts_in_message":[],"_revision_number":2},{"id":"511e9b743f4933104abb5f16b340212733b7e4be","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 19:58:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"cac3df34275b1b5d43f3b88d6bd5359bb3a6d28c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 20:00:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2dc8d75946941f4b594de1092ca47fe\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d978e015906c44668acfbe0f0c1276cd : SUCCESS in 51s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"66eb0808527895e76a77b210093ebab5176a1467","revisions":{"3d34afff6d180239d2b87687cbbbaaaf287ef6bf":{"kind":"REWORK","_number":1,"created":"2026-01-07 11:06:35.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/53/972453/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/972453/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/972453/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/972453/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/972453/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/972453/1"}}},"commit":{"parents":[{"commit":"af2aefeb87f6e2cdacf2b164338e440e131692ad","subject":"Merge \"[Docs] Remove note about FWaaS not being supported by OVN backend\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/af2aefeb87f6e2cdacf2b164338e440e131692ad"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-01-07 11:04:44.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-01-07 11:06:25.000000000","tz":60},"subject":"Add ``path_separator`` to alembic configuration","message":"Add ``path_separator`` to alembic configuration\n\nAdded the configuration variable ``path_separator`` for alembic:\n  [alembic]\n  path_separator \u003d os\n\nThe value \"os\" is the recommended default value by the library.\n\nCloses-Bug: #2137631\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I2de68420284e31afcb749e07e0103a36bb5b0099\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3d34afff6d180239d2b87687cbbbaaaf287ef6bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3d34afff6d180239d2b87687cbbbaaaf287ef6bf"}]},"branch":"refs/heads/master"},"66eb0808527895e76a77b210093ebab5176a1467":{"kind":"REWORK","_number":2,"created":"2026-01-07 13:47:42.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/53/972453/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/972453/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/972453/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/972453/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/972453/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/972453/2"}}},"commit":{"parents":[{"commit":"af2aefeb87f6e2cdacf2b164338e440e131692ad","subject":"Merge \"[Docs] Remove note about FWaaS not being supported by OVN backend\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/af2aefeb87f6e2cdacf2b164338e440e131692ad"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-01-07 11:04:44.000000000","tz":60},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2026-01-07 13:47:42.000000000","tz":0},"subject":"Add ``path_separator`` to alembic configuration","message":"Add ``path_separator`` to alembic configuration\n\nAdded the configuration variable ``path_separator`` for alembic:\n  [alembic]\n  path_separator \u003d space\n\nThe value \"space\" depends on the way Neutron is building the\n\"version_locations\" string, using a space between paths [1].\n\n[1]https://github.com/openstack/neutron/blob/af2aefeb87f6e2cdacf2b164338e440e131692ad/neutron/db/migration/cli.py#L558\n\nCloses-Bug: #2137631\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I2de68420284e31afcb749e07e0103a36bb5b0099\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/66eb0808527895e76a77b210093ebab5176a1467"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/66eb0808527895e76a77b210093ebab5176a1467"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
