)]}'
{"id":"openstack%2Fneutron~972491","triplet_id":"openstack%2Fneutron~master~I9a231ad8ffbca46c8e71d35bc20d9762df0970f9","project":"openstack/neutron","branch":"master","topic":"oslo.config_10.2.0","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-01-21 10:25:50.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9a231ad8ffbca46c8e71d35bc20d9762df0970f9","subject":"Bump oslo.config to 10.2.0","status":"MERGED","created":"2026-01-07 14:34:57.000000000","updated":"2026-01-21 10:27:06.000000000","submitted":"2026-01-21 10:25:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"972491-oslo.config_10.2.0","meta_rev_id":"12c24162341dfde95caf5c5f894a4e14d8d3fe3f","_number":972491,"virtual_id_number":972491,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-21 10:25:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2026-01-21 08:22:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2026-01-20 23:11:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-01-21 08:22:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2026-01-20 14:05:45.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-07 14:44:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-01-07 15:51:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-07 19:44:54.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2026-01-20 14:08:44.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2026-01-20 14:43:58.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2026-01-20 23:11:55.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"14bb14b3f945e078898b1f71288ad38d4d423b72","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-01-07 14:34:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b9c94985879865c75703509448ecd22654c9b63f","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-01-07 14:35:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6c662cd84fb4d3a5d8f52333b02e6d595224d5e0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 14:44:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dbde39b0819548a7947d01db3d2ca035\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/208b3379972942a0b3df4c9fbb10072a : FAILURE in 5m 38s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/928afc6519d04e4fa524e684eab10833 : FAILURE in 5m 28s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f6715db457a541ab8f8744ffc9bbc6ee : FAILURE in 6m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"94854c66b296712793d0162f66d28299e9542974","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 15:51:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5861801877d748389371ea78ba3298bd\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/10b8926e1bb844cd89085b37aef7f5e0 : FAILURE in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba0ac96bf47447348a9ebc8357881e17 : FAILURE in 4m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d815635154234759a24e0e9dcc73e84a : FAILURE in 2m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/07270e0ec1614ed197f957dd0db27c08 : FAILURE in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81e4b2a6de1f4e96a92ba35420cbd8b4 : FAILURE in 2m 55s\n- requirements-check https://zuul.opendev.org/t/openstack/build/9a94a5f8936f46e4bd346acf41763874 : SUCCESS in 2m 34s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/cf1b5ab2410a409eb59e15e8855ec982 : FAILURE in 36m 32s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/2af17fe1b7154529aad2718c2bf833b6 : FAILURE in 43m 45s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a29b80d1ca6e4c468826ddf26ad20812 : FAILURE in 17m 24s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/6dd7df69ea6c4c81b161572b6fe33dd4 : FAILURE in 59m 25s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/a5982f1fec1e425497e3cc33d86c0296 : FAILURE in 1h 05m 37s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/6ab224c891774613bd1fc804c536bcef : FAILURE in 11m 47s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/99ba334aff7644e8a790c76822df234e : FAILURE in 17m 26s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/9d7dbf137e6c461daa092e325ea3f94e : FAILURE in 14m 38s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/c27fe1f9514442dbad767ab097f34a2f : FAILURE in 7m 48s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0b30985cff9043af803d411c4f016241 : FAILURE in 7m 42s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/bf1f5166136f45bf932c42e48f8bebc1 : FAILURE in 15m 01s","accounts_in_message":[],"_revision_number":1},{"id":"9c536ab94c672cf8df61b808cdcdf6c16bbd66b0","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-01-07 15:53:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"74c18c9d3a35966cdadffdde8d5d9349546c7948","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-01-20 14:05:45.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"b9932d3f6d69d7303ad1c4c0d77da2c695f28537","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-01-20 14:08:44.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"416cdc8cc518858104b3840eb2e065d23dc8a420","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-01-20 14:09:00.000000000","message":"Patch Set 1: -Code-Review","accounts_in_message":[],"_revision_number":1},{"id":"fd44a66c975dcf3bc9a979e6c2f5d1ec77f12b3b","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-01-20 23:11:55.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0a3d57d9f05bf45fe8a57b71c2432d8942b2e610","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 23:49:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4cf6305662a84ff1b5b3da4be74732e4\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2397deebf59d439ca8f999e4e854bc11 : SUCCESS in 31m 54s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bb6930816ceb4c6fbe6718e311619fa3 : SUCCESS in 35m 05s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/52db036704624bfc840eafaf730c1cc3 : RETRY_LIMIT in 4m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f675adc8e189fe81a221faa1ce01fec3187e78d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 01:18:26.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/68ceb3277c4045a193c13cef50c8a2fa\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6a9d007bb675468fa92538bd1c486a2b : SUCCESS in 1h 03m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f21d3e050374cd18a3d246b1cd2e650 : SUCCESS in 26m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b2bfe680d9e84a74a56dc838fd37f539 : SUCCESS in 28m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a80ccceccf214d2bba18e50beae7b851 : SUCCESS in 24m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/503b2d83a21a416da195df7824754d47 : SUCCESS in 6m 31s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1aa971d928d240e5a2949179c1a407fd : SUCCESS in 2m 26s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/e8151c315fa14fbd8a4f6c4d1d8e6146 : SUCCESS in 46m 38s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/840155670f5d498fb04a104e253b35c9 : SUCCESS in 1h 09m 52s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/12cf1d76ec2445ff970b89936372a487 : SUCCESS in 1h 53m 30s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c6f2c60d66df4bdaa48328175eb37757 : SUCCESS in 1h 06m 06s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/549c16fa79114d53ab756e1e8fae94a4 : SUCCESS in 1h 10m 15s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/c88d2eea6e0c487f8b843965ef98e46c : SUCCESS in 1h 11m 42s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/afc85a5e7a9944cd9975b5e8f11c08a4 : SUCCESS in 1h 41m 45s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/12d89633b38d489f9c99038aea3d73f0 : SUCCESS in 59m 09s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/c9922f0110fd43deb23f6d581cc85add : SUCCESS in 1h 06m 32s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/88bb480fc62e41af90efbd10996b990d : SUCCESS in 1h 24m 45s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/76504f8b9d504bea8631e87be4d5dc7a : FAILURE in 56m 47s","accounts_in_message":[],"_revision_number":1},{"id":"987dc3f28d3add9a2b40b97b417b6f715a3a9906","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-01-21 01:33:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f670e442719c83ef748d2a319e4de09119c0848c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 02:39:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ccbe040096b3446b8d3ba38a61a1cde5\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/85663e1900554a3e8834c0dff2e0938f : SUCCESS in 1h 03m 55s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6214f7f7c5dc401d918b79c4ca8a6499 : SUCCESS in 49m 39s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/985ce682195c4e26bee32a9348f1069d : RETRY_LIMIT in 4m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fb6b6282f9a817ba76d6354ce67383d833d220f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 03:28:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e63b4f61d8e4666acec63d4646b4ee2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/71dfbfb142064a54b249df1e99b6024c : SUCCESS in 38m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/995c3da23acb461484c994a3488d30b2 : SUCCESS in 21m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/157bc652080842149abd3d23a5cf3cb8 : SUCCESS in 24m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/29cdd7ab162147d2ba217f9fd5294b11 : SUCCESS in 24m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92a00a899c1a407087880fb4dca598ac : SUCCESS in 12m 24s\n- requirements-check https://zuul.opendev.org/t/openstack/build/2aac2d1f166d41b5a9c5ae13dad03d7e : SUCCESS in 2m 19s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/7550d6e3fb9446baa648f27888a011ac : SUCCESS in 33m 59s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/2408f3bb0d8a4380bf28e81f5b8b426c : SUCCESS in 1h 10m 20s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/21b8e2f73bcd42f78aec65d87f24e2f5 : SUCCESS in 1h 15m 35s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/5244a960746f4b1ea8e34fdf12625da7 : SUCCESS in 1h 09m 26s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/c8e784de20e64edb83a1e3c5106c7a12 : SUCCESS in 1h 10m 07s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/b28c4a379656436f87121358fc590815 : SUCCESS in 1h 47m 43s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/536070187ac54a87bf4188f161c30df3 : SUCCESS in 1h 23m 08s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/bfc7d715f2b14cd197083abb85be2f75 : SUCCESS in 1h 04m 27s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/300a8e7bef1a4fd680cd047790153b2b : SUCCESS in 31m 46s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/336e0b9d32ec465ab15efdc3f13f0297 : SUCCESS in 1h 31m 11s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e192d4f7d63c41fa8abd0221a3e26fbb : SUCCESS in 1h 26m 52s","accounts_in_message":[],"_revision_number":1},{"id":"da05d982ac6dbd93d5efae293b0614a20d93f031","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-01-21 08:22:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3c06b2f86ede3c85bd4d133036841f4da79dac47","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 08:23:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ee9fbba4977bbb9160a1ae3eca174e5f87c0b2bb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 10:25:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f25da0d692ab43c0aafdeb8ba2759765\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f701f19536247ef97dab5913ca1b1a3 : SUCCESS in 23m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1b68edca20414cc190df13530fe5526e : SUCCESS in 21m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ff9483325e8c4d9091811b0f15d8d405 : SUCCESS in 26m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27b419483a4b46e39050472070b43821 : SUCCESS in 12m 59s\n- requirements-check https://zuul.opendev.org/t/openstack/build/be4814464ac54c72b44f3c7ddbfcb1c6 : SUCCESS in 2m 36s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/bd56b92908944ea491ab9c5b81591688 : SUCCESS in 1h 58m 13s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/2c65b910ef504869b7605d0da14e10ad : SUCCESS in 1h 15m 05s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/b1862f7cbd474355afea95f12b9870ea : SUCCESS in 1h 12m 49s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/0306bd09a7084ed08a40c2fbb5b6c979 : SUCCESS in 1h 22m 50s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/f049301566164b7296f19e7df5fd6b3e : SUCCESS in 54m 29s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/951174738937462e864dc2106f14685f : SUCCESS in 1h 23m 48s","accounts_in_message":[],"_revision_number":1},{"id":"7cb441346495e3b1c4e87ebad3d36bf8d56ad4bd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 10:25:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"12c24162341dfde95caf5c5f894a4e14d8d3fe3f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 10:27:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca7b7eb3ffde4bd9b5cf694b72091063\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7f87923f08842c39301cb5ef64f8a4d : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9b0087fc0c4dc41988407172da49b2ef3dfb0811","revisions":{"9b0087fc0c4dc41988407172da49b2ef3dfb0811":{"kind":"REWORK","_number":1,"created":"2026-01-07 14:34:57.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/91/972491/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/91/972491/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/972491/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/972491/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/972491/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/91/972491/1"}}},"commit":{"parents":[{"commit":"af2aefeb87f6e2cdacf2b164338e440e131692ad","subject":"Merge \"[Docs] Remove note about FWaaS not being supported by OVN backend\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/af2aefeb87f6e2cdacf2b164338e440e131692ad"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-01-07 14:33:57.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-01-07 14:33:57.000000000","tz":60},"subject":"Bump oslo.config to 10.2.0","message":"Bump oslo.config to 10.2.0\n\nThis new version calls ``NamedExtensionManager`` with\n\"on_missing_entrypoints_callback\u003dNone\" parameter instead of\n\"warn_on_missing_entrypoint\u003dFalse\", that is deprecated.\n\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I9a231ad8ffbca46c8e71d35bc20d9762df0970f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9b0087fc0c4dc41988407172da49b2ef3dfb0811"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9b0087fc0c4dc41988407172da49b2ef3dfb0811"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
