)]}'
{"id":"openstack%2Fneutron~972986","triplet_id":"openstack%2Fneutron~master~If61989948a0457450782f0cf7207c52f78393f14","project":"openstack/neutron","branch":"master","topic":"bug/2136827","attention_set":{},"removed_from_attention_set":{"8655":{"account":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"last_update":"2026-04-07 08:26:29.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"If61989948a0457450782f0cf7207c52f78393f14","subject":"DNM: Trigger functional tests","status":"ABANDONED","created":"2026-01-09 23:51:01.000000000","updated":"2026-04-07 08:26:29.000000000","total_comment_count":1,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":false,"meta_rev_id":"0b1d9199e55302830a0465502a68f9a1d9817ebe","_number":972986,"virtual_id_number":972986,"owner":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-01-13 10:20:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}]},"reviewer_updates":[{"updated":"2026-01-10 00:19:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-13 10:36:12.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2026-04-07 08:26:29.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"}],"messages":[{"id":"7c52ab479617062625692fdecf5d1cc8f4c9c441","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-01-09 23:51:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"12ca2172e0c3cf4a0a1645e1589fa4e11c78604b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-10 00:19:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/398f6b6eaf8f4b9fad9bf96abdbc47e7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f6ae24aa4814dfab17de7f0a3ea81ec : SUCCESS in 20m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45aba17612fc4a43951cc617b14d4046 : SUCCESS in 11m 28s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/fb2f60ad7efb4e46b3a3e7112c897b14 : FAILURE in 18m 24s","accounts_in_message":[],"_revision_number":1},{"id":"166fe1b2edd3317e2a110905297b6d2dcdf8ec3e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-01-12 17:43:23.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"a5c7af8f1589155cd3901832b6419100f734210a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-01-12 18:40:32.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"64123cde676961820ce015bc616b56ce21c07948","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 19:04:44.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0045d24cdff747deb6437c9621684038\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6137c6d968954d1aa6f983878e6ac820 : SUCCESS in 17m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/073d0a2b57204914bfe3f92aac82e611 : SUCCESS in 6m 32s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/6c1d73550d4843cfb40ecdcd18dac46e : FAILURE in 22m 31s","accounts_in_message":[],"_revision_number":3},{"id":"968eb43c7067d058b486f515a2137f94bdb4fdcb","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-01-12 19:50:39.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"31e0042fc0c3860d2a7a1dfa655ad2346f99d07a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 20:58:51.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ad4b95a39444573a2563e6c3cdaa99a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3f61934387447aa9b354ba1c5b54d26 : SUCCESS in 24m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2462275a1bcd4025ad82511f70889b6a : SUCCESS in 13m 06s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/256ccd6ddb0448c1a692aecbb9081de0 : SUCCESS in 1h 06m 25s","accounts_in_message":[],"_revision_number":4},{"id":"b0ff70c08f7f789b05ee2478ab9cc7705b8a3096","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-01-13 08:57:56.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"b52c2240fd79c4ef12a5947c94e1929f267de358","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 10:20:57.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3cac45fafc4b49e89dca66336059e47b\n\n- neutron-functional-01 https://zuul.opendev.org/t/openstack/build/9212e4fd8bf848b69bb28c92aed3ae0f : SUCCESS in 1h 10m 43s\n- neutron-functional-02 https://zuul.opendev.org/t/openstack/build/93764194db604b6b834632e495aee27a : SUCCESS in 1h 11m 49s\n- neutron-functional-03 https://zuul.opendev.org/t/openstack/build/436718fba18544c7a19e6584d257c6a8 : SUCCESS in 1h 14m 08s\n- neutron-functional-04 https://zuul.opendev.org/t/openstack/build/523643a6f98647af8a073de6fa9fbbb2 : SUCCESS in 57m 20s\n- neutron-functional-05 https://zuul.opendev.org/t/openstack/build/8d738b7d3f9242c5a5751f295348c506 : SUCCESS in 46m 25s\n- neutron-functional-06 https://zuul.opendev.org/t/openstack/build/56c6f654abde4e6eada2cc4f546b6ee9 : FAILURE in 31m 08s\n- neutron-functional-07 https://zuul.opendev.org/t/openstack/build/d6ec4c9253e944ee8924fb9cff704a30 : SUCCESS in 45m 57s\n- neutron-functional-08 https://zuul.opendev.org/t/openstack/build/ccbb517b4a8145c1a981c819b9c8dc8d : SUCCESS in 47m 27s\n- neutron-functional-09 https://zuul.opendev.org/t/openstack/build/20262d4644914fba937d3b66979145fb : SUCCESS in 1h 18m 03s\n- neutron-functional-10 https://zuul.opendev.org/t/openstack/build/9d03bf1ddc254bccb717ba250ade700d : SUCCESS in 46m 03s","accounts_in_message":[],"_revision_number":5},{"id":"123f9610e8e42239291a3b98c3555b35c43cd8cb","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-01-13 10:36:12.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"2af8b94554d41dd1ceeb8353b312956f7e3e8264","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 11:28:24.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/99ece2dd6fc84bf1aae09f342c84496b\n\n- neutron-functional-01 https://zuul.opendev.org/t/openstack/build/246a181428fc4cda98ef1a285fc2e76a : SUCCESS in 47m 52s\n- neutron-functional-02 https://zuul.opendev.org/t/openstack/build/f3dea8c8a5014af5ad62649920dc23fa : SUCCESS in 50m 39s\n- neutron-functional-03 https://zuul.opendev.org/t/openstack/build/f6277cb5ecf845b7bf6e077b50c8123d : SUCCESS in 51m 15s\n- neutron-functional-04 https://zuul.opendev.org/t/openstack/build/6257469908a24df29bd7a5d0b3de96f0 : SUCCESS in 47m 58s\n- neutron-functional-05 https://zuul.opendev.org/t/openstack/build/87c244ad453d428195e7519b9c9ff0b0 : FAILURE in 22m 26s\n- neutron-functional-06 https://zuul.opendev.org/t/openstack/build/a05636428baa47fc9a7cdf470ef69980 : SUCCESS in 48m 28s\n- neutron-functional-07 https://zuul.opendev.org/t/openstack/build/b5e1b7920f344d2a83e6bc9b21643007 : SUCCESS in 50m 49s\n- neutron-functional-08 https://zuul.opendev.org/t/openstack/build/791766d6b2bc4b2880840d8d67dc81a2 : SUCCESS in 49m 34s\n- neutron-functional-09 https://zuul.opendev.org/t/openstack/build/4d674b9a458b44299b291987e8928f88 : SUCCESS in 48m 08s\n- neutron-functional-10 https://zuul.opendev.org/t/openstack/build/bf5935484ae74a31803e7b59c18ade01 : SUCCESS in 47m 09s","accounts_in_message":[],"_revision_number":5},{"id":"0b1d9199e55302830a0465502a68f9a1d9817ebe","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-04-07 08:26:29.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"61dc436b4ab0764e258aa685bc8057364022a2b5","revisions":{"0b0f6fb9ddf02baec20f3a58be21b57275a8d4a5":{"kind":"REWORK","_number":1,"created":"2026-01-09 23:51:01.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/86/972986/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/86/972986/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/972986/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/972986/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/972986/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/86/972986/1"}}},"commit":{"parents":[{"commit":"eb8205a118843da34691c902edbdd35d1ea28b60","subject":"func: Do not teardown DB if not dirty","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/eb8205a118843da34691c902edbdd35d1ea28b60"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-01-09 23:50:25.000000000","tz":0},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-01-09 23:50:53.000000000","tz":0},"subject":"DNM: Trigger functional tests","message":"DNM: Trigger functional tests\n\nChange-Id: If61989948a0457450782f0cf7207c52f78393f14\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0b0f6fb9ddf02baec20f3a58be21b57275a8d4a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0b0f6fb9ddf02baec20f3a58be21b57275a8d4a5"}]},"branch":"refs/heads/master"},"f1cc3af43c02f5c894d1feeb9c859a9882f49455":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-01-12 17:43:23.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/86/972986/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/86/972986/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/972986/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/972986/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/972986/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/86/972986/2"}}},"commit":{"parents":[{"commit":"47950e0976be87ed286059af6d4e648701faa679","subject":"func: Do not teardown DB if not dirty","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/47950e0976be87ed286059af6d4e648701faa679"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-01-09 23:50:25.000000000","tz":0},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-01-12 17:38:02.000000000","tz":0},"subject":"DNM: Trigger functional tests","message":"DNM: Trigger functional tests\n\nChange-Id: If61989948a0457450782f0cf7207c52f78393f14\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f1cc3af43c02f5c894d1feeb9c859a9882f49455"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f1cc3af43c02f5c894d1feeb9c859a9882f49455"}]},"branch":"refs/heads/master"},"dd0298eb4ea48c4da7b45d71b21ad1bb68c54d65":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-01-12 18:40:32.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/86/972986/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/86/972986/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/972986/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/972986/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/972986/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/86/972986/3"}}},"commit":{"parents":[{"commit":"fb13954404dacf79672410e49bb738a8a677f21c","subject":"func: Do not teardown DB if not dirty","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fb13954404dacf79672410e49bb738a8a677f21c"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-01-09 23:50:25.000000000","tz":0},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-01-12 18:40:23.000000000","tz":0},"subject":"DNM: Trigger functional tests","message":"DNM: Trigger functional tests\n\nChange-Id: If61989948a0457450782f0cf7207c52f78393f14\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dd0298eb4ea48c4da7b45d71b21ad1bb68c54d65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dd0298eb4ea48c4da7b45d71b21ad1bb68c54d65"}]},"branch":"refs/heads/master"},"3649c9f236af40c5ce433bdb365f4bbb2ef7d935":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-01-12 19:50:39.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/86/972986/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/86/972986/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/972986/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/972986/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/972986/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/86/972986/4"}}},"commit":{"parents":[{"commit":"fe9c4584b52b167bb1b1d64c6a57915910f59799","subject":"func: Do not teardown DB if not dirty","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fe9c4584b52b167bb1b1d64c6a57915910f59799"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-01-09 23:50:25.000000000","tz":0},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-01-12 19:50:24.000000000","tz":0},"subject":"DNM: Trigger functional tests","message":"DNM: Trigger functional tests\n\nChange-Id: If61989948a0457450782f0cf7207c52f78393f14\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3649c9f236af40c5ce433bdb365f4bbb2ef7d935"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3649c9f236af40c5ce433bdb365f4bbb2ef7d935"}]},"branch":"refs/heads/master"},"61dc436b4ab0764e258aa685bc8057364022a2b5":{"kind":"REWORK","_number":5,"created":"2026-01-13 08:57:56.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/86/972986/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/86/972986/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/972986/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/972986/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/972986/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/86/972986/5"}}},"commit":{"parents":[{"commit":"fe9c4584b52b167bb1b1d64c6a57915910f59799","subject":"func: Do not teardown DB if not dirty","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fe9c4584b52b167bb1b1d64c6a57915910f59799"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-01-09 23:50:25.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-01-13 08:57:34.000000000","tz":60},"subject":"DNM: Trigger functional tests","message":"DNM: Trigger functional tests\n\nChange-Id: If61989948a0457450782f0cf7207c52f78393f14\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/61dc436b4ab0764e258aa685bc8057364022a2b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/61dc436b4ab0764e258aa685bc8057364022a2b5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
