)]}'
{"id":"openstack%2Fneutron~973366","triplet_id":"openstack%2Fneutron~master~Ibea3307dd20b38112bca69b4536e05b5940de979","project":"openstack/neutron","branch":"master","topic":"bug/2136864","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-02-13 16:42:58.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2026-02-13 16:42:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibea3307dd20b38112bca69b4536e05b5940de979","subject":"[OVN] Sync segment host mapping reading the agents from the SQL DB","status":"MERGED","created":"2026-01-14 11:57:37.000000000","updated":"2026-02-13 16:44:08.000000000","submitted":"2026-02-13 16:42:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"973366-bug/2136864","meta_rev_id":"24b3602668871e7419fe3288bf6d354bfa312fd4","_number":973366,"virtual_id_number":973366,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-13 16:42:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"recommended":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-02-02 10:38:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2026-02-13 14:20:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2026-02-02 19:19:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":1,"date":"2026-02-05 14:59:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2026-02-13 14:20:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2026-01-30 08:22:02.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}],"CC":[{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-14 13:12:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-01-14 13:53:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-14 14:24:57.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"state":"CC"},{"updated":"2026-01-14 14:24:57.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"},{"updated":"2026-01-14 15:51:49.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2026-01-26 09:33:18.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2026-02-02 10:38:27.000000000","updated_by":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"reviewer":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"state":"REVIEWER"},{"updated":"2026-02-02 19:19:37.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2026-02-05 14:59:06.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2026-02-09 16:13:41.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2026-02-12 14:38:38.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"CC"},{"updated":"2026-02-13 14:20:46.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"71ea2e70e57d66f89b96079649496711513585fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-01-14 11:57:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"60c397543783ea04c14276e07bca0758d233ba01","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-01-14 12:20:03.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"14b26e3381fe9cd3190c4961f842781092c00fc7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 13:12:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c3e47c615d441458f3638e8e9f39fe0\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/dd3784b455154f049513a42d0ecfe509 : FAILURE in 1h 12m 17s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a29e56e4cc5d48d385c69dd5f1f907d9 : FAILURE in 57m 13s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0b29c4c9420f4a8fa535ab83afe5ab62 : FAILURE in 1h 11m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b1cc941f2278a210d6a4f5dbd538b6462bc0ca14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 13:53:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0b3990e8b8824279b59cc18c4b1b8a03\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ff1179c17537443980c581214446020b : SUCCESS in 1h 49m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/43db7402760b4821bd93320d9061b6d9 : FAILURE in 44m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c47e0344c6e949ccbbeed726502a6602 : SUCCESS in 24m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/119da9f5e60245a4875b0e53a128b21e : FAILURE in 32m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/962001dd8cc54a79a834441a9c47a183 : FAILURE in 23m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/959363e86d5042e1aab8711d16ee1fa2 : SUCCESS in 13m 06s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/1651a81f4825409db1f3e9e2a23dd9eb : SUCCESS in 1h 13m 38s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/3e49306790b643ca9bdf038f04824cb7 : FAILURE in 1h 51m 18s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/1df9fce07c0142f0a2cd54891b749b21 : SUCCESS in 1h 09m 09s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a830a58016c24c25ac53b2aef17a0895 : SUCCESS in 1h 06m 57s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5b24b41402e04c6892a2244f8c63f78a : SUCCESS in 1h 20m 47s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f677f725a4ff4f0cb7e49ade6df60dd3 : SUCCESS in 1h 07m 08s","accounts_in_message":[],"_revision_number":1},{"id":"57ed2df43d14db1f0370a6f99d1fc863fc6e48fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-01-14 14:23:08.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"b8702ea28c74ba577addd1f753b6dcc27ec2686a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 15:21:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8bc81a23cd274ac1b408977412b0c9eb\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8d50fb3cb4e141c585e9e3816c73811f : SUCCESS in 56m 26s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a31f15418691491da705a0caad4973b5 : SUCCESS in 24m 49s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f2c7d03fe0434063be7bda623f1983eb : SUCCESS in 30m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d2619366e4a97a590d0b9f683fac289a26b22dde","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 16:28:19.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4cc8ba93653c40a19d06ff9c4eaa2488\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/3ee62c34c02947e0a153c06e29582adf : SUCCESS in 2h 02m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/042b45d44c6447da9cd55f825653bf4c : SUCCESS in 37m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d5b737e61c047b58ce24ec7623c4ecf : SUCCESS in 12m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4d99e2f78ef648f7809260b8c7bbeecc : SUCCESS in 20m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4a67491680f247dd80019e2fbfea818f : SUCCESS in 37m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0248baf6ce140f2a6d591ba97736326 : SUCCESS in 13m 44s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/7a1e86b95d0b475c8c15b5ec7b41364b : SUCCESS in 1h 13m 05s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c7b02ea81ef941b1938b00328b046576 : FAILURE in 1h 21m 10s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/474356058c784ca3b53be2c2b91e0474 : SUCCESS in 1h 16m 57s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/f37216cbefa044e2b40065c6016dd39f : SUCCESS in 49m 23s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d558b7d37f9443d991e499642ffee2bf : SUCCESS in 1h 03m 34s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/8ff1172305974f5489974ceadb8b75cc : SUCCESS in 1h 28m 39s","accounts_in_message":[],"_revision_number":2},{"id":"10b05bbe2f628759b871c48f019afa788d08e82f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-01-15 13:52:34.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"1a23d7b8709cdd1a86b650135a14c7eeb580ae35","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-15 15:26:27.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/12bf06804f2048f8a9923d9e2ce507df\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e8d41f4d147b49b9b657e752a1c0e9f8 : SUCCESS in 52m 27s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/cbb82f2252d74eb2a8dadfe9881998f2 : SUCCESS in 22m 21s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1c1c81c5914642cfb4ccf5d01c0349bc : SUCCESS in 38m 05s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"453bc8e6e34d0796ded422addcaa9edd2c72d421","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-15 16:28:42.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e16de50e6d3047ae8ee6a0a4bdcd5d7b\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/8034c226be7f4e8389223f3fdf6bf837 : SUCCESS in 2h 25m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8e11c8913b5b47d9b0d74986077c6f2f : SUCCESS in 38m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9eae82db259d4afb868f2cfd8f617e5b : FAILURE in 7m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/340d476033824c558ce2df5f34da1c53 : SUCCESS in 21m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6ba9dd18202f4e9394bf6af9c0d3c810 : SUCCESS in 27m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d979031e8fd84c65af2655e81a4e3689 : SUCCESS in 11m 29s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/a92d61b269a54b9f99d2f8d24249c182 : SUCCESS in 1h 11m 25s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/61076226e5b647a9b4dc7527a7f8b5b3 : TIMED_OUT in 2h 33m 12s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/2f2e4115baa34d68a92bcff534c4ecb4 : SUCCESS in 1h 09m 08s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/108ab43866b546d1926f9f2e40946c77 : SUCCESS in 1h 14m 19s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0c23e01ca3a94777a4846dfe55e810ed : SUCCESS in 1h 24m 02s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/233fb4dc2f594620ad472d23717f13ce : SUCCESS in 1h 23m 59s","accounts_in_message":[],"_revision_number":3},{"id":"cb54ece1d6ddf32cadf9e28da53f252367c8f738","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-01-26 08:19:47.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"b14525bc6cf6ec674f1eb4bf3cbef3e41aade26f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 09:14:52.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19e284adf9804fa3addee7a7b35fb022\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/521d87fbe8744862aa9492e1c2c89403 : SUCCESS in 53m 16s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/35a36338c23047b3a31e1f9c2b67bbfc : SUCCESS in 41m 47s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/84cf4fe657624ab6865c8f2ba6640670 : RETRY_LIMIT in 4m 52s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"665ac6fbbfbcedd368ae760b253a9fe483387c7b","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-01-26 09:33:18.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"00aec66afaacf70221ed6b89b6c45fcfe51a7dd7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 11:00:21.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3977bff121c14892b21af7b971a711f9\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/bef7b8b3f6b842fd9b0f703251c7b4af : SUCCESS in 2h 06m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a0b88cca97f143abbd5bec0acd765694 : SUCCESS in 39m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3789116ba2614195be6bb5fdf6fadeef : SUCCESS in 19m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/63415ad05cd14cbb8835e24ae6cb64b5 : SUCCESS in 25m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/35308d04c7c94e00a146a9b80d0a10d5 : SUCCESS in 22m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/948a900d718846b5af18fc6d55bd3490 : SUCCESS in 13m 34s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/4aa7b4f4be914895b51d6198eb5513a0 : SUCCESS in 1h 10m 15s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/77ca0288a8d64a38b39ba65cdceb1508 : TIMED_OUT in 2h 35m 23s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/892ee5978d774f65a712fb44901cd394 : SUCCESS in 34m 59s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/777cf27d7a7b43419abdee37b8ada5b5 : SUCCESS in 1h 19m 26s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/26d3965132304cfdbd1bf48c5bf8a4dd : SUCCESS in 1h 22m 11s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f4794f1cd42549a7b6946013a3860738 : SUCCESS in 1h 17m 42s","accounts_in_message":[],"_revision_number":4},{"id":"05bc831ef80af16035aadef001c6b9ec129dc87e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-01-28 13:52:05.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"52b3e19d90ecaac5283300f638c11b259b162afb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 16:04:32.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2224a76b59244fa8ae041c78c7052bb6\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/27d81fac725645e5b893fb55aed29132 : SUCCESS in 2h 10m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0d3aadde41b74c45a9d53c08c29142bf : SUCCESS in 37m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f01fd1bd43f4b5da3625cacaf28f26d : SUCCESS in 25m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0758336b97c64d28b5a32ebaf26b0cd8 : SUCCESS in 33m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/16f4cca6bcec460981b34e2e81f76909 : SUCCESS in 25m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca29034d038742a992b3013debcf88e7 : SUCCESS in 15m 04s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/eed30a855a674eedbeb0dbe298df52c7 : SUCCESS in 51m 05s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/525966388f00436cba6e5a975fbb8420 : FAILURE in 47m 45s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/1b74927c090d4c93a4a686ffeba7440d : SUCCESS in 1h 12m 51s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/2f0c423b4372459099cf0cf58b400866 : SUCCESS in 46m 56s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/11f92d58e5a8434484843e7ab4046231 : SUCCESS in 1h 05m 31s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f7d9752fadf6418792e31d19f2a47ff7 : SUCCESS in 1h 22m 29s","accounts_in_message":[],"_revision_number":5},{"id":"e8905326368092fbe44c30fd052b026288055f79","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 23:59:46.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35c54f03c2da458fa0878d9cd0ecdf9d\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/21192bd96328420bba0f14a9bc8c9603 : SUCCESS in 55m 10s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d502b50085c74da7a3ae407191f6e126 : SUCCESS in 43m 14s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/96c7e02e2a2d47faa7aa244bd248d054 : RETRY_LIMIT in 5m 22s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"e4354e9ac57fafc4caf0f548e6acd3c6b5e6ace4","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-01-29 12:07:45.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"0c3d9f6f95eb8ed8a64fd638ba2fbf86cbcf928f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 14:17:36.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/17b1c9547f3b431b9ff23cde05f75a20\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/33679e1efcd6474c851241f5cccc8b57 : SUCCESS in 2h 06m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e8f99fe3e4ed47118732350cf01965a1 : TIMED_OUT in 1h 21m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc058615da7a4946b5fbbd4339b3b7e9 : SUCCESS in 10m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d09abf9035c742a09c0f4d063a542db8 : SUCCESS in 33m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/553828e09c60432d8a46ba524934091e : SUCCESS in 29m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2ad9e7beb6b42f4a2833770be683e8f : SUCCESS in 13m 29s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/caa3a1ac7ef64d618a5dfa9027e0c08f : SUCCESS in 1h 08m 25s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/e12b50c27b304d3ca36100a4473183cd : SUCCESS in 1h 14m 14s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/14c7b49316de45cfb3bf83bcc6ff89b3 : SUCCESS in 34m 10s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/16c56697c8ef41e787b83d39d4413400 : SUCCESS in 1h 25m 30s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/cd54b578e1a34e428d006a219b174814 : SUCCESS in 1h 27m 56s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f4cc193f7bdd49078fa57b17f177581c : SUCCESS in 1h 22m 03s","accounts_in_message":[],"_revision_number":5},{"id":"a145520ade59e42a68133f3780075e84e879c690","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-01-29 15:37:03.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"7a466c727fc20d8f45fa2646145ebef686109e38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 17:47:28.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8f739772b304eae81841b7662ef343c\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/2d215d34deeb4016b8d289432d4adbb0 : SUCCESS in 2h 09m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/696c6a49d72d48bb91e629d3b1b69eab : SUCCESS in 1h 03m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6cbd83cc571a449aa3e0aaa52bdedca9 : SUCCESS in 10m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/529273e51b014f39b0f749f96893f77c : SUCCESS in 31m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/522c2d0376d548ed80f5952efb1fd541 : SUCCESS in 26m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58fad2d230ef4d6ba55b51fccf4df0da : SUCCESS in 6m 08s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/adb355634fab4ef79db79b24278cf8a0 : SUCCESS in 1h 08m 54s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/812a9d562cd44c21bb93d7ce768e109d : SUCCESS in 1h 28m 52s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/195cdae6cb754a699c209b3cc091123c : SUCCESS in 1h 09m 53s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/582bd7a34619401cae1588ed36d2cedb : SUCCESS in 1h 05m 35s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4eba563f1b22468ba8fa902d7831042b : SUCCESS in 1h 23m 24s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f7af0695537e4ffd811b241c3ae9de4d : SUCCESS in 1h 21m 05s","accounts_in_message":[],"_revision_number":5},{"id":"99c52eb6bbda2b0d377994fdde4ffd3f02279176","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-30 00:31:38.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85170a3aa8c9429ab1e1d27d6f62fb3a\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a09e05266e0b40bd9b7050de325da828 : SUCCESS in 57m 04s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d52fdc60e47145e3b6728a5d8d583125 : SUCCESS in 44m 27s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a9879cbdcc2746359d41dc49d4b74b3c : RETRY_LIMIT in 4m 36s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"368097d5568723b580d92314f6235a21d4caeb7f","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-01-30 08:22:02.000000000","message":"Patch Set 5: Review-Priority+1","accounts_in_message":[],"_revision_number":5},{"id":"b4bc14bbd4e5326f68c832e1dec30da401882f81","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2026-02-02 10:38:27.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"1dfde4b928789c5ea83c3504f6f682e9776b63fd","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-02-02 19:19:37.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"9682ec1ac6ecd0a14bbaf19ddba520256ad9ea59","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-02-05 14:59:06.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"0a9eaf318163027f0cc20d35f4738c85184d0c19","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-02-13 14:20:46.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"64a28ff6a960305123fc87bf2daeb526b85d0be1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 14:21:20.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"f84a858a16bdec0384da2a9129e3d007501f2fe2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 16:42:58.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0df5c4692771481db0b548cf763e82f5\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/2bb110da43cc41849cabadf6bd801c8d : SUCCESS in 2h 19m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c30d738e1e1a4044b3d2df779b255c3e : SUCCESS in 26m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/76d5ea6bb6a345468e1986146ed6d839 : SUCCESS in 25m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b015c854e67549b3b792ad5c6e3ebba4 : SUCCESS in 26m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd328b7e3b454cc099aa5b3c3fc47aaa : SUCCESS in 13m 34s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/052a4788d9634558bc72cdd5d80892ed : SUCCESS in 1h 26m 57s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/3f67a749771c45fcbfe329570c51e697 : SUCCESS in 1h 13m 33s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/b86561ab3e5340b0ba7e02cfbd7071cf : SUCCESS in 50m 02s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/fd87219705f8446ea1d038cde926a523 : SUCCESS in 1h 10m 24s","accounts_in_message":[],"_revision_number":5},{"id":"deea2a66ff55fec5b89bd57cb9781bd5b5f6d6b4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 16:42:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"24b3602668871e7419fe3288bf6d354bfa312fd4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 16:44:08.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/864e139b4d064d429792b1cfd2097bb5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/279ba9f035914ae1ad47fd2720738b93 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"eea85b6530de6aa88c4ec24a3583618b9599cef4","revisions":{"99bbe11926c20811eb635336d849e203b641e5ff":{"kind":"REWORK","_number":1,"created":"2026-01-14 11:57:37.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/66/973366/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/66/973366/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/66/973366/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/66/973366/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/66/973366/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/66/973366/1"}}},"commit":{"parents":[{"commit":"9656a5fcf2c706d2f159655f9fa40c9768bbf5c1","subject":"Merge \"bgp: Introduce BGP chassis commands\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9656a5fcf2c706d2f159655f9fa40c9768bbf5c1"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-01-14 11:46:07.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-01-14 11:46:07.000000000","tz":60},"subject":"[OVN] Sync segment host mapping reading the agents from the SQL DB","message":"[OVN] Sync segment host mapping reading the agents from the SQL DB\n\nThe OVN agents are not stored in the SQL database. Instead, the OVN\nagents are locally cached in each Neutron API worker in a singleton\ninstance ``AgentCache``.\n\nSince [1], the method ``sync_hostname_and_physical_networks`` was using\na SQL filter to retrieve the current host mapped segments, using the SQL\nstored agents. This is incorrect for ML2/OVN.\n\nThis patch removes this filter, retrieving all the current host mapped\nsegments. The method ``sync_hostname_and_physical_networks``:\n* Removes the hosts not present in the OVN SB but still present in the\n  SQL database, comparing the current ``Chassis`` physnet mapping.\n* Creates new mappings based on the new ``Chassis`` registers created.\n* Updates the present ones if the physnet mappings have been updated.\n\nThe functional test ``test_ovn_sb_sync`` has been updated to reflect the\nchanges in the code base.\n\n[1]https://review.opendev.org/q/I2cfd52a2657fad989e24e974fda470ecd960262b\n\nCloses-Bug: #2136864\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: Ibea3307dd20b38112bca69b4536e05b5940de979\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/99bbe11926c20811eb635336d849e203b641e5ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/99bbe11926c20811eb635336d849e203b641e5ff"}]},"branch":"refs/heads/master"},"2677aea372ea7f9543cd96cfc4aac9311deda5db":{"kind":"REWORK","_number":2,"created":"2026-01-14 14:23:08.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/66/973366/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/66/973366/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/66/973366/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/66/973366/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/66/973366/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/66/973366/2"}}},"commit":{"parents":[{"commit":"9656a5fcf2c706d2f159655f9fa40c9768bbf5c1","subject":"Merge \"bgp: Introduce BGP chassis commands\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9656a5fcf2c706d2f159655f9fa40c9768bbf5c1"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-01-14 11:46:07.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-01-14 14:21:40.000000000","tz":60},"subject":"[OVN] Sync segment host mapping reading the agents from the SQL DB","message":"[OVN] Sync segment host mapping reading the agents from the SQL DB\n\nThe OVN agents are not stored in the SQL database. Instead, the OVN\nagents are locally cached in each Neutron API worker in a singleton\ninstance ``AgentCache``.\n\nSince [1], the method ``sync_hostname_and_physical_networks`` was using\na SQL filter to retrieve the current host mapped segments, using the SQL\nstored agents. This is incorrect for ML2/OVN.\n\nThis patch removes this filter, retrieving all the current host mapped\nsegments. Then retrieves the OVN controller hostnames from the local\ncache and returns the ones present in both sets.\n\n The method ``sync_hostname_and_physical_networks``:\n* Removes the hosts not present in the OVN SB but still present in the\n  SQL database, comparing the current ``Chassis`` physnet mapping.\n* Creates new mappings based on the new ``Chassis`` registers created.\n* Updates the present ones if the physnet mappings have been updated.\n\nThe functional test ``test_ovn_sb_sync`` has been updated to reflect the\nchanges in the code base.\n\n[1]https://review.opendev.org/q/I2cfd52a2657fad989e24e974fda470ecd960262b\n\nCloses-Bug: #2136864\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: Ibea3307dd20b38112bca69b4536e05b5940de979\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2677aea372ea7f9543cd96cfc4aac9311deda5db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2677aea372ea7f9543cd96cfc4aac9311deda5db"}]},"branch":"refs/heads/master"},"2573e54d895d054987f666ba7e1e5f0d1dcc086f":{"kind":"REWORK","_number":3,"created":"2026-01-15 13:52:34.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/66/973366/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/66/973366/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/66/973366/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/66/973366/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/66/973366/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/66/973366/3"}}},"commit":{"parents":[{"commit":"9656a5fcf2c706d2f159655f9fa40c9768bbf5c1","subject":"Merge \"bgp: Introduce BGP chassis commands\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9656a5fcf2c706d2f159655f9fa40c9768bbf5c1"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-01-14 11:46:07.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-01-15 13:52:13.000000000","tz":60},"subject":"[OVN] Sync segment host mapping reading the agents from the SQL DB","message":"[OVN] Sync segment host mapping reading the agents from the SQL DB\n\nThe OVN agents are not stored in the SQL database. Instead, the OVN\nagents are locally cached in each Neutron API worker in a singleton\ninstance ``AgentCache``.\n\nSince [1], the method ``sync_hostname_and_physical_networks`` was using\na SQL filter to retrieve the current host mapped segments, using the SQL\nstored agents. This is incorrect for ML2/OVN.\n\nThis patch removes this filter, retrieving all the current host mapped\nsegments. Then retrieves the OVN controller hostnames from the local\ncache and returns the ones present in both sets.\n\n The method ``sync_hostname_and_physical_networks``:\n* Removes the hosts not present in the OVN SB but still present in the\n  SQL database, comparing the current ``Chassis`` physnet mapping.\n* Creates new mappings based on the new ``Chassis`` registers created.\n* Updates the present ones if the physnet mappings have been updated.\n\nThe functional test ``test_ovn_sb_sync`` has been updated to reflect the\nchanges in the code base.\n\n[1]https://review.opendev.org/q/I2cfd52a2657fad989e24e974fda470ecd960262b\n\nCloses-Bug: #2136864\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: Ibea3307dd20b38112bca69b4536e05b5940de979\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2573e54d895d054987f666ba7e1e5f0d1dcc086f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2573e54d895d054987f666ba7e1e5f0d1dcc086f"}]},"branch":"refs/heads/master"},"1cae690beb2f4594e29031cbf0c0e04697c41a07":{"kind":"REWORK","_number":4,"created":"2026-01-26 08:19:47.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/66/973366/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/66/973366/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/66/973366/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/66/973366/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/66/973366/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/66/973366/4"}}},"commit":{"parents":[{"commit":"d384b7f5f8cebcfd3f67708d5e129e16b656dc62","subject":"Merge \"Change agent and Ipam code to use project_id\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d384b7f5f8cebcfd3f67708d5e129e16b656dc62"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-01-14 11:46:07.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-01-26 08:19:36.000000000","tz":60},"subject":"[OVN] Sync segment host mapping reading the agents from the SQL DB","message":"[OVN] Sync segment host mapping reading the agents from the SQL DB\n\nThe OVN agents are not stored in the SQL database. Instead, the OVN\nagents are locally cached in each Neutron API worker in a singleton\ninstance ``AgentCache``.\n\nSince [1], the method ``sync_hostname_and_physical_networks`` was using\na SQL filter to retrieve the current host mapped segments, using the SQL\nstored agents. This is incorrect for ML2/OVN.\n\nThis patch removes this filter, retrieving all the current host mapped\nsegments. Then retrieves the OVN controller hostnames from the local\ncache and returns the ones present in both sets.\n\n The method ``sync_hostname_and_physical_networks``:\n* Removes the hosts not present in the OVN SB but still present in the\n  SQL database, comparing the current ``Chassis`` physnet mapping.\n* Creates new mappings based on the new ``Chassis`` registers created.\n* Updates the present ones if the physnet mappings have been updated.\n\nThe functional test ``test_ovn_sb_sync`` has been updated to reflect the\nchanges in the code base.\n\n[1]https://review.opendev.org/q/I2cfd52a2657fad989e24e974fda470ecd960262b\n\nCloses-Bug: #2136864\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: Ibea3307dd20b38112bca69b4536e05b5940de979\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1cae690beb2f4594e29031cbf0c0e04697c41a07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1cae690beb2f4594e29031cbf0c0e04697c41a07"}]},"branch":"refs/heads/master"},"eea85b6530de6aa88c4ec24a3583618b9599cef4":{"kind":"REWORK","_number":5,"created":"2026-01-28 13:52:05.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/66/973366/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/66/973366/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/66/973366/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/66/973366/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/66/973366/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/66/973366/5"}}},"commit":{"parents":[{"commit":"f86b34dcdf4b977c0739ebc73f34e7092e070c1e","subject":"Merge \"bgp: Add event to create BGP chassis resources\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f86b34dcdf4b977c0739ebc73f34e7092e070c1e"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-01-14 11:46:07.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-01-28 13:51:52.000000000","tz":60},"subject":"[OVN] Sync segment host mapping reading the agents from the SQL DB","message":"[OVN] Sync segment host mapping reading the agents from the SQL DB\n\nThe OVN agents are not stored in the SQL database. Instead, the OVN\nagents are locally cached in each Neutron API worker in a singleton\ninstance ``AgentCache``.\n\nSince [1], the method ``sync_hostname_and_physical_networks`` was using\na SQL filter to retrieve the current host mapped segments, using the SQL\nstored agents. This is incorrect for ML2/OVN.\n\nThis patch removes this filter, retrieving all the current host mapped\nsegments. Then retrieves the OVN controller hostnames from the local\ncache and returns the ones present in both sets.\n\n The method ``sync_hostname_and_physical_networks``:\n* Removes the hosts not present in the OVN SB but still present in the\n  SQL database, comparing the current ``Chassis`` physnet mapping.\n* Creates new mappings based on the new ``Chassis`` registers created.\n* Updates the present ones if the physnet mappings have been updated.\n\nThe functional test ``test_ovn_sb_sync`` has been updated to reflect the\nchanges in the code base.\n\n[1]https://review.opendev.org/q/I2cfd52a2657fad989e24e974fda470ecd960262b\n\nCloses-Bug: #2136864\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: Ibea3307dd20b38112bca69b4536e05b5940de979\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/eea85b6530de6aa88c4ec24a3583618b9599cef4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/eea85b6530de6aa88c4ec24a3583618b9599cef4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
