)]}'
{"id":"openstack%2Fneutron~973473","triplet_id":"openstack%2Fneutron~master~Ib8b738e9512fcfde60c80b81af672fcb65c5f38d","project":"openstack/neutron","branch":"master","topic":"bug/2138430","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2026-01-20 18:57:30.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib8b738e9512fcfde60c80b81af672fcb65c5f38d","subject":"Remove stale methods from the OvnNbSynchronizer class","status":"MERGED","created":"2026-01-15 13:29:02.000000000","updated":"2026-01-20 18:58:33.000000000","submitted":"2026-01-20 18:57:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"973473-bug/2138430","meta_rev_id":"2e390282d7a2b99b65c26beebf4f5492b3f14a32","_number":973473,"virtual_id_number":973473,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-20 18:57:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-01-19 09:59:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2026-01-20 16:47:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-01-20 16:47:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-15 14:02:45.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2026-01-15 14:31:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-01-15 15:42:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-15 17:46:56.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2026-01-19 09:59:39.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2026-01-20 16:47:23.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"cefd8e06b5ef3c9caf57d52978a39c17c82b93ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-01-15 13:29:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"33a829c9749183fd54096244a610027b862101c7","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-01-15 14:02:45.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"b86e227f2875a90ba2057224e69473d2bdcbb5fa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-15 14:31:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d89d1aa742984fd5aa56852c49d1257c\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/13d0a5c8a11d45e392125521167d056f : SUCCESS in 1h 00m 33s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5503addb3d644b8e9e70c2a3aba618cb : SUCCESS in 47m 41s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ba9b80035623434aa5723ffea3a51a02 : SUCCESS in 1h 00m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e11f6c2c76f62a807a46c9db076af3a0a04ca44d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-15 15:42:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/68674b5906ad48c18c505d8237aaad36\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/eff5631482ee48aba92928fef9540387 : SUCCESS in 2h 10m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7749a50108e249aab881c04f30f0e039 : SUCCESS in 52m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b4050b496a645b18ca238e8a66ace43 : FAILURE in 4m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/60897d122e794e1f91f9bce141cb9ef7 : SUCCESS in 24m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1ea5eede2c4343dbaa77ff3a8caa3a02 : SUCCESS in 29m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98b0095c84d640d88d3afee6958d8c38 : SUCCESS in 6m 51s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/eac806cc88704e77b925ac610be641b4 : SUCCESS in 1h 03m 29s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/154867bfe7ec481da65a16c3c278372a : SUCCESS in 1h 26m 06s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/2ca308247a774fd98fc64ce219ff4f31 : SUCCESS in 1h 13m 33s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/55a54475b64942c5b883a396437aca7a : SUCCESS in 1h 06m 37s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2136fd0dd9d240deb54f27f159b00612 : SUCCESS in 1h 06m 52s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/25e9d39bdc81444fb6a42aad79cde07c : SUCCESS in 1h 13m 44s","accounts_in_message":[],"_revision_number":1},{"id":"cfe643a03b41e97c8e1d5106b179e545d09d7404","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-01-19 09:52:47.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"b5dd617e2645e9ffc4da29fe6639b611d27bfc95","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-01-19 09:59:39.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"68121f55775064029216e9b3b92fd12de303f5fb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 10:22:24.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a91de8f81d84aa5a546f9cd6efc41d1\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/80906de41146471c99f152b39895b472 : SUCCESS in 27m 58s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/29841e9d581d4d1eb8f0577e5f4e4809 : SUCCESS in 22m 41s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/286a015b4dec46198d20c5a7cc13d3f0 : RETRY_LIMIT in 3m 51s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8b9a2aba32c613e1c4e5c9db5076c9d0054503b4","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-01-19 11:09:19.000000000","message":"Topic I12f018d29e7d00d1f0fb0272dc905d1026633cbb removed","accounts_in_message":[],"_revision_number":2},{"id":"bbf00c324349b8f636f36b9503ad251e6eab05f4","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-01-19 11:09:24.000000000","message":"Topic set to bug/2138430","accounts_in_message":[],"_revision_number":2},{"id":"083ac6fd5557086412447b4fea0ab1b083eac7c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 11:39:44.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e1e8459657a84ac39ff5ff656f3bf950\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/6e75e2c42564471c944a0b0f84680328 : SUCCESS in 1h 43m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6731d1b4637a4b66862f5f18fa2ec871 : SUCCESS in 36m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/900251b41a194b91a9529e02e733fe3f : SUCCESS in 12m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/93ad88f1606e4ef79f3b3053a89463d7 : SUCCESS in 21m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3d837e9ff6374dba8c03651fb68d3b9d : SUCCESS in 24m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/170355377d7742ea8563372d3ec66983 : SUCCESS in 13m 14s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/ffd03ca5b4424a0eaaba1e9f862b3712 : FAILURE in 19m 47s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c5868f56aab44ca6a907c6d0552e2b13 : SUCCESS in 1h 35m 43s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/32d7bf4e579848988182b8601eb05863 : SUCCESS in 1h 09m 47s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/7628596f4e04497b8eb680308be194d4 : SUCCESS in 1h 24m 07s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b9ee00def0d3417bb08b51677357b025 : SUCCESS in 47m 09s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/ab896653ab2647ef90e2c6b1250bd2ff : SUCCESS in 59m 24s","accounts_in_message":[],"_revision_number":2},{"id":"2ab99b5ceccf7f18c9c5e3147c622f145dea05d8","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-01-20 14:14:53.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fd6a4ef7f726554fb53572db0d366c4e0dc399bf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 15:27:22.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5b3e62a81104f5ead156bd6f8890ae8\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e68421854c3f49b6957100afc418693c : SUCCESS in 38m 19s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/23c8f2f204674f95ba3443b1714a7df5 : SUCCESS in 25m 59s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/898d9732e023484db781f745cd62d9eb : RETRY_LIMIT in 4m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6aa2b2389505989d538b303b82851d7173a4d914","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 16:12:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e09c6cbba8d428781b20c841b03ba06\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/7f3da4241cc3427ca8311a7f431d6a6d : SUCCESS in 1h 49m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c3d42703d0a6439e9b86f55a0cac2e4a : SUCCESS in 1h 00m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c10c6dbb3d14b00ae47f8ab93657bdf : SUCCESS in 23m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d49f01699f02457e84b0ac20b5d1f7f5 : SUCCESS in 36m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4542fcd6d17d4406a72d24e58d81062a : SUCCESS in 30m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3df64d037c8d46abacb577d7b031f58c : SUCCESS in 12m 13s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/501725d2c2b94472899d5f4e4bc046be : SUCCESS in 1h 11m 03s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/1271b0466dca4e88af7aae264e06217d : SUCCESS in 1h 47m 05s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/1ca7770b35824ce9946a22e2b2c5e56a : SUCCESS in 1h 15m 13s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/71a3ceeaa9b746d8bd0816abffc8273f : SUCCESS in 56m 58s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3dd890b270894dac961e45f769080e08 : SUCCESS in 36m 35s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/408944887a82430f8c423b242e1b1a0d : SUCCESS in 1h 12m 37s","accounts_in_message":[],"_revision_number":2},{"id":"db2322e092885defddaf04069a3c6ae68ea58e68","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-01-20 16:47:23.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1231f02f0cf15fdf1ce7ff40da77c8bfcac7d244","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 16:47:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"eb977326bbad511a6a29883a504d1e15b1a6ebd4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 18:57:30.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/636c41eed7ed41d7814d28ad52b7b72d\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d1e411f795bf479580f485e8eb3f4324 : SUCCESS in 2h 08m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/307c08c2a7e24615ab5b58c8559e8bc1 : SUCCESS in 12m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/799cc5f8320b4a589b4c49b0922be610 : SUCCESS in 28m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6759385f451e43269cf7b0cb1da524a9 : SUCCESS in 25m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d94cba09bc9435cbeb11f4bce789289 : SUCCESS in 7m 05s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/4cac0707069d43edb13d34363bad9696 : SUCCESS in 1h 02m 47s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/c3f91bb6901449118094c12aa9c4cbf2 : SUCCESS in 56m 52s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/9835e3e22c6848aaa256b2a1332c047d : SUCCESS in 52m 19s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/28dd69f82af84104820ae020c817a9f0 : SUCCESS in 1h 31m 16s","accounts_in_message":[],"_revision_number":2},{"id":"68f56731a9eb513144d587e18ad609b5dd47c9c7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 18:57:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"2e390282d7a2b99b65c26beebf4f5492b3f14a32","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 18:58:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28a8caee5ce54c5aae76969c5b1c9cda\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/186bb6c6cd86405bab1fe67b958dc578 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b243c1b2d7a5bdbfc1ef50d9839d33d07d423cae","revisions":{"74c8c90774941ce0dbd689408b131a6c771b426b":{"kind":"REWORK","_number":1,"created":"2026-01-15 13:29:02.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/73/973473/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/73/973473/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/973473/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/973473/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/973473/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/73/973473/1"}}},"commit":{"parents":[{"commit":"9656a5fcf2c706d2f159655f9fa40c9768bbf5c1","subject":"Merge \"bgp: Introduce BGP chassis commands\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9656a5fcf2c706d2f159655f9fa40c9768bbf5c1"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-01-15 13:24:11.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-01-15 13:28:58.000000000","tz":60},"subject":"Remove stale methods from the OvnNbSynchronizer class","message":"Remove stale methods from the OvnNbSynchronizer class\n\nThere were old methods: `remove_common_acls` and `get_acls` in the\nOvnNbSynchronizer class which were note really used since patch [1]\nwhich we have merged long time ago.\nThis patch just makes follow up cleanup and removes not used code.\n\n[1] https://review.opendev.org/c/openstack/neutron/+/733886\n\nChange-Id: Ib8b738e9512fcfde60c80b81af672fcb65c5f38d\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/74c8c90774941ce0dbd689408b131a6c771b426b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/74c8c90774941ce0dbd689408b131a6c771b426b"}]},"branch":"refs/heads/master"},"b243c1b2d7a5bdbfc1ef50d9839d33d07d423cae":{"kind":"REWORK","_number":2,"created":"2026-01-19 09:52:47.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/73/973473/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/73/973473/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/973473/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/973473/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/973473/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/73/973473/2"}}},"commit":{"parents":[{"commit":"9656a5fcf2c706d2f159655f9fa40c9768bbf5c1","subject":"Merge \"bgp: Introduce BGP chassis commands\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9656a5fcf2c706d2f159655f9fa40c9768bbf5c1"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-01-15 13:24:11.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-01-19 09:52:20.000000000","tz":60},"subject":"Remove stale methods from the OvnNbSynchronizer class","message":"Remove stale methods from the OvnNbSynchronizer class\n\nThere were old methods: `remove_common_acls` and `get_acls` in the\nOvnNbSynchronizer class which were note really used since patch [1]\nwhich we have merged long time ago.\nThis patch just makes follow up cleanup and removes not used code.\n\n[1] https://review.opendev.org/c/openstack/neutron/+/733886\n\nChange-Id: Ib8b738e9512fcfde60c80b81af672fcb65c5f38d\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b243c1b2d7a5bdbfc1ef50d9839d33d07d423cae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b243c1b2d7a5bdbfc1ef50d9839d33d07d423cae"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
