)]}'
{"id":"openstack%2Fneutron~973584","triplet_id":"openstack%2Fneutron~master~Idc54d1b307fa88a4113228a63049f5fc7ddaa7cb","project":"openstack/neutron","branch":"master","topic":"bug/2138468","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-01-21 10:25:58.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2026-01-21 08:23:19.000000000","reason":"\u003cGERRIT_ACCOUNT_11975\u003e replied on the change","reason_account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2026-01-20 15:53:44.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"Idc54d1b307fa88a4113228a63049f5fc7ddaa7cb","subject":"Fix ``LogicalRouterPortEvent`` when a LR is deleted","status":"MERGED","created":"2026-01-16 07:52:39.000000000","updated":"2026-01-21 10:28:14.000000000","submitted":"2026-01-21 10:25:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"973584-bug/2138468","meta_rev_id":"d438c5d8e3ee9363c4ecd6e95a2c137bd7610b3c","_number":973584,"virtual_id_number":973584,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-21 10:25:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2026-01-21 08:23:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2026-01-20 15:53:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-01-21 08:23:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-16 08:51:17.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2026-01-16 08:58:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-01-16 09:09:38.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2026-01-16 10:10:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-19 11:03:23.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2026-01-20 15:53:44.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"e7755be8231be03c8d89462f486692f1779e4760","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-01-16 07:52:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cd5fdb7bdd8bd4c1afb8ea76cd9e8cddd46ffb31","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-01-16 08:51:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5afbe13f4cb87567d95cffb8b3297ad514a27b50","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 08:58:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fffb6b7fb0fc40ffa4745de50a3d9161\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6c91047c9f554a9e8e415757466f7f0f : SUCCESS in 57m 29s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0d442faea90a40a0bf193592c2c04aee : SUCCESS in 23m 17s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/38cc9533509b4524918c726c806e071f : SUCCESS in 54m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4cbc90abc462302cac6ccd101b79f0a16c2ddb89","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-01-16 09:55:51.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ec9f10f2886e121fe32d5a6aa354614854373043","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-01-16 09:56:00.000000000","message":"Removed Code-Review+2 by \u003cGERRIT_ACCOUNT_11975\u003e\n","accounts_in_message":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}],"_revision_number":1},{"id":"575d8418c83197cc59edda575f4ba4d4715d38d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 10:10:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/318c7b79026a405588ee240fb07e2f03\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/cabaac564eb14bddb9d7c54f7fa308ed : SUCCESS in 2h 09m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d0e25c025ad641df9afc12ab0c5bbecf : SUCCESS in 42m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41537b6a6092465aadac80ec65a5a691 : SUCCESS in 11m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/15d2594ff6714a30a13ac25f6324b24e : SUCCESS in 26m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/64426b60548447f08d522a23e52dd06d : SUCCESS in 24m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c8bf51f8fb042a1bfa08923d60db774 : SUCCESS in 10m 42s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/474567d19e8c4e72978d078d43e580a9 : SUCCESS in 1h 21m 54s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f496a91feac840318113a7d9afc3a1b8 : SUCCESS in 1h 50m 15s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/63b6e73cd4cb4488b5b40d937ddd49e1 : SUCCESS in 1h 10m 41s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/84c1c47e4e7c4ff5b137d275aad40fb6 : SUCCESS in 48m 13s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f96692975a0049b5ab884feabf1f1a73 : SUCCESS in 1h 01m 53s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a47392eb1e924a29b0b3f98404e27436 : SUCCESS in 1h 27m 42s","accounts_in_message":[],"_revision_number":1},{"id":"b5fa1ab5efba594e6a06a0d94cea2f6968501185","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-01-16 10:13:10.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"069e723917fb43d648a9e59a117b49e41d3fb9e8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 11:11:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1d30de3891c4b59b033f917b32286f4\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/adc4b61e93db429eac037d5a86cff495 : SUCCESS in 25m 36s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/276d31f5c59047e4a0842b82afc43a15 : SUCCESS in 49m 57s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c1d4186eeff1440b972520242026dc44 : SUCCESS in 27m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ae961981078ce5b61b1d99c7e35bacaad36abe6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 12:05:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/83d0305a361341a58a3a942b266d87fe\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c93ba4ab2f734314a7f99638dd32800e : SUCCESS in 1h 45m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c21b91b61c734befbf90ae22e8b54058 : SUCCESS in 1h 02m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4686bfe75104b29a303517b0f1dd50a : SUCCESS in 21m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bf8350751d0e462980dd32d104f1cfdd : SUCCESS in 23m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/79d2bd26ca8948e4ab6ee6fe79ffda53 : SUCCESS in 22m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ce3c0186d23497d992b467f5cc30f5c : SUCCESS in 5m 49s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/f77def17fec940aa8afb0e0fcaec76c5 : SUCCESS in 1h 07m 16s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a800069e6c964e128a33758c32e70a38 : SUCCESS in 1h 47m 35s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/c13edccf5fc6465d8c8f1582d7042bf8 : SUCCESS in 52m 57s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/894b5143bda8456896244471ca86d260 : SUCCESS in 51m 29s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/33d04f4456fe48798a87dce979f1c4e9 : SUCCESS in 1h 16m 26s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/97f6ac1eb95a44c381b9173e0c0e2791 : SUCCESS in 1h 03m 39s","accounts_in_message":[],"_revision_number":2},{"id":"4112d366b4a9a6416f0d91d7c31bf2ce48c8d207","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-01-16 15:38:28.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"5d981942c71b196b49752b718ac91b6a679c763c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 17:39:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a925fa7425114230b779d9f0929a2376\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/0ccfa358dd1d4309928afd931f21e4c6 : SUCCESS in 1h 56m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/30c043582bb74f2a8962efe00f97e1b3 : SUCCESS in 53m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44546e9e86c04fa49844ce0cf2e57e26 : SUCCESS in 11m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/81ecec5e704f4bf98eb5945ddd098071 : SUCCESS in 34m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/de0d77655bef4cca9cf5360373ed48de : SUCCESS in 31m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f7e3d2b87b64d2a8247db9b30cabeb5 : SUCCESS in 11m 58s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/d7965bd3cbfd4c85a4084c8c1872c868 : SUCCESS in 1h 12m 26s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d41a0b9b49484c47a05ff852cfb6f146 : SUCCESS in 1h 04m 41s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/69be266f3eba49c9801a264b65efaf28 : SUCCESS in 40m 50s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/09685917c7874a00846597ac81c2d467 : SUCCESS in 1h 09m 05s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/89fcebc2c13342a38d623f4e9bd752eb : SUCCESS in 1h 32m 43s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/7e75b990c8fb4262960396a363206662 : SUCCESS in 1h 22m 44s","accounts_in_message":[],"_revision_number":3},{"id":"2d75f844a1b498efbd0425be54ce34c1da7f35c9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 17:51:45.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9aa46b6932b44f1d9b2d1c95fcc16a3d\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2383ad68545b48f88cf13395c2f5865f : SUCCESS in 24m 23s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7735595183884bd881417a95b715aad3 : SUCCESS in 21m 42s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/13ec1a4b2ce1424fa6502b102b3ca1c5 : SUCCESS in 30m 32s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"06cbe836f123a90b8bfbc4833b61302dfb2151b8","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-01-17 18:33:22.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"d8930b72ccbfccd51a8f172dd8db42536850d084","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-01-19 08:20:40.000000000","message":"Patch Set 3: -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6ccf883645769a273097579c553a501893a45ae7","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-01-19 10:52:49.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"d650f40eafdec596667760753717c3ff8c16b01a","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-01-19 18:41:48.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"474b96576835eab371394df7f463c5934a216a8c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-01-20 08:38:40.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0eac8695efcb9e4749587851a9710e4ccaaa62da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-01-20 09:04:30.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"40fca95e70af0007df3d389a25d17d986a412e83","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 09:59:56.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99a9c9ae370a43a3a70ff3246350fa48\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/eb7dbdc13a234532bb4bd2b2717e8aca : SUCCESS in 52m 54s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/69f6193485ec4b2895c8f54b4317e46a : SUCCESS in 49m 30s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4d39fb54e4814809953c6123328edbb9 : RETRY_LIMIT in 3m 57s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"231c4f96a28062053a7f55c62e6bec8628ad34fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 11:05:16.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b48509e76c96426b8beaa347a8519b0c\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/94c220f333c748e4a83ffb7d70008e55 : SUCCESS in 1h 59m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/836697ca78154c699cbe4e71414724cd : SUCCESS in 36m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eda7873e71654c539d59dc704ed0f5b2 : SUCCESS in 10m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5f48616397ad4387b0d17c88d336ed16 : SUCCESS in 25m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ddf43cc336b542b7aaccadbe581ca165 : SUCCESS in 21m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fdd3605ad4d747d2b54329c7d8b5eafb : SUCCESS in 9m 16s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/e1c67b5c594745158119fde399991f75 : SUCCESS in 1h 18m 40s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/89ca7b00a7f04e98b02dd92337b5399e : SUCCESS in 1h 47m 18s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/7e48d9347ce340ee936b5a88d7ba56f9 : SUCCESS in 52m 57s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/90d7758b63c14bb3a5c07911a72d5354 : SUCCESS in 54m 25s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9e6f52c354fa44868bb018e429494c89 : SUCCESS in 1h 22m 56s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/caeab338f53e475d97822a7e64d5ab5e : SUCCESS in 1h 14m 13s","accounts_in_message":[],"_revision_number":4},{"id":"7fab82794aea4b59ce0e04d3651b62b42f0bb325","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-01-20 15:53:44.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"b2e4de179b4e24c8054367b978882d19891238d8","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-01-21 08:23:19.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"e9cafe0a46720d2190d0415fc91295fff0654444","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 08:23:42.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"3cf1cb7fbe70742d773e675c7ddece7fda122737","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 10:25:58.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c8095b9c97548858d2ca17b3cd454ea\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/4db572eccc9e4004b28e922f2d2a949e : SUCCESS in 1h 53m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02941cabbf5c4df090bf5249554a738b : SUCCESS in 25m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/21761407ae7b4f40a4724cdc9b9c2ff1 : SUCCESS in 24m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f5c05a9bdcca4d03b199e29ad98d36c0 : SUCCESS in 25m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1699493a4c2c407585dbc3ed23453c8d : SUCCESS in 5m 58s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a773a44f1ec942e4b0719df0ad83fc6d : SUCCESS in 1h 44m 25s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/ab0be0df40594b15a798d5d8cf186cfe : SUCCESS in 1h 10m 01s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/2e5824ee6a70426e96ca5427ad73c1cb : SUCCESS in 1h 06m 26s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/47a3ab9ba634410aad32292c8f5dd075 : SUCCESS in 1h 06m 32s","accounts_in_message":[],"_revision_number":4},{"id":"9a5b0775f725bf1082cf7a4da08b12a081eebed8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 10:25:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"d438c5d8e3ee9363c4ecd6e95a2c137bd7610b3c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 10:28:14.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c92660396484ba282b74dcca7af921d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbcb09fe51f94b0ea0d2e7e75dfee996 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"7b7d0ad6fa2cf883d93483e963887a6d5fddf5f5","revisions":{"aabd9b85b5023e2877f6619329fa14515a72bb2f":{"kind":"REWORK","_number":1,"created":"2026-01-16 07:52:39.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/84/973584/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/84/973584/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/973584/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/973584/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/973584/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/84/973584/1"}}},"commit":{"parents":[{"commit":"9656a5fcf2c706d2f159655f9fa40c9768bbf5c1","subject":"Merge \"bgp: Introduce BGP chassis commands\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9656a5fcf2c706d2f159655f9fa40c9768bbf5c1"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-01-16 06:56:23.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-01-16 07:49:41.000000000","tz":60},"subject":"Fix ``LogicalRouterPortEvent`` when a LR is deleted","message":"Fix ``LogicalRouterPortEvent`` when a LR is deleted\n\nIf a ``Logical_Router`` is deleted, the only ``Logical_Router_Ports``\nthat could be present are the gateway ports. The\n``Logical_Router_Ports`` deletion will trigger a\n``LogicalRouterPortEvent`` event. In that case, no action is needed;\nit is needed to check if the ``Logical_Router`` still exists and\nexit if not.\n\nThis bug was introduced in [1].\n\n[1]https://review.opendev.org/c/openstack/neutron/+/962155/\n\nCloses-Bug: #2138468\n\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: Idc54d1b307fa88a4113228a63049f5fc7ddaa7cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/aabd9b85b5023e2877f6619329fa14515a72bb2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/aabd9b85b5023e2877f6619329fa14515a72bb2f"}]},"branch":"refs/heads/master"},"7381ee14aba3c1a4783cfd7b4afabe4ec1a7409b":{"kind":"REWORK","_number":2,"created":"2026-01-16 10:13:10.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/84/973584/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/84/973584/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/973584/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/973584/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/973584/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/84/973584/2"}}},"commit":{"parents":[{"commit":"9656a5fcf2c706d2f159655f9fa40c9768bbf5c1","subject":"Merge \"bgp: Introduce BGP chassis commands\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9656a5fcf2c706d2f159655f9fa40c9768bbf5c1"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-01-16 06:56:23.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-01-16 09:25:10.000000000","tz":60},"subject":"Fix ``LogicalRouterPortEvent`` when a LR is deleted","message":"Fix ``LogicalRouterPortEvent`` when a LR is deleted\n\nIf a ``Logical_Router`` is deleted, the only ``Logical_Router_Ports``\nthat could be present are the gateway ports. The\n``Logical_Router_Ports`` deletion will trigger a\n``LogicalRouterPortEvent`` event. In that case, no action is needed;\nit is needed to check if the ``Logical_Router`` still exists and\nexit if not.\n\nThe same is happening when ``update_router_ha_chassis_group`` is called\nfrom the ``run`` method. If the ``Logical_Router`` has been deleted,\nthe method needs to fast exit.\n\nThis bug was introduced in [1].\n\n[1]https://review.opendev.org/c/openstack/neutron/+/962155/\n\nCloses-Bug: #2138468\n\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: Idc54d1b307fa88a4113228a63049f5fc7ddaa7cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7381ee14aba3c1a4783cfd7b4afabe4ec1a7409b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7381ee14aba3c1a4783cfd7b4afabe4ec1a7409b"}]},"branch":"refs/heads/master"},"552651cff5a38d20c5687b79a48b5f1079879301":{"kind":"REWORK","_number":3,"created":"2026-01-16 15:38:28.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/84/973584/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/84/973584/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/973584/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/973584/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/973584/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/84/973584/3"}}},"commit":{"parents":[{"commit":"4bbd427d9f58ac6bc35a2dcdc07584cc5ab415c6","subject":"Merge \"Make sure OVN PG is present before adding a SG rule\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4bbd427d9f58ac6bc35a2dcdc07584cc5ab415c6"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-01-16 06:56:23.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-01-16 15:37:56.000000000","tz":60},"subject":"Fix ``LogicalRouterPortEvent`` when a LR is deleted","message":"Fix ``LogicalRouterPortEvent`` when a LR is deleted\n\nIf a ``Logical_Router`` is deleted, the only ``Logical_Router_Ports``\nthat could be present are the gateway ports. The\n``Logical_Router_Ports`` deletion will trigger a\n``LogicalRouterPortEvent`` event. In that case, no action is needed;\nit is needed to check if the ``Logical_Router`` still exists and\nexit if not.\n\nThe same is happening when ``update_router_ha_chassis_group`` and\n``link_network_ha_chassis_group`` are called from the ``run`` method.\nIf the ``Logical_Router`` has been deleted, the methods need to fast\nexit.\n\nThis bug was introduced in [1].\n\n[1]https://review.opendev.org/c/openstack/neutron/+/962155/\n\nCloses-Bug: #2138468\n\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: Idc54d1b307fa88a4113228a63049f5fc7ddaa7cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/552651cff5a38d20c5687b79a48b5f1079879301"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/552651cff5a38d20c5687b79a48b5f1079879301"}]},"branch":"refs/heads/master"},"7b7d0ad6fa2cf883d93483e963887a6d5fddf5f5":{"kind":"REWORK","_number":4,"created":"2026-01-20 09:04:30.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/84/973584/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/84/973584/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/973584/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/973584/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/973584/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/84/973584/4"}}},"commit":{"parents":[{"commit":"4bbd427d9f58ac6bc35a2dcdc07584cc5ab415c6","subject":"Merge \"Make sure OVN PG is present before adding a SG rule\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4bbd427d9f58ac6bc35a2dcdc07584cc5ab415c6"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-01-16 06:56:23.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-01-20 09:04:17.000000000","tz":60},"subject":"Fix ``LogicalRouterPortEvent`` when a LR is deleted","message":"Fix ``LogicalRouterPortEvent`` when a LR is deleted\n\nIf a ``Logical_Router`` is deleted, the only ``Logical_Router_Ports``\nthat could be present are the gateway ports. The\n``Logical_Router_Ports`` deletion will trigger a\n``LogicalRouterPortEvent`` event. In that case, no action is needed;\nit is needed to check if the ``Logical_Router`` still exists and\nexit if not.\n\nThe same is happening when ``update_router_ha_chassis_group`` and\n``link_network_ha_chassis_group`` are called from the ``run`` method.\nIf the ``Logical_Router`` has been deleted, the methods need to fast\nexit.\n\nNOTE: the test ``TestLogicalRouterPortEvent.test_delete_router`` was\nactually testing the router deletion and the ``LogicalRouterPortEvent``\nevent call. This call was generating an exception when trying to\nretrieve the ``Logical_Router`` that was actually deleted. Because\nthe processing of the event is done in a separate thread and nothing\nwas expected to be called, the test was passing. Now this patch is\nremoving the exception by carefully checking if the router is present\nor not.\n\nThis bug was introduced in [1].\n\n[1]https://review.opendev.org/c/openstack/neutron/+/962155/\n\nCloses-Bug: #2138468\n\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: Idc54d1b307fa88a4113228a63049f5fc7ddaa7cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7b7d0ad6fa2cf883d93483e963887a6d5fddf5f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7b7d0ad6fa2cf883d93483e963887a6d5fddf5f5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
