)]}'
{"id":"openstack%2Fneutron~973632","triplet_id":"openstack%2Fneutron~stable%2F2024.2~If6a60eb086b9bcdfd90e154879561ba569f45750","project":"openstack/neutron","branch":"stable/2024.2","topic":"bug/2090921-stable/2024.2","attention_set":{},"removed_from_attention_set":{"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2026-01-20 12:05:21.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If6a60eb086b9bcdfd90e154879561ba569f45750","subject":"Make sure OVN PG is present before adding a SG rule","status":"MERGED","created":"2026-01-16 14:29:54.000000000","updated":"2026-01-20 12:06:28.000000000","submitted":"2026-01-20 12:05:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"973632-bug/2090921-stable/2024.2","cherry_pick_of_change":946216,"cherry_pick_of_patch_set":3,"meta_rev_id":"68870902560e9bb129be5d3d963fb3fec6ce74a9","_number":973632,"virtual_id_number":973632,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-20 12:05:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-01-20 09:56:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-01-19 11:31:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2026-01-19 12:43:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2026-01-20 09:56:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-16 16:31:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-19 11:31:21.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2026-01-19 12:43:15.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2026-01-20 09:56:52.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"405be81c7f7118ef0e8611769f90c6ea704562a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-01-16 14:29:54.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"3644d774eb245158d1732f1bf96d3ed2cde238a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 16:31:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b5384ad6d1ec403fb84f6789edfd49e7\n\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/e24e0045a7b6408daf8b97bbd8f3b9dd : SUCCESS in 1h 27m 21s\n- neutron-tempest-plugin-designate-scenario-2024-2 https://zuul.opendev.org/t/openstack/build/b0539160d8454185a159d8a5b25c8720 : SUCCESS in 29m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c41ec989518a4c9185f87a52919ca59b : SUCCESS in 41m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69d1a2bfb49345758460bf7e3a688d2a : SUCCESS in 9m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/72e412acf9614aef92898c17ded22449 : SUCCESS in 19m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cd3ee2823d3e4fe9bbed7ce99b2b527b : SUCCESS in 20m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b3ec0a198d9249cd9c1a0947d986b026 : SUCCESS in 15m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f51586ecc14434e84aa06e319a64473 : SUCCESS in 10m 52s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/baffb4257da44ea190a5d15c2eead740 : SUCCESS in 1h 51m 04s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ec4d221b51e54b84a4127b73a227cd8a : SUCCESS in 1h 56m 06s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/b859f2832a5c45029d6f481ddb8a7748 : SUCCESS in 58m 43s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/33de7319f9a44eb9a337ca11692db43c : SUCCESS in 1h 01m 32s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7cbc56324cd0466ca75323510280049c : SUCCESS in 29m 49s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3740a9b3791d45f9b8475c55a85937f9 : SUCCESS in 1h 18m 05s","accounts_in_message":[],"_revision_number":1},{"id":"b046c0560930a4559adad6b60d06364e981a2d34","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2026-01-19 11:31:21.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ffd0484fe2e15f1454e31f2fd1996d87e92a64d9","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-01-19 12:43:15.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d0b4b0441ccb549825861759d55a53c293393cbb","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-01-20 09:56:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"33f3c39633cb977b9d2a7c380bba235ebbefae8f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 09:57:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"99554987307596488e66755ac5a4f4dc6452cf2f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 12:05:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32fccddc6fb54a1eb64afd8a1397f818\n\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/e7a904c0787c4a1bac2b28f14686bfa9 : SUCCESS in 1h 50m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7eb02e18b5c34a72aaf97d821496b430 : SUCCESS in 7m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/253fc9fb969f455ca4bdc4c66a93d0cd : SUCCESS in 23m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c98f15369d3b48b98bf2e93f237a3371 : SUCCESS in 18m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5048a7ce1c4242b6931232d7ec01cc71 : SUCCESS in 12m 11s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4fae47be567f4a1b986f6fd957b609d5 : SUCCESS in 1h 31m 25s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/71aa9616f5ad40509b6d5b217c5eafb5 : SUCCESS in 1h 52m 54s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/1efa40848518496287e8f815d58934b4 : SUCCESS in 54m 46s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/11f7c6b157194e579bb36de6f33b6e41 : SUCCESS in 49m 41s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/22b8f3a070d540739ca487450fdbb877 : SUCCESS in 1h 15m 44s","accounts_in_message":[],"_revision_number":1},{"id":"5d3df9c53b4d6d1b4f9bfd479d37de7fb372f748","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 12:05:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"68870902560e9bb129be5d3d963fb3fec6ce74a9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 12:06:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6400dff249084abf974639e825b0ab82\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f4a185769d54fb2bcb63cdf6f132bd0 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5bc0e42acb6bea92c45fcec04d1f12c837008df5","revisions":{"5bc0e42acb6bea92c45fcec04d1f12c837008df5":{"kind":"REWORK","_number":1,"created":"2026-01-16 14:29:54.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/32/973632/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/32/973632/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/32/973632/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/32/973632/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/32/973632/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/32/973632/1"}}},"commit":{"parents":[{"commit":"796278e55567d7226188735af99e49243daf9260","subject":"Merge \"[Docs] Add doc about Virtual IP addresses in Neutron ML2/OVN\" into stable/2024.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/796278e55567d7226188735af99e49243daf9260"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2025-04-02 21:56:27.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-01-16 14:29:54.000000000","tz":0},"subject":"Make sure OVN PG is present before adding a SG rule","message":"Make sure OVN PG is present before adding a SG rule\n\nIt\u0027s possible to have a security group created on one\ncontroller and then a security group rule created on a\ndifferent controller quickly enough that the second\ncontroller does not yet see that security group in its\nlocal cache of the OVN northbound database. Check if\nthe port group is present or not in the idl\u0027s local copy\nof the database before creating the security group rule.\n\nCloses-bug: #2090921\nChange-Id: If6a60eb086b9bcdfd90e154879561ba569f45750\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n(cherry picked from commit be4fb73d77135d76f3b25b85c80bc63e3fa54a30)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5bc0e42acb6bea92c45fcec04d1f12c837008df5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5bc0e42acb6bea92c45fcec04d1f12c837008df5"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
