)]}'
{"id":"openstack%2Fneutron~973633","triplet_id":"openstack%2Fneutron~unmaintained%2F2024.1~If6a60eb086b9bcdfd90e154879561ba569f45750","project":"openstack/neutron","branch":"unmaintained/2024.1","topic":"bug/2090921-unmaintained/2024.1","attention_set":{},"removed_from_attention_set":{"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2026-01-22 07:56:53.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If6a60eb086b9bcdfd90e154879561ba569f45750","subject":"Make sure OVN PG is present before adding a SG rule","status":"MERGED","created":"2026-01-16 14:30:24.000000000","updated":"2026-01-22 07:57:54.000000000","submitted":"2026-01-22 07:56:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"973633-bug/2090921-unmaintained/2024.1","cherry_pick_of_change":946216,"cherry_pick_of_patch_set":3,"meta_rev_id":"2ff1f89a7010ad609005f58261be367c94e232a6","_number":973633,"virtual_id_number":973633,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-22 07:56:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2026-01-20 09:56:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2026-01-22 06:11:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":2,"date":"2026-01-19 12:43:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-01-22 06:13:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-16 16:17:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-19 12:43:26.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2026-01-20 09:56:30.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2026-01-22 06:11:47.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"}],"messages":[{"id":"be96f87a07c083fe1e4002263a70300c25f1b3a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-01-16 14:30:24.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"f76a603f06cf5a5931c05eb2bb025e8dff51bd02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 16:17:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/609074c4e5fd41f6a05c1c60802b49b5\n\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/38ad0ffc6667467796aba30b9e1a2302 : SUCCESS in 1h 26m 06s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/d9b0cb9c46e24fc5bbb45da9998f624f : SUCCESS in 27m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/640c39afcd5a4130949399f131b94072 : SUCCESS in 37m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92771540b41743a380487fdbeecf1f94 : SUCCESS in 10m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c58921dd5ad245e3b3783acd3db79ce8 : SUCCESS in 25m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/53f2474ff801402da81907b91a763ffc : SUCCESS in 24m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b876c7f88d4c4765a9783bc04ab1fb4d : SUCCESS in 11m 40s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/cc99f585d2044cbebfdc5bb99b114dd9 : SUCCESS in 1h 38m 23s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/90837a0df1034db0ba2e7b58708f3978 : SUCCESS in 1h 39m 00s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/02e07817319549f3b25c097261d60203 : SUCCESS in 54m 44s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d12adaa20cc747e0a158435da4d6cc5e : SUCCESS in 1h 20m 49s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/56d60728d97e46b59cd6e4b408decde9 : SUCCESS in 1h 25m 06s","accounts_in_message":[],"_revision_number":1},{"id":"d080f4a9db34174b9d1d6c58aa93b7fa31749d89","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-01-19 12:43:26.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0cb54476ad4a02ed2ced5d425d0eae26ff963e3e","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-01-20 09:56:30.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"5fd6e1de1438479e92da977811908b9e6aed143e","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2026-01-22 06:11:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ed0571dbddb997dd1b4453235970ac9ee70b8a3b","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2026-01-22 06:13:00.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4abdc747d71dc73cafd66c41dbd0164daa1e9474","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 06:13:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"99d88b35feaf2e23415e31285f146e2e5a612459","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 07:56:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e43f93838e5140779909586e729154e2\n\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/250cf6bf832b410cb91d429dcbb3bb11 : SUCCESS in 1h 33m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/522dccb855b549a282103ff1f76333f6 : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/acd41223ade44ba0a620a86a039ffd87 : SUCCESS in 27m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4064b9e373b14beba60129e58de64d8f : SUCCESS in 23m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35f953c0c3514cd4befa65db61a4dd74 : SUCCESS in 10m 28s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b529bb06b6c041e8a4bb7b684849ccea : SUCCESS in 1h 28m 00s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c6684bca9f404af6b0c21374f318f8c5 : SUCCESS in 1h 38m 53s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/dbf3c8c45dc04a08a91ff686c755c2a9 : SUCCESS in 1h 01m 35s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/b9a8d28f4a0647deafd833903109ad1f : SUCCESS in 58m 02s","accounts_in_message":[],"_revision_number":1},{"id":"5a0a2799e1dbb8ea32c66d4754858f61a1b03502","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 07:56:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2ff1f89a7010ad609005f58261be367c94e232a6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 07:57:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7deae2fbf5bb4316bd03ec1ead9bda36\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a31fc62b4e8044ba88c11ebfd0314aa1 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"37553716eac01ad1c9d6c02005753867ecaae971","revisions":{"37553716eac01ad1c9d6c02005753867ecaae971":{"kind":"REWORK","_number":1,"created":"2026-01-16 14:30:24.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/33/973633/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/33/973633/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/973633/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/973633/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/973633/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/33/973633/1"}}},"commit":{"parents":[{"commit":"7576e92c92e303cafaeb790bd2da87b1f131c993","subject":"Search for active port binding when updating LSP host info","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7576e92c92e303cafaeb790bd2da87b1f131c993"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2025-04-02 21:56:27.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-01-16 14:30:24.000000000","tz":0},"subject":"Make sure OVN PG is present before adding a SG rule","message":"Make sure OVN PG is present before adding a SG rule\n\nIt\u0027s possible to have a security group created on one\ncontroller and then a security group rule created on a\ndifferent controller quickly enough that the second\ncontroller does not yet see that security group in its\nlocal cache of the OVN northbound database. Check if\nthe port group is present or not in the idl\u0027s local copy\nof the database before creating the security group rule.\n\nCloses-bug: #2090921\nChange-Id: If6a60eb086b9bcdfd90e154879561ba569f45750\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n(cherry picked from commit be4fb73d77135d76f3b25b85c80bc63e3fa54a30)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/37553716eac01ad1c9d6c02005753867ecaae971"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/37553716eac01ad1c9d6c02005753867ecaae971"}]},"branch":"refs/heads/unmaintained/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
