)]}'
{"id":"openstack%2Fneutron~974835","triplet_id":"openstack%2Fneutron~stable%2F2025.1~I12f018d29e7d00d1f0fb0272dc905d1026633cbb","project":"openstack/neutron","branch":"stable/2025.1","topic":"bug/2138430-stable/2025.2-stable/2025.1","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2026-01-28 16:00:20.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I12f018d29e7d00d1f0fb0272dc905d1026633cbb","subject":"Don\u0027t delete from OVN DB ACLs not managed by Neutron","status":"MERGED","created":"2026-01-27 11:46:23.000000000","updated":"2026-01-28 16:02:05.000000000","submitted":"2026-01-28 16:00:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"974835-bug/2138430-stable/2025.2-stable/2025.1","cherry_pick_of_change":974834,"cherry_pick_of_patch_set":2,"meta_rev_id":"d06aa5a24568ff8426d9da023d04f4888ae4fd7d","_number":974835,"virtual_id_number":974835,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-28 16:00:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-01-28 13:35:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2026-01-28 11:31:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2026-01-28 13:35:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-27 13:48:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-28 11:31:54.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2026-01-28 13:35:43.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"3d2d4fd2355f67ba0fba4fe74a9c822be1fca190","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-01-27 11:46:23.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2025.2.","accounts_in_message":[],"_revision_number":1},{"id":"ad90a888625aa6366d724ad7257af37eb603d110","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 13:48:20.000000000","message":"Patch Set 1: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/74506134faef4bf08d3a5fc0280e1b37\n\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/c0772f36738042f89248d6d282b5214e : SUCCESS in 2h 00m 47s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/0d185276de214713ac40cc546e7ac952 : SUCCESS in 1h 19m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7961c36ebfaf47ff9b28de1e662b2211 : FAILURE in 44m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0680559c5934019a00379039d331b67 : FAILURE in 5m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ee9944516ab7438880c237b60ceb1419 : FAILURE in 22m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c8f1f0a4101f431cb02891d32d4f42d1 : FAILURE in 22m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebfcad9436f94807a2551fc669a8b1ec : SUCCESS in 12m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1e3c5bee0abf4df09a6b6d76d5517d30 : SUCCESS in 26m 17s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/a5260be5058d41a38f2f8bbcebf80f2d : SUCCESS in 1h 10m 22s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/11cc4827d64b4c7b9eca7d76a88d7853 : SUCCESS in 1h 02m 07s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/05b4c8edeaa543d29279ba6a5b246dfd : SUCCESS in 58m 10s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/0a80ac57dbd54d2695864e1f7324bdfa : SUCCESS in 1h 17m 33s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f469bf6080d64d4eb4f42f7c03017990 : SUCCESS in 1h 13m 48s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/978712e5d54c4dbc929de84c362a9ad6 : SUCCESS in 1h 16m 01s","accounts_in_message":[],"_revision_number":1},{"id":"1e9358e95f6d2d276d2a413170164a53b04ac008","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-01-27 14:57:42.000000000","message":"Patch Set 2: Cherry Picked from branch stable/2025.2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"8d755911ef7b969eb708ac5ca6669bf1f7497dff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 17:07:07.000000000","message":"Patch Set 2: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b764fd86e45c425485873b4154e350ac\n\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/e99dea151bc54004acd12cb6f744493f : SUCCESS in 2h 08m 03s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/bac9406b7e6b47c59bd33ccda97af244 : SUCCESS in 1h 16m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cb43e6a8e91a4798ac6d0ebee8c92d0b : SUCCESS in 42m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7ce1fb9a8af4012a3318da3b0ffc9e5 : FAILURE in 3m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e58daa9b821449e2aed5deef9b1fdb0d : SUCCESS in 23m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3dc3d622fb734a35a9e46e5dd21c8919 : SUCCESS in 19m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4debfb0523e94f368e13cfd03d9aa9ea : SUCCESS in 14m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d6f00436ed1c427389e3c5067751c8ef : SUCCESS in 28m 14s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/26e39f4283b24f25a8f7992b0b6ef9db : SUCCESS in 38m 10s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/6989662b953b4a978086850d3ecc0589 : SUCCESS in 1h 08m 10s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/bcc4c6766db84c1aa303579d6b7ff83a : SUCCESS in 41m 49s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/90fd373823f44866bd5abf6e70a8707c : SUCCESS in 1h 07m 24s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/40a9990a3c834e10aca03b1be31adfbd : SUCCESS in 1h 16m 04s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c68ad58c89d84d20ae629995e5e42816 : SUCCESS in 1h 18m 06s\n\nWarning:\n  Comments left for invalid file neutron/tests/functional/plugins/ml2/drivers/ovn/mech_driver/ovsdb/test_ovn_db_sync.py\n  Comments left for invalid file neutron/tests/unit/plugins/ml2/drivers/ovn/mech_driver/ovsdb/test_impl_idl_ovn.py\n  Comments left for invalid file neutron/tests/unit/plugins/ml2/drivers/ovn/mech_driver/ovsdb/test_maintenance.py","accounts_in_message":[],"_revision_number":2},{"id":"2baae3700c4eb0631d524e08efc10f8e3e28706c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-01-28 10:03:50.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"78c7ed20177828f2137b3e2ad6541ad2c0e52464","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-01-28 11:31:54.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"7bf648dd448f44c87f2a95d4d2d0a8c31b9af5a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 12:05:46.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/898217c774da4819aeee4b837e05addd\n\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/dcbd1fcf9a8643439f25cb986dca3c10 : SUCCESS in 1h 47m 57s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/d3bfdc2e975b4f9ea099d6a7e4b5515b : SUCCESS in 1h 15m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/971ddcee83804942bba541ec2220cd3f : SUCCESS in 45m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5670e7017a124e168e88552730e66702 : SUCCESS in 12m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c270c8ec40ae4546901954b4a422f380 : SUCCESS in 19m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/743306d357e843f69bc426efffaa02cc : SUCCESS in 20m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6022bf7d96d94338859f2b0a341c7484 : SUCCESS in 12m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/97ee2e3be83343e88845fe32c65632c8 : SUCCESS in 22m 58s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/b8c02f3c0ce141d69518b3c90db13c30 : SUCCESS in 1h 06m 36s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/bc93ab5b79e044cb8bdd7def628a2011 : SUCCESS in 1h 24m 03s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/15653cd510c947499e7bd9ba329e25d5 : SUCCESS in 1h 08m 09s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/76222385d66a4eb688b9b9b5870313e3 : SUCCESS in 1h 02m 13s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d77da4fe58074784b6a494c1c79e671b : SUCCESS in 1h 10m 11s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e0af4690a1354436939c21d0e5a62713 : SUCCESS in 1h 18m 21s","accounts_in_message":[],"_revision_number":3},{"id":"844718b566b3a77d252a513222509d2669c8cc24","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-01-28 13:35:43.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"42597fafc8bbf6baa1dadba040c408d061f7acb7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 13:36:07.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"8df5ed635f299c6b5348ab46c1d9cd6888d80de5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 16:00:20.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aad5ce5a4f4e40d68c6b34b883b5564a\n\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/cda61e42186c4a3c97c35c76fe4ff4d9 : SUCCESS in 2h 23m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b59e1510f33e41be83b4768e0226ce47 : SUCCESS in 11m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4fd47ece1cd14c76a15b7bcf34a567bc : SUCCESS in 20m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c7eb86852f134230bd002aa918b27980 : SUCCESS in 24m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d89731cba6942398b8172fe6ee9ad6e : SUCCESS in 6m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/59dac3e9e26c46d39260901190ecdefc : SUCCESS in 25m 52s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/807c23225dde43c8a47d1222ffa4a87a : SUCCESS in 1h 15m 45s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/63f8cf6adae54709ad9c840f4ae05b91 : SUCCESS in 43m 05s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/3daad50d1b2c4900b6b8a7bf927539a4 : SUCCESS in 58m 02s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c72f9c59448043edb0a5e98775299f52 : SUCCESS in 1h 03m 33s","accounts_in_message":[],"_revision_number":3},{"id":"aab7ab4fa9cf4d26984e7431c9a24aa076b277d3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 16:00:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"d06aa5a24568ff8426d9da023d04f4888ae4fd7d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 16:02:05.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ecbc9a01e854f14bcf0b1c6bb4549fc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8282a32202d34d08b67d251d13171889 : SUCCESS in 1m 06s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cb33b1ed6b8c4d139a6ec8fed6bc62d1 : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ed988ed2b6acbc98e4eca23893a05c389af97f61","revisions":{"8d5865f07c2d89c1f4883615979beaa2bda8623c":{"kind":"REWORK","_number":1,"created":"2026-01-27 11:46:23.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/35/974835/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/35/974835/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/35/974835/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/35/974835/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/35/974835/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/35/974835/1"}}},"commit":{"parents":[{"commit":"37efcf3bf8fe4cca1dc3d5b460e4dae47875e127","subject":"Make sure OVN PG is present before adding a SG rule","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/37efcf3bf8fe4cca1dc3d5b460e4dae47875e127"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-01-16 16:24:20.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-01-27 11:46:23.000000000","tz":0},"subject":"Don\u0027t delete from OVN DB ACLs not managed by Neutron","message":"Don\u0027t delete from OVN DB ACLs not managed by Neutron\n\nNeutron ML2/OVN driver is using ACLs to implement Security Group rules\nin the OVN backend but neutron-ovn-db-sync util was trying to clean OVN\nDB slightly too agressively and when it was run in the \"repair\" mode it\nwas removing from OVN DB all ACLs which did not had security group rule\nid in the external_ids.\nThat could break e.g. ACLs created by the neutron-fwaas as its OVN\ndriver is also using ACLs to implement rules but is adding different key\nin the external_ids.\n\nThis patch changes logic of the neutron-ovn-db-sync-util tool so that it\nwill skip all ACL rules which don\u0027t have in external_ids key related to\nthe SG rule.\nThis is similar change to what was already done e.g. for Port Groups in\n[1].\n\n[1] https://review.opendev.org/c/openstack/neutron/+/958679\n\nCloses-bug: #2138430\n\nConflicts:\n    neutron/tests/unit/plugins/ml2/drivers/ovn/mech_driver/ovsdb/test_ovn_db_sync.py\n\nChange-Id: I12f018d29e7d00d1f0fb0272dc905d1026633cbb\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n(cherry picked from commit 7524142d7599d1ad7764a8e79a48130e8a631199)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8d5865f07c2d89c1f4883615979beaa2bda8623c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8d5865f07c2d89c1f4883615979beaa2bda8623c"}]},"branch":"refs/heads/stable/2025.1"},"b309dd2e79a27a5e2ccf2c0409dc81f8b36f9ce6":{"kind":"REWORK","_number":2,"created":"2026-01-27 14:57:42.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/35/974835/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/35/974835/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/35/974835/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/35/974835/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/35/974835/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/35/974835/2"}}},"commit":{"parents":[{"commit":"49dfd242da5efe8312fd3a28e7534351e5d8ed91","subject":"ovn_db_sync: Improve coexistence support.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/49dfd242da5efe8312fd3a28e7534351e5d8ed91"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-01-16 16:24:20.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-01-27 14:57:42.000000000","tz":0},"subject":"Don\u0027t delete from OVN DB ACLs not managed by Neutron","message":"Don\u0027t delete from OVN DB ACLs not managed by Neutron\n\nNeutron ML2/OVN driver is using ACLs to implement Security Group rules\nin the OVN backend but neutron-ovn-db-sync util was trying to clean OVN\nDB slightly too agressively and when it was run in the \"repair\" mode it\nwas removing from OVN DB all ACLs which did not had security group rule\nid in the external_ids.\nThat could break e.g. ACLs created by the neutron-fwaas as its OVN\ndriver is also using ACLs to implement rules but is adding different key\nin the external_ids.\n\nThis patch changes logic of the neutron-ovn-db-sync-util tool so that it\nwill skip all ACL rules which don\u0027t have in external_ids key related to\nthe SG rule.\nThis is similar change to what was already done e.g. for Port Groups in\n[1].\n\n[1] https://review.opendev.org/c/openstack/neutron/+/958679\n\nCloses-bug: #2138430\n\nConflicts:\n    neutron/tests/unit/plugins/ml2/drivers/ovn/mech_driver/ovsdb/test_ovn_db_sync.py\n\nChange-Id: I12f018d29e7d00d1f0fb0272dc905d1026633cbb\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n(cherry picked from commit 7524142d7599d1ad7764a8e79a48130e8a631199)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b309dd2e79a27a5e2ccf2c0409dc81f8b36f9ce6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b309dd2e79a27a5e2ccf2c0409dc81f8b36f9ce6"}]},"branch":"refs/heads/stable/2025.1"},"ed988ed2b6acbc98e4eca23893a05c389af97f61":{"kind":"REWORK","_number":3,"created":"2026-01-28 10:03:50.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/35/974835/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/35/974835/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/35/974835/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/35/974835/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/35/974835/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/35/974835/3"}}},"commit":{"parents":[{"commit":"7590dff9b17ae8a889fce1dc7a0636a4e58b61de","subject":"ovn_db_sync: Improve coexistence support.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7590dff9b17ae8a889fce1dc7a0636a4e58b61de"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-01-16 16:24:20.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-01-28 09:49:01.000000000","tz":60},"subject":"Don\u0027t delete from OVN DB ACLs not managed by Neutron","message":"Don\u0027t delete from OVN DB ACLs not managed by Neutron\n\nNeutron ML2/OVN driver is using ACLs to implement Security Group rules\nin the OVN backend but neutron-ovn-db-sync util was trying to clean OVN\nDB slightly too agressively and when it was run in the \"repair\" mode it\nwas removing from OVN DB all ACLs which did not had security group rule\nid in the external_ids.\nThat could break e.g. ACLs created by the neutron-fwaas as its OVN\ndriver is also using ACLs to implement rules but is adding different key\nin the external_ids.\n\nThis patch changes logic of the neutron-ovn-db-sync-util tool so that it\nwill skip all ACL rules which don\u0027t have in external_ids key related to\nthe SG rule.\nThis is similar change to what was already done e.g. for Port Groups in\n[1].\n\n[1] https://review.opendev.org/c/openstack/neutron/+/958679\n\nCloses-bug: #2138430\n\nConflicts:\n    neutron/tests/unit/plugins/ml2/drivers/ovn/mech_driver/ovsdb/test_ovn_db_sync.py\n\nChange-Id: I12f018d29e7d00d1f0fb0272dc905d1026633cbb\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n(cherry picked from commit 7524142d7599d1ad7764a8e79a48130e8a631199)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ed988ed2b6acbc98e4eca23893a05c389af97f61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ed988ed2b6acbc98e4eca23893a05c389af97f61"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
