)]}'
{"id":"openstack%2Fneutron~974892","triplet_id":"openstack%2Fneutron~stable%2F2025.1~I0660d88ac12596858678ac6cf64de7ed7743d7e3","project":"openstack/neutron","branch":"stable/2025.1","topic":"bug/2138430-stable/2025.2-stable/2025.1","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2026-01-28 17:00:25.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2026-01-28 15:21:32.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}},"hashtags":[],"change_id":"I0660d88ac12596858678ac6cf64de7ed7743d7e3","subject":"Cleaning debug import of the pdb module","status":"MERGED","created":"2026-01-27 15:53:19.000000000","updated":"2026-01-28 17:01:55.000000000","submitted":"2026-01-28 17:00:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"974892-bug/2138430-stable/2025.2-stable/2025.1","cherry_pick_of_change":974891,"cherry_pick_of_patch_set":1,"meta_rev_id":"4e73b69f6cdea1e41ba7bb9a0b5cfa33055f76d3","_number":974892,"virtual_id_number":974892,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-28 17:00:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-01-28 13:35:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2026-01-28 15:21:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-01-28 15:21:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-27 15:57:04.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2026-01-27 16:43:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-28 13:35:30.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"64735fdd3c62018edb31fe1e768a7562e98c1d24","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-01-27 15:53:19.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2025.2.","accounts_in_message":[],"_revision_number":1},{"id":"18300d6046d7f7f551fd0aec4ef851406f1707e4","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-01-27 15:57:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b6e2c0d9923924dea9b3f4e3c9334379233e0429","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 16:43:04.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5efc4f619212482b95eb4d789347cf80\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ef6b5c0d5f574115a7e894f19d1d7591 : FAILURE in 44m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fa9ccfc473a4ec9a2aacae5615e5d23 : FAILURE in 4m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6a8c1833f1f44ddfa340f4b8e0646eef : SUCCESS in 18m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2d2fd01c1d7d48a5af855004b97ced3c : SUCCESS in 17m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70862c24da6448cbbb230cf360245e8a : SUCCESS in 11m 24s\n\nWarning:\n  Comments left for invalid file neutron/tests/functional/plugins/ml2/drivers/ovn/mech_driver/ovsdb/test_ovn_db_sync.py\n  Comments left for invalid file neutron/tests/unit/plugins/ml2/drivers/ovn/mech_driver/ovsdb/test_impl_idl_ovn.py\n  Comments left for invalid file neutron/tests/unit/plugins/ml2/drivers/ovn/mech_driver/ovsdb/test_maintenance.py","accounts_in_message":[],"_revision_number":1},{"id":"ea742e4bef8e3332306a8ce0019b78ca9f255bb8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-01-28 10:03:50.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"f9c7f13f065e12b84800bee1636ea46dc41f87fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 10:55:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c55534d7dde043cf89ce9068a9f4e530\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/efcfc659b1f44e61a92548772328de1b : SUCCESS in 44m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7ec4d4275c2426dabe1e741606c4ba6 : SUCCESS in 13m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3e85d47eeb51442988420abe750a4240 : SUCCESS in 25m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/96e958c9db014d4d94d7b4987e7dcb5a : SUCCESS in 19m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/702578cb00be417585c51ca317e8891c : SUCCESS in 12m 52s","accounts_in_message":[],"_revision_number":2},{"id":"5ddf93b0480f690a4b2c9aebfb234fd395f53e15","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-01-28 13:35:30.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"81708c3986fbcc8ea3e9310d393ac3760c5672da","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-01-28 15:21:32.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"108a1a09af4d7bd99675dd98e4845940c1465f73","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 15:22:00.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"758915764e9b80a5147a6456a1a17c1e499543eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 17:00:24.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25bd5d120e9e43c1867a515569e3220b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2814b984618246ebb127d1a235cd7f3b : SUCCESS in 12m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/843f5d3c1ceb4f73aecb2fcd691da563 : SUCCESS in 22m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/85ea2ab24b1c4334b517c26802883b15 : SUCCESS in 19m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f785e44054b14fff9e651ef9e329dc38 : SUCCESS in 14m 25s","accounts_in_message":[],"_revision_number":2},{"id":"f7e71e39671e78e2b14af1dd6ff051d1d9e221b4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 17:00:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"4e73b69f6cdea1e41ba7bb9a0b5cfa33055f76d3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 17:01:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37cc2ac979454ad192bcd8a2271b906f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76243a98d7d244a0a2269dc4e8c3c53a : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8d32160a99cb745e94aa0a4c894e68b61614caa2","revisions":{"71143eb16b8e9e37eb05ac93a5a43ec8bbbdf109":{"kind":"REWORK","_number":1,"created":"2026-01-27 15:53:19.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/92/974892/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/92/974892/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/974892/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/974892/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/974892/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/92/974892/1"}}},"commit":{"parents":[{"commit":"b309dd2e79a27a5e2ccf2c0409dc81f8b36f9ce6","subject":"Don\u0027t delete from OVN DB ACLs not managed by Neutron","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b309dd2e79a27a5e2ccf2c0409dc81f8b36f9ce6"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-01-27 15:44:25.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-01-27 15:53:19.000000000","tz":0},"subject":"Cleaning debug import of the pdb module","message":"Cleaning debug import of the pdb module\n\nIn patch [1] there was accidentally merged commented line which imports\nremote_pdb. This was used only for the debugging process and should\nnever be merged.\n\nTrivialFix\n\n[1] https://review.opendev.org/c/openstack/neutron/+/973648\n\nChange-Id: I0660d88ac12596858678ac6cf64de7ed7743d7e3\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/71143eb16b8e9e37eb05ac93a5a43ec8bbbdf109"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/71143eb16b8e9e37eb05ac93a5a43ec8bbbdf109"}]},"branch":"refs/heads/stable/2025.1"},"8d32160a99cb745e94aa0a4c894e68b61614caa2":{"kind":"REWORK","_number":2,"created":"2026-01-28 10:03:50.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/92/974892/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/92/974892/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/974892/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/974892/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/974892/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/92/974892/2"}}},"commit":{"parents":[{"commit":"ed988ed2b6acbc98e4eca23893a05c389af97f61","subject":"Don\u0027t delete from OVN DB ACLs not managed by Neutron","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ed988ed2b6acbc98e4eca23893a05c389af97f61"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-01-27 15:44:25.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-01-28 09:55:59.000000000","tz":60},"subject":"Cleaning debug import of the pdb module","message":"Cleaning debug import of the pdb module\n\nIn patch [1] there was accidentally merged commented line which imports\nremote_pdb. This was used only for the debugging process and should\nnever be merged.\n\nTrivialFix\n\n[1] https://review.opendev.org/c/openstack/neutron/+/973648\n\nConflicts:\n    neutron/tests/unit/plugins/ml2/drivers/ovn/mech_driver/ovsdb/test_ovn_db_sync.py\n\nChange-Id: I0660d88ac12596858678ac6cf64de7ed7743d7e3\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8d32160a99cb745e94aa0a4c894e68b61614caa2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8d32160a99cb745e94aa0a4c894e68b61614caa2"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
