)]}'
{"id":"openstack%2Fneutron~975159","triplet_id":"openstack%2Fneutron~unmaintained%2F2023.1~I72f1939fc81739987c1c26fccd602fe2e02cf45c","project":"openstack/neutron","branch":"unmaintained/2023.1","topic":"fix-zuul-config-error-neutron","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2026-01-29 20:03:25.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2026-01-29 20:03:25.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I72f1939fc81739987c1c26fccd602fe2e02cf45c","subject":"[CI][stable-only] Remove neutron-ovn-tempest-with-uwsgi-loki","status":"MERGED","created":"2026-01-29 14:19:16.000000000","updated":"2026-01-29 20:05:28.000000000","submitted":"2026-01-29 20:03:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"975159-fix-zuul-config-error-neutron","meta_rev_id":"7237ebff93881d12c2d1c44df9c5ad2bdd348ffa","_number":975159,"virtual_id_number":975159,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-29 20:03:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":2,"date":"2026-01-29 18:28:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":1,"date":"2026-01-29 18:28:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-29 15:47:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-29 16:04:06.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2026-01-29 18:28:24.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"1892f4bc442d4d1b7251350f961e5d02c3e55411","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-01-29 14:19:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f1675d7f69ccdb6b667935134bdf64c718bc42b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 15:47:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91367284ab564bd08c5c14bcc9e01b46\n\n- neutron-tempest-plugin-linuxbridge-2023-1 https://zuul.opendev.org/t/openstack/build/1079e62b685442e18481707de073e8cd : SUCCESS in 1h 23m 48s\n- neutron-tempest-plugin-openvswitch-2023-1 https://zuul.opendev.org/t/openstack/build/c75a53bac7fd44bb9e4a37d9b89ce8c1 : SUCCESS in 1h 05m 39s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-1 https://zuul.opendev.org/t/openstack/build/6e1b285723fa47bea9480ec367b17f27 : SUCCESS in 1h 22m 10s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/21a8ed0cc10e4db4a0a6fdedc9b0192f : SUCCESS in 1h 26m 14s\n- neutron-tempest-plugin-designate-scenario-2023-1 https://zuul.opendev.org/t/openstack/build/bc2945cd259e45c38147e9d6e946b6d8 : SUCCESS in 28m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3353a7f291264492af740a550dc9f162 : SUCCESS in 23m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b4f9d41dbb64367ab55f421970358ff : SUCCESS in 10m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9a4a454dd0c64129b1ebf3d82e88f8a8 : SUCCESS in 15m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eb950ebc90ba4fbb995117ae15e11535 : SUCCESS in 19m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55688901395d4e67b6cdd681918b4702 : SUCCESS in 8m 30s","accounts_in_message":[],"_revision_number":1},{"id":"1ff45161aa43f972d16820af86addc0704c4552e","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-01-29 18:28:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4a694d470d8ba34dacf04a65a9d0d7eb3a3fa6c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 18:29:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"02ea0d15f25677b570e6739d2887b7515b10a884","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 20:03:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/21dc9990f2184459ad93b1439a06c227\n\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/d331d38c15cf44d5bf3c774b6e4d7aa0 : SUCCESS in 1h 33m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e963a7305d3e48209d16d314775bb0ad : SUCCESS in 13m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/40df84f11367442988f1689703bed8e8 : SUCCESS in 22m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6a4722958b264f928953e2a1bf3e1c64 : SUCCESS in 22m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f9a1550c3d44a8da0bc52bd6dd22c83 : SUCCESS in 11m 23s","accounts_in_message":[],"_revision_number":1},{"id":"5f6787ef5a4430250e3c487cf25f27d60f943d93","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 20:03:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7237ebff93881d12c2d1c44df9c5ad2bdd348ffa","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 20:05:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a1d912eb01a40b3b6ecf24a3f1bd277\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f988ed91eb7344f88f91784284bbfe87 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"95cadbd379667c8520c84f124913b868eaa9ae5c","revisions":{"95cadbd379667c8520c84f124913b868eaa9ae5c":{"kind":"REWORK","_number":1,"created":"2026-01-29 14:19:16.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/59/975159/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/59/975159/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/59/975159/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/59/975159/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/59/975159/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/59/975159/1"}}},"commit":{"parents":[{"commit":"b8e53c24fd69db25ce76e7927f0b5b0e1b6737b3","subject":"Merge \"Limit trunk ACTIVE state hack to OVN\" into unmaintained/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b8e53c24fd69db25ce76e7927f0b5b0e1b6737b3"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-01-29 14:12:49.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-01-29 14:18:47.000000000","tz":60},"subject":"[CI][stable-only] Remove neutron-ovn-tempest-with-uwsgi-loki","message":"[CI][stable-only] Remove neutron-ovn-tempest-with-uwsgi-loki\n\nneutron-ovn-tempest-with-uwsgi-loki job\u0027s parent is\nneutron-tempest-with-uwsgi, which had definition only on older\nbranches, that were deleted recently (victoria, wallaby, and\nxena moved to End of Life), hence this is now causing zuul config\nerrors [1]. The job\u0027s parent is fixed in newer branches [2], but\nsince this is an experimental job, probably it\u0027s enough to simply\nremove the job.\n\n[1] https://zuul.opendev.org/t/openstack/config-errors?project\u003dopenstack%2Fneutron\u0026severity\u003derror\u0026skip\u003d0\n[2] If865a863971bb655ddc59a5237c4d1e57e46e407\n\nChange-Id: I72f1939fc81739987c1c26fccd602fe2e02cf45c\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/95cadbd379667c8520c84f124913b868eaa9ae5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/95cadbd379667c8520c84f124913b868eaa9ae5c"}]},"branch":"refs/heads/unmaintained/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
