)]}'
{"id":"openstack%2Fneutron~975251","triplet_id":"openstack%2Fneutron~master~I0c40fcf06314c605b3dd2a63d195187f7368b16c","project":"openstack/neutron","branch":"master","topic":"bug/port_binding_additional_chassis","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-02-06 12:34:22.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0c40fcf06314c605b3dd2a63d195187f7368b16c","subject":"OVN ``Port_Binding.additional_chassis`` is supported since v22.09.0","status":"MERGED","created":"2026-01-30 09:42:16.000000000","updated":"2026-02-06 12:36:18.000000000","submitted":"2026-02-06 12:34:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"975251-bug/port_binding_additional_chassis","meta_rev_id":"33775f936ca1b80f1f9b92467b3f3d74694420bf","_number":975251,"virtual_id_number":975251,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-06 12:34:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":2,"date":"2026-02-06 10:20:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2026-02-04 18:02:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":1,"date":"2026-02-06 10:20:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-30 10:37:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-01-30 11:43:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-30 17:07:26.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2026-02-04 18:02:47.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2026-02-06 10:20:06.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"05c0c0b2e9d42a647113539157a05f359963bb2a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-01-30 09:42:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"433c6967228e8061f01755bbe6723a3b3b08b7ea","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-30 10:37:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e86683770c1044fabd9951a231a5d832\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/11a44868859643a288dc7a283d297a39 : FAILURE in 53m 12s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/71d3d6a425b648518d8be81541fd4e33 : FAILURE in 44m 23s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5bcb1078499241408479878fea53b936 : RETRY_LIMIT in 4m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bd67e7af46d23cddc3b8f63c828f5da9a5e1b751","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-30 11:43:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/292a812614b64edda9036fe163b960c0\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/fa44a1f5e9874b70a43043dc29cb143d : SUCCESS in 1h 58m 59s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1ce5aa4cfc7a4bdca53a15264f00efc8 : SUCCESS in 33m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1380e69f87464a5ebdc67da15724f870 : FAILURE in 39m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d745f1d62184edf944bbefae1739703 : SUCCESS in 24m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/63509ba8bd474d5cb423ebca0f61ffe1 : FAILURE in 34m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0048c82a805542b4ac062cad55067b41 : FAILURE in 20m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ff929e63be44b69a1c02d11cf5c6397 : SUCCESS in 12m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ef4fd63fba1644869598bb0e2ea06f09 : SUCCESS in 23m 02s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/19863ab7caad4e8f9014c13d2ad2d7e0 : SUCCESS in 1h 11m 31s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/cf04840f60e34f2f98c4f9bcf4e64a50 : FAILURE in 11m 56s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/f4cfb5dc6d8b449384200f8346274d61 : SUCCESS in 1h 02m 15s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/632eb856414b419bb25209df51376f72 : SUCCESS in 54m 23s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4afe14031931479b90be6fee2ef6ae8f : SUCCESS in 44m 42s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a073f90c958242cc8934fbf85bb8dc91 : SUCCESS in 1h 22m 05s","accounts_in_message":[],"_revision_number":1},{"id":"775f08d59f8372e3e779396240d3ef1ef7826c96","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-01-30 14:27:41.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"3186e21be22b3f859bc4419f3dff593fd3729869","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-30 15:23:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4419f06ce80b49479843efb119d0c992\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/04c140784862441f84adb9dc978e49f3 : SUCCESS in 54m 19s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a64486a576dc4c10a63b774bfc21f020 : SUCCESS in 42m 28s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8036daf7b827488fbf12aaa0ada24d35 : RETRY_LIMIT in 4m 45s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"72cf095b5d961c402bc7c791aca66878ea09e9ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-30 16:17:50.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c3d3018869a7446ca1973c8e1cffc984\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d2e5b0ef3ac24ebf98d3e88288eba027 : SUCCESS in 1h 48m 52s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a41cfe89838f4dd3a02501a014f25315 : SUCCESS in 31m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8cad910f63654f4abcb15e15f29f8bdc : SUCCESS in 1h 01m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75bef870fabe4904a40e7b37aea4ed39 : FAILURE in 7m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cdff45dbfad1410caa508a7bfd0010ac : SUCCESS in 32m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/965510fc5b8d43faa289caa5cbabe177 : SUCCESS in 24m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a522e3c548104c788c5d2b38e3c1a87d : SUCCESS in 14m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2afdec08bf9b40dfae2926f5f74823f4 : SUCCESS in 25m 43s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/38c652ff4f75478d9a006b2f5360ea37 : SUCCESS in 1h 19m 45s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c3218ae260c9456c9b37140d495a375e : FAILURE in 43m 02s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/0dc8eaf4183142e49a343845c0179969 : SUCCESS in 1h 07m 52s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/edae3f4d38a94f3cb24bbf556ff59b80 : SUCCESS in 1h 02m 56s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/071d37fcde9c42528cae3480ff002050 : SUCCESS in 1h 33m 18s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/114280504f174102ac093c9187f863c3 : SUCCESS in 1h 28m 39s","accounts_in_message":[],"_revision_number":2},{"id":"7a6e89f81672247ec41e43662d7cd1ce58c18e83","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-02-02 08:41:58.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"011e079f73768a58b7a617fe031574e49a3553b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-02-02 08:42:02.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"e260b06e4ae93213ee146ce0d3edbcd76e56aedb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-02 09:38:58.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe9cdfa7d4c74adb8f48f7b38a8456d4\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/256b62ca419d496a911c01bad8268c32 : SUCCESS in 55m 23s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0e003f62d28f41c7ba5d15ad33ea19d7 : SUCCESS in 39m 47s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b0cd0892de3244779e4941b0e36106c6 : RETRY_LIMIT in 3m 33s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c2b65025dcb357bea637830d3f03b65ae14b698b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-02 10:58:24.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a601b8102317414ebc6eab02b6dc088b\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/be0e6ccb934a4d73a2c640f35348a495 : SUCCESS in 2h 11m 31s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f32890d4fd3c4bfab08fa9339a172367 : SUCCESS in 30m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c07c965c7ecf4435a7132ecf81a21bd2 : SUCCESS in 36m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a977170b465947d0a822cb5e5d360fc0 : SUCCESS in 23m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eeb81f0d32a04fedb293ed7eae3960b8 : SUCCESS in 28m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4fa0df6da39e4ad2bfa1de04348d973a : SUCCESS in 24m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ab9b6beb83e4972bbb48257614883be : SUCCESS in 5m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d6894ce0ecc048dea04c3ba043a0c1b1 : SUCCESS in 25m 29s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/e9a47452bf3e43f28539c728bd2d0bbc : SUCCESS in 1h 01m 05s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/dd056dfe915a45d18bc518dd91ac686f : FAILURE in 42m 30s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/453aa6d66f8a4951a8596884c97b0f80 : SUCCESS in 58m 00s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/b99971613fa44b99a8e256e316fd2fae : SUCCESS in 53m 10s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e7706098bc4844ff9d7f8197fd9379ad : SUCCESS in 1h 22m 12s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/232c88ce7528414a8b869b9fa8110b22 : SUCCESS in 1h 08m 29s","accounts_in_message":[],"_revision_number":4},{"id":"856f3f1e63df27edb166023bc026090c5147f37f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-02-02 16:10:59.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"87c7a02a2f36f04929a88e12a2761f7e490fa273","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-02 18:44:19.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e0ed5140c9a4ab881faafa7138b9d29\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/af1c744a0aaa4b53a576e75beab1f6d1 : SUCCESS in 58m 58s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/021eaf302f794554b560d2ea6ed6b3ab : SUCCESS in 45m 01s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/95ee8b17255b4d7e83e37b0cddc1ce04 : RETRY_LIMIT in 4m 12s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"66d871ffff48145e19125f3780c4a108fcc7009f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-02 19:25:42.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/53b5b61fd2774bfb8075b4b1786e8687\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/5ce305be502c4f72bacc89b4185f9d3c : SUCCESS in 1h 56m 55s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d9ff6338154b4caba24dc757b669cb10 : SUCCESS in 34m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/36dc54c8da614457b32a6dac03dd65c3 : SUCCESS in 58m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf00d13f5eda4f33a0ed73d23af60ede : FAILURE in 3m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5a1e5d57bf5c4a9da7bbb99bbd18389d : SUCCESS in 34m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/298e9ac6d0e44bfcb3a9875670280e37 : SUCCESS in 28m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b86634d52fc46dc8204226169400968 : SUCCESS in 12m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/62ea328a1ccd488abd6c5cdaf1d31886 : SUCCESS in 13m 38s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/81e73d7261084e539a578fe6207ccaab : SUCCESS in 1h 17m 26s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c94ddb5ece8c4956beafdbf35a8941a6 : SUCCESS in 1h 22m 47s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/cd31f097769d4aa596d8db5b14a38204 : SUCCESS in 41m 07s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/eff518c75b1e408ba8cfc099d76b816f : SUCCESS in 1h 18m 59s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ded09d05a7f44a65ac8f019188bac9bf : TIMED_OUT in 2h 46m 10s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/2d2c0eec45f3439592bd34b96bd0e41c : SUCCESS in 1h 31m 25s","accounts_in_message":[],"_revision_number":5},{"id":"b6d97d40f8192524c3e439bc69f706a6639bc878","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-02-03 07:07:24.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"14ce690229337f01a91c5edf537b452d2349e3e8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-03 08:02:09.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1211f046d41e4cf78aab19e6d8cc132d\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/27c700c3003b46ca852f6c7fbf1f76c2 : SUCCESS in 51m 26s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/410143da8e7a40e9bc82b80d7655629b : SUCCESS in 43m 49s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9f9f379fc0fa43629227766d86cee110 : RETRY_LIMIT in 4m 05s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"96ef850aa19f725dae01c21e1a2aebdc7ea652b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-03 08:57:31.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d83b0fcf8d514142a52ee637cad30b3a\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/93926bfda77740acbbc25e7975cb1d5c : SUCCESS in 1h 49m 06s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f1e0ee0e7dcc41e6848e9bb6d9a54b5f : SUCCESS in 31m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/16bec5557ccc4bc082e2db021ebe3d8c : SUCCESS in 41m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/132ee466176647ca9e892008698891eb : SUCCESS in 25m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7ad62823a6cb43bd8bba6b98fded3620 : SUCCESS in 32m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/718071e7a18142f8b085d0c45228198b : SUCCESS in 24m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ace131aa468f47bfa8f85e39551e8018 : SUCCESS in 6m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3ac22c41af3745459bfc2435173a70a5 : SUCCESS in 18m 39s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/258d5b82e86b4b6db447c05b28d125e0 : SUCCESS in 1h 15m 50s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/83034ddc123544b0bd974bb57034a27c : SUCCESS in 40m 07s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/bbee52e74b5045079b3b55c84e8d6fa8 : SUCCESS in 1h 17m 21s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/1af10a32e4e745e2b623c476835751c1 : SUCCESS in 1h 09m 50s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9738a5139d9f476fb26094509d05c38c : SUCCESS in 41m 45s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/289fc86cf5cb4d77bf7e1c5c8eee049b : SUCCESS in 59m 58s","accounts_in_message":[],"_revision_number":6},{"id":"835aa8b928fea04fd8bafbf37435ea5260ffebeb","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-02-04 18:02:47.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"83ca334256ab4a516e3d74fe4ba369ce0284b993","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-02-06 10:20:06.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"5ae19534084c7c29d05e8ef5014bf1a7beecace1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-06 10:20:36.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"b5452ef49ebf9399cedc16d224653b0b9a370d4f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-06 12:34:22.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28e71f460cc64ad3957968adbb6e0831\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/54964816773a4a61a8cd202d54af3c30 : SUCCESS in 1h 56m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7be3b2cd509a44838b2639b4811b92e1 : SUCCESS in 10m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/054c35eb7bd94d01b1ad34dacc9af6ac : SUCCESS in 33m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6abe3e8ee8b146bdaa98f98741210154 : SUCCESS in 27m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d06208c33ca442e86369100cb3d7311 : SUCCESS in 12m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7069b0a0162741ffa97946e6dbfb96e7 : SUCCESS in 31m 46s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/897c3ba93c6641129a1062e3908d67e9 : SUCCESS in 52m 12s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/8a3a7b64192d4457986adb8e22a08a12 : SUCCESS in 56m 57s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/131d5813891c4f36a04002a5d2b41293 : SUCCESS in 46m 53s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/1a4e910ca01848fc85e46da74849fb5d : SUCCESS in 1h 06m 15s","accounts_in_message":[],"_revision_number":6},{"id":"dfdc979e59efc5508b0b2350b7dfde3df6c4c7bc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-06 12:34:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"33775f936ca1b80f1f9b92467b3f3d74694420bf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-06 12:36:18.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b61cbca7e42346f09017c8e1134dc1c9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39c8bd90be884a5e926a3dd274915274 : SUCCESS in 51s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aba2468d842f48e2ac77617669f4a35e : SUCCESS in 46s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"d1f42cd00dd11a37b711b67ea3da61f008b66ea5","revisions":{"dcb84774e3fa1351b172c4cc87763674a0f1d9c9":{"kind":"REWORK","_number":1,"created":"2026-01-30 09:42:16.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/51/975251/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/51/975251/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/51/975251/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/51/975251/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/51/975251/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/51/975251/1"}}},"commit":{"parents":[{"commit":"7a80f5afa7299a632ccc78eac92c6c040a9259ad","subject":"Merge \"Bump os-ken to 4.1.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7a80f5afa7299a632ccc78eac92c6c040a9259ad"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-01-30 08:53:40.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-01-30 09:42:10.000000000","tz":60},"subject":"OVN ``Port_Binding.additional_chassis`` is supported since v22.09.0","message":"OVN ``Port_Binding.additional_chassis`` is supported since v22.09.0\n\nThis patch removes all checks to confirm the presence of\n``Port_Binding.additional_chassis``. Since Dalmatinan (2024.2) [1], the\nminimum OVN version required is v24.03.\n\n[1]https://docs.openstack.org/neutron/latest/install/ovs-ovn-requirements.html\n\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I0c40fcf06314c605b3dd2a63d195187f7368b16c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dcb84774e3fa1351b172c4cc87763674a0f1d9c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dcb84774e3fa1351b172c4cc87763674a0f1d9c9"}]},"branch":"refs/heads/master"},"216a7778e89fef1bbfc55890b4b45ce5da89ff5f":{"kind":"REWORK","_number":2,"created":"2026-01-30 14:27:41.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/51/975251/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/51/975251/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/51/975251/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/51/975251/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/51/975251/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/51/975251/2"}}},"commit":{"parents":[{"commit":"6752b702aa97c575026f043e64f36350085d4ce5","subject":"Merge \"Add parity gap limitation in OVN: packet rate rules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6752b702aa97c575026f043e64f36350085d4ce5"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-01-30 08:53:40.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-01-30 14:27:24.000000000","tz":60},"subject":"OVN ``Port_Binding.additional_chassis`` is supported since v22.09.0","message":"OVN ``Port_Binding.additional_chassis`` is supported since v22.09.0\n\nThis patch removes all checks to confirm the presence of\n``Port_Binding.additional_chassis``. Since Dalmatinan (2024.2) [1], the\nminimum OVN version required is v24.03.\n\n[1]https://docs.openstack.org/neutron/latest/install/ovs-ovn-requirements.html\n\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I0c40fcf06314c605b3dd2a63d195187f7368b16c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/216a7778e89fef1bbfc55890b4b45ce5da89ff5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/216a7778e89fef1bbfc55890b4b45ce5da89ff5f"}]},"branch":"refs/heads/master"},"352820f22c9ede1c5e0552e9d68665b6ac4aa7bb":{"kind":"REWORK","_number":3,"created":"2026-02-02 08:41:58.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/51/975251/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/51/975251/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/51/975251/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/51/975251/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/51/975251/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/51/975251/3"}}},"commit":{"parents":[{"commit":"6752b702aa97c575026f043e64f36350085d4ce5","subject":"Merge \"Add parity gap limitation in OVN: packet rate rules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6752b702aa97c575026f043e64f36350085d4ce5"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-01-30 08:53:40.000000000","tz":60},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2026-02-02 08:41:58.000000000","tz":0},"subject":"OVN ``Port_Binding.additional_chassis`` is supported since v22.09.0","message":"OVN ``Port_Binding.additional_chassis`` is supported since v22.09.0\n\nThis patch removes all checks to confirm the presence of\n``Port_Binding.additional_chassis``. Since Dalmatinan (2024.2) [1], the\nminimum OVN version required is v24.03.\n\n[1]https://docs.openstack.org/neutron/latest/install/ovs-ovn-requirements.html\n\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I0c40fcf06314c605b3dd2a63d195187f7368b16c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/352820f22c9ede1c5e0552e9d68665b6ac4aa7bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/352820f22c9ede1c5e0552e9d68665b6ac4aa7bb"}]},"branch":"refs/heads/master"},"7abf51998ca2027f442f347fe188d63f1acb0e52":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-02-02 08:42:02.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/51/975251/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/51/975251/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/51/975251/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/51/975251/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/51/975251/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/51/975251/4"}}},"commit":{"parents":[{"commit":"b0afd69f2512a30b8cd1fb5c0dd9370c9d93ea45","subject":"Merge \"Replace \"Inspector.from_engine()\" with \"sqlalchemy.inspect()\" (2)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b0afd69f2512a30b8cd1fb5c0dd9370c9d93ea45"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-01-30 08:53:40.000000000","tz":60},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2026-02-02 08:42:02.000000000","tz":0},"subject":"OVN ``Port_Binding.additional_chassis`` is supported since v22.09.0","message":"OVN ``Port_Binding.additional_chassis`` is supported since v22.09.0\n\nThis patch removes all checks to confirm the presence of\n``Port_Binding.additional_chassis``. Since Dalmatinan (2024.2) [1], the\nminimum OVN version required is v24.03.\n\n[1]https://docs.openstack.org/neutron/latest/install/ovs-ovn-requirements.html\n\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I0c40fcf06314c605b3dd2a63d195187f7368b16c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7abf51998ca2027f442f347fe188d63f1acb0e52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7abf51998ca2027f442f347fe188d63f1acb0e52"}]},"branch":"refs/heads/master","description":"Rebase"},"50a48099c6e3b4385184634635bcc0bd700e3e03":{"kind":"REWORK","_number":5,"created":"2026-02-02 16:10:59.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/51/975251/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/51/975251/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/51/975251/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/51/975251/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/51/975251/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/51/975251/5"}}},"commit":{"parents":[{"commit":"6e32b2cd36452ac28c55fd18ec5bbc4ca4587861","subject":"Merge \"Optimize ``get_subnet[s]_dhcp_options`` methods\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6e32b2cd36452ac28c55fd18ec5bbc4ca4587861"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-01-30 08:53:40.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-02-02 16:10:39.000000000","tz":60},"subject":"OVN ``Port_Binding.additional_chassis`` is supported since v22.09.0","message":"OVN ``Port_Binding.additional_chassis`` is supported since v22.09.0\n\nThis patch removes all checks to confirm the presence of\n``Port_Binding.additional_chassis``. Since Dalmatinan (2024.2) [1], the\nminimum OVN version required is v24.03.\n\n[1]https://docs.openstack.org/neutron/latest/install/ovs-ovn-requirements.html\n\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I0c40fcf06314c605b3dd2a63d195187f7368b16c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/50a48099c6e3b4385184634635bcc0bd700e3e03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/50a48099c6e3b4385184634635bcc0bd700e3e03"}]},"branch":"refs/heads/master"},"d1f42cd00dd11a37b711b67ea3da61f008b66ea5":{"kind":"REWORK","_number":6,"created":"2026-02-03 07:07:24.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/51/975251/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/51/975251/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/51/975251/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/51/975251/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/51/975251/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/51/975251/6"}}},"commit":{"parents":[{"commit":"6e32b2cd36452ac28c55fd18ec5bbc4ca4587861","subject":"Merge \"Optimize ``get_subnet[s]_dhcp_options`` methods\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6e32b2cd36452ac28c55fd18ec5bbc4ca4587861"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-01-30 08:53:40.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-02-03 07:07:02.000000000","tz":60},"subject":"OVN ``Port_Binding.additional_chassis`` is supported since v22.09.0","message":"OVN ``Port_Binding.additional_chassis`` is supported since v22.09.0\n\nThis patch removes all checks to confirm the presence of\n``Port_Binding.additional_chassis``. Since Dalmatinan (2024.2) [1], the\nminimum OVN version required is v24.03.\n\n[1]https://docs.openstack.org/neutron/latest/install/ovs-ovn-requirements.html\n\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I0c40fcf06314c605b3dd2a63d195187f7368b16c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d1f42cd00dd11a37b711b67ea3da61f008b66ea5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d1f42cd00dd11a37b711b67ea3da61f008b66ea5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
