)]}'
{"id":"openstack%2Fneutron~976010","triplet_id":"openstack%2Fneutron~master~I1f5678f1e5bdf0512a1d089ce3a9e03479f75fa3","project":"openstack/neutron","branch":"master","topic":"2026-02-07-remote-managed-pf","attention_set":{},"removed_from_attention_set":{"13686":{"account":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"last_update":"2026-08-10 08:35:16.000000000","reason":"Change was abandoned"},"24824":{"account":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"last_update":"2026-08-10 08:35:16.000000000","reason":"Change was abandoned"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2026-02-12 08:57:13.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2026-08-10 08:35:16.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I1f5678f1e5bdf0512a1d089ce3a9e03479f75fa3","subject":"Add support for remote-managed PFs","status":"ABANDONED","created":"2026-02-07 20:41:37.000000000","updated":"2026-08-10 08:35:16.000000000","total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"967cc37829f0f50145cfc38be38760b3a4f97e6c","_number":976010,"virtual_id_number":976010,"owner":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-07-07 15:34:04.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":385,"name":"Jason Kölker","email":"jason@koelker.net","username":"jason-koelker"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-07 20:43:09.000000000","updated_by":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"reviewer":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"state":"REVIEWER"},{"updated":"2026-02-07 20:43:09.000000000","updated_by":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2026-02-07 20:43:34.000000000","updated_by":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"reviewer":{"_account_id":385,"name":"Jason Kölker","email":"jason@koelker.net","username":"jason-koelker"},"state":"CC"},{"updated":"2026-02-07 21:36:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-02-07 22:43:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-11 14:12:11.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2026-08-10 08:35:16.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"}],"messages":[{"id":"7a2af89b6998528f9c3c78cec0daf6056e218cdb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2026-02-07 20:41:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"052f6795424ee8f0d0b0df1c3f600b1cef1c895f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-07 21:36:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8edf74a6393048ba9fa02849dcdc0974\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5983cf586a1747e69291e847c8a1a0e3 : SUCCESS in 53m 12s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/059d8d61ce9b45de91d0b3b160e7debb : SUCCESS in 40m 55s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8550a7921dff4a4e86104020d144e5b2 : RETRY_LIMIT in 3m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"abdcf61e8a8ed5f2522867c0f8ca13d24c4d4bb2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-07 22:43:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30621a45e343490b9f0ee9701fb6bb7f\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/40cdca22beb04664916d0b5eea40d61f : SUCCESS in 1h 59m 10s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/bbd871a732fa449aaa99f20f0df805fc : SUCCESS in 30m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/da579aeb3ffb4632b45b4685a263a839 : SUCCESS in 45m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/138bae766f154cbdb10977506e0b57c3 : SUCCESS in 17m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3d4b1b68f44d4f19a2dabaaf7bf38b71 : SUCCESS in 36m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e610e665431242afa61831deade7f236 : SUCCESS in 15m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80b571c3e8c041d691e2e73a603cae63 : SUCCESS in 11m 45s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/d402d841c9ec40389c1bf95c25857ae7 : SUCCESS in 34m 13s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/df38025c53f94780add59c92828b840f : SUCCESS in 1h 26m 42s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/c0ea03c5428341e091675cf9d635fbdf : SUCCESS in 41m 24s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/6f2c343dc4c14ac0b2f2eb3426048acb : SUCCESS in 49m 10s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1e9d0e59016943e0beef3997f679a825 : SUCCESS in 1h 23m 21s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/1734b50f2e214f92a8d15464c85556ce : SUCCESS in 1h 18m 31s","accounts_in_message":[],"_revision_number":1},{"id":"1d30e51238185888b143e11b3bc1cdcf2e834610","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-02-11 14:12:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ef056bcaacef2d8159f505703dbda704f8b0eb91","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2026-02-12 04:52:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"494d645ab46d04f9a907253fd471b5642f799af7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2026-02-12 04:52:17.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"21411c8a9f389eac88e6997af4aca9bb05688dde","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 06:06:28.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eeaf14eca69e492790c8bb817af34572\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5208b45caf6b4f6492e11c27eae8b9fb : SUCCESS in 1h 12m 07s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/25ba7371797b4d79858273d5416adb40 : SUCCESS in 55m 12s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/eb5fc883e5be4cbd8a73f739a4ad09d7 : RETRY_LIMIT in 4m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"16ed049865f09f20b1238edfd59c2ac174ce6c99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 07:13:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c7ad7cb8427e4a279174db2a5dc91167\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c64045bab4c842229671ae388af5569f : SUCCESS in 2h 20m 20s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8af7a9822215425eb1d5c00b32e664f4 : SUCCESS in 32m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/526de0645bce41d492106de94838b059 : SUCCESS in 1h 02m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/420399bd12334cea80eae80ae80101f6 : SUCCESS in 11m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/649e7869b6cf4d6183eafe264885b966 : SUCCESS in 23m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/456473cffa774345949dcc1ce310775a : SUCCESS in 15m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34e8f077968a4ea1aa124267fce66ec5 : SUCCESS in 13m 29s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/4cdab6e857564e9f98e17cd4cb984932 : SUCCESS in 1h 10m 22s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/cc4987b71231429a8f051b18011f48d6 : SUCCESS in 1h 24m 00s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/d1f91e11f59b40009e850e385470ecb3 : SUCCESS in 1h 09m 34s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/9ca6c1ddef194a4ba15d01494b2bb822 : SUCCESS in 1h 04m 48s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d6945773c43744fdbd39424c18e7e131 : SUCCESS in 1h 22m 43s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c37fbe4839f34766aeddb4c76acb9d46 : SUCCESS in 59m 00s","accounts_in_message":[],"_revision_number":2},{"id":"ed1582c2bf25c9fde5f35fbf629c2d2f81ab8f15","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-02-12 08:57:13.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8f949e8184bf3aa9fe9b31945993b3ac2f939436","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2026-07-07 13:21:57.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"5ccda5103095827129dab5bf2d51ed351529d060","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 15:34:04.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c7a65f6c5ea84343ab559e6bf992ee63\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/bc89f57d634940c1a767e165757253d3 : SUCCESS in 1h 37m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8b63f23b5be448b9970baf891346220d : SUCCESS in 1h 16m 15s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/507e044f1af8415fb94acea54e667e97 : FAILURE in 2h 10m 16s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b05ee7f52f0142358aed27dfb1d03c6c : SUCCESS in 32m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f80ab1de35344a54b5f8370816292a80 : SUCCESS in 1h 05m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f9f89738bc24c8c9d7b98a8207414d7 : FAILURE in 9m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/59672d4417ec44fcb070f73c2576b142 : SUCCESS in 28m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d96e9f30cc5f4cfa95eec71fd0dab1c8 : SUCCESS in 25m 14s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/71b22926157140f68bbe2c1db8e03e04 : SUCCESS in 33m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c9a46fcb0cd4ce984fa4a1e9b4293cf : SUCCESS in 12m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bad7a161599c4f8fbab7e4a9eb6d09b1 : SUCCESS in 7m 29s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/cd415c144faa4c0eb925cbe5240cb284 : SUCCESS in 53m 04s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/32b7b6f563154206ae2398500bd72cd0 : SUCCESS in 1h 20m 30s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/23d8b15fe3cd433596e7038587e43c64 : SUCCESS in 1h 18m 01s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/fafef7bb069c409080b9a9a18dab7f34 : SUCCESS in 1h 33m 32s","accounts_in_message":[],"_revision_number":3},{"id":"c8aae25665b7e10ca12e87b76da37a2f68cda1e0","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2026-07-07 18:19:01.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f3c2527c3111bc5b5aed5200d311d4aaf639284c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 20:19:22.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8bff72c51ac542bab78af1d3f311a79b\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/8033904f548d47b282d3f988fac711c5 : SUCCESS in 1h 07m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9c259dd6977c4928b17bfb56c95d68be : SUCCESS in 1h 11m 42s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/1a372f2f42574a20ab35f195e452a242 : FAILURE in 1h 58m 05s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c129e408cd954c9fa860eca64e32c122 : SUCCESS in 36m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c5c3b3429d9f4785a6080ab943c53d2a : SUCCESS in 47m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a25ee665367d421193ee8c19557c0b7f : FAILURE in 9m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3697e398a06347fd8f44382fc92631c1 : SUCCESS in 20m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/01327deea67f43d3818bf99758b1ea96 : SUCCESS in 18m 20s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/bfc0129af09747668a9c2991209e41b9 : SUCCESS in 28m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f00d580dadac43139fa400eb93daa2f9 : SUCCESS in 10m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b40a60548a9e415bb47d1f61d2a9a5ae : SUCCESS in 20m 52s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/6c44695c6eae48b386559ae023c0e957 : SUCCESS in 1h 34m 52s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/5086b7fb310c4ac29acea4c5d13d391a : SUCCESS in 1h 22m 27s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/bcaa9cfbfaf34699a0ca2ce6f93ddd84 : SUCCESS in 1h 13m 35s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/cc469a7f92344fcfa956f818d6b562e1 : SUCCESS in 1h 27m 35s","accounts_in_message":[],"_revision_number":3},{"id":"67c18946c837ac4d9fd86e46aa8c04e56977f54c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 22:02:06.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8726899b3aeb40d9965c7f260b5a2292\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/42cd45d8f8594e86a399761dcdfff98b : SUCCESS in 1h 15m 46s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/cd2161e1ccc84e05a9e4d4774d7404df : SUCCESS in 37m 42s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/c08df07b4632451db12e16aef131a669 : SUCCESS in 1h 04m 48s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"967cc37829f0f50145cfc38be38760b3a4f97e6c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-08-10 08:35:16.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"4a9d7b6a16591cba8be872a4fcba1df1d796710d","revisions":{"7c7c7e0c1a788c2e306bee8a8a864943f8c1c0db":{"kind":"REWORK","_number":1,"created":"2026-02-07 20:41:37.000000000","uploader":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"ref":"refs/changes/10/976010/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/10/976010/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/976010/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/976010/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/976010/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/10/976010/1"}}},"commit":{"parents":[{"commit":"dd1148d71f671d52d6626813f8e370296d937a89","subject":"Merge \"[OVN] Wait for Port_Binding additional chassis in live-migration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dd1148d71f671d52d6626813f8e370296d937a89"}]}],"author":{"name":"Dmitrii Shcherbakov","email":"Dmitrii.S@protonmail.com","date":"2026-02-06 23:36:44.000000000","tz":180},"committer":{"name":"Dmitrii Shcherbakov","email":"Dmitrii.S@protonmail.com","date":"2026-02-07 20:41:10.000000000","tz":180},"subject":"Add support for remote-managed PFs","message":"Add support for remote-managed PFs\n\nCurrently only remote-managed VF representor plugging can be requested\nsince vf-num is a mandatory parameter.\n\nHowever, PF representor plugging is also a valid case where pf-mac and\ncard-serial-number are specified but not vf-num.\n\nBoth PF and VF representors can be plugged at the same time - the\nactual DPU-side use-case for this may vary, although it is likely to\nbe used for bare metal instance provisioning.\n\nThis change allows for both cases to be handled and updates unit and\nfunctional tests to account for that.\n\nGiven that parameter sets may overlap, a longest parameter set match is\ndone.\n\nNot specifying vif-plug:representor:vf-num is documented as a valid\ncase:\nhttps://github.com/ovn-org/ovn-vif/blob/main/Documentation/topics/vif-plug-providers/vif-plug-representor.rst\n\"When OVN_Northbound:Logical_Switch_Port:options key\nvif-plug:representor:vf-num is not set this option is used to locate a\nPF representor port.\"\n\nContext:\nhttps://docs.ovn.org/en/stable/topics/vif-plug-providers/vif-plug-providers.html\nhttps://specs.openstack.org/openstack/neutron-specs/specs/yoga/off-path-smartnic-dpu-port-binding-with-ovn.html\n\nSee also: https://github.com/ovn-org/ovn-vif/pull/17\n\nChange-Id: I1f5678f1e5bdf0512a1d089ce3a9e03479f75fa3\nSigned-off-by: Dmitrii Shcherbakov \u003cDmitrii.S@protonmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7c7c7e0c1a788c2e306bee8a8a864943f8c1c0db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7c7c7e0c1a788c2e306bee8a8a864943f8c1c0db"}]},"branch":"refs/heads/master"},"f47ab7fd7a18ddf1c93c3fe964c65d8a3d7b2200":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-02-12 04:52:17.000000000","uploader":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"ref":"refs/changes/10/976010/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/10/976010/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/976010/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/976010/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/976010/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/10/976010/2"}}},"commit":{"parents":[{"commit":"dd1148d71f671d52d6626813f8e370296d937a89","subject":"Merge \"[OVN] Wait for Port_Binding additional chassis in live-migration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dd1148d71f671d52d6626813f8e370296d937a89"}]}],"author":{"name":"Dmitrii Shcherbakov","email":"Dmitrii.S@protonmail.com","date":"2026-02-06 23:36:44.000000000","tz":180},"committer":{"name":"Dmitrii Shcherbakov","email":"Dmitrii.S@protonmail.com","date":"2026-02-12 04:52:17.000000000","tz":0},"subject":"Add support for remote-managed PFs","message":"Add support for remote-managed PFs\n\nCurrently only remote-managed VF representor plugging can be requested\nsince vf-num is a mandatory parameter.\n\nHowever, PF representor plugging is also a valid case where pf-mac and\ncard-serial-number are specified but not vf-num.\n\nBoth PF and VF representors can be plugged at the same time - the\nactual DPU-side use-case for this may vary, although it is likely to\nbe used for bare metal instance provisioning.\n\nThis change allows for both cases to be handled and updates unit and\nfunctional tests to account for that.\n\nGiven that parameter sets may overlap, a longest parameter set match is\ndone.\n\nNot specifying vif-plug:representor:vf-num is documented as a valid\ncase:\nhttps://github.com/ovn-org/ovn-vif/blob/main/Documentation/topics/vif-plug-providers/vif-plug-representor.rst\n\"When OVN_Northbound:Logical_Switch_Port:options key\nvif-plug:representor:vf-num is not set this option is used to locate a\nPF representor port.\"\n\nContext:\nhttps://docs.ovn.org/en/stable/topics/vif-plug-providers/vif-plug-providers.html\nhttps://specs.openstack.org/openstack/neutron-specs/specs/yoga/off-path-smartnic-dpu-port-binding-with-ovn.html\n\nSee also: https://github.com/ovn-org/ovn-vif/pull/17\n\nChange-Id: I1f5678f1e5bdf0512a1d089ce3a9e03479f75fa3\nSigned-off-by: Dmitrii Shcherbakov \u003cDmitrii.S@protonmail.com\u003e\nCloses-Bug: #2141626\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f47ab7fd7a18ddf1c93c3fe964c65d8a3d7b2200"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f47ab7fd7a18ddf1c93c3fe964c65d8a3d7b2200"}]},"branch":"refs/heads/master","description":"Edit commit message"},"4a9d7b6a16591cba8be872a4fcba1df1d796710d":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-07-07 13:21:57.000000000","uploader":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"ref":"refs/changes/10/976010/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/10/976010/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/976010/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/976010/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/976010/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/10/976010/3"}}},"commit":{"parents":[{"commit":"51042ce322be72410cc11a90aa26e38d5397e86c","subject":"Merge \"ovn: Handle ``PortNotFound`` in ``update_virtual_port_parent_host``\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/51042ce322be72410cc11a90aa26e38d5397e86c"}]}],"author":{"name":"Dmitrii Shcherbakov","email":"Dmitrii.S@protonmail.com","date":"2026-02-06 23:36:44.000000000","tz":180},"committer":{"name":"Dmitrii Shcherbakov","email":"Dmitrii.S@protonmail.com","date":"2026-07-07 13:21:57.000000000","tz":0},"subject":"Add support for remote-managed PFs","message":"Add support for remote-managed PFs\n\nCurrently only remote-managed VF representor plugging can be requested\nsince vf-num is a mandatory parameter.\n\nHowever, PF representor plugging is also a valid case where pf-mac and\ncard-serial-number are specified but not vf-num.\n\nBoth PF and VF representors can be plugged at the same time - the\nactual DPU-side use-case for this may vary, although it is likely to\nbe used for bare metal instance provisioning.\n\nThis change allows for both cases to be handled and updates unit and\nfunctional tests to account for that.\n\nGiven that parameter sets may overlap, a longest parameter set match is\ndone.\n\nNot specifying vif-plug:representor:vf-num is documented as a valid\ncase:\nhttps://github.com/ovn-org/ovn-vif/blob/main/Documentation/topics/vif-plug-providers/vif-plug-representor.rst\n\"When OVN_Northbound:Logical_Switch_Port:options key\nvif-plug:representor:vf-num is not set this option is used to locate a\nPF representor port.\"\n\nContext:\nhttps://docs.ovn.org/en/stable/topics/vif-plug-providers/vif-plug-providers.html\nhttps://specs.openstack.org/openstack/neutron-specs/specs/yoga/off-path-smartnic-dpu-port-binding-with-ovn.html\n\nSee also: https://github.com/ovn-org/ovn-vif/pull/17\n\nChange-Id: I1f5678f1e5bdf0512a1d089ce3a9e03479f75fa3\nSigned-off-by: Dmitrii Shcherbakov \u003cDmitrii.S@protonmail.com\u003e\nCloses-Bug: #2141626\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4a9d7b6a16591cba8be872a4fcba1df1d796710d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4a9d7b6a16591cba8be872a4fcba1df1d796710d"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"dd1148d71f671d52d6626813f8e370296d937a89","ours":"f47ab7fd7a18ddf1c93c3fe964c65d8a3d7b2200","theirs":"51042ce322be72410cc11a90aa26e38d5397e86c","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
