)]}'
{"id":"openstack%2Fneutron~976158","triplet_id":"openstack%2Fneutron~master~Iacbd4bb69e7bc4951cabeb6aa464f87410a931a4","project":"openstack/neutron","branch":"master","topic":"bug/2111276","attention_set":{},"removed_from_attention_set":{"8655":{"account":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"last_update":"2026-02-16 13:31:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iacbd4bb69e7bc4951cabeb6aa464f87410a931a4","subject":"bgp: Support IPv6 coming from fabric","status":"MERGED","created":"2026-02-09 18:14:22.000000000","updated":"2026-02-16 13:32:43.000000000","submitted":"2026-02-16 13:31:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"976158-bug/2111276","meta_rev_id":"aba037900b4ea0ae78469be18c775bcac2d6589f","_number":976158,"virtual_id_number":976158,"owner":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-16 13:31:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-02-13 06:56:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2026-02-16 11:17:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-02-16 11:17:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-09 18:43:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-02-09 20:19:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-13 06:56:06.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2026-02-13 06:56:50.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2026-02-16 11:17:31.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"0e47d499384bea661e89cb9eea4fa1bf1f3ab370","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-02-09 18:14:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c77c0ca40e0531d3eb5655d87d4e5e6efc2d295b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 18:43:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2eefbe1ab5ff4f4abd9e1c4a60c753e7\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/cb090d25b5a3417885c5e181d109af41 : SUCCESS in 27m 18s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1e410f38974b48debbf51fe4714f62e8 : SUCCESS in 25m 00s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/da2f40cb211248e99f3b4d21a62ce358 : RETRY_LIMIT in 4m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9d3dca7daec8cc7a935d75d4c20bc55c2349af68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 20:19:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/73a3969d9e174b1cb0afb5f233c43a97\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/78b9aa8e5b424bb3a44307ae18036286 : SUCCESS in 2h 04m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/895fd188bdfc483a81e1ea834f0d7416 : SUCCESS in 1h 11m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c7d654ce8394a2f9a1b688feeb27a8e : SUCCESS in 23m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1bd7075e9d664b74bf2e0d7d7f25cd17 : SUCCESS in 27m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9b2ecb5727df44d0bca867b5e476e057 : SUCCESS in 27m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc01ad2ef11444e1b7a0103ceb894cc4 : SUCCESS in 13m 01s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/5f2f89c2f14f40ecbe10f23d40113cc7 : FAILURE in 10m 22s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/9afcba1adf3444c288513e41822ed541 : SUCCESS in 40m 35s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/a6f285bdb40c42b48f2b4487a0ade00b : FAILURE in 18m 14s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/db03b93dfe8c4f59b74393f30c07bd08 : SUCCESS in 53m 34s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/aadd2472b7a44efabd3bb25c7ac0aed5 : SUCCESS in 1h 23m 15s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/dc2e13b90c66433484a9bda15e62a0ec : SUCCESS in 1h 04m 11s","accounts_in_message":[],"_revision_number":1},{"id":"dcfd60bf50019a0fa635d4216e7ce1b5519de518","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-02-13 06:56:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0df976896fc63738b034e29a8d3368782cfc3c90","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-02-13 06:56:50.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"00a889307e59092cda464f51573e54f7b0ff1974","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 07:29:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0fdf760cd410478dbdc69c85d50ffa47\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/60e9100b57ae431eb9babc26eae9151c : SUCCESS in 25m 36s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1d7402ccf8e24cc2878e931e262e0a0c : SUCCESS in 31m 15s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/cab773ead54f4fb09745e822438c4e7d : RETRY_LIMIT in 3m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"418b3879f57b3ef6c5a8a816e4db8dec7aa6a657","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 08:58:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3051013ca7e54218bf424e34206595cb\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a6a1b3d727af4099bef179854abbb99f : SUCCESS in 2h 00m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3b684544c0f4445f87e0c01b1e3ba101 : SUCCESS in 1h 03m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9338b58e3ff4ce08176eff4ae18b85a : SUCCESS in 18m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a2024d0839ac424cb35827cf16cdc0a9 : SUCCESS in 25m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/14fdab581f674533bfd0cfa8f8a2d8f9 : SUCCESS in 18m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/177ed535b4664b388abdf1da34ee9e59 : SUCCESS in 9m 03s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/37272d60ccc740aab2fa23eaf6ae0f06 : SUCCESS in 1h 11m 11s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/5b1e6102c05c403a85c9c608d9e08fbc : SUCCESS in 1h 15m 46s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/cc424ff2a6f44ef6b23843ddeffbd740 : SUCCESS in 1h 09m 39s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/06b6ffdc82f0494aa695ff0a76dab701 : SUCCESS in 52m 56s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/df6a0519713a439093e36602bdd5252a : SUCCESS in 1h 24m 42s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/2a839870f3c64c1784aa656629d280dc : SUCCESS in 1h 24m 27s","accounts_in_message":[],"_revision_number":1},{"id":"d16f956a7fbe915bb2863aab1caed6ba700257cf","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-02-16 11:17:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2af90b1671768c3f01100ec5cb348cfb17ec3cc9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 11:17:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"794454a6b3541efa19571a408dd402e7902aec87","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 13:31:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5206b23394364cd39ab641b45041cbd5\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/61d6172a9e4f48cb815e5d77ec907cf9 : SUCCESS in 2h 13m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15ed1cc6ec1f41599784e543586fc258 : SUCCESS in 22m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6a750c9f124941cf8475c1eab831adf9 : SUCCESS in 23m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5b9448d491bc4f3381214601f0cbc45c : SUCCESS in 22m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2bbc7da824924963b5a6d95108cd7b2d : SUCCESS in 12m 37s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/e8a41f0d72974dddadb108c60aeee356 : SUCCESS in 47m 24s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/bbed49c829bf451892b0feddb370e3aa : SUCCESS in 1h 10m 26s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/812ea4eb80934275b191844a5bf226d1 : SUCCESS in 1h 05m 50s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/202f4383bf644651a52464dfa5039f0f : SUCCESS in 1h 14m 39s","accounts_in_message":[],"_revision_number":1},{"id":"3e333b6804479d04a57947cc99fa06936ea74276","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 13:31:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"aba037900b4ea0ae78469be18c775bcac2d6589f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 13:32:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/05fdd39f2c3242bb8ca5d052411ef5bf\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e2f3e4a44364871a4aeb2affef6650b : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"73c5380a044ca614152ed259b0f54c404642a578","revisions":{"73c5380a044ca614152ed259b0f54c404642a578":{"kind":"REWORK","_number":1,"created":"2026-02-09 18:14:22.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/58/976158/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/58/976158/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/58/976158/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/58/976158/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/58/976158/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/58/976158/1"}}},"commit":{"parents":[{"commit":"aa84b8986c7d3994769ac7802ac0b5b588b6b3d4","subject":"bgp: Use port mapping for learned routes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/aa84b8986c7d3994769ac7802ac0b5b588b6b3d4"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-02-05 21:34:49.000000000","tz":0},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-02-09 18:13:58.000000000","tz":0},"subject":"bgp: Support IPv6 coming from fabric","message":"bgp: Support IPv6 coming from fabric\n\nAs there are lot of incoming ICMPv6 packets that are broadcasted we do\nnot know if it\u0027s inteded for the host or the OVN workloads, so we must\nflood the traffic to both the host and the OVN.\n\nRelated-Bug: #2111276\n\nChange-Id: Iacbd4bb69e7bc4951cabeb6aa464f87410a931a4\nSigned-off-by: Jakub Libosvar \u003cjlibosva@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/73c5380a044ca614152ed259b0f54c404642a578"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/73c5380a044ca614152ed259b0f54c404642a578"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
