)]}'
{"id":"openstack%2Fneutron~976159","triplet_id":"openstack%2Fneutron~master~I544d0d617176e96ede41761b63bd0a8e49756128","project":"openstack/neutron","branch":"master","topic":"bug/2111276","attention_set":{},"removed_from_attention_set":{"5756":{"account":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"last_update":"2026-02-12 15:12:23.000000000","reason":"\u003cGERRIT_ACCOUNT_5756\u003e replied on the change","reason_account":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}},"8655":{"account":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"last_update":"2026-02-16 13:32:49.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I544d0d617176e96ede41761b63bd0a8e49756128","subject":"bgp: Fix configuring vrf id on BGP routers","status":"MERGED","created":"2026-02-09 18:14:22.000000000","updated":"2026-02-16 13:33:57.000000000","submitted":"2026-02-16 13:32:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"976159-bug/2111276","meta_rev_id":"22801e5ed6dd7351b8e2d7e84b34582182b5b4ef","_number":976159,"virtual_id_number":976159,"owner":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-16 13:32:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-02-13 06:56:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2026-02-12 15:12:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2026-02-13 06:56:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-09 19:08:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-02-09 19:55:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-10 15:46:14.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2026-02-13 06:56:21.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2026-02-13 06:56:26.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"b7ea30a6bda629b352ab3ec2aa5f1ecb73674f9b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-02-09 18:14:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d3cae6c58510400d6cd4e0f3da89b6cabcf428a8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 19:08:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9288305f573742e5aa701e005622081c\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0ecef17429fe499180d92cc5df523dcc : SUCCESS in 51m 59s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c716e0ecabfe4415bdbbe76371f7930b : SUCCESS in 29m 08s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1c7ef00d35eb40e19bde874d9e89bc47 : RETRY_LIMIT in 3m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2680059a0097cb7b51576c11d2f601f83bb6cdb1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 19:55:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6083d835b2fb4d3cb77e7f58c6084187\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/0cabd011cab54c6a88df051d3c3b1e31 : SUCCESS in 1h 26m 53s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/2a8e629ece8343acbb9ee2a43e7c3195 : SUCCESS in 1h 37m 57s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/44d939c4380a4fe4b1a17dcca37257a7 : SUCCESS in 31m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/79764abd10914e158b7b28eb748c6832 : SUCCESS in 1h 07m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76f3d4be648c4a498b78cbd06d041e5a : SUCCESS in 26m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/aa19832decce408babdb13b2cbd8474d : SUCCESS in 38m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/432ed82b95e14094abc7a43f7bbc8e2b : SUCCESS in 24m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06264dd097cc458ba8f4d94fee926dee : SUCCESS in 14m 06s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/7ebe5e122a9a4fae9429c41167a8ee18 : FAILURE in 15m 55s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/80da80b15f03493898cd930e271dfabc : FAILURE in 13m 48s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/0c3c65d43dcf4fe4a61b264b4a8d0eb1 : SUCCESS in 47m 03s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b56a3d3efc394ff48d0b5bb94780ff0e : FAILURE in 17m 21s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/bb0ecdbadf02458286d6cb8b51a61362 : FAILURE in 11m 05s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/d2e4b5411e7549e192a869ef1df1e35d : SUCCESS in 1h 23m 01s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b16d1ade477e40e09b3923e8d653cc37 : SUCCESS in 1h 08m 33s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/27144b6b7cea4040954fc96eec281718 : SUCCESS in 1h 00m 43s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/d399489524d64ae78411837b56e4afda : SUCCESS in 35m 35s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4ef0f1c35cf9456f803cbe22d109c21b : SUCCESS in 36m 48s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/7cc049ee1674446cb827e447a383763f : SUCCESS in 1h 19m 10s","accounts_in_message":[],"_revision_number":1},{"id":"e43799994f1c76fa28a893abfc1e615d07b047c2","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2026-02-10 15:46:14.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"41f52771c3f0bcf16c42f98ff39772142b5c9fca","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-02-12 15:11:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0af57a59b972a764639b5e4f8679a4aaaf00023a","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2026-02-12 15:12:23.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"83497bc11451277d9eaac8a532b61c6bd0dc41e9","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-02-13 06:56:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"49c97414fb5da29ad69eb39f8d17560e37acae24","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-02-13 06:56:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c9aaa297256a3e60e17b263fc240786f6d90bcb0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 07:51:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d2a6cec4f2734d7a8d485982cf0364ca\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0facc20e521b486e88f4557cbfbf19b9 : SUCCESS in 53m 19s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a8da543c01f345e482ba8a951cc05f43 : SUCCESS in 41m 11s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f0e6092340b34033a461e01d8416b709 : RETRY_LIMIT in 3m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"96f9d6af9b685d281a8382428289b5645f7196d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 09:08:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d963fcb0ef564d319e8c1e71930a4382\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/053e993d1acc415e982f9a49b4c63499 : SUCCESS in 1h 25m 34s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/eab3ffa1eaa74c88ab3aac6ebd4ff6b1 : SUCCESS in 1h 51m 02s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/365fe148e98a4f809a0942a84a94e301 : SUCCESS in 33m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ca0f70ebd220438fbf72c2cf83935779 : SUCCESS in 40m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/205317e63ada4f17883f309baf7ef488 : SUCCESS in 24m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b84c77ccfde84186b0e2273dbba7ce05 : SUCCESS in 29m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d1f213b36dd94b67926ba97ed0781e71 : SUCCESS in 15m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a15a9780d3f046be8260badcc20cb59a : SUCCESS in 12m 13s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/d439b42757b44744b119c6d3c4bf36b2 : SUCCESS in 31m 19s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/2cfc2e6478d5468ea18e1f421cc88b57 : SUCCESS in 1h 11m 26s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/95303809f9174c4ea7381359b8831ed2 : SUCCESS in 45m 31s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/2f0510677118446ebdb93fea545b658f : SUCCESS in 1h 10m 08s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/9494120d27114607993e2f886b51b1c4 : SUCCESS in 1h 09m 52s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/8024bfd906ad43a0abddab2125438e98 : SUCCESS in 2h 01m 20s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/05aab0f2c436426588b157d135295e05 : SUCCESS in 1h 25m 26s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/6a37f11cc040429aa89e5ebdfa5043d1 : SUCCESS in 1h 06m 09s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/4a7157087aa343378d9dd419b504536c : SUCCESS in 1h 14m 49s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/61182a72303740949b6381e5e8e8ef7d : SUCCESS in 1h 01m 20s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/84f3bdd32c104c77b71816f74c579f95 : SUCCESS in 1h 21m 57s","accounts_in_message":[],"_revision_number":1},{"id":"414eb9280237ec2014ad12e0b8c5f991c02288f8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 11:17:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4bae950a5ae1e5182937ebebadb1e74f516c953d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 13:32:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0368178ffe8b43109ecb060560a6c6ee\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/99cd10db59b044458049f992fb6acbe1 : SUCCESS in 1h 23m 07s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/debd4cfeff98402cbf7198f5596b7062 : SUCCESS in 2h 10m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8857313b98c34f8781ec6e43dd37d028 : SUCCESS in 25m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e646bf2f75a54e6189a4427a6509c879 : SUCCESS in 27m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/80ba3eb03f324e058c1c786c6a1d6ed8 : SUCCESS in 25m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a74ed77fe3ee45cfbbdc0247456b99e4 : SUCCESS in 11m 48s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/090b3c36f4354a9488d8556da47231ad : SUCCESS in 43m 54s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/d6e4580412ee4ed09f73c6f4b6f842c7 : SUCCESS in 1h 43m 46s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/9e3839381b5242c99db1e8bab74e8b0f : SUCCESS in 1h 04m 44s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1a8b69c7675142bd83d50a400465edc7 : SUCCESS in 44m 14s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/7e2b48afcfb74cc2af1aa080cd06f81e : SUCCESS in 49m 26s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/8164914d3532441a951c06a043013672 : SUCCESS in 1h 12m 09s","accounts_in_message":[],"_revision_number":1},{"id":"7c78b04c6df51b496a244d09c1b2dec233f7082d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 13:32:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"22801e5ed6dd7351b8e2d7e84b34582182b5b4ef","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 13:33:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/09d76fcc880440ae9dcd5d7d62bd2cc1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d26a66c1f48b42e2af80c0c437b82ece : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a3095f4372c18f581c5fb0f75b5516ff05d4805f","revisions":{"a3095f4372c18f581c5fb0f75b5516ff05d4805f":{"kind":"REWORK","_number":1,"created":"2026-02-09 18:14:22.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/59/976159/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/59/976159/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/59/976159/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/59/976159/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/59/976159/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/59/976159/1"}}},"commit":{"parents":[{"commit":"73c5380a044ca614152ed259b0f54c404642a578","subject":"bgp: Support IPv6 coming from fabric","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/73c5380a044ca614152ed259b0f54c404642a578"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-02-06 14:53:02.000000000","tz":0},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-02-09 18:13:58.000000000","tz":0},"subject":"bgp: Fix configuring vrf id on BGP routers","message":"bgp: Fix configuring vrf id on BGP routers\n\nThe value for VRF ID must be a string which worked only for the main\nBGP router. The patch makes the Config Opt an Int for both per-chassis\nand main routers. Then before writing to ovsdb the value is converted\nto string.\n\nRelated-Bug: #2111276\n\nChange-Id: I544d0d617176e96ede41761b63bd0a8e49756128\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a3095f4372c18f581c5fb0f75b5516ff05d4805f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a3095f4372c18f581c5fb0f75b5516ff05d4805f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
