)]}'
{"id":"openstack%2Fneutron~976204","triplet_id":"openstack%2Fneutron~master~I1834642e9c8de971f560c50d3a0a8b76e051f41b","project":"openstack/neutron","branch":"master","topic":"integrated-gate-networking","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2026-02-26 03:04:28.000000000","reason":"\u003cGERRIT_ACCOUNT_8556\u003e replied on the change","reason_account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}}},"hashtags":[],"change_id":"I1834642e9c8de971f560c50d3a0a8b76e051f41b","subject":"Add integrated-gate-networking testing in neutron gate","status":"MERGED","created":"2026-02-10 03:08:35.000000000","updated":"2026-02-26 08:37:25.000000000","submitted":"2026-02-26 08:35:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"976204-integrated-gate-networking","meta_rev_id":"2d58c4b462010b054abf3cf3260824e6bc1613de","_number":976204,"virtual_id_number":976204,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-26 08:35:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-02-24 14:14:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":-1,"date":"2026-02-10 03:08:49.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":2,"date":"2026-02-15 01:06:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":1,"date":"2026-02-24 14:54:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":1,"date":"2026-02-18 14:40:50.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}],"CC":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-10 03:08:49.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"},{"updated":"2026-02-10 05:28:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-10 08:23:12.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2026-02-10 23:15:03.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2026-02-15 01:06:43.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2026-02-24 14:14:52.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"f2d3f4f165613c1ee6073bf56226320af6da98e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-02-10 03:08:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a8187d9cc1d1c696f1b588805168214128d3bc2d","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-02-10 03:08:49.000000000","message":"Patch Set 1: Code-Review-1\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":1},{"id":"8489f3b93b6f79c171b789bcdc9b839208a6bb7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-10 05:28:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3aeffb4eeed34fe8a234e131c5bb2d6b\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/eb1bf6dbddd74af68ddda838819b26c0 : SUCCESS in 1h 02m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5489289c19014a47bc361800b052403a : FAILURE in 33m 39s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/6cbf4dbb24234d5885e3dedd98ba0e89 : SUCCESS in 1h 26m 30s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/0a6a97395e784863969f94c020b91292 : SUCCESS in 2h 17m 26s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/82d393c4eb75482fb22c4071f18f556e : SUCCESS in 32m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/841a93e539bf417eb7cd815e905f6390 : SUCCESS in 21m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c91924066c2b47e1acfb51b625db9229 : SUCCESS in 6m 27s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/88c4afb16ce248f9aed5fd9acfa42d5c : FAILURE in 17m 14s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/5d5aa45303d540968d8752e8cac21dad : FAILURE in 16m 02s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/438d15ebd1c446948c4023963bbabc80 : SUCCESS in 41m 25s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/208e78d0e2cf4b0f8f44383d0851cc95 : FAILURE in 8m 42s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/9070519abea24b5bac695f4b89231df2 : FAILURE in 8m 45s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/07812959ace04e4087624aa51c02fe6e : SUCCESS in 1h 57m 39s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/69c64347efa44e8eb44a5ad71f3eb922 : SUCCESS in 1h 20m 04s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/1c46f86db61140ceb4897259a8dc7886 : SUCCESS in 1h 00m 18s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/667843f7db6243bf87361fe6d1d2679d : SUCCESS in 56m 08s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/27d310c0926b4376b9eb00958d776b84 : SUCCESS in 36m 29s","accounts_in_message":[],"_revision_number":1},{"id":"9f0620db987d7525f17a93f67b44e6fd4b6bed92","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-02-10 23:15:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1ea161210199baa3b151972d9163bee2a2258fcf","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-02-10 23:15:17.000000000","message":"Patch Set 1:\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":1},{"id":"cfe9e2dda3548791a99f41de696a3101eea428d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 01:06:19.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0ca8c7e0f2c84fc5bf4bfd7c38809889\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f2d66dec9a8046f7828daa804de593f3 : SUCCESS in 1h 28m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a99a33edc1f54762a05f3951f2138d33 : FAILURE in 32m 35s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/800a9431049a406c8acfe105e83b05ca : SUCCESS in 1h 21m 24s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/fdae7a53f72147e2a497d48652ca8d6f : SUCCESS in 1h 48m 30s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a8684a1d8db74f66a351505f478a6314 : SUCCESS in 32m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66f337d7d7a24b4c9d3ce544d8bcff62 : SUCCESS in 10m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e5bbb808a4b473ea8ff913eb1f9d416 : SUCCESS in 13m 32s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/9d850a7cbbbe42b28a1accc1e65196be : SUCCESS in 33m 39s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/ac4bc16016ad442ca19353e96330c714 : SUCCESS in 1h 07m 27s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/be3cf1bd0bfd4ae3825323ef8eda6584 : SUCCESS in 1h 01m 49s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b9700e9b2112459bbabbb9349bea5f02 : SUCCESS in 1h 15m 18s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/582e60e3a3de4df69dd9c3e5f745d283 : SUCCESS in 1h 14m 06s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/f4f2b0cf06e24feeadea063483e89271 : SUCCESS in 1h 05m 20s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7f4950b377174e1f9e2636bc50b18709 : SUCCESS in 1h 41m 36s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/884fcee98db6447cba127d027e85556f : SUCCESS in 1h 02m 03s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/a944336dae654610a17ace8fdee3d561 : SUCCESS in 1h 15m 29s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4de04c8dc1ec4cb3bfb3a1e0a3474aae : SUCCESS in 1h 25m 44s","accounts_in_message":[],"_revision_number":1},{"id":"d5ae40f5f0a68943840d863221c181120ccc8b99","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-02-11 15:11:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b03e224a9beadbcb937566f555fe24b02d0dcf8b","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-02-11 15:11:45.000000000","message":"Patch Set 1:\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":1},{"id":"3756920f991151bf43be231157ea84a57299192e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 17:31:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ae5668aa2fd47feba0e490c53c041a8\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/23b812ef25304ee09d055deb322f7b74 : SUCCESS in 1h 07m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/651e3873b69649e49e154dd88b43ca1a : SUCCESS in 1h 20m 44s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/be4e327cd8fd48429c2b66d84abae4a8 : SUCCESS in 1h 27m 40s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e4074fdfbfac450797ce7c21681c25ef : SUCCESS in 2h 17m 06s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e243a27346e544fd92b35c553e24b9c5 : SUCCESS in 34m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e79c875b4be47eba7941d6b82eda429 : SUCCESS in 18m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50f2ab2ba9b6424092c912a9b9694cf3 : SUCCESS in 13m 03s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/8dc79daedd204825a6bdfc48095b70f4 : SUCCESS in 56m 46s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/31055b779efe4f4eba6fd620a54280c7 : SUCCESS in 39m 34s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/cd81d3d5be5a42cfa7a3c905b0d26fad : SUCCESS in 1h 16m 29s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/4b55168189be4e8c9fad24153acf9875 : SUCCESS in 56m 47s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/38961fe6d0304702bc64d370610214db : SUCCESS in 44m 35s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/69e16da090664fc0b95bd30949dfc175 : SUCCESS in 1h 08m 13s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ecc9b200a3bb44e38f7006bc97c5bf12 : SUCCESS in 1h 02m 09s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/bd0bc4711de14b8f9190e90f8b1d940d : SUCCESS in 1h 18m 02s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/0d9a6d610bf244649fcbbffbea85768e : SUCCESS in 58m 45s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0d1cebfa0a5f41cea52ba0827be26a93 : SUCCESS in 1h 47m 02s","accounts_in_message":[],"_revision_number":1},{"id":"8217e3878a22c426cc4196bd12d9b350ba5ae56f","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-02-12 03:54:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1879be30c460513a90c873c30e9d3da596a2fa49","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-02-12 03:54:21.000000000","message":"Patch Set 1:\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":1},{"id":"36e9bad43d1c074e4bf13bf2b424b8d0c8b3803d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 06:00:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/53071da67b4d4c0e839b947012347496\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c85b1e0a60344fcf94a5bc709139b869 : SUCCESS in 1h 24m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7639ccea74e74137b1e487d78a24cf52 : SUCCESS in 51m 07s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/3dee792e9159474789be5ebd17d63509 : SUCCESS in 1h 36m 01s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d117f99106784336a612695f6b799e77 : SUCCESS in 1h 54m 06s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4b7d750183864c138c2551edec754297 : SUCCESS in 34m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/604097f08f974aa7ae42c521fe3f8c07 : SUCCESS in 26m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f09a7524a17549008cd72223e71618f4 : SUCCESS in 14m 58s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/60ad9efd1bd04f7ab230677f3e5f3220 : SUCCESS in 56m 37s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/0dc3fe1b26f54d538c017f4158051b43 : SUCCESS in 1h 04m 50s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/af375f4c05b049e287c6ffde66a93db4 : FAILURE in 33m 52s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e08a7f1ce51f45cba09da6c2e35d28b1 : SUCCESS in 1h 05m 20s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/5454266652914dd9a47f1ac7bffb7ed1 : SUCCESS in 1h 16m 44s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/3f834371fdf14214bb51f94e7452ff67 : SUCCESS in 1h 59m 44s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c8609ab19e984db9b368a890f66407c6 : SUCCESS in 1h 31m 38s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/5cbfb89faae14ff485af924c937add40 : SUCCESS in 1h 04m 15s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/b6ac31b6588146528ccb864adb9241f4 : SUCCESS in 1h 14m 02s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ed08f72a896e49eab17624f992fc7d64 : SUCCESS in 1h 23m 58s","accounts_in_message":[],"_revision_number":1},{"id":"ceb3a1dae4f08eb69b08acbc145b9a3ee0767495","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-02-12 14:20:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"418b66d465fd23490fe719f87d46a141f1734661","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-02-12 14:20:37.000000000","message":"Patch Set 1:\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":1},{"id":"d27f80fe71533453ce1b60f751963a391b110f49","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 16:33:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc191a9eb8b049cd842c1219c3b2ae31\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/76c6aff962ed4f27a961d33576de4086 : SUCCESS in 1h 06m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9a4be6933f7544a5a4665af01736798a : SUCCESS in 1h 06m 02s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/51edcdfbbba745ab9f3c75aeef867a80 : SUCCESS in 1h 31m 30s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/9803f15366f644aca65b2c4f83e37dfe : SUCCESS in 2h 10m 35s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/187b6583ef534d4b9961e72e922af93d : SUCCESS in 33m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6fbd863bcbf4de0b1f7121ae9e8f6fb : SUCCESS in 24m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/215289fe75864e1593af225188fdcf5c : SUCCESS in 13m 34s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/4553f9557e9f44f98f12a393032353c5 : SUCCESS in 1h 05m 07s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/b17762c9bb6147bfa6ef70be7daff897 : SUCCESS in 1h 14m 47s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/bc7558f27d0c440f895ff9a7941b738b : SUCCESS in 1h 16m 24s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a27c9a04b3394e79ac6aec7c15ee67fa : SUCCESS in 1h 16m 02s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/272a04f0aab74742ab15decae228d434 : SUCCESS in 42m 01s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/832f5346bac54504b4acbada5f1e22ac : SUCCESS in 1h 42m 46s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f878b557043b4d1f8fb5f3b76481f73a : SUCCESS in 1h 59m 57s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/f5a94bfd0c9746639c4883ed87f09822 : SUCCESS in 1h 06m 40s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/af06a86d3de44679b21ffb8fece03920 : SUCCESS in 1h 13m 01s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2388c2c296464e5e9465fdbecfff16db : SUCCESS in 1h 45m 07s","accounts_in_message":[],"_revision_number":1},{"id":"ccb321564a656792c54fa8aa7ee9b295cc0c52a5","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-02-15 01:06:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0e0994e44daf5ac76609f223bac812ab99b46662","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-02-18 14:40:50.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"d548f2000da04e471a13ce08cef35093896fb55f","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-02-24 14:14:52.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5437d959afc4acfb3ea4113b67adf8a6b1e2d7fb","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-02-24 14:54:15.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7cfee322a46dede7dc2d02db1ca884d82efb37d4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 14:55:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"238453619d6d1a8be0f786c547edfa89ac00e2a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 16:47:53.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c4f2412ee43945e39e4b8f930eb5c2ce\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/2ce9bbd44b7743008076cc6739ff8a58 : SUCCESS in 1h 36m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/52e61566167b4cc5b7e281a8c5eeab14 : SUCCESS in 36m 04s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/12ed5b9f909147fcaffe57ed85fa215b : SUCCESS in 1h 24m 56s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a18f45de8c56400093d90f4fe1315d48 : SUCCESS in 1h 52m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d427b2bea0f540d3b3ea6a3524f441c5 : SUCCESS in 20m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/464e459644cd45469d18d30622c1f6fb : SUCCESS in 6m 53s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a7e59c053b01491f8436dd9fd5180a47 : FAILURE in 36m 10s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/9852c1e3f7724d5db9e1256ce15b135a : SUCCESS in 1h 06m 26s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/43a4a883a73c49f2b6275340f06d435a : SUCCESS in 1h 15m 34s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/fc12a17209314923a9625097ab7b4598 : SUCCESS in 1h 21m 02s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/587310263f2042d397efd1f982885274 : SUCCESS in 1h 03m 21s","accounts_in_message":[],"_revision_number":1},{"id":"c4c70d39d9d39e8219b584f31fd68bc653fee76a","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-02-24 17:34:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"01727b478778bb3b47f1d80beab179b67a7a1bbb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 20:26:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/046224680c4549138003955a00cf8e00\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f55fff859e7e4e06acec9b967752bd6b : SUCCESS in 1h 29m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3264d1a4d9c8415d895e6c8ad082a442 : SUCCESS in 51m 46s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/9d7a4bea327f49eca352e1bd8a478553 : TIMED_OUT in 2h 50m 06s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/00791bf155f243cdab990ca5a43f126d : SUCCESS in 2h 03m 05s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/27408146f77c474c8902847aad6a1f88 : SUCCESS in 30m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d2bf10c50d944d0a178ec7d1573cb6b : SUCCESS in 23m 39s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7f4095da18ef42f2aaf27f5b11e56cf5 : SUCCESS in 23m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e935498bda54d58b369117c4899c381 : SUCCESS in 12m 46s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/afb25f4165be42c08c5620a6ee95d27d : SUCCESS in 34m 54s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/41cfbfc32aee4ffe98f66f227cacaad9 : SUCCESS in 1h 12m 37s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/fae3a24f47b54750a0ea72e84e9c6ae9 : SUCCESS in 40m 53s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/2659df0dbd46410385506c3455ff9f6c : SUCCESS in 43m 02s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/7b9f7ad019a54c40a8e6ebda1d75ce0f : SUCCESS in 1h 00m 00s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/0e67d57193cb4ba0890a2405c1fbfbe6 : SUCCESS in 1h 59m 25s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2c4cb3382ce94c379dcc537f75f34742 : SUCCESS in 1h 45m 13s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/37b78abaacc146f58e05704396561096 : SUCCESS in 1h 29m 27s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/1ed4f65d6fe84a19a46e3616eeb3f136 : SUCCESS in 1h 16m 29s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6ac4529f8e2f472b887a1df15fbf15ff : SUCCESS in 1h 28m 14s","accounts_in_message":[],"_revision_number":1},{"id":"2c383d0ca337dfa9decfeb39e88a59189523e69c","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-02-25 00:17:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"309acc5220fd0610c77b9b80255f975da9755146","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-25 00:46:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13eebe1f46144be0be587720784631cf\n\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/3539b8d249014b94bbb91ba5927c6905 : SUCCESS in 26m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7ab368369453e9f111c04f72d313865b956b2043","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-25 02:36:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1fa501faae7b44dbaa699c43cb06ecad\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/2ab646bdc3594d6c9025e358fc32486e : SUCCESS in 1h 06m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/66d93fe00a35492f859a40a2fb172daa : SUCCESS in 37m 05s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/351afcdea5b24a76963d32be6474d7f0 : SUCCESS in 1h 15m 04s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/9baf64406cc24d4a828da53f86fec9e5 : SUCCESS in 2h 16m 19s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/96065eff7b6c4bc1a0a298d44c8b4643 : SUCCESS in 29m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fab07a30ac084d97b7ffedce6ee7d2a1 : SUCCESS in 10m 27s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/52ecbae7b60b41a682d8c98a1dd09ff3 : SUCCESS in 26m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a06bd10fe40f4b95b353ed0f2b58836c : SUCCESS in 10m 46s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/8e8d086f3e414fef80fefde2c48e1249 : SUCCESS in 1h 00m 22s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/c735ed125d1c4b179c3b9d346910c177 : SUCCESS in 39m 51s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/44e943d0d0bb410c8b9ddc1a8957481d : SUCCESS in 45m 55s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/fbf60f9e32a646cfa568fd9a5b856a8a : SUCCESS in 41m 45s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/eb463b60eaa1455b903f49ec0e165fe1 : SUCCESS in 1h 10m 17s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/52c0632d870b492894a123ba57897048 : SUCCESS in 1h 07m 32s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e907a6147e16448484a2fe240a0a1bae : SUCCESS in 1h 03m 56s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/ff7bda311d81484d9763dbd3bd6776df : SUCCESS in 56m 25s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/90d4686531ff41b4b5f8612c3f6532f0 : SUCCESS in 1h 12m 02s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7eb491451e0844d29d4264ea6fde002a : SUCCESS in 43m 08s","accounts_in_message":[],"_revision_number":1},{"id":"7b49fef131f6d1ee2c7efb6a2b959435e587a1ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-25 02:37:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"77caa47eca90ba09700ee5096f8062be13051269","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-25 04:49:45.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4cf84c73d3a64e4b843878a6d0bd73e6\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/831eced357a546b4a1b3ab8d12d4bcf4 : SUCCESS in 1h 30m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/906827a206bd4a00911e8a9eccd87f01 : SUCCESS in 1h 04m 14s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/8926a34aef2645cc9dc31d044b1167e2 : SUCCESS in 1h 30m 09s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d04b69304b754ddda599d287a2145502 : SUCCESS in 2h 02m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4ce3f4c190748b0bcb8b1c572f5f006 : SUCCESS in 26m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c10b70dcc714e40ba1718894425496f : SUCCESS in 12m 16s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/33e09636e494479d9f6d27d1e9f02458 : FAILURE in 43m 47s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/5b0b5fcdfed84106a5e1a39fa4006062 : FAILURE in 1h 10m 15s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/b655605e1e794f219e6f5bf235b4ea94 : SUCCESS in 40m 53s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/6975672eadb94057afb08eacd63a0b91 : SUCCESS in 47m 48s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/e6df7d3d0ad94dbea20519605352656e : SUCCESS in 1h 11m 33s","accounts_in_message":[],"_revision_number":1},{"id":"04071902ab56f570aa8bf4535c85f1149a0b68bb","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-02-25 05:04:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"19086be55a92b144b666fc3e306ab1874bb81c52","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-25 05:41:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ab83df14eb042f39a348eaf1f5ec92b\n\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/0e8b7de7eb8c482f9e8e3867df90715e : SUCCESS in 35m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"004196a36282331c3fa5c8c296cddb275d98db81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-25 06:53:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/51aa692a89cc4f7ebcd9e8abb2dc1cad\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/7ceb43e460654192af907c6e825e01c6 : SUCCESS in 1h 06m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/85325aec59c345978d13a39c0a3df9dc : SUCCESS in 33m 41s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/87fae5242cfb43ac9041c999ca6bc05b : SUCCESS in 1h 29m 00s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/0aa541eee4794583bc2a9b071fc98b9c : SUCCESS in 1h 47m 32s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c1f9bf7a83e2414684344dfbab417cae : SUCCESS in 30m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f2a0af8ec8e4172943512e6c71faeda : SUCCESS in 24m 38s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/11dcd003da8f496b8eafd294c71cea89 : SUCCESS in 35m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb8e77c5718d45de969251316532124c : SUCCESS in 7m 26s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/91e55dbe989f400297f104eec2681de3 : SUCCESS in 55m 10s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/286715659ec544d3ab5f113926c22437 : SUCCESS in 1h 09m 54s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/4cc0ffb90b3f42e6b2c7d71782e44ae4 : FAILURE in 45m 14s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/8b252329a3d245d996efc4edbc5a7d19 : SUCCESS in 40m 05s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/eb330682f2b848299ac89ccbcb49fc72 : SUCCESS in 1h 13m 01s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/0db292b44afb44308a9d5c26b962a0f0 : FAILURE in 1h 14m 54s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/5a0fe77611c846628d9d6191e80ebc2c : FAILURE in 1h 27m 40s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/3197aee8e4b54e4bab5ba636b85ee1d2 : SUCCESS in 1h 21m 07s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/28e7fa84c82e49a8834f989b4e179454 : SUCCESS in 1h 11m 05s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0cd45281d11b434fbf9b47f7106331ba : SUCCESS in 43m 44s","accounts_in_message":[],"_revision_number":1},{"id":"8ab17d2a3710845715f748756fe5ffeaf8647fda","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-02-25 18:23:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8e83a0621634e630d1340749b7a90cc0ad3f72cb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-25 19:02:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/448bed5e61cf4c43b45c55602098f4b6\n\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/22e57af1d002438c8b8ef994b5f79b4c : SUCCESS in 31m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2496875b69de505347eff4703cba99c065af9151","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-25 20:22:26.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/51aef3aaeb364be88a5218589b2e961e\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/2d421713654f4890af5bb3743425bb49 : FAILURE in 51m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/938967d511494da1915f1bba068cd795 : SUCCESS in 1h 02m 38s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/1f64e99fdd8f4d2f9c1a1b267e96a3ba : SUCCESS in 1h 32m 03s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/157655c85c234a7b81849d4377fd24ca : SUCCESS in 1h 54m 49s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/bfc8e6cf77c54cc4b38c0ff5ad8f0c80 : SUCCESS in 36m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a5d87ac7286424d97450f7b493b88ca : SUCCESS in 24m 23s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/46c4d107f7934f52b2ffa2664cab27dd : SUCCESS in 22m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4489df8f250a47f1a608e5c12550d6fe : SUCCESS in 10m 59s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/be9454e23e504047bc328aec535de751 : SUCCESS in 59m 43s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/c853bcce04ac4ceeaac7e1eeaded67de : SUCCESS in 45m 27s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/4b24df811d9049559711b8938f416a8e : SUCCESS in 1h 21m 24s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c4c88bc5af8043479d4b01866861d872 : SUCCESS in 1h 11m 16s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/26a1647be0174f9fb36b3b8226c50f19 : SUCCESS in 1h 10m 20s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/8739995598cc4fc4ab98513584889dbf : SUCCESS in 1h 48m 09s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7cac557bd73049f58f13ce9c99242872 : SUCCESS in 1h 36m 04s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/3db8181966d6458da3e6248d2b153aaa : SUCCESS in 54m 31s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/2d03caac0c82413d9c65b157dc55be8d : SUCCESS in 55m 21s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/68b15a5b71a8432886bd0439a1974ca9 : SUCCESS in 44m 03s","accounts_in_message":[],"_revision_number":1},{"id":"e262aaafa8f0fbac90794506e8bf692012f6e3e1","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-02-26 03:04:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4dbf7ef6ac1de41e2179d4ec8c7a5195f9da63b1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-26 04:46:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c847eb2bc9004660937e20eb83998e3f\n\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/fb23567524eb415b85323f9a596720ef : SUCCESS in 40m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"208009e15f4ac09969fe193e6608d119636909e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-26 05:10:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f87d2cea35674ae2b1ddb0067a0666d8\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/73394ee2f40146ab9f5c8365d6bdf1ae : SUCCESS in 1h 27m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e3bbbe0d59624e3d9ede22dba7fe652e : SUCCESS in 53m 20s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/0e9e569ff8d24811be032209a24d7910 : SUCCESS in 1h 38m 27s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d0e05114616b417a8e5d57356be62f60 : SUCCESS in 1h 49m 24s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fc62a439452f429d934b450b0f821284 : SUCCESS in 33m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1a4a9f22221442f80ea3c06ceca41c2 : SUCCESS in 11m 05s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b528fab0d21c402a92afa2c4ccc5621f : SUCCESS in 33m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c69e0190cd945bf8fdb8a34725c0d46 : SUCCESS in 11m 58s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/3e645d9d680041af816af1d479cc392b : SUCCESS in 53m 15s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/17f4bccc46ae4a94bb7b4395e5a0484a : SUCCESS in 55m 49s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/3a7ccceac85749f1b60184a8514c645b : SUCCESS in 51m 18s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/2e5a079234574cf8ae9fe9f87e10495a : SUCCESS in 1h 11m 24s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/69fd810bdab54770b44ee76da2bc2ec7 : SUCCESS in 42m 36s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/02786d09aa07443bb1849d394f6a7db2 : SUCCESS in 1h 56m 34s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f7e4346e47b74854a7b46234f0b92751 : SUCCESS in 1h 00m 28s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/5ec0f3f656004dcbacd57c8139fd2dc9 : SUCCESS in 1h 04m 23s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/e0b953f7c4b14be990000182bbd30f02 : SUCCESS in 54m 16s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7bbdb0f5972c40baa49b8d7a6a8f35e7 : SUCCESS in 1h 31m 23s","accounts_in_message":[],"_revision_number":1},{"id":"2b6e31ad4ae65f2182509d6dd9928dd6d61d5c7e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-26 05:11:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"be4d89eb66a8f7a4450e6273718dfc6744ca7f5c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-26 08:35:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a401457af0a44869762a61e8321605e\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/5c699669cbbc4b8ebde467371a47d1e2 : SUCCESS in 1h 07m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3c871188d98946eca5532370fdac1544 : SUCCESS in 1h 01m 50s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f632f0727e4d43df973a8dbd20350672 : SUCCESS in 1h 22m 30s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ea36a7d4a8014b99b02d80663c44d920 : SUCCESS in 1h 48m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b6371885bd0429f838e4e9b0c89b0ea : SUCCESS in 24m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10ba9b95be504d8b81cf0d12b3bc3b7a : SUCCESS in 15m 01s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/30ebd9e878804453b4b19beb925a8f61 : SUCCESS in 1h 03m 57s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/476859e5b8264146bfcd2d1e990d0ef0 : SUCCESS in 2h 06m 03s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/6afe65a3d38c4cb79308482145fbe37a : SUCCESS in 1h 06m 53s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/773647c49db7495db5b5ad07cae2f0da : SUCCESS in 1h 05m 42s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a7697468bd1449169d9f0fbb30dc9ef8 : SUCCESS in 1h 08m 32s","accounts_in_message":[],"_revision_number":1},{"id":"797ecd6507658231564e01b3d71ef6ae14035146","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-26 08:35:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2d58c4b462010b054abf3cf3260824e6bc1613de","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-26 08:37:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4fd7dccd47dc470aaf47f5794973052b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2d02523b0dc48bca21ed047ba7b6c31 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f478408a7498614cd2d7264303f4e4a7eb530f2d","revisions":{"f478408a7498614cd2d7264303f4e4a7eb530f2d":{"kind":"REWORK","_number":1,"created":"2026-02-10 03:08:35.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/04/976204/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/04/976204/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/976204/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/976204/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/976204/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/04/976204/1"}}},"commit":{"parents":[{"commit":"e1836382bde820fd4ca6819f2533247b651c7ed4","subject":"Merge \"Enable ovs_create_tap in ovn multinode jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e1836382bde820fd4ca6819f2533247b651c7ed4"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2026-02-10 02:57:50.000000000","tz":0},"committer":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2026-02-10 03:08:33.000000000","tz":0},"subject":"Add integrated-gate-networking testing in neutron gate","message":"Add integrated-gate-networking testing in neutron gate\n\nTempest defined the integrated testing template with some set of\nconfiguration that helps to avoid any breaking change across different\nprojects. For networking, integrated-gate-networking template defined\nthe necessary jobs for networking integration testing.\n\nThere was one valid concern in past about duplicating the grenade testing\ndone via grenade base jobs present in integrated-gate-networking and the\ngreande jobs defined in neutron repo. Tempest depends on change solve that\nconcern by removing the grenade and grenade-skip-level from\nintegrated-gate-networking template\n\nThis adds integrated-gate-networking template to run on neutron gate so\nthat we can run other integrated jobs on their gate to avoid any breaking\nchange.\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/976203\n\nChange-Id: I1834642e9c8de971f560c50d3a0a8b76e051f41b\nSigned-off-by: Ghanshyam Maan \u003cgmaan@ghanshyammann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f478408a7498614cd2d7264303f4e4a7eb530f2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f478408a7498614cd2d7264303f4e4a7eb530f2d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
