)]}'
{"id":"openstack%2Fneutron~976230","triplet_id":"openstack%2Fneutron~master~I4b4768e0ee3c607dcc8511d7f1b5764d1fd0a63a","project":"openstack/neutron","branch":"master","topic":"bug/2002839","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-02-11 14:05:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4b4768e0ee3c607dcc8511d7f1b5764d1fd0a63a","subject":"Fix ``ChassisAgentWriteEvent`` match condition","status":"MERGED","created":"2026-02-10 11:15:12.000000000","updated":"2026-02-11 14:06:39.000000000","submitted":"2026-02-11 14:05:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"976230-bug/2002839","meta_rev_id":"338bc699331886d250e4844a96de9df2697388fb","_number":976230,"virtual_id_number":976230,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-11 14:05:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2026-02-10 11:43:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2026-02-10 16:23:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-02-11 04:41:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-02-10 11:18:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-10 11:34:53.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2026-02-10 11:41:26.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"CC"},{"updated":"2026-02-10 11:41:26.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"CC"},{"updated":"2026-02-10 11:43:30.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2026-02-10 12:18:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-02-10 13:11:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-10 16:23:17.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"02cb0da2d21845748861713f6db0cdca8babcc3d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-02-10 11:15:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b02b438037f62aee6e3e0c07d73c8eb4fa1490f7","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-02-10 11:18:38.000000000","message":"Patch Set 1: Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"8cb7d68864594fd992151dc2a72cf25919e5c28e","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2026-02-10 11:43:30.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8b7ac06a6a9b7c5892f6b60cea96f70b9e344f5e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-10 12:18:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe91b43d53f04cf3b3b43eef6853e0aa\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/48f575bff66c4d38ba5fe1b8e8b8c4b1 : SUCCESS in 1h 01m 41s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6ac5eaee3cac4fb7bb26812b3a431560 : SUCCESS in 21m 51s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/fbb4ed20d80748a49de099c11bb9c44b : RETRY_LIMIT in 3m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e63ada2687029e258d8850d6ab8a07fcb61b7dc2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-10 13:11:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/86351cc00b0746f993acb2f93f877466\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a1447e5f957e4807820b6441fd1f301e : SUCCESS in 1h 54m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a4565612513a4a109d50fa7fe688499f : SUCCESS in 43m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4130698c4d042318585fc8980cee53f : SUCCESS in 21m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/26f266389e0d48f0990306272dda3b3f : SUCCESS in 33m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/67faddb367e14dd499dd757dd55842d4 : SUCCESS in 19m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1b040ad5682418aaa129530a00646ad : SUCCESS in 13m 04s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/fb9721255e9140ea88cbb47f7631a396 : FAILURE in 8m 57s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f9093655069841a797749fbd773c3dd1 : SUCCESS in 42m 14s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/35e6abc2153a400fb915126aa8c3d95f : FAILURE in 11m 09s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/ea7504d27717470e81508b5d2b1591ac : SUCCESS in 1h 06m 30s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/70710848d0b545919ce9151733899b6e : SUCCESS in 1h 31m 29s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/0a25a9d401eb4a72bddfa1384c600577 : SUCCESS in 1h 01m 48s","accounts_in_message":[],"_revision_number":1},{"id":"a7b8aa261a92f30b9c349515ef878f16c0d3bdf4","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2026-02-10 13:33:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a564ef665ca74f13d337933e7e5ab8a927b48404","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-02-10 16:23:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7e741a641a2932f75655be86053c7ff9a336d538","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2026-02-11 04:41:49.000000000","message":"Patch Set 1: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1c36d6fc423dae4f06f282ae764839c8fac56b96","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 05:39:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2a56469b4ad496f8543511645da8b36\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8c1027dbd8a64fe5b9ac1c68d8b572b8 : SUCCESS in 23m 53s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/010013a900224b3089563907b5ff0efb : SUCCESS in 55m 41s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5c747b6fac5546dbb9e0ed070fb41f44 : RETRY_LIMIT in 3m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d9e1e8f36ac4c4b7c8ce589061de8a7b2732790f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 06:32:15.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5fca4d627e374cbdae7e253045d2d37c\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/267a7738cc3f4e4fbce760e50fa1f9f6 : SUCCESS in 1h 49m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2a49345281374a75bc905382c0195511 : TIMED_OUT in 1h 21m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46a3eebd21974fab941f1c4eadc4eee6 : SUCCESS in 25m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3e174691995c475c83e8145aa1fa1d1d : SUCCESS in 25m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f49003faa6aa4cee9d8e5f27dbdb3d94 : SUCCESS in 24m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbd5637d82144665a8014d3e6fbdcd7a : SUCCESS in 13m 25s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/a759b678c7b64506b58ce820cc82c227 : SUCCESS in 1h 16m 13s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/239b1ff3a2c946f68314b631b81c7730 : SUCCESS in 1h 03m 09s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/6fd19ec64a0c46b3afcbcef01697703f : SUCCESS in 1h 08m 04s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/c49b4138737b4fb987c3a70552601b50 : SUCCESS in 54m 23s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8cca467643c6494aa307118576863b38 : SUCCESS in 1h 28m 43s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/15cef6c893664bf092cc3570ac5b46aa : SUCCESS in 1h 00m 35s","accounts_in_message":[],"_revision_number":1},{"id":"22295acad23f150b758867226f4c201d1d23b24d","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2026-02-11 07:05:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bf6286b6a068cdfc7ecef1e22c063392da388965","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 07:39:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e41f285f422482b8e89120ec2d7ae46\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8d49cf333865451295065d9a92bf24de : SUCCESS in 31m 12s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e43607d0db8f4d128ea9fde8abc6e521 : SUCCESS in 26m 51s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f33235c94145498186dc4d03649436b6 : RETRY_LIMIT in 3m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7d8296986cd2f298e00d58e97e249f9a805283c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 09:13:04.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6e58ed0ec94146499cffc934a7f712a2\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/7776c5365bd64f398f89331bf9719985 : SUCCESS in 2h 06m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5900cd31219644dbb6261cf91bcf3aad : SUCCESS in 41m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1b8ed3ee86c4515b758cb304d166d98 : TIMED_OUT in 30m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e6e69e6fcd834e6c863d3645b3a4d373 : SUCCESS in 28m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ef1ad6a017be4a378c658d05f9863fbc : SUCCESS in 15m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1bfc9e35e0d74bd0841331dec5046183 : SUCCESS in 11m 30s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/ba815ed536f044d89040695be09b5b5c : SUCCESS in 1h 07m 09s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/14bb4ee408d7467f87f5d20e40a53ab5 : SUCCESS in 45m 42s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/ca73fc9e83ad4d7a94879103ba0e6117 : SUCCESS in 1h 08m 15s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a80cd28df7dc4365807e3e9225eadf45 : SUCCESS in 1h 05m 34s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2f910ed02f82486d9646fe3a4098772f : SUCCESS in 1h 20m 36s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4c0b7cf3ffe14510a869fbcdaaa87a0d : SUCCESS in 1h 11m 30s","accounts_in_message":[],"_revision_number":1},{"id":"d5e043ce262b50e6ea69a95eace80c84458fff94","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2026-02-11 10:30:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d034cbcd49437700279c4186162b5ba89a4bcac9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 11:34:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d60aa56cd6ce44aca10a2d6f4a612343\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6fb947b4d21342f89b439a882c64fc26 : SUCCESS in 1h 00m 05s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6a99f3bdf6424bdb9e1212d6bc159024 : SUCCESS in 44m 40s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/44a915c9d6c548d98431f59dea516ab7 : RETRY_LIMIT in 3m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ebcb40b59a6dfefe42c3f19c4874f2283f4b8749","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 12:20:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a3c3f1d23c34dbd90c052144de4dea8\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/8cf7874660464edc8157b5a09239dda2 : SUCCESS in 1h 46m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/462302e4f74f4808b8be170b564f93c3 : SUCCESS in 1h 18m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/849ce55966b045369560f6b14c2a3b1c : SUCCESS in 12m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/41a96a3c79d24abcb6d07973ab43b95d : SUCCESS in 32m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9d80880aa30d4624a30ad48bf722421a : SUCCESS in 24m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51991b8847e24e7e8d54072264cfa530 : SUCCESS in 13m 14s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/91c2e7e1cae64ab282d4f81dcf92b92b : SUCCESS in 1h 14m 49s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/92116135b49c49a58a74216b62d3895c : SUCCESS in 1h 11m 54s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/411710b43f354b269de084a1df9e9902 : SUCCESS in 37m 11s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/32edac2c2c694abcbcabfff51a8f4a98 : SUCCESS in 1h 04m 57s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b83a196ab5f342eeb27d114191adc2ac : SUCCESS in 37m 11s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/fbcac76ee3d04aa1aab130285fcf6b25 : SUCCESS in 1h 20m 08s","accounts_in_message":[],"_revision_number":1},{"id":"2641824fcfbe1246e39e9da8efc64ad787dcb693","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 12:20:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a8dc86a58de4d0440047a55aaa6c5f1396e2f6e0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 14:05:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/720586b9e2e14645801f449df69bdcdf\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/6ad80eb07b0e44ba950f6f98a7df9da2 : SUCCESS in 1h 44m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/837467806d2a41b0a26b98bc0454592b : SUCCESS in 24m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b2809adbe71e42b893c87da4a94f27d9 : SUCCESS in 23m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/af2e0edf08ba4dcf9db2e5ccebd3d7b0 : SUCCESS in 15m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea2227d4c72648f7a79deea5dd4c7e00 : SUCCESS in 12m 37s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/ada65c8d77374e33b63649501169ef91 : SUCCESS in 41m 09s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/f1649a0b08044b658df6167525946e60 : SUCCESS in 1h 14m 23s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/4fe799b1aa77466cb0468e0316f77029 : SUCCESS in 57m 21s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f023c02d55964e48b3aedb28a9dd8fdf : SUCCESS in 1h 10m 39s","accounts_in_message":[],"_revision_number":1},{"id":"6c4736a9286dc7824e81a61ed8041c205c49e7ef","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 14:05:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"338bc699331886d250e4844a96de9df2697388fb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 14:06:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d33a625286514bf9a7a60e65abe119fd\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/283b66858ecd4625a6a08ff6a3e8e789 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bf9e039e2fe5beb84b764242e4d293e90b56e13a","revisions":{"bf9e039e2fe5beb84b764242e4d293e90b56e13a":{"kind":"REWORK","_number":1,"created":"2026-02-10 11:15:12.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/30/976230/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/30/976230/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/30/976230/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/30/976230/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/30/976230/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/30/976230/1"}}},"commit":{"parents":[{"commit":"e1836382bde820fd4ca6819f2533247b651c7ed4","subject":"Merge \"Enable ovs_create_tap in ovn multinode jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e1836382bde820fd4ca6819f2533247b651c7ed4"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-02-10 11:11:17.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-02-10 11:11:17.000000000","tz":60},"subject":"Fix ``ChassisAgentWriteEvent`` match condition","message":"Fix ``ChassisAgentWriteEvent`` match condition\n\nAs the method description says, the ``ChassisAgentWriteEvent`` uses the\n``Chassis_Private.chassis`` information to update the local OVN agent\ncache. If there is no chassis, the match should not pass.\n\nThis method was updated in [1], during the backwards compatibility\nremoval. This condition was incorrectly refactored.\n\n[1]https://review.opendev.org/c/openstack/neutron/+/904238\n\nRelated-Bug: #2002839\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I4b4768e0ee3c607dcc8511d7f1b5764d1fd0a63a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bf9e039e2fe5beb84b764242e4d293e90b56e13a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bf9e039e2fe5beb84b764242e4d293e90b56e13a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
