)]}'
{"id":"openstack%2Fneutron~976300","triplet_id":"openstack%2Fneutron~master~Ie05e0dc4fff9327fd0a52cf48028a923c4e82330","project":"openstack/neutron","branch":"master","topic":"bug/2111276","attention_set":{},"removed_from_attention_set":{"5756":{"account":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"last_update":"2026-02-17 23:38:06.000000000","reason":"Change was submitted"},"8655":{"account":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"last_update":"2026-02-17 23:38:06.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie05e0dc4fff9327fd0a52cf48028a923c4e82330","subject":"bgp: Small fixes","status":"MERGED","created":"2026-02-10 20:08:09.000000000","updated":"2026-02-17 23:39:07.000000000","submitted":"2026-02-17 23:38:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"976300-bug/2111276","meta_rev_id":"7d39282546100eda342312a1ad9179af4474c411","_number":976300,"virtual_id_number":976300,"owner":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-17 23:38:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-02-17 13:51:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2026-02-17 20:10:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-02-17 20:10:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":1,"date":"2026-02-17 19:21:01.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"}],"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-10 20:39:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-02-10 22:12:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-12 16:15:20.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2026-02-16 14:21:58.000000000","updated_by":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"reviewer":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"state":"CC"},{"updated":"2026-02-17 13:51:55.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2026-02-17 20:10:35.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"6601b24d72b6d6ac21de98d682bdc99dc0ac5b8b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-02-10 20:08:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ef6f9882b815562c0f5921cd384ee8ce570f3ffb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-02-10 20:08:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a15cea085b18f9cdf40bf64110ecbbd98ef0df73","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-02-10 20:11:13.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"429eef40a8147bf2c91f68581df40720318d43ae","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-10 20:39:48.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5cdacf14608e445f8ba80f080480e566\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/dc5bab73081c45449f890247fd4e1b39 : SUCCESS in 26m 46s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f51ed60244ed4ec3a6f0e68aa77098de : SUCCESS in 21m 33s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/452ef5b6e4ef4daca8fa20dafd781fd3 : RETRY_LIMIT in 3m 44s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2148b8d02c5d233ff85ce28f1cb20c2cdc3dc294","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-10 22:12:37.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1591a41f0781461c87ab7893f290b79b\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/a6a0c80da03d48f9b0f2c8febc568817 : SUCCESS in 1h 20m 14s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/42e3aa9de0cc483fab24b71c3fb27816 : SUCCESS in 1h 47m 03s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/876bea8e54ba4c0abc6526cd972d1bdc : SUCCESS in 30m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/09938095ce424b5db3926a391a3433b4 : SUCCESS in 1h 01m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/acc04bfc2fa04ef282c4b518f91e8c24 : FAILURE in 8m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5b4a4de313b840f899f67bd936cb3a5a : SUCCESS in 22m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cea29949a1e44825b2ba4262a6cac6d9 : SUCCESS in 23m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27f3e6cdefc044f7a79ce3df4dd540c1 : SUCCESS in 12m 01s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/19af73bf1e3c490fb4134026eac7f0e0 : SUCCESS in 1h 06m 41s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/6bfe3f4f80b141369bc8df17a6a1d87f : SUCCESS in 57m 08s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/efdef96b47ac43dbb25dc2cba56194d6 : SUCCESS in 41m 08s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/db602d8a2fc14da1bf8db1dc97150348 : FAILURE in 7m 47s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/b515ef333b164edc8e651932623c1040 : FAILURE in 18m 52s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/592cbd5b6f5a4df4b197d9e77f66d167 : SUCCESS in 1h 58m 38s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6e128ee24e1e43969d3b492fb151d8c8 : SUCCESS in 56m 25s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a70f5baa785842d6bf5015f8c6f134d6 : SUCCESS in 53m 07s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/854b8b7bd6e64f549f5caa0a9d6b2d0a : SUCCESS in 1h 18m 32s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/af5022db1a24409fa933c87c37c38811 : SUCCESS in 1h 23m 26s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/0e37834909c347d78e27738093924523 : SUCCESS in 1h 23m 13s","accounts_in_message":[],"_revision_number":3},{"id":"819815611b3af62cc35826deb3c2ce2ea726225c","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2026-02-12 16:15:20.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"0ef74b426732db7f9fd48a83e53187c45b3ff5c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-02-13 15:13:10.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"511e4f315108f07b309527886e4762c91f610ec1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 16:05:35.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ce874f4553d407f821a4482d6cdb0c9\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/067d4132cbde4373b49160e2bb77b02d : SUCCESS in 25m 31s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/546715c316df461288a66bea588d26e7 : SUCCESS in 50m 15s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/583ec4597cd141429d2b20a74e771dd1 : RETRY_LIMIT in 3m 41s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"8eca7212feabf43a91529ad8b9747bcceb733e72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 17:22:32.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/933e337f90b04ee1869546bc43d422b1\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/4ecb35adc8834bf2b3e440c45f8e6a6c : SUCCESS in 1h 18m 14s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/316eeef1f7264dc2b436f6a8b9869cfc : SUCCESS in 1h 39m 00s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1dac7f9e986b4ce8be3c0203d9bf689d : SUCCESS in 32m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cdd520fc9f8940c78a554908bffea3ce : SUCCESS in 36m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e14b03d422f14a81a0ef5cf43fe1b8a0 : SUCCESS in 21m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ffd493d8d4d84ae7b12d049d9d63d370 : SUCCESS in 20m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/86ed3b0d57f9452b84ac3f97d1935f29 : SUCCESS in 15m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d3126feefaf4a11a7fe5a425c08029e : SUCCESS in 10m 26s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/93bd0156e5e8413fb5875b3d72739d17 : SUCCESS in 50m 00s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/9f9131acc08e4a9da2ac63c3f71b4138 : SUCCESS in 38m 34s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f2adae99f93649bd87966894fbfc3cad : SUCCESS in 1h 08m 06s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c1c872dcdae64400a65689be3ee9c562 : SUCCESS in 51m 43s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/2fc71807f85742f99fef547f6ace1727 : SUCCESS in 1h 30m 22s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/a68125daab754362af7f45e169ea382b : SUCCESS in 2h 06m 13s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b378616ac61e40079da08bb319c559fc : FAILURE in 1h 53m 23s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/1bb3ab6866b64954b10f1dbbcedaf80d : SUCCESS in 56m 09s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/e864d6d1784b4f44bc315fb7cab2ee38 : SUCCESS in 30m 40s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/54f0a19124924f39bbe87e603ad69905 : SUCCESS in 1h 06m 45s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/0ccdb44690bc4cb9a5cfd57554d535f1 : SUCCESS in 57m 32s","accounts_in_message":[],"_revision_number":4},{"id":"5c7f4e0d78f0797e0ffd4fc1a0f3b6a9704c3468","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-02-16 14:21:58.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"0d4596cc24beb1cf969329abec2e4852deee5748","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 16:04:02.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c934e63382bd41a185886af8a30b36ec\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/2610567865c34012a9a5523a7645ff24 : SUCCESS in 1h 30m 17s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/019eb364e4984b55a0a12f73013f3a2e : SUCCESS in 1h 40m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/37651e5fa48d4a06b3b56fa2d7d9480a : SUCCESS in 31m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d615d33921a04d7c8afe3ab5af18cf8b : SUCCESS in 45m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7fd262ac2724265a255288459fb40ef : SUCCESS in 13m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a3d1eabda93b467b9f4586729a0477ee : SUCCESS in 25m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bbf11a4d51d14da890e6295347f02887 : SUCCESS in 15m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c955ac2edd34ea9be7c49452edbf160 : SUCCESS in 11m 29s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/5c6039e602554180b2e1e94335fb42a4 : SUCCESS in 1h 00m 11s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/45f489674af44213bbd977c47ad8489c : SUCCESS in 1h 09m 50s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/0aee911024e94b188206cfe0c8696d0f : SUCCESS in 1h 11m 16s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/7b9b6937be6148f7b753346d4236a1be : SUCCESS in 41m 05s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/06ae4e11edfe45fd8fc514d4260995f4 : SUCCESS in 1h 07m 51s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/b2f0b3a096f4473ba6ebb45940a92b84 : SUCCESS in 1h 09m 30s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b12592d4a6d942e9b468f035c537af95 : SUCCESS in 1h 05m 41s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/0afdfac5351e4e76af863e6ddc1919cf : SUCCESS in 47m 40s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/a8780cb217c64d4e90b8402160886100 : SUCCESS in 1h 04m 36s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0ddd37987a7d404c8b14f050c83c05f2 : SUCCESS in 1h 29m 33s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e8983f2454d94d41bdc018764654fa46 : SUCCESS in 1h 07m 53s","accounts_in_message":[],"_revision_number":4},{"id":"ce9d3f90bedc78fe00f618ed787da4468236df90","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 19:44:48.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c8d44b949fbf40b2a1abc1cc14fd0747\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/450e230addd24cdc9fbbd1c9cd6c7370 : SUCCESS in 25m 44s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e857d05d03604fafb4188154ab8c768e : SUCCESS in 27m 35s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8c1ce5c527144c2392a5e439c712877c : RETRY_LIMIT in 3m 37s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"4e243231d9408290e7dc88e4ad249bc24e0e716d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-02-17 13:51:55.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"ef540b2d349e9648e349ac866f4f2d266e0f2523","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-02-17 19:21:01.000000000","message":"Patch Set 4: Review-Priority+1","accounts_in_message":[],"_revision_number":4},{"id":"7d5b178b7b001c430db85a705e6a99c1e2ce2e87","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-02-17 20:10:35.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"6e471c7e779c1add285a1cf031428041b01a680d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 20:11:06.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"88a036ccea1fa1fa26e67cc638030376fe0700b1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 23:38:06.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/625adfaf0c5340b49c6f8894cdadb92c\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/6f1a58d4413e44018cc3eef59e6f341d : SUCCESS in 1h 24m 46s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/0e513f07e3194a0a910eb9dfb0fcfe52 : SUCCESS in 2h 02m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de59e8a1cef0450ba30431836b8b51e0 : SUCCESS in 21m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/65f601324b554ad6bb3b85a601cd22a1 : SUCCESS in 28m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bfd339394e0549bc97c691e1557b0d84 : SUCCESS in 20m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f91f0a9de8ca4b3cbce4d11ccfd483e8 : SUCCESS in 7m 35s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/9f1b10120bc94d9a874617d77db39824 : SUCCESS in 44m 52s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/488d99663ae34c6395a587a4a9809a5d : SUCCESS in 1h 53m 12s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/735f96ecac79427ebe14b2669b4aa1a7 : SUCCESS in 1h 12m 39s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/8740b0294b1f4b24ac97c74f99e3860b : SUCCESS in 44m 03s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/1620791b8fe4457b9dedfc4c6d48cd76 : SUCCESS in 1h 17m 48s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/0dad7f1170cd48349ad1d1836472d45b : SUCCESS in 1h 23m 06s","accounts_in_message":[],"_revision_number":4},{"id":"345688cf86a24fdd858a2f84e3eff2fee19e3b42","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 23:38:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"7d39282546100eda342312a1ad9179af4474c411","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 23:39:07.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0cd4b6fb9bf4e639e3473ca3b45fbbd\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6284d64dc1a34de2a143a1e7479ae7b7 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"c073bc4e15e0f0f310b6189a2d20a68af5882781","revisions":{"1e7610e43284643dba035fb70802fd4499197ca5":{"kind":"REWORK","_number":1,"created":"2026-02-10 20:08:09.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/00/976300/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/00/976300/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/00/976300/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/00/976300/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/00/976300/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/00/976300/1"}}},"commit":{"parents":[{"commit":"a3095f4372c18f581c5fb0f75b5516ff05d4805f","subject":"bgp: Fix configuring vrf id on BGP routers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a3095f4372c18f581c5fb0f75b5516ff05d4805f"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-02-10 20:05:54.000000000","tz":0},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-02-10 20:06:30.000000000","tz":0},"subject":"bgp: Small fixes","message":"bgp: Small fixes\n\nTrivialFix\n\nChange-Id: Ie05e0dc4fff9327fd0a52cf48028a923c4e82330\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1e7610e43284643dba035fb70802fd4499197ca5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1e7610e43284643dba035fb70802fd4499197ca5"}]},"branch":"refs/heads/master"},"41e97aaf36ed2485293a69f39ab9c5ea681e3f96":{"kind":"REWORK","_number":2,"created":"2026-02-10 20:08:36.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/00/976300/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/00/976300/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/00/976300/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/00/976300/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/00/976300/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/00/976300/2"}}},"commit":{"parents":[{"commit":"a3095f4372c18f581c5fb0f75b5516ff05d4805f","subject":"bgp: Fix configuring vrf id on BGP routers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a3095f4372c18f581c5fb0f75b5516ff05d4805f"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-02-10 20:05:54.000000000","tz":0},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-02-10 20:08:27.000000000","tz":0},"subject":"bgp: Small fixes","message":"bgp: Small fixes\n\nTrivialFix\n\nChange-Id: Ie05e0dc4fff9327fd0a52cf48028a923c4e82330\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/41e97aaf36ed2485293a69f39ab9c5ea681e3f96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/41e97aaf36ed2485293a69f39ab9c5ea681e3f96"}]},"branch":"refs/heads/master"},"d47409e9ee9426de1ddd3ff86dc637ece28fafbe":{"kind":"REWORK","_number":3,"created":"2026-02-10 20:11:13.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/00/976300/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/00/976300/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/00/976300/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/00/976300/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/00/976300/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/00/976300/3"}}},"commit":{"parents":[{"commit":"a3095f4372c18f581c5fb0f75b5516ff05d4805f","subject":"bgp: Fix configuring vrf id on BGP routers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a3095f4372c18f581c5fb0f75b5516ff05d4805f"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-02-10 20:05:54.000000000","tz":0},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-02-10 20:10:23.000000000","tz":0},"subject":"bgp: Small fixes","message":"bgp: Small fixes\n\nTrivialFix\n\nAssisted-By: claude-4.5-sonnet\n\nChange-Id: Ie05e0dc4fff9327fd0a52cf48028a923c4e82330\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d47409e9ee9426de1ddd3ff86dc637ece28fafbe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d47409e9ee9426de1ddd3ff86dc637ece28fafbe"}]},"branch":"refs/heads/master"},"c073bc4e15e0f0f310b6189a2d20a68af5882781":{"kind":"REWORK","_number":4,"created":"2026-02-13 15:13:10.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/00/976300/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/00/976300/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/00/976300/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/00/976300/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/00/976300/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/00/976300/4"}}},"commit":{"parents":[{"commit":"a3095f4372c18f581c5fb0f75b5516ff05d4805f","subject":"bgp: Fix configuring vrf id on BGP routers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a3095f4372c18f581c5fb0f75b5516ff05d4805f"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-02-10 20:05:54.000000000","tz":0},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-02-13 13:19:56.000000000","tz":0},"subject":"bgp: Small fixes","message":"bgp: Small fixes\n\nTrivialFix\n\nAssisted-By: claude-4.5-sonnet\n\nChange-Id: Ie05e0dc4fff9327fd0a52cf48028a923c4e82330\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c073bc4e15e0f0f310b6189a2d20a68af5882781"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c073bc4e15e0f0f310b6189a2d20a68af5882781"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
