)]}'
{"id":"openstack%2Fneutron~976562","triplet_id":"openstack%2Fneutron~master~I36cde1222cf2f671bee117afc1e895ebbf3f310d","project":"openstack/neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2026-02-12 10:09:44.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I36cde1222cf2f671bee117afc1e895ebbf3f310d","subject":"Bump pre-commit-config versions","status":"MERGED","created":"2026-02-12 00:04:38.000000000","updated":"2026-02-12 10:11:10.000000000","submitted":"2026-02-12 10:09:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"976562","meta_rev_id":"9f15983aaee25dd0987241560e049883702f7839","_number":976562,"virtual_id_number":976562,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-12 10:09:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":2,"date":"2026-02-12 05:17:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2026-02-12 09:01:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2026-02-12 09:01:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-12 01:13:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-12 05:17:42.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2026-02-12 09:01:29.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"9551bda7b65f2c76027abcfdd65b3ddfc920bdf6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-02-12 00:04:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"98937e2036642ef725e85ab5785ebe0ed1d6864a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 01:13:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e9719726e4534fd5ac6bbf10d31677f6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bfa0ea79130b4d0dbf27283ded0274f4 : SUCCESS in 1h 03m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/533d08c8e2234997b32460f4977b4f19 : SUCCESS in 26m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1bdd8286a3274a8f8027ece1489940db : SUCCESS in 33m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/559eae589fce4646be96dc9b82c1cdd7 : SUCCESS in 22m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/112ea11a6ebd4500a94a43719021ff89 : SUCCESS in 12m 01s\n- requirements-check https://zuul.opendev.org/t/openstack/build/fc785e26e77c400bbe029649919604e8 : SUCCESS in 3m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/150ae6b49b8849a4b016172bf1463b56 : SUCCESS in 29m 27s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/80ff2f1388414a04a4f993c2ac6d6c59 : FAILURE in 33m 10s","accounts_in_message":[],"_revision_number":1},{"id":"2d7cc80f2ae52b0a6b814b9cc68ea6f3c2f8e9ef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 01:16:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8fc6baba9eeb40f981d3d2b971c733dc\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/cca920b7c4dc41f7b595f9bb6ce5a831 : SUCCESS in 1h 09m 26s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a94ae74c10664b568b9a0d8e3e187076 : SUCCESS in 52m 03s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a71486490aa0478bacea04a54735b6fa : RETRY_LIMIT in 4m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fd027114e2bd5702ac20273c3200fbfa65b20830","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-02-12 01:22:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7a651f79b4eea1c0b487cb4088a1c6ffffed3979","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 02:30:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e200e723e284791b9a372b0e43d21f0\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/228c9e7b4c754807b8b16d116028b364 : SUCCESS in 1h 06m 00s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/842dbbae3f5b4d66b936470b1436b97c : SUCCESS in 51m 15s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8d54d6d7ccd742389cd62a74ab6ccf60 : RETRY_LIMIT in 4m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6296c163e7de2dc187939305e0fa670995e847d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 02:42:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed35b723367c4b03a0fda97ba3daf4ea\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1ded8b9813284a0cb8f613d316d08759 : SUCCESS in 35m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bdfc2261f99c4d4ca0f90bc933246d79 : SUCCESS in 10m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/26113e831b4c44d3ba049143d268ec09 : SUCCESS in 19m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/504083caee694d89b363dc8667aa72c5 : SUCCESS in 22m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8b3447f0b344ac9aa5feedf7732cbb5 : SUCCESS in 9m 22s\n- requirements-check https://zuul.opendev.org/t/openstack/build/292a8229af6a408f8f7c6a77b337d4c0 : SUCCESS in 5m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/25ee7fa0bc804d13b9488534c34966c9 : SUCCESS in 23m 40s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/3e678f0dbfa54bcc8b41707a98f09e6b : SUCCESS in 1h 15m 02s","accounts_in_message":[],"_revision_number":1},{"id":"71e7c15667eea3e6707d21ee684cc2fe445d1fc4","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2026-02-12 05:17:42.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9e69ad99456657e6bcf4faa9aed25809791c2272","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-02-12 09:01:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9604995eab384b8ef7c3cbef434e97abbe874804","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 09:01:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f96d330a333d863a1a811b9d6f19b417dde85e5a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 10:09:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f55376553da49bfa4ac7c48f1cd8052\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcbc0a702e164284a91c2fdfa82f18cd : SUCCESS in 12m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6328e5d8cf844b3aaa923dd8e6328513 : SUCCESS in 29m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1549eea027b44936b73c57af1e4a121e : SUCCESS in 20m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfbd4f661fd441bfa678af85fa059df8 : SUCCESS in 10m 01s\n- requirements-check https://zuul.opendev.org/t/openstack/build/545df17d2f264589b715005398d89557 : SUCCESS in 2m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9a5cf6b55f664ce2a99602ce5db13c6c : SUCCESS in 18m 28s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/feb75715fe3b4ee8b04284ae07841778 : SUCCESS in 1h 06m 38s","accounts_in_message":[],"_revision_number":1},{"id":"f26a35363bf1a3dd5fc2270aa78f7c9a950d67de","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 10:09:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9f15983aaee25dd0987241560e049883702f7839","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 10:11:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/715e486d246b4fde95c230a9c7587f73\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7021ee5723e045f3b325e8f34a574492 : SUCCESS in 1m 03s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4613fd85105c4a3cb6e88d1ad6ac4762 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"825d79593ae951e56b54de5cb4878bb4432e9999","revisions":{"825d79593ae951e56b54de5cb4878bb4432e9999":{"kind":"REWORK","_number":1,"created":"2026-02-12 00:04:38.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/62/976562/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/62/976562/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/976562/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/976562/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/976562/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/62/976562/1"}}},"commit":{"parents":[{"commit":"025ea04a3919cdc1ef4111574ea46d6f4d1dce6b","subject":"Merge \"Remove configuration variable ``enable_signals``.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/025ea04a3919cdc1ef4111574ea46d6f4d1dce6b"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-02-12 00:02:00.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-02-12 00:02:00.000000000","tz":-300},"subject":"Bump pre-commit-config versions","message":"Bump pre-commit-config versions\n\nBumped hacking, etc, versions in .pre-commit-config.yaml\n\nAdded some additional checks and fixed any complaints\nthey found.\n\nTrivialFix\n\nChange-Id: I36cde1222cf2f671bee117afc1e895ebbf3f310d\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/825d79593ae951e56b54de5cb4878bb4432e9999"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/825d79593ae951e56b54de5cb4878bb4432e9999"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
