)]}'
{"id":"openstack%2Fneutron~976582","triplet_id":"openstack%2Fneutron~stable%2F2025.1~I4b4768e0ee3c607dcc8511d7f1b5764d1fd0a63a","project":"openstack/neutron","branch":"stable/2025.1","topic":"bug/2002839","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-02-13 16:47:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4b4768e0ee3c607dcc8511d7f1b5764d1fd0a63a","subject":"Fix ``ChassisAgentWriteEvent`` match condition","status":"MERGED","created":"2026-02-12 07:41:22.000000000","updated":"2026-02-13 16:48:27.000000000","submitted":"2026-02-13 16:47:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"976582-bug/2002839","meta_rev_id":"37b9aabaeb8227da3018aae6e7b89c7c227e6994","_number":976582,"virtual_id_number":976582,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-13 16:47:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":2,"date":"2026-02-13 09:47:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2026-02-13 14:54:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2026-02-13 14:54:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-12 09:44:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-12 23:29:24.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2026-02-13 09:47:19.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2026-02-13 14:54:50.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"666e4c295adf6e547d9c55e61ff8fea7e779cea2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-02-12 07:41:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"391fa0acb1b1ba07b274328277e71234a4ffc290","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 09:44:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9b63d69726eb43928ef61a006302c1bd\n\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/b650eef5f29447b488057c189e4011cd : SUCCESS in 2h 01m 41s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/fdb42dd9829c41a39c81b74da02f0246 : SUCCESS in 1h 10m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/39c28d8276c749a7bb320aff786fd558 : SUCCESS in 44m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2664520a240e4ff3a24255ec8d09b111 : FAILURE in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9a2aae2b5af84a1f8ad3966e9f19464f : SUCCESS in 20m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/061d58a7f4074bbb9471b261be32ace5 : SUCCESS in 22m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff42851744ec4760a2a8935fdac531a1 : FAILURE in 6m 15s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/d3feaee46a644db58547e3ca11e7301b : SUCCESS in 1h 15m 41s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/9e4247f39f124da286f53b861d4d99eb : SUCCESS in 1h 07m 34s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/b0b297c5b6a8472eb1fe20891eb40e40 : SUCCESS in 45m 20s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/ae935d049a7a41f9b3ae9a345c17be62 : SUCCESS in 1h 00m 05s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/442e4b67374647b5beaaef84465ab516 : SUCCESS in 1h 15m 03s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/8150474d8d444e1aafb33c1a411603d4 : SUCCESS in 1h 23m 59s","accounts_in_message":[],"_revision_number":1},{"id":"d65a7de00403fe19245b4e0935d03839e63b9c80","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-02-12 23:29:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"08311e5f4c10e5fde8b7c194b80729805ce9efcc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-02-13 09:03:30.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"99b01d012ad4a97020fece1b5d2d28fe9d70778e","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-02-13 09:47:19.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"d5e6954284ffd43b90b3dbf59b5c2a5ddbe80e7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 10:57:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a1aa3cf1d7344bdb8049d5a25c30f4c\n\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/69bcb6fc0405416c89e1d6f2abfc1248 : SUCCESS in 1h 50m 59s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/78f69866870c4f8f9d2d9040eb0322ab : SUCCESS in 1h 20m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a23de99d25714063803710e93ddff58d : SUCCESS in 48m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e1d218b590143a58d10ad1e7b3fb0f6 : SUCCESS in 5m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a301afa42174492496495701b7ab613a : SUCCESS in 26m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/76c25cb68f3441fbb41edff6475c849a : SUCCESS in 21m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0159aade3b043f7b33f0bd51f05e99a : SUCCESS in 12m 44s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/d90bc072af0e41c1af908aa47708121d : SUCCESS in 1h 21m 00s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/4853f06f1fba4d30b7502142b45e6326 : SUCCESS in 59m 11s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/93a0b8bd2eba4400bd347724fdf8f5e8 : SUCCESS in 1h 05m 09s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/4e546b97509549db954cd48fc5acd5df : SUCCESS in 1h 00m 49s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/84d66e50dec84aabb35717401784d6eb : SUCCESS in 1h 14m 09s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/68d787549b614f0e844f66248ede8897 : SUCCESS in 1h 16m 59s","accounts_in_message":[],"_revision_number":2},{"id":"af2888ec8ca77b12ac571f9e89ebb7d3a573d11f","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-02-13 14:54:50.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"64e6b48f20ad30140f2c7d79c8800817f0e7f87d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 14:55:33.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e1ffecfc5637df13657403892fa3ad925fcf259a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 16:47:14.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de7c9ebfde524808b2b63aea963ce4ad\n\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/ada2674d758348afbf7a9f812896af8b : SUCCESS in 1h 51m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33a3a7914a5f41e4bde743d816094655 : SUCCESS in 11m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6ca91d95e9334a238c1fb55219d2e283 : SUCCESS in 23m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0974225e493a46c08f5eabbce4d85279 : SUCCESS in 16m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0387ee09519b4b89b9ffb36b2c1dce38 : SUCCESS in 14m 58s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/7616db5265f940688320471dd086bf5e : SUCCESS in 1h 05m 10s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/248bed55598740b687ef8eb3187e2d6a : SUCCESS in 1h 21m 46s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/3ac20807f56941c3a958ce0c4a8f5abf : SUCCESS in 1h 11m 52s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a1a890cd16f34323a57b9ee1374144cd : SUCCESS in 1h 30m 53s","accounts_in_message":[],"_revision_number":2},{"id":"24ee65b706b85aad6e67db8ed066b64d046d0dd8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 16:47:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"37b9aabaeb8227da3018aae6e7b89c7c227e6994","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 16:48:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77fd248dfabd4f5da96bc69b32aa8666\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be31e83c05f74301a9f48c5ba53d4e9b : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7f1bc37dda84c90747a301542f56b2b4680ae615","revisions":{"100adbd77799343fe24ddb922351733b41725758":{"kind":"REWORK","_number":1,"created":"2026-02-12 07:41:22.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/82/976582/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/82/976582/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/976582/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/976582/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/976582/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/82/976582/1"}}},"commit":{"parents":[{"commit":"8d32160a99cb745e94aa0a4c894e68b61614caa2","subject":"Cleaning debug import of the pdb module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8d32160a99cb745e94aa0a4c894e68b61614caa2"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-02-10 11:11:17.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-02-10 15:59:09.000000000","tz":60},"subject":"Fix ``ChassisAgentWriteEvent`` match condition","message":"Fix ``ChassisAgentWriteEvent`` match condition\n\nAs the method description says, the ``ChassisAgentWriteEvent`` uses the\n``Chassis_Private.chassis`` information to update the local OVN agent\ncache. If there is no chassis, the match should not pass.\n\nThis method was updated in [1], during the backwards compatibility\nremoval. This condition was incorrectly refactored.\n\n[1]https://review.opendev.org/c/openstack/neutron/+/904238\n\nRelated-Bug: #2002839\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I4b4768e0ee3c607dcc8511d7f1b5764d1fd0a63a\n(cherry picked from commit bf9e039e2fe5beb84b764242e4d293e90b56e13a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/100adbd77799343fe24ddb922351733b41725758"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/100adbd77799343fe24ddb922351733b41725758"}]},"branch":"refs/heads/stable/2025.1"},"7f1bc37dda84c90747a301542f56b2b4680ae615":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-02-13 09:03:30.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/82/976582/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/82/976582/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/976582/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/976582/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/976582/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/82/976582/2"}}},"commit":{"parents":[{"commit":"9835a8d00bc733e6c85bd6c5e431c7c6e4a4b589","subject":"Fix issues related to pkg_resources module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9835a8d00bc733e6c85bd6c5e431c7c6e4a4b589"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-02-10 11:11:17.000000000","tz":60},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2026-02-13 09:03:30.000000000","tz":0},"subject":"Fix ``ChassisAgentWriteEvent`` match condition","message":"Fix ``ChassisAgentWriteEvent`` match condition\n\nAs the method description says, the ``ChassisAgentWriteEvent`` uses the\n``Chassis_Private.chassis`` information to update the local OVN agent\ncache. If there is no chassis, the match should not pass.\n\nThis method was updated in [1], during the backwards compatibility\nremoval. This condition was incorrectly refactored.\n\n[1]https://review.opendev.org/c/openstack/neutron/+/904238\n\nRelated-Bug: #2002839\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I4b4768e0ee3c607dcc8511d7f1b5764d1fd0a63a\n(cherry picked from commit bf9e039e2fe5beb84b764242e4d293e90b56e13a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7f1bc37dda84c90747a301542f56b2b4680ae615"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7f1bc37dda84c90747a301542f56b2b4680ae615"}]},"branch":"refs/heads/stable/2025.1","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
