)]}'
{"id":"openstack%2Fneutron~977050","triplet_id":"openstack%2Fneutron~stable%2F2024.2~I652344fe6629ac3d308cd4d4c4e68d267f6f2efc","project":"openstack/neutron","branch":"stable/2024.2","topic":"tobiko-devstack","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-03-13 12:49:56.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2026-03-13 12:49:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I652344fe6629ac3d308cd4d4c4e68d267f6f2efc","subject":"Use the `devstack-tobiko` job","status":"MERGED","created":"2026-02-17 09:05:22.000000000","updated":"2026-03-13 12:52:04.000000000","submitted":"2026-03-13 12:49:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"977050-tobiko-devstack","meta_rev_id":"68fb833fc11b759a291cf68a07d2465127ee61f1","_number":977050,"virtual_id_number":977050,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-13 12:49:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"recommended":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-03-12 11:28:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2026-03-13 11:21:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2026-03-10 11:47:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2026-03-13 11:21:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2026-03-12 11:03:08.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-17 09:52:52.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2026-02-17 10:50:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-02 14:41:24.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"CC"},{"updated":"2026-03-10 11:47:36.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2026-03-10 12:13:46.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2026-03-12 11:28:30.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2026-03-13 11:21:40.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"dfdccc2305147eeab71ab883ab60c7acc2a917f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-02-17 09:05:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bc2d8c01480b8eb3b52308fa17ac1eaa4e133b60","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-02-17 09:52:52.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"87b0150a2d350680c0257d0a74553a96dc440f1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 10:50:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6033861bc6fe4ed8af26bcdff85a69ee\n\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/09754169374347608e10ba70c9c7cdcf : SUCCESS in 1h 26m 13s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/de7bf923a2d543f28200e10c102efb6d : SUCCESS in 1h 25m 41s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/a6d3ec86b9ec480585e3db19059766f1 : SUCCESS in 1h 36m 38s\n- neutron-tempest-plugin-designate-scenario-2024-2 https://zuul.opendev.org/t/openstack/build/f46d1244d6e44ac9b5f2db07fd497c73 : SUCCESS in 27m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ebdbca08f6a14c28ab316d18e0a35b02 : FAILURE in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c50e3b336df540cebeb5942e80a8aeb9 : FAILURE in 2m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d661b5f25d0e4577b3ac7395324b84dc : FAILURE in 2m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a09e663cef434aba9fd03b62903a8fc6 : FAILURE in 4m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/44c8f063502d4cf3b6818e2b924d52ed : FAILURE in 4m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9ae09fdd211451b976f46128750f0cb : FAILURE in 5m 32s","accounts_in_message":[],"_revision_number":1},{"id":"b7e1f34889cbcfc99bb731e44737232a3f225eb1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-02-27 00:05:09.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"29e9d1dd8fa20185987cf54c27ab117cef0d2436","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 01:55:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d9a0b6ff1c02463fb1de03bab06a16e5\n\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/35192222c19b44edab71dea432130459 : SUCCESS in 1h 13m 51s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/ffac2dc5c6994e2f9ec2296112a66b00 : SUCCESS in 1h 30m 16s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/4e4bcbf22b814b24acdafe018ec6df2d : SUCCESS in 1h 48m 24s\n- neutron-tempest-plugin-designate-scenario-2024-2 https://zuul.opendev.org/t/openstack/build/7aed9b8fef3a4138826dab0cdc785e13 : SUCCESS in 27m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/80218133e7994b90a8f899e3cad47c09 : SUCCESS in 42m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6bdfafdb61a42639d543c5b673ca750 : SUCCESS in 10m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5ce3dadb7c7d4fb181ce24385e0934e4 : SUCCESS in 23m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e1b5a13a6d1f431385dd1cd9b7553bd3 : SUCCESS in 20m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/01425e71120e4cbc849de07a23c00bf5 : SUCCESS in 14m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0aa41d36f99d42f4a6d547a1240ce02a : SUCCESS in 10m 38s","accounts_in_message":[],"_revision_number":2},{"id":"63e98484d81c05093e1875fa4ce5da2780321e00","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2026-03-02 14:41:24.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9b509bc4ed6edee19269636fd7ed3a4afef15773","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-03-02 15:17:37.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"6e326853affbe7004ab7b660872ff596f6695ceb","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-02 16:31:27.000000000","message":"Patch Set 2:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27cf6f2333c245f884f522787f9d2349\n\n- neutron-tempest-plugin-linuxbridge-2024-2 https://zuul.opendev.org/t/openstack/build/f050878b7a5c49998542a20f6e134b36 : SUCCESS in 1h 39m 00s\n- neutron-tempest-plugin-api-ovs-wsgi https://zuul.opendev.org/t/openstack/build/ec65af774e53472b850f80c64a483c4c : FAILURE in 1h 14m 38s (non-voting)\n- neutron-tempest-plugin-api-ovn-wsgi https://zuul.opendev.org/t/openstack/build/356d95efc46b415c83501908f5405fe9 : SUCCESS in 1h 44m 00s (non-voting)\n- devstack-tobiko https://zuul.opendev.org/t/openstack/build/598b80a61ade4f5c910eff47b1fdb2e9 : FAILURE in 45m 10s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/c2a557524f844fd9a88f9cb3b9809f22 : SUCCESS in 52m 09s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/d12e7bee2ce047b58992cafca5805a0d : SUCCESS in 1h 29m 14s\n- neutron-linuxbridge-tempest-plugin-nftables https://zuul.opendev.org/t/openstack/build/97e27c501add4a02987268ae102fc704 : SUCCESS in 1h 46m 57s\n- neutron-ovs-tempest-plugin-iptables_hybrid-nftables https://zuul.opendev.org/t/openstack/build/36ceb4f710854a5fb4e19a2a12d947c3 : SUCCESS in 1h 47m 42s\n- devstack-enforce-scope https://zuul.opendev.org/t/openstack/build/80a9e9b3c541451599c658aee9abf111 : SUCCESS in 20m 18s","accounts_in_message":[],"_revision_number":2},{"id":"43254425776443b59437e874643ec0f4164e6c4b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-03-10 09:24:05.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":3},{"id":"dcca1407e2488e23c5541f669752d88d3c526cfb","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-03-10 09:24:17.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ef923f81d6e94336d5ce33f3a8a83b9cbcc7ec48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-10 11:10:11.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d809794d6fbf4508b0ac377547520822\n\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/763dce057a084f69835ea65a42ee4a8a : SUCCESS in 1h 17m 33s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/1b8f142f682e4ee0a3ce3d0737ab402d : SUCCESS in 1h 38m 51s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/8d5495537a0e4a1095a6c9e26ad57076 : SUCCESS in 1h 43m 15s\n- neutron-tempest-plugin-designate-scenario-2024-2 https://zuul.opendev.org/t/openstack/build/033c1fe066854137ad3738cc89ec924c : SUCCESS in 29m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6e4487d2389c47bba2bf47f61a7ebcbd : SUCCESS in 31m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03c8f575326c43cba239b7576a3b4a41 : SUCCESS in 9m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/71d5a6516148423088b032a90ab708a0 : SUCCESS in 22m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a730ce484e6d445b84246aeead402b47 : SUCCESS in 22m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/36fb538e42bf463c8a52f8ab1ba45377 : SUCCESS in 22m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/035d96ed4fee400db294cb560ce39bca : SUCCESS in 13m 32s","accounts_in_message":[],"_revision_number":3},{"id":"14973e8dcb46d54784988eb3d05a3df082cbe3c1","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-10 11:31:26.000000000","message":"Patch Set 3:\n\nBuild succeeded (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4e8db5fe05d4a9abe69be202142f2b2\n\n- neutron-tempest-plugin-linuxbridge-2024-2 https://zuul.opendev.org/t/openstack/build/fa6ca42dd54c46cca9af880f9bdd40f2 : SUCCESS in 2h 04m 09s\n- neutron-tempest-plugin-api-ovs-wsgi https://zuul.opendev.org/t/openstack/build/eb6436e42f3d4609b25533f970ae02b4 : FAILURE in 1h 27m 09s (non-voting)\n- neutron-tempest-plugin-api-ovn-wsgi https://zuul.opendev.org/t/openstack/build/25fa0318347c4d7490a0cd2a67d900da : SUCCESS in 1h 59m 38s (non-voting)\n- devstack-tobiko https://zuul.opendev.org/t/openstack/build/398c5123f1ed44a885e2a20bbb867ef1 : SUCCESS in 41m 43s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/b8886eab400d4c9aa67bf9361b4aac2c : SUCCESS in 56m 36s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/aa2cba7baa8e4e18ba44ac7a7226ce63 : SUCCESS in 1h 28m 44s\n- neutron-linuxbridge-tempest-plugin-nftables https://zuul.opendev.org/t/openstack/build/79482bc3cacf44519e5ba7931ec574fb : SUCCESS in 1h 31m 00s\n- neutron-ovs-tempest-plugin-iptables_hybrid-nftables https://zuul.opendev.org/t/openstack/build/3d285cf7635f4b75a8b6b7319ad93e9b : SUCCESS in 1h 33m 56s\n- devstack-enforce-scope https://zuul.opendev.org/t/openstack/build/1d5fe170ea6d448ca15f7f07f54a59b6 : SUCCESS in 23m 18s","accounts_in_message":[],"_revision_number":3},{"id":"7b206452eb594758d5ae6656eb293684d30fcb26","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-03-10 11:47:36.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"fedbddaa62c028384987c0de5d1866fea2a5aacd","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-03-12 11:03:08.000000000","message":"Patch Set 3: Review-Priority+1","accounts_in_message":[],"_revision_number":3},{"id":"b3e0a0f0a98e76ec358bbd3421fc65f7510c7b8f","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2026-03-12 11:28:30.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"f69b3d9c6a10dd4c8f37a4de4188295c99165952","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2026-03-13 11:21:40.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"472f8b1b9bfb1a7d7155a1e92bacb0e8c3987ad7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-13 11:23:02.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"d84deadbdc6f04d96f63458dd028c774ed0aed20","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-13 12:49:56.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/75b9d1588121479e88f22b3b0b0d1a07\n\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/746624994d3c4b2b99d859906f50def1 : SUCCESS in 1h 26m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23823cddc1264707a98bc0165aabcb61 : SUCCESS in 6m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7ebb08cb9eac49e09d1e50cd3d5d39fb : SUCCESS in 15m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/95ec38b159a041d680ec71dfbc72ab75 : SUCCESS in 17m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da947cc5bc3d44849df3a1414a311cd1 : SUCCESS in 9m 23s","accounts_in_message":[],"_revision_number":3},{"id":"cd8317f1368d05011c5a80f0bf3bfa427891e44e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-13 12:49:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"68fb833fc11b759a291cf68a07d2465127ee61f1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-13 12:52:04.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b7246dec5ff5437c867e906f2c83cf2a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/630b803fdd8b43ad8272b87b37e9c51b : SUCCESS in 58s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"25c5202a609e2e20c596b248374f47283e28e6cb","revisions":{"711797101ef60101232e8830d31ac3c11260149f":{"kind":"REWORK","_number":1,"created":"2026-02-17 09:05:22.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/50/977050/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/50/977050/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/50/977050/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/50/977050/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/50/977050/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/50/977050/1"}}},"commit":{"parents":[{"commit":"8c301a0c97cc5c30076d249c98f232f02d9181d5","subject":"Fix ``ChassisAgentWriteEvent`` match condition","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8c301a0c97cc5c30076d249c98f232f02d9181d5"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-02-06 10:07:36.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-02-16 09:32:34.000000000","tz":60},"subject":"Use the `devstack-tobiko` job","message":"Use the `devstack-tobiko` job\n\nThe Tobiko project will deprecate and remove some jobs, like for\nexample `devstack-tobiko-neutron`.\n\nConflicts:\n  zuul.d/job-templates.yaml\n\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I652344fe6629ac3d308cd4d4c4e68d267f6f2efc\n(cherry picked from commit 33adbccc812dc59193fa5a6ef96558e750f27efd)\n(cherry picked from commit 560d624488c0933a4c32935a47ba0741724e81d5)\n(cherry picked from commit 69b62ddad75fed379f8b0700cb92f6c4cafc5cf0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/711797101ef60101232e8830d31ac3c11260149f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/711797101ef60101232e8830d31ac3c11260149f"}]},"branch":"refs/heads/stable/2024.2"},"29326f47ebab4e003c4dff03ed8a8cdc58608331":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-02-27 00:05:09.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/50/977050/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/50/977050/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/50/977050/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/50/977050/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/50/977050/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/50/977050/2"}}},"commit":{"parents":[{"commit":"7f31aa17ddeb3e30775897444f497b2ec54af403","subject":"Get rid of pkg_resources","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7f31aa17ddeb3e30775897444f497b2ec54af403"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-02-06 10:07:36.000000000","tz":60},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2026-02-27 00:05:09.000000000","tz":0},"subject":"Use the `devstack-tobiko` job","message":"Use the `devstack-tobiko` job\n\nThe Tobiko project will deprecate and remove some jobs, like for\nexample `devstack-tobiko-neutron`.\n\nConflicts:\n  zuul.d/job-templates.yaml\n\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I652344fe6629ac3d308cd4d4c4e68d267f6f2efc\n(cherry picked from commit 33adbccc812dc59193fa5a6ef96558e750f27efd)\n(cherry picked from commit 560d624488c0933a4c32935a47ba0741724e81d5)\n(cherry picked from commit 69b62ddad75fed379f8b0700cb92f6c4cafc5cf0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/29326f47ebab4e003c4dff03ed8a8cdc58608331"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/29326f47ebab4e003c4dff03ed8a8cdc58608331"}]},"branch":"refs/heads/stable/2024.2","description":"Rebase"},"25c5202a609e2e20c596b248374f47283e28e6cb":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-03-10 09:24:05.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/50/977050/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/50/977050/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/50/977050/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/50/977050/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/50/977050/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/50/977050/3"}}},"commit":{"parents":[{"commit":"7f31aa17ddeb3e30775897444f497b2ec54af403","subject":"Get rid of pkg_resources","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7f31aa17ddeb3e30775897444f497b2ec54af403"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-02-06 10:07:36.000000000","tz":60},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2026-03-10 09:24:05.000000000","tz":0},"subject":"Use the `devstack-tobiko` job","message":"Use the `devstack-tobiko` job\n\nThe Tobiko project will deprecate and remove some jobs, like for\nexample `devstack-tobiko-neutron`.\n\nDepends-On: https://review.opendev.org/c/x/devstack-plugin-tobiko/+/979677\n\nConflicts:\n  zuul.d/job-templates.yaml\n\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I652344fe6629ac3d308cd4d4c4e68d267f6f2efc\n(cherry picked from commit 33adbccc812dc59193fa5a6ef96558e750f27efd)\n(cherry picked from commit 560d624488c0933a4c32935a47ba0741724e81d5)\n(cherry picked from commit 69b62ddad75fed379f8b0700cb92f6c4cafc5cf0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/25c5202a609e2e20c596b248374f47283e28e6cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/25c5202a609e2e20c596b248374f47283e28e6cb"}]},"branch":"refs/heads/stable/2024.2","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
