)]}'
{"id":"openstack%2Fneutron~977072","triplet_id":"openstack%2Fneutron~stable%2F2024.2~I1b478a63ad1d73f9f3528939362797ea1fc68534","project":"openstack/neutron","branch":"stable/2024.2","topic":"pkg_resources","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-02-24 10:31:03.000000000","reason":"Change was submitted"},"21798":{"account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"last_update":"2026-02-24 10:31:03.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2026-02-24 10:31:03.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1b478a63ad1d73f9f3528939362797ea1fc68534","subject":"Get rid of pkg_resources","status":"MERGED","created":"2026-02-17 12:08:40.000000000","updated":"2026-02-24 10:33:02.000000000","submitted":"2026-02-24 10:31:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"977072-pkg_resources","meta_rev_id":"fb73ae92dc400cbf4b9ae1abf5dac7a876d8deb6","_number":977072,"virtual_id_number":977072,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-24 10:31:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"recommended":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-02-24 03:30:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2026-02-24 08:14:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2026-02-23 16:20:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2026-02-24 08:14:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2026-02-23 21:03:34.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2026-02-24 03:30:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-17 12:08:40.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2026-02-17 14:19:28.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2026-02-17 15:44:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-23 16:20:05.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2026-02-24 03:30:17.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2026-02-24 03:30:17.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2026-02-24 03:30:17.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2026-02-24 08:14:33.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"586a44360b321e5a5ebfbaa14761c93c7f92538f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-02-17 12:08:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2d4e35848234ad9f892a7d06c7c170994cd8bf4c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-02-17 13:46:15.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9807656703b5cc68a56e942055e7a25967d27791","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-02-17 14:19:28.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"dd162e57ee2abb7a349af0e8d2b8f807d4ae9f1d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 15:44:39.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4dd9c47d514d4f10b83cb227afc8042e\n\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/7296d12a754c402280fb61539400a588 : SUCCESS in 1h 28m 24s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/db3c0c219ce14ccead84e2f06d6b621b : SUCCESS in 1h 47m 48s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/be48d1b9d07c40ad80f9174dee245d0d : SUCCESS in 1h 41m 13s\n- neutron-tempest-plugin-designate-scenario-2024-2 https://zuul.opendev.org/t/openstack/build/dc34d9e9bd9e43c1baa83d356c0b33ee : SUCCESS in 28m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6c60cbcee1ee43948e0155c16a78a06e : SUCCESS in 42m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8aea8b1d1fe54c8e86480eb34c6013a9 : FAILURE in 5m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/de4684398ca9434ab8e11bb9d339d108 : SUCCESS in 21m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/399d2387d6fd47e496c278c6d8095df9 : SUCCESS in 14m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/43a59383be3c4e079859f655bc7b708c : SUCCESS in 15m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6c57add5d4644c9bb472123037ba373 : FAILURE in 4m 34s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c6765b0ee5014f86b18c7b80ac224959 : SUCCESS in 2m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c2bc4d4aaa854dfc9482fdf08b15bfde : SUCCESS in 24m 16s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/451bd974e3a942c4908229fd12315889 : FAILURE in 20m 53s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/88bbe5c2685c4786a2feb4572961a0a6 : FAILURE in 15m 40s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/2aa0ea7033554203bfb6210f1e81d9fd : SUCCESS in 38m 30s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/b078eb659ab94486900c95b9dd05b6de : SUCCESS in 41m 10s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/beeafc1e2806414ba58b54c6d1945ee2 : SUCCESS in 1h 01m 36s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/2e7866d34638407e8241be8709a0fb53 : SUCCESS in 1h 00m 00s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9d95f969618941229f2a92bbb94aaff8 : SUCCESS in 1h 05m 54s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/90476775af094c8ebab1ef570b72ac0d : SUCCESS in 1h 04m 41s","accounts_in_message":[],"_revision_number":2},{"id":"ce96f7f9ab4add7f314b9b9be62f06076f159355","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-02-23 07:09:15.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"68d888f53b035b985b51aa5562fa35ae13dee953","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-02-23 08:29:31.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"cf4ed28430c97bd235b86510a8440f646790ed3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-23 10:39:14.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5ce2d5f489a945a1a76731021a86f4b8\n\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/d774654b5fb54e5d8d911573e8902131 : SUCCESS in 1h 06m 14s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/36e886e4c51449cab9188f5e6e6d8f46 : SUCCESS in 1h 55m 40s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/5a5779eda59b4c4b88709e68d405038a : SUCCESS in 1h 36m 29s\n- neutron-tempest-plugin-designate-scenario-2024-2 https://zuul.opendev.org/t/openstack/build/e5db2351a4d24e56851739cbff742ba8 : SUCCESS in 27m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c8bd95cb34b34a52a760f1b76d2bec7a : SUCCESS in 23m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6161265d3405460396571130aa2dce5f : FAILURE in 4m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3a5aa070b4b6455998e7a3382b31cd7e : SUCCESS in 14m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/de6a875da6e04f14a2df1470c1a149da : SUCCESS in 19m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c6351a9d5b6e4ee99838fbf6123b8cb6 : SUCCESS in 21m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/300549c845614841963f6d71751d9070 : SUCCESS in 11m 40s\n- requirements-check https://zuul.opendev.org/t/openstack/build/99dfea7a2e6e4a6da635f7cf903f78ee : SUCCESS in 3m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/01bf80aacd1f4b66a6258baca1de977a : SUCCESS in 28m 42s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/fbbbcee945514c38adc27b4a758ade7b : SUCCESS in 1h 35m 32s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2854c46a694a4c7e83d7f4a5b8bb9822 : SUCCESS in 1h 52m 07s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/10f70bf0c2c94d6fafd8409b00e95aae : SUCCESS in 51m 58s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/c6a37f2ba3f44b62a9a1119217143d58 : SUCCESS in 1h 08m 29s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/58e38ed672a34500916dd53c6ef606a5 : SUCCESS in 1h 05m 36s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/1e02f6fd1ff74074ae31bc6888778ddb : SUCCESS in 47m 51s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a0c3e074396d4715911e38857e73b229 : SUCCESS in 28m 54s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/9448c96c31864d44bfe29e9a7fe7fe24 : SUCCESS in 1h 14m 08s","accounts_in_message":[],"_revision_number":4},{"id":"2c5d80584694e8146be745088a9b1e3aa9518a97","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-02-23 15:50:03.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"3c8dd05dfb41752025e0f10992c922aa420629ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-02-23 16:03:49.000000000","message":"Patch Set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"78da1434db245b20cce985ecb7e28d940e345eac","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-02-23 16:20:05.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"9e18fd51e14ebcb308575b00e4a9448809ee0048","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-23 17:53:37.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5920453463fd476c97045fbb5fdbc7f7\n\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/0e5bf53d794d40f199a552569a2ceee3 : SUCCESS in 1h 24m 40s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/ab9dba90f65f4629aeea171469f4cdbf : SUCCESS in 1h 44m 11s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/625f18984d7b4cbe9d6193111a96fbfa : SUCCESS in 1h 39m 20s\n- neutron-tempest-plugin-designate-scenario-2024-2 https://zuul.opendev.org/t/openstack/build/0030f8f007e04111b43ce9f9c6bb566b : SUCCESS in 27m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2e8bc1fce7604c87ab020b93204c3df8 : SUCCESS in 30m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d6067bf5e354590b0e4c7f5c35a0ab5 : SUCCESS in 6m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a7c4bce734a34980a2ee5cf6cce0650b : SUCCESS in 15m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/153c2e1062054b8d8be4dd01d38aa63e : SUCCESS in 14m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/78c4624c54844f38a92104db37e81a77 : SUCCESS in 23m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18d1d8a69c5146a2845e2e7296b7c476 : SUCCESS in 10m 16s\n- requirements-check https://zuul.opendev.org/t/openstack/build/14f48861da6247ca9e90800260229bf4 : SUCCESS in 2m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8c8aacb8bfb34489bfbc392a1889495a : SUCCESS in 22m 32s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2eeef2fcc1fb429698c91d59d86e74aa : SUCCESS in 1h 05m 39s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/11a4ebc16888480d9179cb7de95ca9f5 : SUCCESS in 1h 44m 07s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/72c18c085fc94ed988bb1e14cb60502d : SUCCESS in 1h 04m 58s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/ff60c9c776384538b59befd003afa4db : SUCCESS in 35m 39s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/6506ba676af74258a641ef2dd16db11f : SUCCESS in 59m 11s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/433f1ebc1937451780dcd5e49e9ce9b6 : SUCCESS in 53m 02s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f0db611c4c7a4ad5b8c12e5fca368ccc : SUCCESS in 57m 18s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/48e7e8f79a7b4987b3822c3bdbb3c0d0 : SUCCESS in 1h 15m 36s","accounts_in_message":[],"_revision_number":6},{"id":"8e29b5c3689131cff3ad51ba14f6cca3c66acd62","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-02-23 21:03:34.000000000","message":"Patch Set 6: Review-Priority+1","accounts_in_message":[],"_revision_number":6},{"id":"ddbe333bd6bee019729b92c6cddfd51498ff4e66","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2026-02-24 03:30:17.000000000","message":"Patch Set 6: Code-Review+1 Review-Priority+2","accounts_in_message":[],"_revision_number":6},{"id":"6f7ea0d64aa84a2c0fd67f30fcbef81371efe1d2","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-02-24 08:14:33.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"d15c642b955337a03523d5b76ec8c982d7f9cdd1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 08:14:55.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"35514f227302bd78382243cc96bd0fa35825c8b1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 10:31:03.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2777acb2b830456185a9e6fe1b835310\n\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/ee59d1d74cfe4aa68beb11cfe5b1fbb7 : SUCCESS in 2h 13m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42cf400d91004e59ad64ba6097b1636e : SUCCESS in 10m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b443a866a4aa43f6b191ded2b8a15cce : SUCCESS in 19m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0fbfefe4d62d46b8b5bd9107b486dfef : SUCCESS in 19m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f2a260fc3754ecdaa085c351886e0a8 : SUCCESS in 10m 37s\n- requirements-check https://zuul.opendev.org/t/openstack/build/7895e640c89642008ec47a72eaf4547d : SUCCESS in 3m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5a582c3ce4fc4516bea4a9647baf3e64 : SUCCESS in 22m 45s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/080db9ac3e3e434ca1f2fb566745dce9 : SUCCESS in 1h 29m 01s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c34f61ae333f4bed86b23d33c5762dd7 : SUCCESS in 1h 16m 58s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/d3f78232abcd412797930741c8401f3a : SUCCESS in 51m 36s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/7b19c83e0c154ed49a357997c7ac69ea : SUCCESS in 30m 50s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/aad484865c7d411896a7332aa8477109 : SUCCESS in 42m 47s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/5aaab633312346849ca441d419025c4f : SUCCESS in 1h 19m 27s","accounts_in_message":[],"_revision_number":6},{"id":"204adadacb7ef2ee321b9926312014586cdf99b5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 10:31:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"fb73ae92dc400cbf4b9ae1abf5dac7a876d8deb6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 10:33:02.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6fe2ae261c3e47faa6baa9552af3008b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9c66172c70d48ce8fec7e5cab7d2e5a : SUCCESS in 1m 10s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3473f5412ee84b0493273a2ede6b1bd2 : SUCCESS in 1m 22s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"7f31aa17ddeb3e30775897444f497b2ec54af403","revisions":{"0c5571158a13c8de7e1a2016d1a3a298ce379090":{"kind":"REWORK","_number":1,"created":"2026-02-17 12:08:40.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/72/977072/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/72/977072/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/977072/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/977072/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/977072/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/72/977072/1"}}},"commit":{"parents":[{"commit":"8c301a0c97cc5c30076d249c98f232f02d9181d5","subject":"Fix ``ChassisAgentWriteEvent`` match condition","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8c301a0c97cc5c30076d249c98f232f02d9181d5"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-03 12:26:30.000000000","tz":540},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-02-16 12:35:16.000000000","tz":60},"subject":"Get rid of pkg_resources","message":"Get rid of pkg_resources\n\n... because it was removed in Python 3.12 [1].\n\n[1] https://docs.python.org/3/whatsnew/3.12.html#ensurepip\n\nNote:\nSelectable entry points were introduced in importlib_metadata 3.6 and\nPython 3.10 .\n\nThis patch also bumps flake8-import-order to 0.19.0.\n\nChange-Id: I1b478a63ad1d73f9f3528939362797ea1fc68534\n(cherry picked from commit b7fe561d0980db5c3d6d5ef9fb5b9a036d5cdbd1)\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0c5571158a13c8de7e1a2016d1a3a298ce379090"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0c5571158a13c8de7e1a2016d1a3a298ce379090"}]},"branch":"refs/heads/stable/2024.2"},"c186605fb01d34d2940208b477d87676e7c5c6bc":{"kind":"REWORK","_number":2,"created":"2026-02-17 13:46:15.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/72/977072/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/72/977072/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/977072/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/977072/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/977072/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/72/977072/2"}}},"commit":{"parents":[{"commit":"8c301a0c97cc5c30076d249c98f232f02d9181d5","subject":"Fix ``ChassisAgentWriteEvent`` match condition","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8c301a0c97cc5c30076d249c98f232f02d9181d5"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-03 12:26:30.000000000","tz":540},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-02-16 14:13:23.000000000","tz":60},"subject":"Get rid of pkg_resources","message":"Get rid of pkg_resources\n\n... because it was removed in Python 3.12 [1].\n\n[1] https://docs.python.org/3/whatsnew/3.12.html#ensurepip\n\nNote:\nSelectable entry points were introduced in importlib_metadata 3.6 and\nPython 3.10 .\n\nThis patch also bumps flake8-import-order to 0.19.0.\n\nChange-Id: I1b478a63ad1d73f9f3528939362797ea1fc68534\n(cherry picked from commit b7fe561d0980db5c3d6d5ef9fb5b9a036d5cdbd1)\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c186605fb01d34d2940208b477d87676e7c5c6bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c186605fb01d34d2940208b477d87676e7c5c6bc"}]},"branch":"refs/heads/stable/2024.2"},"f3929233fcafacf8d68a1f34ad0bbfbc49d4efdf":{"kind":"REWORK","_number":3,"created":"2026-02-23 07:09:15.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/72/977072/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/72/977072/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/977072/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/977072/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/977072/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/72/977072/3"}}},"commit":{"parents":[{"commit":"8c301a0c97cc5c30076d249c98f232f02d9181d5","subject":"Fix ``ChassisAgentWriteEvent`` match condition","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8c301a0c97cc5c30076d249c98f232f02d9181d5"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-03 12:26:30.000000000","tz":540},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-02-21 08:21:57.000000000","tz":60},"subject":"Get rid of pkg_resources","message":"Get rid of pkg_resources\n\n... because it was removed in Python 3.12 [1].\n\n[1] https://docs.python.org/3/whatsnew/3.12.html#ensurepip\n\nNote:\nSelectable entry points were introduced in importlib_metadata 3.6 and\nPython 3.10 .\n\nThis patch also bumps flake8-import-order to 0.19.0.\n\nChange-Id: I1b478a63ad1d73f9f3528939362797ea1fc68534\n(cherry picked from commit b7fe561d0980db5c3d6d5ef9fb5b9a036d5cdbd1)\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f3929233fcafacf8d68a1f34ad0bbfbc49d4efdf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f3929233fcafacf8d68a1f34ad0bbfbc49d4efdf"}]},"branch":"refs/heads/stable/2024.2"},"6ecd3b71992cd643b8cbde89bd5435ffb5dd0a77":{"kind":"REWORK","_number":4,"created":"2026-02-23 08:29:31.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/72/977072/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/72/977072/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/977072/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/977072/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/977072/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/72/977072/4"}}},"commit":{"parents":[{"commit":"8c301a0c97cc5c30076d249c98f232f02d9181d5","subject":"Fix ``ChassisAgentWriteEvent`` match condition","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8c301a0c97cc5c30076d249c98f232f02d9181d5"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-03 12:26:30.000000000","tz":540},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-02-23 08:29:07.000000000","tz":60},"subject":"Get rid of pkg_resources","message":"Get rid of pkg_resources\n\n... because it was removed in Python 3.12 [1].\n\n[1] https://docs.python.org/3/whatsnew/3.12.html#ensurepip\n\nNote:\nSelectable entry points were introduced in importlib_metadata 3.6 and\nPython 3.10 .\n\nThis patch also bumps flake8-import-order to 0.19.0.\n\nChange-Id: I1b478a63ad1d73f9f3528939362797ea1fc68534\n(cherry picked from commit b7fe561d0980db5c3d6d5ef9fb5b9a036d5cdbd1)\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6ecd3b71992cd643b8cbde89bd5435ffb5dd0a77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6ecd3b71992cd643b8cbde89bd5435ffb5dd0a77"}]},"branch":"refs/heads/stable/2024.2"},"a72a6a75ffecbe2d05e67674c9f7541866d1f9f7":{"kind":"REWORK","_number":5,"created":"2026-02-23 15:50:03.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/72/977072/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/72/977072/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/977072/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/977072/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/977072/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/72/977072/5"}}},"commit":{"parents":[{"commit":"8c301a0c97cc5c30076d249c98f232f02d9181d5","subject":"Fix ``ChassisAgentWriteEvent`` match condition","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8c301a0c97cc5c30076d249c98f232f02d9181d5"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-03 12:26:30.000000000","tz":540},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2026-02-23 15:50:03.000000000","tz":0},"subject":"Get rid of pkg_resources","message":"Get rid of pkg_resources\n\n... because it was removed in Python 3.12 [1].\n\n[1] https://docs.python.org/3/whatsnew/3.12.html#ensurepip\n\nNote:\nSelectable entry points were introduced in importlib_metadata 3.6 and\nPython 3.10 .\n\nThis patch also bumps flake8-import-order to 0.19.0.\n\nChange-Id: I1b478a63ad1d73f9f3528939362797ea1fc68534\n(cherry picked from commit b7fe561d0980db5c3d6d5ef9fb5b9a036d5cdbd1)\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a72a6a75ffecbe2d05e67674c9f7541866d1f9f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a72a6a75ffecbe2d05e67674c9f7541866d1f9f7"}]},"branch":"refs/heads/stable/2024.2"},"7f31aa17ddeb3e30775897444f497b2ec54af403":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2026-02-23 16:03:49.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/72/977072/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/72/977072/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/977072/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/977072/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/977072/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/72/977072/6"}}},"commit":{"parents":[{"commit":"8c301a0c97cc5c30076d249c98f232f02d9181d5","subject":"Fix ``ChassisAgentWriteEvent`` match condition","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8c301a0c97cc5c30076d249c98f232f02d9181d5"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-03 12:26:30.000000000","tz":540},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2026-02-23 16:03:49.000000000","tz":0},"subject":"Get rid of pkg_resources","message":"Get rid of pkg_resources\n\n... because it was removed in Python 3.12 [1].\n\n[1] https://docs.python.org/3/whatsnew/3.12.html#ensurepip\n\nNote:\nSelectable entry points were introduced in importlib_metadata 3.6 and\nPython 3.10 .\n\nThis patch also bumps flake8-import-order to 0.19.2.\n\nChange-Id: I1b478a63ad1d73f9f3528939362797ea1fc68534\n(cherry picked from commit b7fe561d0980db5c3d6d5ef9fb5b9a036d5cdbd1)\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7f31aa17ddeb3e30775897444f497b2ec54af403"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7f31aa17ddeb3e30775897444f497b2ec54af403"}]},"branch":"refs/heads/stable/2024.2","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
