)]}'
{"id":"openstack%2Fneutron~977173","triplet_id":"openstack%2Fneutron~master~Icb97630615cbcf39f6c04dcb5489449ebaaab351","project":"openstack/neutron","branch":"master","topic":"bug/2111276","attention_set":{},"removed_from_attention_set":{"36998":{"account":{"_account_id":36998,"name":"Ilia Baikov","display_name":"Ilia Baikov","email":"ilia.baikov@ib.systems","username":"frct1"},"last_update":"2026-02-27 02:28:06.000000000","reason":"\u003cGERRIT_ACCOUNT_36998\u003e replied on the change","reason_account":{"_account_id":36998,"name":"Ilia Baikov","display_name":"Ilia Baikov","email":"ilia.baikov@ib.systems","username":"frct1"}},"5756":{"account":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"last_update":"2026-03-03 20:48:05.000000000","reason":"\u003cGERRIT_ACCOUNT_5756\u003e replied on the change","reason_account":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}},"8655":{"account":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"last_update":"2026-03-09 19:11:05.000000000","reason":"\u003cGERRIT_ACCOUNT_8655\u003e replied on the change","reason_account":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}},"34271":{"account":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"last_update":"2026-03-03 15:58:18.000000000","reason":"\u003cGERRIT_ACCOUNT_34271\u003e replied on the change","reason_account":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}}},"hashtags":[],"change_id":"Icb97630615cbcf39f6c04dcb5489449ebaaab351","subject":"bgp: Connect BGP and Neutron worlds","status":"MERGED","created":"2026-02-17 23:44:00.000000000","updated":"2026-03-09 23:27:24.000000000","submitted":"2026-03-09 23:26:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":41,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"977173-bug/2111276","meta_rev_id":"8a88e6920079787706c1248d4b61e4844bada874","_number":977173,"virtual_id_number":977173,"owner":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-09 23:26:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-03-09 06:51:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-03-04 20:17:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":1,"date":"2026-03-05 13:59:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2026-03-09 06:51:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},{"value":1,"date":"2026-03-06 19:33:07.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}],"CC":[{"_account_id":36998,"name":"Ilia Baikov","display_name":"Ilia Baikov","email":"ilia.baikov@ib.systems","username":"frct1"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-18 00:39:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-02-18 01:44:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-19 16:51:20.000000000","updated_by":{"_account_id":36998,"name":"Ilia Baikov","display_name":"Ilia Baikov","email":"ilia.baikov@ib.systems","username":"frct1"},"reviewer":{"_account_id":36998,"name":"Ilia Baikov","display_name":"Ilia Baikov","email":"ilia.baikov@ib.systems","username":"frct1"},"state":"CC"},{"updated":"2026-02-26 18:58:09.000000000","updated_by":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"reviewer":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"state":"REVIEWER"},{"updated":"2026-02-27 17:20:00.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2026-03-09 06:51:41.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"13b6b1a9fa43b58908806464b83cd7b18a8b2769","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-02-17 23:44:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"de0a5ffab650b1f8089945f5a3e01ae3c1df90ed","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 00:39:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9843682a13bb4cd9a517b10606d0e19d\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/02fdcc05b580465298c8d28b62c2aa82 : SUCCESS in 26m 22s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ebd7565408db4766952790dd24441e96 : SUCCESS in 20m 57s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8c32db90379640018f4ec7815cf44815 : RETRY_LIMIT in 4m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8f894c430c90f79cb64a2bc1ad3486caf6996441","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 01:44:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/eee7cec1fbe8424d943f7fb2f9c9c0a1\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/484452de355941fa90001330911da95c : SUCCESS in 1h 30m 31s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d662dbc7de2b46aa8f1a481d1633454d : SUCCESS in 1h 55m 30s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e7f00600361945488ec312af2eea0319 : SUCCESS in 32m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f6a1d84badd64ddfb6886f93d223c208 : SUCCESS in 1h 16m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd3692e830c8460aa60afe5323efa6f4 : SUCCESS in 13m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e77772833fff472b8ca004ca5050c1c1 : SUCCESS in 25m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2e4ca1894d714145bc1a70c3416bc805 : SUCCESS in 30m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/727cb4b837834fccb94643b3d7e7c795 : SUCCESS in 11m 38s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/361e076f294347e7aa49e3f51bfb1062 : SUCCESS in 47m 57s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/5208f4d039964d04b3fa18ae54350baf : SUCCESS in 1h 10m 21s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f21121ec23d8463f8f0c6163653ee501 : FAILURE in 1h 15m 11s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/0cda0c2ba7e14db6a0e6e679c9b0dd5d : SUCCESS in 1h 14m 53s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/838e6fa8badb4a1a90c43be263f7e598 : SUCCESS in 1h 09m 39s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/05214c1b86b84322aa4cae28ccc98411 : SUCCESS in 1h 34m 17s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d180eb60ce724204925df18a5c63ac92 : SUCCESS in 59m 55s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/9ef53471f4544e0bbf5b15b1bca758f8 : SUCCESS in 1h 04m 55s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/0aaeea4610404fcdb52c43ff1af6e02e : SUCCESS in 1h 09m 32s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5d337cb17c5347769a7125426beccea6 : SUCCESS in 1h 16m 30s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/b528a8933ddb4d36a494130b6d3b388e : SUCCESS in 1h 09m 39s","accounts_in_message":[],"_revision_number":1},{"id":"8315bbc7699e60e22403c2643bf190a17705fb79","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-02-19 00:14:06.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"cfb425acc5c88184bf6ceb2565eb6faf56630651","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-19 00:42:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e05480fa5154451ae0a4f26c68502bd\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d0b8a5ccb6474173b77f0819798d0b5e : SUCCESS in 26m 54s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/315a04f409b64ca39cd85be84c515ff6 : SUCCESS in 21m 24s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a115d8697e0940eb885873a06f932abe : RETRY_LIMIT in 2m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e0ed71e98e69d573d194993139d8b58b7e994f30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-19 02:19:34.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4b45376ac0fd4902b37b010bf654cbc1\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/606ab4398f034a4598df962fcdb6c1fd : SUCCESS in 1h 27m 20s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c7d604d1b903403db6bb3c2c7ebf8f85 : SUCCESS in 2h 02m 39s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6ccaddd262e14ae4bc4f55cd9156cacf : SUCCESS in 31m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d3a36b40c92343a0b5fe4c58ed3babc2 : SUCCESS in 1h 21m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12428a3c06de4d99b25c755f11455335 : SUCCESS in 19m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/abd5be417965469d870191a53b59e564 : POST_FAILURE in 39m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e15eb9b88abe4dbc9b73031331c62583 : SUCCESS in 24m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0982456b717a4f8aa6062d95520d3bdb : SUCCESS in 12m 09s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/e2a3d4f1545d4526b9c599c78a39c3e1 : POST_FAILURE in 1h 02m 40s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/4a6c2769f55446a99dc6f90fc5cf2dab : POST_FAILURE in 1h 12m 20s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/fd49353946534471b0aa4e138d640377 : FAILURE in 1h 19m 32s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/20233c16b4df4d03b739aaf4eab04bff : SUCCESS in 1h 06m 57s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/012790863b974b6f9a9afd0f2faf2636 : POST_FAILURE in 1h 16m 04s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/90ccd7c9d6534eccbcdc2fe62d1996e9 : FAILURE in 1h 14m 45s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/355c1ac3c97c4689867712767c85c656 : SUCCESS in 1h 53m 36s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/9a4b4441c82945b78d28311554639a5c : SUCCESS in 1h 16m 31s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/05e955e7cbd94d28b46753eb8c5dd5cd : SUCCESS in 57m 43s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/87d0636a5d294097be903e69ac8d0378 : SUCCESS in 1h 09m 06s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4527676aa20d4cba8d2e215d70bfff45 : SUCCESS in 1h 20m 34s","accounts_in_message":[],"_revision_number":2},{"id":"189f3f641585f179ca58951d4dec6346a4801982","author":{"_account_id":36998,"name":"Ilia Baikov","display_name":"Ilia Baikov","email":"ilia.baikov@ib.systems","username":"frct1"},"date":"2026-02-19 16:51:20.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fd3c390e0b128416cdee7cc92ea1c6de94df7094","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-02-19 21:27:41.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"c0ca0d6d2f6e9f1af7428d7f8f43d4a7d3edb4ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-02-19 22:03:45.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"927f19fad859d398b417b874462fc9094e9dad2a","author":{"_account_id":36998,"name":"Ilia Baikov","display_name":"Ilia Baikov","email":"ilia.baikov@ib.systems","username":"frct1"},"date":"2026-02-19 22:44:36.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"61955246ab9f9bcac29576b5936345310856b5d5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-19 22:45:50.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37e00145705543dba9f7864273e95f80\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e540173183e64d999a4142389d2b4717 : POST_FAILURE in 30m 28s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c4a32e532bb94bbc9ac7f34ee1990669 : SUCCESS in 22m 50s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/21f102e247d14b47ab1748ee0c079e94 : RETRY_LIMIT in 3m 27s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b2c443f802ae7bd18e4e55fd9f32e53a310f0443","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-20 00:24:46.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c718390cfb04c09bfbf77ef36af90ba\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/2ed3385a9f6e4b2ab94502244d240d7f : SUCCESS in 1h 17m 42s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/0f2b97d8238441cc833eb3befada820b : SUCCESS in 2h 19m 58s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/edfb2397a03543948675b38bb84902ef : SUCCESS in 31m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b054898e80f84cfc9a65ed3fdb514905 : SUCCESS in 38m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/717fc9a4f2394ad0a40c3ab2cacb0b0b : SUCCESS in 25m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f55044b5079343bbbe6a82bf3eba0330 : SUCCESS in 21m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6962dd9bfaf64ca288db199e8a8db4ba : SUCCESS in 27m 51s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6c39eb476b584a988ed2630a4e0d102f : SUCCESS in 35m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed132a38823c4af59226c4ae5bdeb9e0 : SUCCESS in 9m 18s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/4b4b3568836e43e8824d4057e8da7e1e : SUCCESS in 1h 03m 34s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/6de16f65ad7a49a9869bba148291965d : SUCCESS in 37m 40s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/2401309cefaa410c99b228fda4c71570 : SUCCESS in 1h 20m 20s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/05bfe80c00e94e9facc51216dcf6b4be : SUCCESS in 1h 13m 42s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/301fa0c332b44187890e1e76f709307f : SUCCESS in 52m 47s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/7d8574ad955f42d582f5035164821da2 : SUCCESS in 1h 49m 21s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/48887a9322fa454fa0f126909ad93640 : SUCCESS in 1h 29m 39s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/21cfeb2c388d4beb9e858c629a62b8d1 : SUCCESS in 50m 46s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/43710fe6000047ca864b229469a7bc10 : SUCCESS in 1h 15m 46s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0628e2659e654e8ca75d7011e2557959 : SUCCESS in 37m 32s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/068680c56ab34f91bb930a7e8aa2c066 : SUCCESS in 1h 21m 45s","accounts_in_message":[],"_revision_number":4},{"id":"274f8bcab5d47ba4fac4e6fd2489c017cfbd14b2","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-02-25 16:55:41.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a1967b4a7c3539e6e2a57417a4758e6e36961431","author":{"_account_id":36998,"name":"Ilia Baikov","display_name":"Ilia Baikov","email":"ilia.baikov@ib.systems","username":"frct1"},"date":"2026-02-25 17:27:02.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d5e60ae99b9916cd8441c324d27f92a537fd68b9","author":{"_account_id":36998,"name":"Ilia Baikov","display_name":"Ilia Baikov","email":"ilia.baikov@ib.systems","username":"frct1"},"date":"2026-02-25 18:54:33.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b56a02a04b385e7544751403ac2557e0d05b6db5","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-02-25 21:55:14.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"efdac75dcb0c1025dfc9e9c24d13a0cf1d33c203","author":{"_account_id":36998,"name":"Ilia Baikov","display_name":"Ilia Baikov","email":"ilia.baikov@ib.systems","username":"frct1"},"date":"2026-02-25 23:04:19.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7e5dc081e1c95ca86613eb9b3da4b3865e6e7beb","author":{"_account_id":36998,"name":"Ilia Baikov","display_name":"Ilia Baikov","email":"ilia.baikov@ib.systems","username":"frct1"},"date":"2026-02-25 23:21:05.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f68c2089e73ab52b5183dab74a5ec05eafb59d9c","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-02-26 13:07:35.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3e7b927f01ae8ffea28099161b10a15e41ac2d80","author":{"_account_id":36998,"name":"Ilia Baikov","display_name":"Ilia Baikov","email":"ilia.baikov@ib.systems","username":"frct1"},"date":"2026-02-26 19:17:51.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"99ec7caeacc0142c6af542f2b6211f159f98d817","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-02-26 23:28:48.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"405cb65947c15f6246f1a0e9dde21e367ba9c443","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-02-26 23:29:29.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"46d96d4f17a7f7c4097a4eb158a57dded43464f3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 00:03:09.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe19a790aa80429bbe00f8fd8c9ba793\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/db819516be0f4e58b14a1a9b91b316a1 : SUCCESS in 32m 32s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/200c20552a5f43b9b7e3bcb95d6390f0 : RETRY_LIMIT in 3m 31s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/849b827ae33944199523b5065e835c59 : SUCCESS in 25m 31s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8fec71da6a4a4e95af7bdfb1dc606726 : SUCCESS in 22m 02s","accounts_in_message":[],"_revision_number":5},{"id":"4eabeeb92c8e4ff71928f657b64752653144a801","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 01:42:46.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/18a4c8a491e943088438e860c4b6d314\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d4a6f5b7710a41f98f1b18ac26bb3d4c : SUCCESS in 1h 14m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/69deb7edc53e42509c7ce05e31a78fd5 : SUCCESS in 32m 27s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/e02d192390eb43f9a1a08af7fbe6247b : SUCCESS in 1h 38m 54s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ef8076e638554da8bc222998d9b39d75 : SUCCESS in 2h 11m 33s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3c71cb4d2cfe494c8138e51dec814795 : SUCCESS in 34m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6558f2cac5c1402aaf742042fadfb9b0 : SUCCESS in 1h 08m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9273b00fc5b14d17ad94abe09ee0d292 : SUCCESS in 13m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4f507aeec7424426b3adee97cd3d6f97 : SUCCESS in 27m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/20f45cbfd2e241468237f8c053833329 : SUCCESS in 15m 36s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/67b0da6be5264310a8c32a2b032985cc : SUCCESS in 22m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8e4bb5473b5464fb256ab50a02a685e : SUCCESS in 12m 15s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/05630d2142ba4703a6fb6fb6bd7debab : SUCCESS in 54m 23s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/476ac1054a75400580904b62b5d48505 : SUCCESS in 1h 09m 26s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/1a1d266f2dae4389a579637180d4b1fd : FAILURE in 31m 29s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/ef284f85b15a4db09bfcd5fb9ad3d86d : SUCCESS in 1h 11m 26s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/53493067b3b04c7e8c159583950a2fe5 : SUCCESS in 1h 09m 05s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/1c80dc82925a4e3ea3959f6052f14205 : SUCCESS in 1h 49m 14s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f0cebce0fac749a49bf4e4cb0e8ef318 : SUCCESS in 1h 32m 04s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/aa2dce733f81426e9e09511fc802dd06 : SUCCESS in 1h 12m 20s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/fc328a89599f4225b8ac44c4a5a962d6 : SUCCESS in 1h 33m 35s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f1fc6bd1bdd24376987f3ac5435af5c7 : SUCCESS in 1h 26m 41s","accounts_in_message":[],"_revision_number":5},{"id":"7d4e66eb8bf5f577e61d7f2e552d60261148815f","author":{"_account_id":36998,"name":"Ilia Baikov","display_name":"Ilia Baikov","email":"ilia.baikov@ib.systems","username":"frct1"},"date":"2026-02-27 02:28:06.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"583016941721b636aead5c74ad3f73dcc73305cc","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2026-02-27 17:20:00.000000000","message":"Patch Set 4: Code-Review-1\n\nCopied votes on follow-up patch sets have been updated:\n* Code-Review-1 has been copied to patch set 5 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\").\n\n(9 comments)","accounts_in_message":[],"_revision_number":4},{"id":"abb8bb5a6b62d0cb0fb01fb1eb3b47e62a3112b9","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-03-02 20:04:51.000000000","message":"Patch Set 5:\n\n(7 comments)","accounts_in_message":[],"_revision_number":5},{"id":"f635afbdb133b62f1547beea8c279da3c281b677","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-03-02 20:13:44.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"e3ab4faf90a556e3e60ba3b2030814af1e256aa5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-03-02 20:19:43.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"512d700c9108bef1a640b91e83379da75c6d1dbd","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2026-03-02 21:34:45.000000000","message":"Patch Set 7: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"f550538eccbb4aba7fd62b74457110f4d6d10ab2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-02 22:17:56.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/37981d56fa454e3ebee6c44a3b1dfc73\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4dbcfc0800c14e68b460e1642905ef5c : SUCCESS in 1h 01m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0f0f0e4e0a844e4a9e1e91faf6f522d0 : SUCCESS in 1h 01m 55s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/abdd2f83b8d14fb8a69c68cb3de844eb : SUCCESS in 1h 32m 27s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/485a27b1b3dd4e3ea85816228a657232 : SUCCESS in 1h 57m 09s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/263d824c434d4a4f9a20167b545ac5aa : SUCCESS in 31m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4a1cabf420764e54a941a24f48829c37 : SUCCESS in 43m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb732d3ae0ef44e0acbb796092a3d5be : SUCCESS in 10m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e27cd4bc7eee4c41aef0a02f8cb7b44a : SUCCESS in 29m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e54691d8afbb4daf849201f973173c4a : SUCCESS in 25m 04s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d891dafcd6da493591b3bfaeb9849971 : SUCCESS in 21m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18056f05fd1b49e5a58373e3d4a090d6 : SUCCESS in 12m 14s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/85e3f56799554d5e88d0679c56c517a4 : SUCCESS in 50m 59s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/10544924e6514630912fc71806a4b9d9 : SUCCESS in 59m 34s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/4615d3fa1d1147248a8d69212530d523 : FAILURE in 25m 27s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/afe23d268b884ff9907ea96bca6ec447 : SUCCESS in 1h 12m 13s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/e8759601e7e74903bdb86a7db49dfcb5 : SUCCESS in 1h 13m 44s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/3180a37ee741438eb9cb239e915b32e2 : SUCCESS in 1h 52m 26s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f6bbc0a4ca5e4867af19a6b1105158b0 : SUCCESS in 1h 06m 02s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/20ff1d51126a4fb3893741021881eb52 : SUCCESS in 1h 05m 14s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/042c5eb46a7c44de82409ee4ae334bd9 : SUCCESS in 37m 40s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/92a6d09ac7e3432988ac73b8b1fb176a : SUCCESS in 1h 06m 59s","accounts_in_message":[],"_revision_number":7},{"id":"519aef482b3befd61a0fca3ff29a65f8cea2310f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-02 23:19:31.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/573f194ca0594301a364ea049c99e829\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f104891830464a209d21d3919ffff3e1 : SUCCESS in 26m 00s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2c8063cd0b1b4973a1300e2d3e772874 : RETRY_LIMIT in 3m 19s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/408a10e8140842faaeaeecf5131bffed : SUCCESS in 44m 32s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f2a75c0c40bb4b7b919a1efb3fe7e114 : SUCCESS in 28m 57s","accounts_in_message":[],"_revision_number":7},{"id":"20ae7c7eb032a18bfb4bef8334553ee1f0b2965c","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-03-03 12:58:55.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"a016e910f10c88a3e2bf22ebb85f440e3e2640de","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2026-03-03 15:58:18.000000000","message":"Patch Set 7: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"c2c0dc2c6f0df6c2c2c74d2eb50ea9b0d5add132","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-03-03 20:43:55.000000000","message":"Patch Set 7: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"b42721342814977c5eb2dab054c14885dcd7f703","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2026-03-03 20:48:05.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"83c2ae259221d2df6bd9ce895482795bfbe8208e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 21:30:03.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/588088ce436f43ef943c37f6166c4b4c\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7ac3edf9f674406a8b98e42f3a66397d : SUCCESS in 24m 54s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/72295371705f4ed0a67b306ef0c1690c : NODE_FAILURE Node(set) request 47648d844ea74dfb9b2de22bcf29a9fd failed in 0s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/8aa96f6bb4a84967a27d72cbfa07c96b : SUCCESS in 40m 15s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ac0bbb2611354d70adfad2c2fa53e588 : SUCCESS in 22m 11s","accounts_in_message":[],"_revision_number":7},{"id":"78464bf84bedb8857ed3169e651a477dc294f6f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-03-03 22:10:13.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":8},{"id":"942e8da397e5528ca1c8bfc2340db18296c5c189","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-03-03 22:10:38.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"32f07476b11c73efb735aa279a4173c63afc8a91","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2026-03-03 22:28:35.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"2bbca3d6fa4d121f74d348408f63aab06a3a696d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 23:22:01.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0632650887c84ec8958979183e1a5286\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/09b6d3c489194646bda292ade954055f : SUCCESS in 1h 09m 41s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/01c8286cdf4c452c91cc032ac93077d8 : RETRY_LIMIT in 3m 46s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/93904a6a470d4efd9a0472154949fd41 : TIMED_OUT in 1h 01m 10s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/293de3c0bbe049a380718bbb361c9b41 : SUCCESS in 57m 39s","accounts_in_message":[],"_revision_number":8},{"id":"204c59e1bcaa0c2523f60e98971074b8c3e9ce5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-04 00:07:49.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a10e0178ccf4c7294cafc3cae11bb77\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/66ee2fff84554733a49789ae43a7a307 : SUCCESS in 1h 07m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/08e2698b6eff4ec1aaae65ee0a3a8d9b : SUCCESS in 1h 03m 14s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/03af88227c524e21bc04862666089ad1 : SUCCESS in 1h 38m 00s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/552b643941a7457a97232892123dd91e : SUCCESS in 1h 45m 29s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5f48e3af29e2453ebf742f705159176a : SUCCESS in 32m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5e179a3c33e14db3a802dbfeda224d4b : SUCCESS in 1h 04m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6da84f1bee594a9ba7642a8afcd47cdd : SUCCESS in 28m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e1ad8cd526114e51b4aac806bf4a9a8b : SUCCESS in 36m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/95f5159b307749cb823cc5d579f8d430 : SUCCESS in 26m 47s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/89399d9237ff4b989b9e66e20bac5652 : SUCCESS in 33m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f53cf4cb491e4d838626f376ea6831be : SUCCESS in 11m 35s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/8662c224998b45c9b5960794fe5c4cb8 : SUCCESS in 1h 01m 48s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/14e9e333cc234479bd667480c8c74748 : SUCCESS in 54m 13s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c33b8c45f2194702ba349df01d46211e : SUCCESS in 55m 03s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b50106a15b524a7998632ae27fb2d21d : SUCCESS in 46m 37s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/fddbc9e3366c4ddb970838ae633c6224 : SUCCESS in 1h 02m 01s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ed3dc315b30940cda8a1027ad1258c27 : SUCCESS in 1h 55m 45s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ee110e0d24544877836ac63a06aefa4b : SUCCESS in 57m 19s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/7c875e11cfc843a2a712e28ed6e4e9ee : SUCCESS in 1h 05m 45s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/13cf12e3dadd41cc8840a175005dc7cc : SUCCESS in 1h 01m 50s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a02c970af5d04ca198b01dd73fc768a0 : SUCCESS in 1h 07m 56s","accounts_in_message":[],"_revision_number":8},{"id":"16f33f7d661667bbfc3b4dba98bd5375824aaa49","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2026-03-04 19:43:52.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"8f78af78e80936cf05de7426ae3a747329182875","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2026-03-04 20:17:00.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"38bbb37dbb33f232b93587e01b3c529a3b55e83d","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2026-03-05 13:59:33.000000000","message":"Patch Set 8: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"316861d32af32ec4c7273582e41fc2091c3d60d1","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2026-03-06 17:40:42.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"ef78a44f10ed50bc433b313d2d681cfa419b1228","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-03-06 19:33:07.000000000","message":"Patch Set 8: Review-Priority+1","accounts_in_message":[],"_revision_number":8},{"id":"3f2e7220e10725ca8a6f7e924ecbddcc506e0af8","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-03-09 06:51:41.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"26e9c93087ca2813c7c09029d36fa9f549e0b38b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 06:52:05.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"8a50eff7551c882ba6d6ec08fe4f4d68c015b46e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 08:35:08.000000000","message":"Patch Set 8: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6e87e24d397445db94b351c90953bc97\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e452b69a054144cd9ed9acaa96645fc3 : SUCCESS in 1h 15m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d1f10d7f1150488db3fe6a9eb4c7fb8a : SUCCESS in 1h 03m 02s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/613ab304959346b4aad73c60abe03a32 : FAILURE in 1h 37m 47s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/5b2e7b509ae14daeb5feafba83f781b2 : SUCCESS in 1h 42m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/687d93aa7f6042edadd964f0640da176 : SUCCESS in 25m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/60ba984c4be5409d93c89996ec8fc0bb : SUCCESS in 34m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0d7fb0d30cd34cb098d6028b33631552 : SUCCESS in 15m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93891a899f0a462983563c868a02a41e : SUCCESS in 12m 32s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/62f5a46b60c7413db90451bf4e1ea4f1 : SUCCESS in 47m 13s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/b0f7dc9976214f198d2cdb512eaa9db4 : SUCCESS in 1h 36m 09s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/7ba9f4a2c1104b25844c757d801b5075 : SUCCESS in 1h 10m 55s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f25edfeeac71495d8567da8662562168 : SUCCESS in 41m 23s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/1917bdf990e34dc1baa37e598519c0e6 : SUCCESS in 47m 17s","accounts_in_message":[],"_revision_number":8},{"id":"b354e16e8823996cc16a917f24b5bb88955cd613","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-03-09 19:11:05.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"604ae2effc764b195f6e3b241fcbba7d7b6f6eb0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 19:57:35.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/86e7d0751e9347ab82b2d0034bc33005\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/018c1658b3b544c98716f0f8e6cc0a51 : SUCCESS in 44m 57s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/464d80ef68ac48aca83259cf804fcb74 : SUCCESS in 17m 56s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/19e64a3c948043c2b10d55f77382e1a5 : SUCCESS in 23m 31s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"69898074d96e822d3d9034fa5da2226a4e5c0941","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 21:27:02.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b5879209faef48c9b34422eb421318cd\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/839d6842b00d40eab5c8bf88a0d56543 : SUCCESS in 1h 08m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/04eb043ac54c45cf82983f60e67eeb29 : SUCCESS in 32m 18s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/467f5b312f934565b06d1eb8e69fa2bc : SUCCESS in 1h 16m 16s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/576787e2e60b48bb9378daeab7a48f01 : SUCCESS in 2h 15m 00s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d01a2a7cbd81484cb3a422ae39205108 : SUCCESS in 31m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9fac137c00e243d1b319114024d7d401 : SUCCESS in 1h 06m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0aebaee1b7554d36a2e52ebc5eb45812 : SUCCESS in 23m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/923defc9afd543dcb43896837112cc4e : SUCCESS in 24m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fb30a0473c214763b6f5dead550369fb : SUCCESS in 27m 09s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/132e1cdcbe6c4a2da441f98b08773d18 : SUCCESS in 21m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50c65dab533e4d2392d0d358cceba299 : SUCCESS in 6m 20s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/bb838ab2a5984b09aa184f6274c79f2e : SUCCESS in 47m 39s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/206dc398f8e84bb1a29ba951ab1bf7d0 : SUCCESS in 39m 41s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/1aad781f57134f26aae4f26a52c3045a : SUCCESS in 1h 20m 31s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/757cfcf496d744a1bafcbaa432941a6d : SUCCESS in 1h 02m 07s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/23802ce0828c4f0c99885e1f9c04b91d : SUCCESS in 1h 22m 05s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/60dadb34f8644737bd1aad6b9aa08ce9 : SUCCESS in 1h 50m 29s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/5b17a3923f95429ea469c4bbd27d1090 : SUCCESS in 1h 19m 37s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/1b410ecbaa2149a18def8b9cd2fe168a : SUCCESS in 1h 04m 53s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/257fdc032b0e4d0881d28663c9ed83c5 : SUCCESS in 1h 14m 55s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/09de84349bc544c39ec83067239e3737 : SUCCESS in 1h 37m 30s","accounts_in_message":[],"_revision_number":8},{"id":"ceefae4bdd28cefab5c256a361d9e11e07668470","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 21:27:28.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"17b2ae7ae3a9d9370b685f6a56c2408a9076f03e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 23:26:15.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a6ef6b046d84606a1ac43e501115cf3\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/18970a8934b949549ee16ac24811d9a5 : SUCCESS in 1h 05m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c5fdce6649b0451598c2360d851c9579 : SUCCESS in 34m 45s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/6825382b803748cc94fbb7095eee0b4b : SUCCESS in 1h 20m 41s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/09ac292478734ebc98f2b9d872866a7a : SUCCESS in 1h 57m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f00446dbf8e646dc85171672e2fef508 : SUCCESS in 29m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/16f485e538cd47aab36d3d8bdeeb0d41 : SUCCESS in 21m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/076f5797f23c45d482c1a65efc2d2d0e : SUCCESS in 20m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5ff69ecab454c42b13fb5405b0c725e : SUCCESS in 12m 54s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c45da47215f94e22a13eb9e1e0689e87 : SUCCESS in 1h 14m 18s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/7e8b99e92c214a5daed6bd7ab8ab5e43 : SUCCESS in 1h 34m 43s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/bcdd432e1bc1428a9206b20936e0d295 : SUCCESS in 51m 26s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/25ead10d68c448638e7731954c4a87ed : SUCCESS in 1h 06m 28s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/20343fc6d1a04a78bc5da0442e93071b : SUCCESS in 1h 04m 08s","accounts_in_message":[],"_revision_number":8},{"id":"1ede58dd6ca916eacd8aaf2b876d796c3808e433","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 23:26:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"8a88e6920079787706c1248d4b61e4844bada874","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 23:27:24.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e90e131dc912474999ee6c5452747381\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abe05e4611784bcc9d1f9d7858b90264 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"0a1fb53097dca5811b95c426b5cb1755ad8278fe","revisions":{"e96ca5dd8b3d83b2429fb115fcdf4ab22c870a62":{"kind":"REWORK","_number":1,"created":"2026-02-17 23:44:00.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/73/977173/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/73/977173/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/977173/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/977173/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/977173/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/73/977173/1"}}},"commit":{"parents":[{"commit":"ba75e06075fe6e4b3d8ceff4edb3a0fd83355ba3","subject":"Add neutron-id to DHCP_Options entries","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ba75e06075fe6e4b3d8ceff4edb3a0fd83355ba3"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-02-10 21:23:09.000000000","tz":0},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-02-17 23:43:13.000000000","tz":0},"subject":"bgp: Connect BGP and Neutron worlds","message":"bgp: Connect BGP and Neutron worlds\n\nThe patch adds commands to connect Neutron and BGP. The commands are\nonly called from the Full sync in this patch, the events will come\nlater to make it more reviewable.\n\nAssisted-By: claude-4.5-sonnet\nRelated-Bug: #2111276\n\nChange-Id: Icb97630615cbcf39f6c04dcb5489449ebaaab351\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e96ca5dd8b3d83b2429fb115fcdf4ab22c870a62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e96ca5dd8b3d83b2429fb115fcdf4ab22c870a62"}]},"branch":"refs/heads/master"},"590afafc5306e414492503a7c06fef476d41a5da":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-02-19 00:14:06.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/73/977173/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/73/977173/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/977173/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/977173/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/977173/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/73/977173/2"}}},"commit":{"parents":[{"commit":"388a90c098e97378764198b5743e813173c90342","subject":"bgp: Poison BGP worker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/388a90c098e97378764198b5743e813173c90342"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-02-10 21:23:09.000000000","tz":0},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-02-19 00:13:46.000000000","tz":0},"subject":"bgp: Connect BGP and Neutron worlds","message":"bgp: Connect BGP and Neutron worlds\n\nThe patch adds commands to connect Neutron and BGP. The commands are\nonly called from the Full sync in this patch, the events will come\nlater to make it more reviewable.\n\nAssisted-By: claude-4.5-sonnet\nRelated-Bug: #2111276\n\nChange-Id: Icb97630615cbcf39f6c04dcb5489449ebaaab351\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/590afafc5306e414492503a7c06fef476d41a5da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/590afafc5306e414492503a7c06fef476d41a5da"}]},"branch":"refs/heads/master"},"5c9d33b39ea71f217918262c3306ebd121cb2541":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-02-19 21:27:41.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/73/977173/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/73/977173/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/977173/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/977173/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/977173/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/73/977173/3"}}},"commit":{"parents":[{"commit":"3e1b2d231e1c75a08301e159d3d6935dc943eef9","subject":"Add neutron-id to DHCP_Options entries","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3e1b2d231e1c75a08301e159d3d6935dc943eef9"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-02-10 21:23:09.000000000","tz":0},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-02-19 21:17:12.000000000","tz":0},"subject":"bgp: Connect BGP and Neutron worlds","message":"bgp: Connect BGP and Neutron worlds\n\nThe patch adds commands to connect Neutron and BGP. The commands are\nonly called from the Full sync in this patch, the events will come\nlater to make it more reviewable.\n\nAssisted-By: claude-4.5-sonnet\nRelated-Bug: #2111276\n\nChange-Id: Icb97630615cbcf39f6c04dcb5489449ebaaab351\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5c9d33b39ea71f217918262c3306ebd121cb2541"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5c9d33b39ea71f217918262c3306ebd121cb2541"}]},"branch":"refs/heads/master"},"422c76f6d871d7d20c68a299bac3015ebaafec77":{"kind":"REWORK","_number":4,"created":"2026-02-19 22:03:45.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/73/977173/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/73/977173/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/977173/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/977173/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/977173/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/73/977173/4"}}},"commit":{"parents":[{"commit":"3e1b2d231e1c75a08301e159d3d6935dc943eef9","subject":"Add neutron-id to DHCP_Options entries","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3e1b2d231e1c75a08301e159d3d6935dc943eef9"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-02-10 21:23:09.000000000","tz":0},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-02-19 22:03:28.000000000","tz":0},"subject":"bgp: Connect BGP and Neutron worlds","message":"bgp: Connect BGP and Neutron worlds\n\nThe patch adds commands to connect Neutron and BGP. The commands are\nonly called from the Full sync in this patch, the events will come\nlater to make it more reviewable.\n\nAssisted-By: claude-4.5-sonnet\nRelated-Bug: #2111276\n\nChange-Id: Icb97630615cbcf39f6c04dcb5489449ebaaab351\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/422c76f6d871d7d20c68a299bac3015ebaafec77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/422c76f6d871d7d20c68a299bac3015ebaafec77"}]},"branch":"refs/heads/master"},"ccd1e29c3d13f7d13a127c5b1cd61a489e96d98a":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-02-26 23:28:48.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/73/977173/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/73/977173/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/977173/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/977173/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/977173/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/73/977173/5"}}},"commit":{"parents":[{"commit":"659027692a2e169ae842ebd2c84b324b1086395a","subject":"Merge \"Add neutron-id to DHCP_Options entries\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/659027692a2e169ae842ebd2c84b324b1086395a"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-02-10 21:23:09.000000000","tz":0},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-02-26 23:28:33.000000000","tz":0},"subject":"bgp: Connect BGP and Neutron worlds","message":"bgp: Connect BGP and Neutron worlds\n\nThe patch adds commands to connect Neutron and BGP. The commands are\nonly called from the Full sync in this patch, the events will come\nlater to make it more reviewable.\n\nAssisted-By: claude-4.5-sonnet\nRelated-Bug: #2111276\n\nChange-Id: Icb97630615cbcf39f6c04dcb5489449ebaaab351\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ccd1e29c3d13f7d13a127c5b1cd61a489e96d98a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ccd1e29c3d13f7d13a127c5b1cd61a489e96d98a"}]},"branch":"refs/heads/master"},"4f6a5c4c16aeea1b634ef2d15b1b8fa3b8abcdf5":{"kind":"REWORK","_number":6,"created":"2026-03-02 20:13:44.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/73/977173/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/73/977173/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/977173/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/977173/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/977173/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/73/977173/6"}}},"commit":{"parents":[{"commit":"659027692a2e169ae842ebd2c84b324b1086395a","subject":"Merge \"Add neutron-id to DHCP_Options entries\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/659027692a2e169ae842ebd2c84b324b1086395a"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-02-10 21:23:09.000000000","tz":0},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-03-02 20:06:33.000000000","tz":0},"subject":"bgp: Connect BGP and Neutron worlds","message":"bgp: Connect BGP and Neutron worlds\n\nThe patch adds commands to connect Neutron and BGP. The commands are\nonly called from the Full sync in this patch, the events will come\nlater to make it more reviewable.\n\nAssisted-By: claude-4.5-sonnet\nRelated-Bug: #2111276\n\nChange-Id: Icb97630615cbcf39f6c04dcb5489449ebaaab351\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4f6a5c4c16aeea1b634ef2d15b1b8fa3b8abcdf5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4f6a5c4c16aeea1b634ef2d15b1b8fa3b8abcdf5"}]},"branch":"refs/heads/master"},"dc571c192f565759bd95030a4a729ee9b966acdb":{"kind":"REWORK","_number":7,"created":"2026-03-02 20:19:43.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/73/977173/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/73/977173/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/977173/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/977173/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/977173/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/73/977173/7"}}},"commit":{"parents":[{"commit":"659027692a2e169ae842ebd2c84b324b1086395a","subject":"Merge \"Add neutron-id to DHCP_Options entries\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/659027692a2e169ae842ebd2c84b324b1086395a"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-02-10 21:23:09.000000000","tz":0},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-03-02 20:19:21.000000000","tz":0},"subject":"bgp: Connect BGP and Neutron worlds","message":"bgp: Connect BGP and Neutron worlds\n\nThe patch adds commands to connect Neutron and BGP. The commands are\nonly called from the Full sync in this patch, the events will come\nlater to make it more reviewable.\n\nAssisted-By: claude-4.5-sonnet\nRelated-Bug: #2111276\n\nChange-Id: Icb97630615cbcf39f6c04dcb5489449ebaaab351\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dc571c192f565759bd95030a4a729ee9b966acdb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dc571c192f565759bd95030a4a729ee9b966acdb"}]},"branch":"refs/heads/master"},"0a1fb53097dca5811b95c426b5cb1755ad8278fe":{"kind":"REWORK","_number":8,"created":"2026-03-03 22:10:13.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/73/977173/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/73/977173/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/977173/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/977173/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/977173/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/73/977173/8"}}},"commit":{"parents":[{"commit":"659027692a2e169ae842ebd2c84b324b1086395a","subject":"Merge \"Add neutron-id to DHCP_Options entries\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/659027692a2e169ae842ebd2c84b324b1086395a"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-02-10 21:23:09.000000000","tz":0},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-03-03 22:08:41.000000000","tz":0},"subject":"bgp: Connect BGP and Neutron worlds","message":"bgp: Connect BGP and Neutron worlds\n\nThe patch adds commands to connect Neutron and BGP. The commands are\nonly called from the Full sync in this patch, the events will come\nlater to make it more reviewable.\n\nAssisted-By: claude-4.5-sonnet\nRelated-Bug: #2111276\n\nChange-Id: Icb97630615cbcf39f6c04dcb5489449ebaaab351\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0a1fb53097dca5811b95c426b5cb1755ad8278fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0a1fb53097dca5811b95c426b5cb1755ad8278fe"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
