)]}'
{"id":"openstack%2Fneutron~977592","triplet_id":"openstack%2Fneutron~master~I0e26feece4e336a9cd1ad40c0840f771c1724e05","project":"openstack/neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-02-26 12:32:23.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I0e26feece4e336a9cd1ad40c0840f771c1724e05","subject":"DNM: Try upper-constraints bump","status":"ABANDONED","created":"2026-02-23 05:18:33.000000000","updated":"2026-02-26 12:32:23.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a2e4c12e435b7e3127643a2ba2c0e19c43946dfa","_number":977592,"virtual_id_number":977592,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-02-23 06:03:15.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-23 05:46:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-02-23 06:03:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9d79ec0dedf4aff93b932a379f50af6aa3a13805","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-02-23 05:18:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"674c89ef21ca9da613038654e28e4c9bff5e335b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-23 05:46:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19b7099614a04aff8398ede83c500111\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/423cc439ac5d46bc9abd591bebab2603 : SUCCESS in 26m 10s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/49fa95e39b05433ca3c018dcb84e855f : SUCCESS in 19m 50s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d6a9c850c59e41c5a0787ffbfcc33a62 : RETRY_LIMIT in 3m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"04bb3bc3edcf275dcae01d2dcfd239a319e4351c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-23 06:03:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0557ae84f3ae4c11954d60138b51ff6d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/37989bbb499949979882bb0d7bd1bbb7 : SUCCESS in 42m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4c7f1d6be4e44a480717433283aa978 : SUCCESS in 20m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0a1b4d41c0554439b4c516e4c091808e : SUCCESS in 20m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b92cee312fe64214b34bf02b144427a3 : SUCCESS in 23m 57s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ce40215081334d609fa2e989d4e3dcda : SUCCESS in 21m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/687ff60535fc483baf86903bf91673ed : SUCCESS in 11m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/096a44aa726f41a292829a6750c5783a : SUCCESS in 23m 56s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f79b91f7ff4a4caeba915527ba2a0cb1 : FAILURE in 31m 12s","accounts_in_message":[],"_revision_number":1},{"id":"a2e4c12e435b7e3127643a2ba2c0e19c43946dfa","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-02-26 12:32:23.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1fe86154a7c81233e1d11aaaa6989b9981fed9fe","revisions":{"1fe86154a7c81233e1d11aaaa6989b9981fed9fe":{"kind":"REWORK","_number":1,"created":"2026-02-23 05:18:33.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/92/977592/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/92/977592/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/977592/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/977592/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/977592/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/92/977592/1"}}},"commit":{"parents":[{"commit":"9656a5fcf2c706d2f159655f9fa40c9768bbf5c1","subject":"Merge \"bgp: Introduce BGP chassis commands\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9656a5fcf2c706d2f159655f9fa40c9768bbf5c1"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-23 05:18:06.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-23 05:18:06.000000000","tz":540},"subject":"DNM: Try upper-constraints bump","message":"DNM: Try upper-constraints bump\n\nDepends-on: https://review.opendev.org/c/openstack/requirements/+/977591\nChange-Id: I0e26feece4e336a9cd1ad40c0840f771c1724e05\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1fe86154a7c81233e1d11aaaa6989b9981fed9fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1fe86154a7c81233e1d11aaaa6989b9981fed9fe"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
