)]}'
{"id":"openstack%2Fneutron~978005","triplet_id":"openstack%2Fneutron~master~I50074496cd88637c620c18d1a07a22e77b25624e","project":"openstack/neutron","branch":"master","topic":"bug/2111276","attention_set":{},"removed_from_attention_set":{"5756":{"account":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"last_update":"2026-03-03 16:41:49.000000000","reason":"\u003cGERRIT_ACCOUNT_5756\u003e replied on the change","reason_account":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}},"8655":{"account":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"last_update":"2026-03-07 17:38:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I50074496cd88637c620c18d1a07a22e77b25624e","subject":"bgp: Fix UpdateLocalOVSEvent match","status":"MERGED","created":"2026-02-25 17:35:44.000000000","updated":"2026-03-07 17:39:21.000000000","submitted":"2026-03-07 17:38:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"978005-bug/2111276","meta_rev_id":"32c62662f5d3a52f5d8a1254da88a8dd1428b57c","_number":978005,"virtual_id_number":978005,"owner":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-07 17:38:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"recommended":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"all":[{"value":1,"date":"2026-03-05 12:21:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},{"value":2,"date":"2026-03-07 15:17:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2026-03-03 16:41:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-03-05 15:50:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},{"value":1,"date":"2026-03-07 15:17:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"all":[{"value":0,"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},{"value":1,"date":"2026-03-06 22:02:17.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-25 19:22:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-26 20:04:46.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"CC"},{"updated":"2026-03-03 16:41:49.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2026-03-05 12:21:09.000000000","updated_by":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"reviewer":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"state":"REVIEWER"},{"updated":"2026-03-05 15:50:32.000000000","updated_by":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"reviewer":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"state":"REVIEWER"},{"updated":"2026-03-07 15:17:38.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"9ee3c5ad2407df89a7563854415723e07259cd96","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-02-25 17:35:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"68a42859bfed6b9ed7893c3f0a01da7edc3f44fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-25 19:22:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/11c8e2ed46294fd298643fc7ffcd6186\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/81e7acb715f4434c9d3df1b0e38ad57d : FAILURE in 29m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f9c1020972904487a55c589411a108c1 : SUCCESS in 36m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd1311c5ff6d44a2a12372bebe289078 : SUCCESS in 10m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ae4e95aa2de14a96a20ef3971aec42e1 : SUCCESS in 33m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/86ff7a58cd634f6b865d4243ec739186 : SUCCESS in 21m 08s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/55f75adc841a4022bb17ce314bc49206 : SUCCESS in 29m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3c1938860d5490aa3239e267e9f27b7 : SUCCESS in 5m 40s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/4d399d5ca8dc419c9b880e5e919bdae6 : SUCCESS in 37m 28s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/1019db28ee4b4bab9e73e77d255f0759 : SUCCESS in 50m 38s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/98d32b74d6f147dba9ad19be62170d5a : SUCCESS in 36m 18s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/028df362c144403f937d5153fe2b789f : SUCCESS in 58m 13s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/bca0eeda58284d59b5377fd78088d3b1 : SUCCESS in 1h 37m 33s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/aa0f786f00c347e590fbed4507d8b46d : SUCCESS in 1h 32m 19s","accounts_in_message":[],"_revision_number":1},{"id":"7509b6a282f8cc96f12b5bcf78821ecd4f207b94","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-25 19:38:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b809ac357b41481f92f8d62ba9a9e944\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a35e5c532d2e404abaf615c021901abe : SUCCESS in 25m 02s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3a1810c8b4944132ac1dc383b10f2260 : RETRY_LIMIT in 3m 42s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/03664169705d4703817ad9752665c9d0 : SUCCESS in 26m 01s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c8a41ef2972149c09299a6c51b3b8692 : SUCCESS in 21m 06s","accounts_in_message":[],"_revision_number":1},{"id":"6084eb073366472b571ba73b621ca9325689b20e","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2026-02-26 20:04:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5d6681f8275eb927f5fe4a5ec235d9711d1c027e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-26 20:42:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c13ac58bf7742048c08232828d27dba\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/bc54411be63f497c92b874b9ade77716 : SUCCESS in 27m 22s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d9c2e20ef6f54aaab054beba4a61eae8 : RETRY_LIMIT in 3m 33s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/69816dc92f1449789c1faafe5f9a6a3a : SUCCESS in 34m 46s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/445ba90757e642f78b5c2cf10a176bfd : SUCCESS in 20m 08s","accounts_in_message":[],"_revision_number":1},{"id":"c15b0d2c8ec77746f30736907c99c8c581a4aa66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-26 22:44:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15e56ce9a0f9466db1a2581387ed6e34\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/264d731c18204d3d86fd00454228597d : SUCCESS in 1h 05m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/45c0f51329d74542bb2ff61a82cab71d : SUCCESS in 1h 02m 49s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/771403168ed34d44b9667e1656297b42 : SUCCESS in 2h 38m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/642c7244476f4e09b19656dd5f7ee810 : SUCCESS in 37m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3e42fc127154f198d8c7a55bbf2100b : SUCCESS in 25m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/636bafd9f848481b93eff19adf47b3a4 : SUCCESS in 38m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/686d1df9f01143be963b29efbe8a8d84 : SUCCESS in 19m 57s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/44910e209fe84a5e82e831d6633d6df3 : SUCCESS in 20m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5089f87167b14344a958e3a00f88ca15 : SUCCESS in 11m 50s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/0e841e3c680f414cb10d3e8fb89d0688 : SUCCESS in 1h 21m 47s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/0f3e42e35fce4ec7b5170662793c5f7b : SUCCESS in 1h 01m 33s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/83b110f51283492ca1cc215a987d1749 : SUCCESS in 52m 56s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/b3dca384a0024e46bb9d893955cb37f8 : SUCCESS in 58m 04s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/eec30ffeabc548b8aad3ef35bd26567a : SUCCESS in 1h 25m 32s","accounts_in_message":[],"_revision_number":1},{"id":"d33aa4aab835a4a560c0159c62bbcb9335d87517","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2026-02-27 15:43:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6df1a0f29b22b002e21effbfe44bb8f66e6b7088","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-02-27 22:14:05.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"e62d5b6e413d2880d8fcbf3720cebf136e0c8841","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-02-27 22:15:21.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f4bc5f0f0961d431bff13fda3c6f997c3e10f020","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 23:06:50.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e30279b2344d41939e17e91c4bd89d21\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/25db88c553634af19b6100d34a94a2c7 : SUCCESS in 24m 12s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/183a40bbc46b4e6294d3fa0e9f728372 : RETRY_LIMIT in 3m 23s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/04609e38a1cb47648b8da857df174113 : SUCCESS in 50m 25s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/63dc08bdb5a244c79d2f90076d78f7f9 : SUCCESS in 21m 19s","accounts_in_message":[],"_revision_number":2},{"id":"965a8288d38703e75995e0712414313cd6ae003b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-28 00:01:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4e0891381b94467a6c814d796b988af\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/50d3bfaab95e4cfd8d4fd384cd8ab5d0 : SUCCESS in 1h 27m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e3b635413230489fa713d4a5d8ccacda : SUCCESS in 1h 06m 45s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/de7c159c97ce47c49bba2d1029aedc9c : SUCCESS in 1h 44m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9d00d6a40a76497b8810fe750df513ef : SUCCESS in 38m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ffc14e5dca5f48eeb1af09c9cb8f39c6 : SUCCESS in 21m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f914c16432384cdaa53e86c929eb5350 : SUCCESS in 36m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6c6ed58d5f4d4e5182f66d53aec6d453 : SUCCESS in 31m 52s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/98209b90837346f7b41cd90ef0a3adf1 : SUCCESS in 35m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2898934acd4845ef8082df0ec4a380db : SUCCESS in 11m 17s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/e433f61b8b634532b1f84c77d7c58b17 : SUCCESS in 50m 30s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/3864fac6a15546b9acf7ddd82d708722 : SUCCESS in 41m 28s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/9a73f25219a14618b1cf1e4738574013 : SUCCESS in 1h 10m 31s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/e0ad0cfc14de442dbbc692ec4fc0618f : SUCCESS in 1h 05m 06s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/51bd27f732c94445b38203041dd5f062 : SUCCESS in 52m 54s","accounts_in_message":[],"_revision_number":2},{"id":"b20672bc111ca3c05493f1598f49910e2a963123","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2026-03-03 16:41:49.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"38254c9e0b07d1601e57a2dccb2977e9a5574398","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2026-03-04 20:18:26.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1192cd7127bdf61d72a1d4cd41645d03e247fcfd","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-03-05 12:21:09.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"cd12a1d7f34284f8bff452c410dd5ea682521ff5","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2026-03-05 15:50:32.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"4e91478518fddc7faf9e77e0d3c99b1b87d56575","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-03-06 22:02:17.000000000","message":"Patch Set 2: Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"cc984f005b036190c24b70fd483d9122a0e96128","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-03-07 15:17:38.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"770f10820ff3544687baa652098545a6a2810cad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-07 15:18:18.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8165f405c29aafbffe3e7b80d11f1ffe42bb097b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-07 17:38:24.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7024940aa15f41f49df1463205539934\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/8763eed2a7cf45f398aa0ae6b4443166 : SUCCESS in 1h 27m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a10adad1ae954200a8577d5c0a9dfe47 : SUCCESS in 1h 05m 34s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/4e9b3a9f99854e498a9071a457dccf79 : SUCCESS in 2h 18m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65a6546ef06f42f59d3f2a5a31d51690 : SUCCESS in 17m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0bcfadc58ccb437c8cc00883de572e52 : SUCCESS in 37m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/61aaa1624d9b4a0884b561b4ff3fc2a6 : SUCCESS in 17m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3d39bbdf8204521aad53d396ef8cd02 : SUCCESS in 11m 35s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/8efb31dae87241e5ba49d12e0306733a : SUCCESS in 1h 22m 37s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/550ba9600f534655a8a29d5730e268df : SUCCESS in 51m 50s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/5e08a8850b18457ab6d077c9c13473e1 : SUCCESS in 1h 09m 43s","accounts_in_message":[],"_revision_number":2},{"id":"b142d8662e11ca90fdfb77b2dca207b13ba8a76e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-07 17:38:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"32c62662f5d3a52f5d8a1254da88a8dd1428b57c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-07 17:39:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c40ce5f817b6498f9ef66697fe9658f2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff98523db0ab44daa5c469a92f183b1b : SUCCESS in 44s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3f4d5dc93f762e78bf1442d94c7578702a156a7e","revisions":{"55e94b6763ae760201bc93a7f4d0c09eb637fec9":{"kind":"REWORK","_number":1,"created":"2026-02-25 17:35:44.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/05/978005/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/05/978005/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/05/978005/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/05/978005/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/05/978005/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/05/978005/1"}}},"commit":{"parents":[{"commit":"07504c605c5c120116311335e4eb9b7f4fc52168","subject":"Merge \"Use project_id key in subnet plugin code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/07504c605c5c120116311335e4eb9b7f4fc52168"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-02-25 17:19:42.000000000","tz":0},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-02-25 17:28:13.000000000","tz":0},"subject":"bgp: Fix UpdateLocalOVSEvent match","message":"bgp: Fix UpdateLocalOVSEvent match\n\nThe event got triggered on any update because of the wrong\nreturned objects when comparing if the bridge mappings should be\nchanged.\n\nRelated-Bug: #2111276\n\nChange-Id: I50074496cd88637c620c18d1a07a22e77b25624e\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/55e94b6763ae760201bc93a7f4d0c09eb637fec9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/55e94b6763ae760201bc93a7f4d0c09eb637fec9"}]},"branch":"refs/heads/master"},"3f4d5dc93f762e78bf1442d94c7578702a156a7e":{"kind":"REWORK","_number":2,"created":"2026-02-27 22:14:05.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/05/978005/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/05/978005/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/05/978005/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/05/978005/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/05/978005/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/05/978005/2"}}},"commit":{"parents":[{"commit":"07504c605c5c120116311335e4eb9b7f4fc52168","subject":"Merge \"Use project_id key in subnet plugin code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/07504c605c5c120116311335e4eb9b7f4fc52168"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-02-25 17:19:42.000000000","tz":0},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-02-27 22:13:48.000000000","tz":0},"subject":"bgp: Fix UpdateLocalOVSEvent match","message":"bgp: Fix UpdateLocalOVSEvent match\n\nThe event got triggered on any update because of the wrong\nreturned objects when comparing if the bridge mappings should be\nchanged.\n\nRelated-Bug: #2111276\n\nChange-Id: I50074496cd88637c620c18d1a07a22e77b25624e\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3f4d5dc93f762e78bf1442d94c7578702a156a7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3f4d5dc93f762e78bf1442d94c7578702a156a7e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
