)]}'
{"id":"openstack%2Fneutron~978030","triplet_id":"openstack%2Fneutron~unmaintained%2F2024.1~I08a7646c5f4bb20b94e121171a2680a8e20dc0a8","project":"openstack/neutron","branch":"unmaintained/2024.1","attention_set":{},"removed_from_attention_set":{"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2026-02-26 19:52:41.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}},"hashtags":[],"change_id":"I08a7646c5f4bb20b94e121171a2680a8e20dc0a8","subject":"DNM: Tesing n-t-p change","status":"ABANDONED","created":"2026-02-25 21:33:40.000000000","updated":"2026-02-27 21:21:08.000000000","total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8ae71880ffd05c788ecb7581691f0cf3bb51d255","_number":978030,"virtual_id_number":978030,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-02-26 21:39:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-25 23:12:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-26 02:24:26.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"CC"}],"messages":[{"id":"c267bc31e4b9cf6b2e01b50721954e92c1758308","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-02-25 21:33:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1523538482162c8d077b40f869c4816fdb964de5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-25 23:12:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3ae19b7577584b709959858da1241e27\n\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/2164e72535114215b052d8045b142534 : FAILURE in 24m 26s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/143e25a8ac294a329195fa981ff73548 : FAILURE in 23m 56s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/d7da9fe8ec7a41daa9fc2e004435048c : FAILURE in 25m 14s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/dac9ad0f124841a7a509760d7703962f : FAILURE in 25m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/07ab6b74dd8146c0a35db07f80484247 : SUCCESS in 42m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a3cbfccb1124e01b701ea85cca8475a : SUCCESS in 10m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b4c631a785144c038f44fed35fb98f07 : SUCCESS in 20m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c525fdba8f6645289ca1951cba2f1c41 : SUCCESS in 18m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/213c2e8089f44fc9abe77bc038987683 : SUCCESS in 11m 41s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e38427b7549948b8b6dc5913f33a515a : FAILURE in 1h 25m 09s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/276fca39ca044deabe8c31203542a4cd : SUCCESS in 1h 07m 25s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/3eafa542e212462e9b4944d95e53b79d : SUCCESS in 49m 57s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/250bd6f776a34d9f836053e711f4ac02 : SUCCESS in 41m 28s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/901caf0ce4b44438beb418502140c1ac : SUCCESS in 1h 27m 11s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d45878da34db4589a0f1c622d8e55e9a : SUCCESS in 1h 11m 05s","accounts_in_message":[],"_revision_number":1},{"id":"2937f48bb914f0eb946bda330d505064cf4c266b","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-02-25 23:21:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c4cfddcb2655c2f43a7ae0fc8b6954be6e9a246e","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-02-26 00:02:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"eef0aceabeea66f6d69fd08608ed338cc8305266","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-26 00:02:48.000000000","message":"Patch Set 1:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/978027 is needed.","accounts_in_message":[],"_revision_number":1},{"id":"a626f80d2dc3f1d9f8d44b0829f509898935cbb1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-02-26 00:03:08.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"ee87073bddda85c9efb087a1ec2063d8f03ba701","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-26 02:10:08.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5e43e2109cc547b882e2dc1aeb377088\n\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/53503030d7f64b3f9fe34a2c0289206b : FAILURE in 22m 47s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/188c20952568427a88d379796fbacaee : FAILURE in 22m 28s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/538a80385bc1422088e79272bf4a6638 : FAILURE in 24m 04s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/cff2024f800e4bd3b1ddd4aa423118be : FAILURE in 23m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/274e432c88ed4eacbb8937ab93e7c2d3 : SUCCESS in 40m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59e106e5241a44a0bc8d773d8aea225e : SUCCESS in 5m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8928f647574c4724b9bdde774ded4e48 : SUCCESS in 26m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/47a0ec610bfb4e7d9fa5942494e8b329 : SUCCESS in 18m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6610d4e9171488d85abf76cdbcd4022 : SUCCESS in 11m 58s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/abad9c7242dd4ada98c041d276621429 : SUCCESS in 1h 53m 56s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d7bfa0ee1521489d898af91252dae450 : SUCCESS in 1h 40m 03s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/5c01e4da7a384a66bd7e4afcee8cec74 : SUCCESS in 58m 50s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/5b5a58fac7b34c8fbd13343b0e2de90f : SUCCESS in 59m 49s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b2e3be8658ac4b1eac1e191e2fdeefb3 : SUCCESS in 36m 16s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/5aad9e91bf4440539e1261ac1d5bf321 : SUCCESS in 1h 19m 23s","accounts_in_message":[],"_revision_number":2},{"id":"faad4f554edc0dc3074e3294b5c726adfcb4dd90","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-02-26 02:24:26.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8ff61eb813109b8359929dbff27b5c555e7229c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-02-26 03:53:06.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"ebf4ef5e23a8040ae8267d1fda754a4547f8a4e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-26 05:16:14.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6f3dba6aa5434a4095da3f4d08c7309f\n\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/040d4f46d33d41a3ba2557673237fc0d : FAILURE in 21m 37s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/c6c350e28e27420dbf8f63f78d6c6ae2 : FAILURE in 24m 21s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/8bec98310b3a454ca875ca88c2f86ca0 : FAILURE in 22m 20s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/b4e46db8fd7148feb0c6e5e5cb08fb22 : FAILURE in 24m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/210987455fce488d8a590fb3985282f0 : SUCCESS in 25m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10b9f555bfcd4046a7e8b84e6f9ebcc0 : SUCCESS in 7m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c23e9b8a0d404bb6a16e5823b9ff9366 : SUCCESS in 23m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e73d4131dc7040c58d86ef896c01f734 : SUCCESS in 18m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1245b7d39a7b4e5d8379a662d15ea8cc : SUCCESS in 12m 28s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c2b3bd3c98c8435783b2cf49f518bfa1 : SUCCESS in 1h 09m 51s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a14239eca0974d38a51a414f273c3d8a : SUCCESS in 1h 20m 36s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/b1c959837ee44828aff000bf9415ccd1 : SUCCESS in 1h 11m 29s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/7203302c948b41368c1fef7b468f16e0 : SUCCESS in 39m 22s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/bdfb37cb9a3441d89c945553183430a6 : SUCCESS in 1h 00m 47s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/92b37af0808c4e6f83d28a09583b2311 : SUCCESS in 1h 07m 50s","accounts_in_message":[],"_revision_number":3},{"id":"4ee242b7dcb20e697e7652843285482ca32ea854","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-02-26 14:41:33.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"363666a7d33d62df38a69c91e07786afec24a463","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-26 15:54:59.000000000","message":"Patch Set 3:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/978027 is needed.","accounts_in_message":[],"_revision_number":3},{"id":"ff5e3f35eddfcdc897b8657a6bb6f97016a4e4b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-02-26 15:55:41.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"2729e4b6aee47d8a612eeefcdc52f736737d9bb8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-26 17:19:26.000000000","message":"Patch Set 4: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/978027 is needed.","accounts_in_message":[],"_revision_number":4},{"id":"91cc37965ef875ee50a08b21c3be7d06f3144658","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-02-26 17:19:48.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"6fc7796d15b6f23d62d87e1a3862bc981d3e2736","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-26 18:56:49.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e2713197620f438bb07aa9e8294399d1\n\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/a9774c9e6a21496396abe084a5307110 : FAILURE in 23m 37s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/045df6aa7a264448b7135e8f2c0d569b : FAILURE in 23m 34s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/c49f97527640498d8ea7f5761e51e724 : FAILURE in 23m 52s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/db29b7397252404387310b79e1de7d71 : FAILURE in 24m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4413e3607fab486cb41a1707dca83304 : SUCCESS in 31m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6cb0e778e654d0f88f9b23729c5852f : SUCCESS in 11m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cab43ff6c5e142f398fa3d9d9045c7c6 : SUCCESS in 22m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1995bd06f9644bf9832190538266bcec : SUCCESS in 26m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/764e5cf8144c45e58570d4aa85171186 : SUCCESS in 11m 39s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c6fe17a51dd74b3b83cfcd7851b8e58f : SUCCESS in 1h 34m 47s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/cf7a65a70c6c43d1bdecc21c04cbf557 : SUCCESS in 1h 09m 59s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/63ccb666a3004ffca166fbd9ec4a5cb1 : SUCCESS in 1h 00m 56s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/0f5251feb222479293eb77cf43d43bba : SUCCESS in 36m 22s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/bf99fc0bf8dc42479bd7d1b2b174b392 : SUCCESS in 1h 31m 29s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/ce58714e4def4befaff9b0d7fb145254 : SUCCESS in 55m 57s","accounts_in_message":[],"_revision_number":4},{"id":"daa42aa8444499a3d070c3641cbae3e084309dfd","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-02-26 19:52:41.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7ce478dfd5699830a0e5468c8a60a77e69614cf6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-26 21:39:18.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa5659bf9dbc4116b124d57d1d77543a\n\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/edc778e8c7fa48d388fb7b4044e6c5c5 : SUCCESS in 31m 57s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/85347d5b613340d7bbed6d5da8311e0b : SUCCESS in 31m 22s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/97756307b0244a88a53f1e74759eef24 : SUCCESS in 32m 13s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/a42093695d5443c3969e4e2aedf3c9fe : SUCCESS in 27m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/75bc944a4a3b45128596fbeb369dbe9c : SUCCESS in 32m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64a614fe7f2d4be5935642f1a64f304e : SUCCESS in 11m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/90490cf2573b46c98ac0680de2cf2977 : SUCCESS in 30m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/394dc0c557da47ab9b1fabb9dc92658f : SUCCESS in 26m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb9199256b63420d95571460d09a8e65 : SUCCESS in 10m 59s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/babb1f74611c45f2ad86a98f44e2e3ec : SUCCESS in 1h 09m 07s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f00f59aa01764dae8c5fc3ba8fb4440e : SUCCESS in 1h 44m 51s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d50cc79457d44d5eb51462cdb2ed82a7 : SUCCESS in 46m 43s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/f6fe82eb79dd449a9737e07bd3ad038f : SUCCESS in 1h 23m 15s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e9e6887147f94fb3add42bb0c4063186 : SUCCESS in 1h 25m 50s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/9264168a6d7d404daf2eac34d176ec64 : SUCCESS in 1h 15m 22s","accounts_in_message":[],"_revision_number":4},{"id":"895ca441f7fd36dbaf6972c6a3f3ab344a287b64","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-02-26 21:45:58.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"fc7bc2823132d3218c37390801c7bd8858856afd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-26 23:37:03.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b718080bb0c8446c96d1b1d829c20962\n\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/6fd4a5177288454e8d821d301d2a150b : SUCCESS in 57m 30s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/497319c304b343a693a78b54a6088678 : SUCCESS in 1h 08m 09s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/514ef5542b194de798e96bcdaaacf55b : SUCCESS in 1h 32m 12s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/84f3168f13d442b09473aad71e436d3b : SUCCESS in 29m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/20a4bb394026482b8d2d664ce4971b11 : SUCCESS in 28m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/872e5a119c3840c38f84634ce3f7bb0e : SUCCESS in 9m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/353a018d7ac240f48f98ba0631858367 : SUCCESS in 23m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1629dc660cfb40d5a91000e0224c7571 : SUCCESS in 18m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6bb39367e2e45b7901fc0d6d5a9350f : SUCCESS in 11m 18s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/54df87b8d5394e079b87c310519f1801 : SUCCESS in 1h 07m 38s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/43c45c161a324af7b2fbf06e26745035 : SUCCESS in 1h 43m 46s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/40e079db9bd1416bb2d21e38060b8a63 : SUCCESS in 46m 57s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/7a2b36a3247546538b9a9e1edbee5dff : SUCCESS in 1h 04m 36s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/46f49fe8ec05413281948204a206bb96 : SUCCESS in 1h 27m 46s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d18649ea59284fdfb36b0e7d757844db : SUCCESS in 1h 16m 40s","accounts_in_message":[],"_revision_number":4},{"id":"8ae71880ffd05c788ecb7581691f0cf3bb51d255","tag":"autogenerated:gerrit:abandon","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-02-27 21:21:08.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"a4b50118b8870231e518ea83a68ed60be4e5b15d","revisions":{"1442648a9d54024ad0ad8a90a0dafd32e7098473":{"kind":"REWORK","_number":1,"created":"2026-02-25 21:33:40.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/30/978030/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/30/978030/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/30/978030/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/30/978030/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/30/978030/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/30/978030/1"}}},"commit":{"parents":[{"commit":"1e27a85f4f1e32f4a40cedd46335131154b89b6c","subject":"[OVN] Allow multiple values in DHCPv4/v6 options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1e27a85f4f1e32f4a40cedd46335131154b89b6c"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-02-25 21:32:30.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-02-25 21:32:30.000000000","tz":-300},"subject":"DNM: Tesing n-t-p change","message":"DNM: Tesing n-t-p change\n\nDepends-on: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/978027\n\nChange-Id: I08a7646c5f4bb20b94e121171a2680a8e20dc0a8\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1442648a9d54024ad0ad8a90a0dafd32e7098473"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1442648a9d54024ad0ad8a90a0dafd32e7098473"}]},"branch":"refs/heads/unmaintained/2024.1"},"c1bab02e755190706fedfaf1e81f905c4c7d208d":{"kind":"REWORK","_number":2,"created":"2026-02-26 00:03:08.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/30/978030/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/30/978030/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/30/978030/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/30/978030/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/30/978030/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/30/978030/2"}}},"commit":{"parents":[{"commit":"1e27a85f4f1e32f4a40cedd46335131154b89b6c","subject":"[OVN] Allow multiple values in DHCPv4/v6 options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1e27a85f4f1e32f4a40cedd46335131154b89b6c"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-02-25 21:32:30.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-02-26 00:02:59.000000000","tz":-300},"subject":"DNM: Tesing n-t-p change","message":"DNM: Tesing n-t-p change\n\nDepends-on: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/978027\n\nChange-Id: I08a7646c5f4bb20b94e121171a2680a8e20dc0a8\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c1bab02e755190706fedfaf1e81f905c4c7d208d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c1bab02e755190706fedfaf1e81f905c4c7d208d"}]},"branch":"refs/heads/unmaintained/2024.1"},"ecc253582fcb0431953c982d2e7b58566e56724b":{"kind":"REWORK","_number":3,"created":"2026-02-26 03:53:06.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/30/978030/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/30/978030/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/30/978030/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/30/978030/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/30/978030/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/30/978030/3"}}},"commit":{"parents":[{"commit":"1e27a85f4f1e32f4a40cedd46335131154b89b6c","subject":"[OVN] Allow multiple values in DHCPv4/v6 options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1e27a85f4f1e32f4a40cedd46335131154b89b6c"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-02-25 21:32:30.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-02-26 03:52:57.000000000","tz":-300},"subject":"DNM: Tesing n-t-p change","message":"DNM: Tesing n-t-p change\n\nDepends-on: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/978027\n\nChange-Id: I08a7646c5f4bb20b94e121171a2680a8e20dc0a8\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ecc253582fcb0431953c982d2e7b58566e56724b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ecc253582fcb0431953c982d2e7b58566e56724b"}]},"branch":"refs/heads/unmaintained/2024.1"},"a4b50118b8870231e518ea83a68ed60be4e5b15d":{"kind":"REWORK","_number":4,"created":"2026-02-26 15:55:41.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/30/978030/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/30/978030/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/30/978030/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/30/978030/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/30/978030/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/30/978030/4"}}},"commit":{"parents":[{"commit":"1e27a85f4f1e32f4a40cedd46335131154b89b6c","subject":"[OVN] Allow multiple values in DHCPv4/v6 options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1e27a85f4f1e32f4a40cedd46335131154b89b6c"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-02-25 21:32:30.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-02-26 15:55:32.000000000","tz":-300},"subject":"DNM: Tesing n-t-p change","message":"DNM: Tesing n-t-p change\n\nDepends-on: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/978027\n\nChange-Id: I08a7646c5f4bb20b94e121171a2680a8e20dc0a8\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a4b50118b8870231e518ea83a68ed60be4e5b15d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a4b50118b8870231e518ea83a68ed60be4e5b15d"}]},"branch":"refs/heads/unmaintained/2024.1"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
