)]}'
{"id":"openstack%2Fneutron~978060","triplet_id":"openstack%2Fneutron~master~I1af50907461045d747ab112fddbdcf4220397f31","project":"openstack/neutron","branch":"master","topic":"rehome-ovn-db-sync-constants","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-03-20 21:21:50.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2026-03-20 23:33:03.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2026-03-20 23:33:03.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2026-03-15 22:02:23.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I1af50907461045d747ab112fddbdcf4220397f31","subject":"Use ovn-db-sync-util related constants from neutron-lib","status":"MERGED","created":"2026-02-26 10:48:31.000000000","updated":"2026-03-20 23:34:10.000000000","submitted":"2026-03-20 23:33:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"978060-rehome-ovn-db-sync-constants","meta_rev_id":"8cdfe0884ae221f7653da3da9a253f6de25848f4","_number":978060,"virtual_id_number":978060,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-20 23:33:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-03-20 21:21:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-03-10 12:49:55.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2026-03-15 22:02:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2026-03-20 21:21:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-26 11:25:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-02-26 13:12:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-27 00:05:23.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2026-03-02 11:30:32.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2026-03-03 15:33:29.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"f3a87bc7b2a67de4b1626987cee468157125e9ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-02-26 10:48:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f28d6a052ef4c95b020338d152cef2bd37db8389","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-26 11:25:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eee26a029ce844e782d0ade16a87f551\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/94d6f430519642958f746ddfc5051c2b : SUCCESS in 25m 09s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f85cf52293834150a05f3f65f2f52bf7 : RETRY_LIMIT in 3m 13s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/6a6f59135ef34aa883c8c8016a360047 : SUCCESS in 25m 20s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/28cad98aee404f309285c5fb66884422 : SUCCESS in 34m 35s","accounts_in_message":[],"_revision_number":1},{"id":"ff8ba3d0f89d233b7e5876a477efeed551ef05ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-26 13:12:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/40e66a50868f41dd9cdf48557ad9abe1\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4cdc4964d64b4e35a61be909e7ef823f : SUCCESS in 1h 34m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/49c9822608984b97b3ef1770b7a984ee : SUCCESS in 1h 08m 33s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/ed8644e89a5e4c6681f5d2f15064fa6c : SUCCESS in 1h 36m 29s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/113a0805602c421c8f51f13f67786947 : SUCCESS in 2h 21m 32s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3341252efd074937b168ab48864bbdb0 : SUCCESS in 32m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3a211a78f9ce42bb937f1ce1c7e8fc8b : SUCCESS in 1h 08m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e48c993e36514ce898e771b00772c54f : SUCCESS in 21m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/19902ae239cf45ffb5e1e8d503d5d4da : SUCCESS in 41m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/87bbfe1956c642189a0223d174e9fe3a : SUCCESS in 15m 58s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/555603734f8b47b180eb232a7ea8a3c9 : SUCCESS in 37m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e476f07507a84c1cb2615c70599c936a : SUCCESS in 12m 55s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/076c99ea00d94819bdc41660c81d01f6 : SUCCESS in 1h 03m 10s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/a6644a1be3094939b0c81f485b0e0df3 : SUCCESS in 37m 19s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/1cb6c22957244d5b9956be7caf8aeafa : SUCCESS in 1h 37m 12s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e81a3fdfa7b4468ca76c6d4b8b591f7f : SUCCESS in 1h 09m 07s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/a544114f130d46ca9521643d88b89058 : SUCCESS in 1h 17m 41s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/785c3646168646999c8ee7e46860f162 : SUCCESS in 1h 53m 20s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0a63f5a624d44150804ab059c719fd1e : FAILURE in 46m 29s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/dbdce31eae154d8eb4443a6d65c1fc17 : SUCCESS in 1h 16m 47s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/f8d1f99b1e194e3ba82b9a58bb5be1c9 : SUCCESS in 1h 10m 18s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ca1c953f4fe04a99b2679cb2505c36ff : SUCCESS in 42m 41s","accounts_in_message":[],"_revision_number":1},{"id":"f924a0f67c0ede1080671566209c20cbe44f5de2","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-02-27 00:05:23.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"af0409b3164ab5e386a5ac23cba38ac79387ab15","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 00:45:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b00408823fec47448b75300fbbe96b12\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/bbcdcd4346da4f2097fb07a7d7d3a117 : SUCCESS in 23m 36s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/892b504cc45d4892b9758737d5987589 : RETRY_LIMIT in 3m 08s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/2c0fccee26b94bb3830002ccffea7a96 : SUCCESS in 37m 06s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/266957732af748f99cb37c926e115c73 : SUCCESS in 19m 20s","accounts_in_message":[],"_revision_number":1},{"id":"743c0dc60219ef8bb7db4b093cdb4af28c241cd6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 02:41:26.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ef252ff925fc4f88876d780f7bbd4261\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f8d02a20e21b45a689e7ca8c2b3af46c : SUCCESS in 1h 30m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/390dbb8f1c864d1586bbafcfd21aa485 : FAILURE in 1h 14m 34s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/1430e5cbab2f4109b9afccdfb5e4a729 : SUCCESS in 1h 31m 59s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d7722786523f4336a403cbe86a107fc9 : SUCCESS in 1h 35m 30s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e670a96b8da743c681ac75e4a5666684 : SUCCESS in 31m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d65fc170649c476d8b0dc266ad4cea82 : SUCCESS in 45m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12bea5fded1e44de99e9e6592788dbaa : SUCCESS in 22m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/882ac6a642664b7ab1762ef14df9f0ca : SUCCESS in 35m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/854429e17b894448a2df9e997997b2b4 : SUCCESS in 23m 53s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/563fe0fb0ec54e398b808a9ead040b5e : SUCCESS in 34m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f123ef90591498b9d4c23abb03194b4 : SUCCESS in 5m 48s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/f9688bef221d455794b4f711637adcd8 : SUCCESS in 57m 56s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/e7e482f557984261af4a0b7f2653e72f : SUCCESS in 1h 01m 34s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/8cfb335deff7413c96fa15ff4a323ee6 : TIMED_OUT in 2h 32m 53s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/3efdd238c2cb4616a0bf477934d01e41 : SUCCESS in 1h 13m 37s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/54515203d4dc463db7d6a8328bd07222 : SUCCESS in 1h 13m 32s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/29684960e20c43bcb417b601a276f634 : SUCCESS in 2h 02m 34s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/cde73a90a03942069a7a88de8573ae10 : SUCCESS in 1h 41m 40s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/dd9a04da684847bcb04ad4ccd1222909 : SUCCESS in 54m 55s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/b67caf7965104f028a4253f2ca1913b8 : SUCCESS in 1h 13m 08s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e7a4e58cddd041a6ab0e11884072ce1e : SUCCESS in 45m 48s","accounts_in_message":[],"_revision_number":1},{"id":"a08346b8946e9d159adc4ceedea9dd81ea78f74b","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-03-02 11:30:32.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7973673f6c866b507044adf55e323c941ec5cdea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-02 14:56:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b2c47aa02dd465ead2f236567de9c4d\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a3b9be2d72be41e7bc2439d9643034a9 : SUCCESS in 1h 24m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/108dfa94c85242b388c95dec4df2178e : SUCCESS in 53m 14s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/0db07267f8a74dce8fa4f30f17e9f871 : SUCCESS in 1h 24m 09s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c7d617d644ea4fb58e0698f440d89aeb : SUCCESS in 1h 51m 59s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/82b35032cb2f4fbbba39de4ad92f25b6 : SUCCESS in 32m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/35ee681ed47740879ad736602213b806 : SUCCESS in 44m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f72fefd9d1a4e8a9fd66a67f1eb8a17 : SUCCESS in 25m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2f2b682841ff467ebe881abbe56cfb6a : SUCCESS in 24m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c4eee194f194470fa95230ffc571f1c6 : SUCCESS in 24m 54s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e4d189dfc4e24fb0bdf61a0e76d9c235 : SUCCESS in 38m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37f96f11ac3d4e3ca5e2038deecdc06d : SUCCESS in 11m 00s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/28d2db76b98e48029113f8ad6fe296f1 : SUCCESS in 58m 34s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/72fd3ba998a64e26842f7dc5973ee276 : SUCCESS in 52m 47s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/5eee9ca9837c468c949f198bacb09212 : SUCCESS in 48m 01s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/8b4d1136b48a4687818b2f9df6f2aeb2 : SUCCESS in 1h 12m 40s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/efe7fe3276514553bcebda57e882cbcf : SUCCESS in 1h 15m 02s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/0fb27fbfda854afa913727dd94005aba : SUCCESS in 2h 21m 43s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b273f94bf7464e6f9f055dca0edb6941 : SUCCESS in 1h 12m 29s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/9e431c49674d4c83b4e930187212a0d4 : SUCCESS in 1h 11m 58s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/982bdf64e9ce4e699d63c34359f3e403 : SUCCESS in 1h 10m 06s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d65ee47e4a7b472899262cde2a08d39c : SUCCESS in 1h 18m 56s","accounts_in_message":[],"_revision_number":1},{"id":"ce6b4658b2a6085ce2d5cc845e4991c297277a3a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-02 18:53:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da8eead996e64b5890a7a9a522f26262\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b3f5e43916f840e5a9cc261b29854888 : SUCCESS in 27m 06s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/34c9c1373dda48a2bec36a2eac8e81d9 : RETRY_LIMIT in 3m 49s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/d617193a3ece4358964c8c76a36273b4 : SUCCESS in 30m 46s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a361576a3d6e495ea9df53941e45a245 : SUCCESS in 22m 38s","accounts_in_message":[],"_revision_number":1},{"id":"faf75acc33600af6c4b667d4b02189f816c75d55","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-03-03 15:33:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5fc418979d427185719021edaa55cc8dfdaaeeb9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 15:33:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2daae2c62ff4ec39bd7ceb90e58e459bb734d8f8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 17:21:22.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/808016efd0c849a696cdb746f629bba3\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/1df9c7f426584128805f3503ec8aecdc : SUCCESS in 1h 14m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/32c5e3a34c364b6fb26317e83d94eed5 : SUCCESS in 1h 06m 07s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/94580a620374414488e6d0bb5379d9a3 : SUCCESS in 1h 30m 57s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/2b19cf4fa5044964b3ee621f8fa81bb7 : SUCCESS in 1h 46m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81634473113c4953baf4ed70c84cd3ab : SUCCESS in 23m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/102c5b2499d74e92a534df01ee3a7569 : SUCCESS in 34m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2d619cf819da4386a397b859370baa27 : SUCCESS in 21m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cafceb1f7c2492cb8b3f396f3495537 : SUCCESS in 12m 10s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d330650623e54ff5b328c3173b34c45a : FAILURE in 43m 58s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/c0d7629c1a014757814253f960d1c0dc : SUCCESS in 1h 21m 55s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/55a5edb6faae4e44a9271c53cd3ccac0 : SUCCESS in 1h 07m 58s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/cb8959a1dc9d4ceb8a1ad7d05a1ca8ca : SUCCESS in 1h 16m 10s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/b6217dd5b51b460ba0f649d5a9accc41 : SUCCESS in 1h 05m 34s","accounts_in_message":[],"_revision_number":1},{"id":"d9ebea63b99008344a51eaf0e9d8947c6e70378f","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-03-04 08:22:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b014394773f032a15c9391498f6cf3ed4ea84c88","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-04 09:45:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c1fb0d54dbb4aefa8cfa585ce01a8fd\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7c9822589a504aa183c6d03a4ff6e0d0 : SUCCESS in 25m 37s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/92a8478c3feb45909c1874d1256285de : RETRY_LIMIT in 3m 00s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/a5595bbca31c47d38c4c6b195e5da9dc : SUCCESS in 30m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cc6b00512b193d4cdd34c514d205259f356dc21b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-04 10:24:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1138ccc21e3141bd8738ef9a07db2529\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/ec03770a4d7d481ab97e4790da7bacc4 : SUCCESS in 1h 27m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/36c75ccbff6545069ca3ede0b4ae01d8 : FAILURE in 1h 10m 40s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/8a820bb2ee354ad3a561178f7c855b35 : SUCCESS in 1h 22m 30s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/7c84d49ef2354ce38b156754faab55d3 : SUCCESS in 1h 52m 43s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d1022e69bad84a8586298578a57d8eb9 : SUCCESS in 31m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/416dc4e120554a7db6f6f11406daa690 : SUCCESS in 40m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5e70e1759d943efac4bdc809a11fb92 : SUCCESS in 22m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8ace4a2a411341058f4f650a8e69a929 : SUCCESS in 21m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/667ff9e499d942ebaf151a049eb9d093 : SUCCESS in 20m 49s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1392b8a7048742809b03aac1cf267075 : SUCCESS in 32m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4ac36f94c5a484b9108346df91d4b25 : SUCCESS in 15m 35s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/69407dbb694a43f1ba625e13e630be07 : SUCCESS in 57m 54s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/2c6d682806e24f559d18c35eafa28654 : SUCCESS in 38m 14s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/fa497de83dfe4648bec6125fe5fceb1d : SUCCESS in 43m 07s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b99cb83ff8aa4d07856dfd98adcdf507 : SUCCESS in 43m 11s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/f0ca648d01c1495da9fe990988076eab : SUCCESS in 1h 13m 25s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/272c74f29c50413fa222b723ca4b7ec6 : SUCCESS in 1h 44m 00s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f001b28844ea40f683be35ceb39a49ce : SUCCESS in 1h 56m 25s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/fa3877acb00349318f20eedff1ea0059 : SUCCESS in 1h 08m 07s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/d61bf4dd784643eb91c9055aa85a7e3e : SUCCESS in 40m 26s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e3c2003fd85b4b51916a5374fe4e7e78 : SUCCESS in 1h 28m 51s","accounts_in_message":[],"_revision_number":1},{"id":"d39e02e561a08bdef2d9c9bd36008e68e2a264d7","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-03-04 14:26:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ed7e311eb0ac4fe160acbca71290d0c133e4beca","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-04 16:18:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ae1156060874412a93a85677cba5316\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a813b537d0154ef6b17c1e5bb7120378 : SUCCESS in 32m 02s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7758f382170d416ca8274550e02dc32f : RETRY_LIMIT in 3m 40s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/c0d44057e86f4625872233f51177b463 : SUCCESS in 28m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7d5510db58901c904afcb58f54d08a29666d16df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-04 16:39:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee0753f528f74a729165d4212afe0674\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3849ab6a52a44ad2ba76a03c707342ea : SUCCESS in 1h 27m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cc4669de072e4c3bb6a45bbabac38b56 : SUCCESS in 1h 10m 47s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c4fb078a77ef467a936412a5e20bae4e : SUCCESS in 1h 36m 13s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/52079a6d42b649cf96e43040b8136bb0 : SUCCESS in 2h 12m 36s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/985bc43226f44ed6bee0d15e8804f665 : SUCCESS in 35m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eff03e698955438fb0971ae2a464fd85 : SUCCESS in 1h 17m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4879b5464ad1482d8c181d105bbfdb97 : SUCCESS in 29m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e5f245a0bb8049a297be650998f2f8e3 : SUCCESS in 35m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3de72bb53cad440980974d9838ed1da7 : SUCCESS in 28m 31s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/bb89464a6ecd4588bae8c12939737951 : SUCCESS in 34m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fea4b67c3d04279a4ebd0e4c6f84617 : SUCCESS in 12m 09s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/4f9cc3295a7e4807817d48cfac783aae : SUCCESS in 1h 01m 22s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/fd50d874137047e7878e64b9f533f90c : SUCCESS in 1h 06m 31s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/bf3df671fb264f9bb57c6af1882db84a : SUCCESS in 1h 16m 57s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/778e75cca77b4024b27bae93c3a9f3eb : SUCCESS in 1h 12m 02s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/97222af7d75b4d77a8defbe2fa3546f0 : SUCCESS in 1h 17m 07s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/3e11b5e299e347f8ab3715a9eee4a5ad : SUCCESS in 1h 52m 02s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4b4c6b4cd1ca4294ae6ddbdd4b386e43 : SUCCESS in 1h 17m 42s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d263a7f14c26442395ecee7d43cdc7e7 : SUCCESS in 1h 00m 08s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/0f2d51ff51d246fd89113d8fad8d85ad : SUCCESS in 1h 00m 16s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/23a02871e0c74ae19fde8874fe395598 : SUCCESS in 1h 26m 03s","accounts_in_message":[],"_revision_number":1},{"id":"86e71d477ddb61c41299f71b5c27ffce58a965c3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-04 16:40:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e4feb4f6a81ac3f38923fbca06b83035275117b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-04 19:49:22.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/37f71b455e1c4e85a84868a1a0c1219c\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/b73f9f26a4104f878cfaf776a6380696 : SUCCESS in 1h 31m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/711d0c1fe7fd4a24903da6e4c7906f88 : SUCCESS in 58m 47s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/364e3dd2ed6540249abbe04ff24bc227 : FAILURE in 1h 05m 27s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d6f57e44202c4a0e8c56bb9c6823f6cd : SUCCESS in 1h 59m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b38bb5141074b91b761d70a31ddd0d7 : SUCCESS in 25m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d24de104cb5e4e02a67049184ce77dae : SUCCESS in 38m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8aa5d776313445569a92026ccf9bb05e : SUCCESS in 25m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ce166108b18481c87d93381d70d5882 : SUCCESS in 12m 57s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/015ffe9ba82d4a73840a1aa4afecf2d8 : SUCCESS in 1h 04m 17s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/472df90fe149431db0209d6569997e8f : SUCCESS in 1h 31m 33s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/04b62aa6524942bba4fe39b17886fa45 : SUCCESS in 53m 13s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/7a9e0efa43be4de3b9920c6b6a1f7a37 : SUCCESS in 1h 09m 27s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/864e69142d72424b894684bd8d14a90e : SUCCESS in 1h 15m 05s","accounts_in_message":[],"_revision_number":1},{"id":"5aa9ddf0c4bf18d17fb9e082b14e82218d58761a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-03-09 15:36:47.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"04c88ab9197bfe051b3913bd7ae263b4cc252eee","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 16:10:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c67d3cc79cf45b6a01f1e0ca9e109cd\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/15e50e89090341178f8a71fdeea83b5a : SUCCESS in 29m 19s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ed89a8f959624a208015114a8c779777 : SUCCESS in 18m 02s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/f12e4e9dcceb44f2b6e5ef9076dd201c : SUCCESS in 31m 08s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6b6175d3916e03ae342d3d623117df5f2361284e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 17:38:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5115401a2884494c88485936b8d8aa35\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/7942e8ff0cd344e99896b0579f0dae7d : SUCCESS in 1h 19m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/782e58c84c0c4a0c8bd4ba5624c0643d : SUCCESS in 35m 58s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/88f0e02d7a594d48942ab1a79281bf21 : SUCCESS in 1h 22m 44s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/8ce01d3874a44aeb9729af6a2aae6844 : SUCCESS in 1h 53m 32s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/dbc7224522ab47e8a797e38126dda72d : SUCCESS in 32m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7c0cb93239474a16a989a950ec190e27 : SUCCESS in 43m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9cdcc11d980e415aac19801a9d2ac4dc : SUCCESS in 25m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5a7ee41a36294b70a38ccee03e4d1339 : SUCCESS in 32m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/658144de08f345518e141c27d87e2834 : SUCCESS in 26m 50s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ea0489cb43d8481c8e9181801da429da : SUCCESS in 35m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d583a23404984238acb001209cb93f0e : SUCCESS in 11m 01s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/8c12867b179c41a0a20df7f50929ad7e : SUCCESS in 1h 00m 07s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/e0d6a88d4c914b21aee01bad01f56978 : SUCCESS in 59m 07s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/9da02d9c2499411ca8e24375ea5e084b : SUCCESS in 1h 17m 52s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a5aab5cfc4a94cef966940d3aaf3f95d : SUCCESS in 1h 06m 54s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/a4396dbc8d754128b4fdfb845c208d7f : SUCCESS in 1h 18m 45s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/3f77c2a547ee4ca485d0f844849ff482 : SUCCESS in 1h 54m 03s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/01a94ebcc3c54cf58d9de55eddfcedf8 : SUCCESS in 1h 40m 42s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/e0f5300da0be4a8e9c98f32495f60351 : SUCCESS in 1h 12m 18s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/179a6aa796db4a2c88556c0eb44c83b1 : SUCCESS in 1h 09m 29s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3c26de11118149c083ece8d83939db17 : SUCCESS in 1h 28m 56s","accounts_in_message":[],"_revision_number":2},{"id":"b0a478b0d75708a371ed3d24a212f790e5fcceae","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-03-10 12:49:25.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"f8b855ccb3a306a131ecffc22d4bb60150e15afa","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-03-10 12:49:48.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c2676cee95c201da1baded14c56ac0dec80568f3","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-03-10 12:49:55.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"bc9a16f1967a3a85f5c7c3bbe9a0a189ae981b2a","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-03-10 15:33:27.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2515bc0fc0089bca6e9cf3dfcfce1594548caa9d","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-03-15 22:02:23.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"49c12fb21bb2f3eb6f4e3adbaef4d1b5a4f341a5","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-03-20 21:21:50.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"38354949cd6d74589356f628556ffe8585f8a7e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-20 21:22:21.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f7e6706dbc1a63b701eab2209e6a0ae5b41c4691","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-20 23:33:03.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c290b662cdf41299d79003b30e81d9b\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/bd0712ad59b348a0af47e70b37c4cb9c : SUCCESS in 1h 34m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f525a02c60224616b270d6e0f3ba7020 : SUCCESS in 32m 56s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b6eed7576ab14f7cb002017259858e68 : SUCCESS in 1h 23m 59s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/080f9037c0254b74a1a4d4a6ffdf6ba7 : SUCCESS in 2h 09m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0d78dafe1d747b0be4e52934c1a6094 : SUCCESS in 11m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a1aad1b30ecf4b95900d5b12e4e5f1d4 : SUCCESS in 20m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8dea4bb415bf4318bcd284a198c37637 : SUCCESS in 19m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39fe4788ca11451cba57b9b17bb5fb92 : SUCCESS in 12m 46s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/500e2de87e6c404c9d7598789199a187 : SUCCESS in 1h 18m 54s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/f82088020178495ea4076299f2a8398f : SUCCESS in 1h 52m 46s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/731a71a87cfa48698706c61f65cd44ce : SUCCESS in 1h 12m 49s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d8c9d77e9c1f4891b8d6730855c803e8 : SUCCESS in 46m 05s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/e9d2e37065854d7c83d70598e8b09f2c : SUCCESS in 54m 05s","accounts_in_message":[],"_revision_number":2},{"id":"b3f811bde24522eaa76d06b6107fb086155bb9ef","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-20 23:33:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"8cdfe0884ae221f7653da3da9a253f6de25848f4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-20 23:34:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3f8c147ed6c407b893f04b58bc7d7a4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3eaedc0ac00469c81f89f6a0e085e3e : SUCCESS in 55s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5ded0ea7855cea3b13180634be8b5409e8dc9813","revisions":{"c2cb922d335ab4d47de03b7e8f44f86ebf1e8f30":{"kind":"REWORK","_number":1,"created":"2026-02-26 10:48:31.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/60/978060/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/60/978060/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/978060/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/978060/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/978060/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/60/978060/1"}}},"commit":{"parents":[{"commit":"5f8a6c2c5e7f791b5a1d0e12567f3fd61c347e39","subject":"Merge \"[OVN] Update the FDB age threshold default value to 300\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5f8a6c2c5e7f791b5a1d0e12567f3fd61c347e39"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-02-26 10:47:14.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-02-26 10:47:14.000000000","tz":60},"subject":"Use ovn-db-sync-util related constants from neutron-lib","message":"Use ovn-db-sync-util related constants from neutron-lib\n\nThose constants were moved to neutron-lib with patch [1] and are now\navailable to be used from that common place.\n\n[1] https://review.opendev.org/c/openstack/neutron-lib/+/973833\n\nRelated-bug: #2130570\n\nChange-Id: I1af50907461045d747ab112fddbdcf4220397f31\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c2cb922d335ab4d47de03b7e8f44f86ebf1e8f30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c2cb922d335ab4d47de03b7e8f44f86ebf1e8f30"}]},"branch":"refs/heads/master"},"5ded0ea7855cea3b13180634be8b5409e8dc9813":{"kind":"REWORK","_number":2,"created":"2026-03-09 15:36:47.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/60/978060/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/60/978060/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/978060/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/978060/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/978060/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/60/978060/2"}}},"commit":{"parents":[{"commit":"f9cb6dc869bdc6b822132e78cda7b92eed4550fd","subject":"Merge \"[OVN] New configuration variables to control the OVN BFD settings\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f9cb6dc869bdc6b822132e78cda7b92eed4550fd"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-02-26 10:47:14.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-03-09 15:36:19.000000000","tz":60},"subject":"Use ovn-db-sync-util related constants from neutron-lib","message":"Use ovn-db-sync-util related constants from neutron-lib\n\nThose constants were moved to neutron-lib with patch [1] and are now\navailable to be used from that common place.\n\n[1] https://review.opendev.org/c/openstack/neutron-lib/+/973833\n\nRelated-bug: #2130570\n\nChange-Id: I1af50907461045d747ab112fddbdcf4220397f31\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5ded0ea7855cea3b13180634be8b5409e8dc9813"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5ded0ea7855cea3b13180634be8b5409e8dc9813"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
