)]}'
{"id":"openstack%2Fneutron~978155","triplet_id":"openstack%2Fneutron~master~I240c5bdebb78419ab62473f8e555df012b21aa56","project":"openstack/neutron","branch":"master","topic":"bug/2111276","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-03-12 07:43:39.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"31291":{"account":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"last_update":"2026-03-12 10:03:11.000000000","reason":"Change was submitted"},"5756":{"account":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"last_update":"2026-03-11 17:08:13.000000000","reason":"\u003cGERRIT_ACCOUNT_5756\u003e replied on the change","reason_account":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}},"8655":{"account":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"last_update":"2026-03-12 10:03:11.000000000","reason":"Change was submitted"},"34271":{"account":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"last_update":"2026-03-11 15:22:40.000000000","reason":"\u003cGERRIT_ACCOUNT_34271\u003e replied on the change","reason_account":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}}},"hashtags":[],"change_id":"I240c5bdebb78419ab62473f8e555df012b21aa56","subject":"bgp: Connect provider switch when created","status":"MERGED","created":"2026-02-26 23:28:48.000000000","updated":"2026-03-12 10:04:33.000000000","submitted":"2026-03-12 10:03:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":32,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"978155-bug/2111276","meta_rev_id":"d969b6aa0914eb81f9db71ef27cfe0643cfe9747","_number":978155,"virtual_id_number":978155,"owner":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-12 10:03:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"value":0,"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-03-12 07:43:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-03-11 17:08:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"value":1,"date":"2026-03-11 15:22:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2026-03-12 07:43:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"value":0,"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"value":0,"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}],"CC":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-27 00:00:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-02-27 01:23:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-03 15:16:08.000000000","updated_by":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"reviewer":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"state":"REVIEWER"},{"updated":"2026-03-05 10:56:19.000000000","updated_by":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"reviewer":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"state":"CC"},{"updated":"2026-03-10 16:54:52.000000000","updated_by":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"reviewer":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"state":"REVIEWER"},{"updated":"2026-03-11 09:36:56.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2026-03-11 11:51:47.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2026-03-11 14:06:33.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"}],"messages":[{"id":"e0b2dd5ba6593f19961c448d68b11a99728ba8a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-02-26 23:28:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"783c2e91784b231407081e480d0a926eec5bc5fe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 00:00:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed6a2f088718490a8be6b31c0a834efa\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2ed526cfdda7450a83cee4b4f5c12ba8 : SUCCESS in 24m 18s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/be29369693684b1bba88228767bc4354 : RETRY_LIMIT in 3m 38s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/160a746e85804632b22105e427968983 : SUCCESS in 29m 37s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9f7d8334756e4029a79ffef8961544e0 : SUCCESS in 20m 00s","accounts_in_message":[],"_revision_number":1},{"id":"fec3463a7160c4291f4f0a79af543907f81e9c50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 01:23:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/82e87724cd9c42d5b513b67def16ae21\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/39c541cb27334565894eb37b4088ef2b : SUCCESS in 1h 01m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/af33e6d9cab64b90a849b093c15c234e : SUCCESS in 1h 02m 46s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/29883ab45488418995fdefa634db22e0 : SUCCESS in 1h 24m 38s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/65cf511e1b1042258d707bf8d38d2e8a : SUCCESS in 1h 51m 18s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/03c4e18c540544c9928e505ed824243a : SUCCESS in 33m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a709c6e21af14284b4d985df689327e0 : SUCCESS in 1h 03m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f981f716faa480bacddcaddc84e5a14 : SUCCESS in 26m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/db519b9722ee4ddb9f9c6126e678dea5 : SUCCESS in 37m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b15996ea2f304a899b5c531389a098b3 : SUCCESS in 22m 35s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ba183d7f42fa473cba973681c8e8630a : SUCCESS in 27m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e0df5e1974b4da8af93c790019dac26 : SUCCESS in 13m 40s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/bd375bc183b94d9187fc2ac7c119dff3 : SUCCESS in 59m 10s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/09c71e3d7bd74867b10f954b61d3dc38 : SUCCESS in 1h 13m 36s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/5373645d2e24427f9d45ca20716f1bd0 : FAILURE in 44m 50s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/9aebf68dfea64e788fe6a73605f045dc : SUCCESS in 45m 24s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/cbfe833c60e0449681d19b7b805185ad : SUCCESS in 34m 07s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ed9575e636324ecebee6621f9cd5cac1 : SUCCESS in 1h 12m 17s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4133527b34a9489f9261d812326716d8 : FAILURE in 50m 03s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/91b3d0fabd01451c9d721d854639af80 : SUCCESS in 1h 02m 02s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/86d5fdbc4af44e2e8bc514dbb4da5705 : SUCCESS in 34m 03s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/034ca7992af34dd69133ec3195dbf367 : SUCCESS in 1h 24m 57s","accounts_in_message":[],"_revision_number":1},{"id":"aa3325fbfd9fc13f49178956877296f27ed8ed11","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-02-27 15:18:59.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a952e44f27e3025607fe18c659db7ad12e3b4343","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-02-27 15:33:25.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"706509dce0c208f0baac6ccdf05059ed5e164e6c","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-02-27 15:36:22.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c58455b599b8a64c856b57a2bcae81d6a2c7d487","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 17:47:24.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b4e2eb82efb34ee680170d708e5c7048\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/59351d5fa2104008b437f8e6902cc0b9 : SUCCESS in 1h 35m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f0b0163d9a0f438ca626a271e25fba29 : SUCCESS in 40m 03s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/cf0a1d0a92884a1da3e5841fa151f8bc : SUCCESS in 1h 24m 21s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/5fd85c633cd94f62a60227ee55dcd5c1 : SUCCESS in 2h 12m 07s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ff720381c6e3416c9f630f9092acd18b : SUCCESS in 32m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7d011fd11d274bfb83342d142f76af17 : SUCCESS in 1h 13m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b76dddc5390e4e41a43b100eea65da7c : SUCCESS in 23m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b0940e6a64fa4033bfc69cfdf771976e : SUCCESS in 28m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/561a53a5dae34240b3cca06a0351efda : SUCCESS in 16m 41s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9d12b2584bb744d8857304d6e88a34cf : SUCCESS in 23m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0feb74d7bf824629a1c8151fe748f657 : SUCCESS in 12m 34s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/555849000ea14ee7b73fbb37374dc2c3 : SUCCESS in 1h 00m 46s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/8c04bdca35784cdbb98539bf72772c74 : SUCCESS in 46m 35s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/cd1f9947f59247b49ca6a5fa7fbf3c8c : FAILURE in 41m 54s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/8e1e8c886fa64c628f7903fc17dece46 : SUCCESS in 44m 24s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/fa7997f143684ebfbd52aaf74abb4f9d : SUCCESS in 1h 06m 38s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/34d6c872419547888044f71106b4ca39 : SUCCESS in 1h 55m 00s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7c1ea293b58142cab86d87032268f795 : SUCCESS in 1h 38m 53s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/8100851eb26b46ebaf6e48d04d3a42ae : SUCCESS in 1h 05m 18s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/bfd6652b473941b780ef3459d7b94552 : SUCCESS in 37m 15s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1fae1a13dba046a4a494eaefbe945a31 : SUCCESS in 1h 26m 57s","accounts_in_message":[],"_revision_number":2},{"id":"c08f1c656e5731709c4232b46edc5bfbf72da817","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 18:22:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76db37b500e44f08af6de2c8b7010471\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/101c40e99627432796fd41b6f7ca84a6 : SUCCESS in 45m 25s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1f26df9352d9490f9e7f1a52ef449728 : RETRY_LIMIT in 3m 43s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/bf13d87c641f4a7985064ba19567fa4b : SUCCESS in 23m 22s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a5732856286e406fbca34638f4180ce2 : SUCCESS in 38m 56s","accounts_in_message":[],"_revision_number":2},{"id":"cc38b1601569b690a32a6a5a57b0e3fa7b8bc407","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-03-05 10:56:19.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a9cc3b3ff4f6a389da5eaa298a4dd4990d625500","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2026-03-05 17:20:42.000000000","message":"Patch Set 2: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"97e2bf6c43906d76e7fd4ec15a90fff7f07ec68e","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-03-05 21:58:20.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"60707b26b9708c9d79bb4db9dbef21497868dbaf","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2026-03-06 14:51:43.000000000","message":"Patch Set 2: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"b4202fc44e1d649d28a354d02787e30b94315268","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-03-06 15:28:04.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a06f53341584fa3fa2109acd849b8ab4fe06ad97","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-03-06 15:28:42.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"c6283359791cb02df316ce29508791b691bf43b5","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-03-06 15:29:09.000000000","message":"Patch Set 3:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"fec994a9ce060cdf2bb796c0f11c8c89ed109a9a","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-03-06 15:29:49.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"430862e24da03dcba1b5926559c7a4bbe4e876f3","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2026-03-06 15:40:22.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e4e37e2b3a187ae1f3ea26378c92bbf1e86085d7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-06 16:43:11.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a4bb54c10d34e16a285eb0cff4cfad9\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/680e30b0dd124700adcf8ae5c0532dbd : SUCCESS in 24m 07s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/97e0e768c3394d06a8541f337ea7cc3a : SUCCESS in 51m 45s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/0bcac780f85046348b814f9aa19955c3 : SUCCESS in 1h 12m 13s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3cf046f263aed106892753a0af5b4643aecc7e12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-06 17:31:57.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ef57b87d0466439f8d9d626c4094dbbf\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c224415aebed4ae49d01af778ac85116 : SUCCESS in 1h 29m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/acee9451aea04c819758a5984bfa08ae : SUCCESS in 1h 12m 44s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/5924926e59124ba080cd882d79dcaf46 : SUCCESS in 1h 29m 55s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/4cc62dfbb62d4f5ba0ff9d0372778cc8 : SUCCESS in 1h 51m 36s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d9e8970daf9646b3a3249fb523d9c074 : SUCCESS in 30m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/11c27184142f4c51b94a058b3d5c153e : FAILURE in 11m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ffe98ef36d84d52b16d6f84bedcdc13 : SUCCESS in 14m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d754a5560e0e4e7285ed28ad63893c71 : SUCCESS in 38m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/585a5909f3e84f0d8d848bb1a2b1ebf7 : SUCCESS in 22m 57s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f98e360b688c46249adaf3a021b33bab : SUCCESS in 22m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/734c4d5474fb43a395757c5b4220a687 : SUCCESS in 5m 45s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/4b47a8acdb0c492e949d93cb17e9414d : SUCCESS in 53m 11s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/6357d2b37f994400a9878ee67a9e6f17 : SUCCESS in 1h 16m 10s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/2fa5bc350f434a78806c37bd354752c9 : FAILURE in 20m 56s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e25f1e7743484c5691d47d3c2c86c521 : SUCCESS in 1h 14m 34s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/4f957012c824477fae23833ed63a0372 : SUCCESS in 1h 12m 20s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/da7419a7869c46c2abc044fab492bc13 : SUCCESS in 1h 55m 19s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b8cd44333e51405e84cf51cad3e2c741 : SUCCESS in 55m 58s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/0d9e1678182446c09abb3a3c4f6836bf : SUCCESS in 1h 03m 44s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/da712f56d8584b6d9e42fb89ebbc3887 : SUCCESS in 1h 13m 37s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ab5c1624b220424da0a13166d5579df4 : SUCCESS in 1h 28m 00s","accounts_in_message":[],"_revision_number":3},{"id":"e3f149c1996217ac6d727d620493a2298b49df23","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-03-06 22:02:51.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ff0582dfb72622fad4ce13be1eaf14af0167e212","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-03-06 22:03:35.000000000","message":"Patch Set 3: Review-Priority+1","accounts_in_message":[],"_revision_number":3},{"id":"1cc85a58ad6298a99ec4d579fb97105345df0920","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-06 23:04:11.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe0e410b16a747229d074ae236b45e17\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/28b09fa4d50c439a9dac050fa4cf8e28 : SUCCESS in 58m 36s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/977b63f7539348beaa0a6962f442a8df : SUCCESS in 25m 48s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/47b60a46485046f1bbfbda319fa9f1b4 : SUCCESS in 26m 25s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0b0f9c76483d3005e8bb1f2f7d6d48051b1f5b55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-07 00:22:24.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/266df9ab9da446bb9eb78d7a48a11e35\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/84cb430db47a4ab2aa6e6ca079fed8d0 : SUCCESS in 1h 24m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b66ce8b9a252419db0b036df823f5838 : SUCCESS in 40m 21s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/80a0929228fc46d282bb8de6525997ac : SUCCESS in 1h 26m 24s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b1d0665b6336409bb1ffec3c298b912e : SUCCESS in 2h 18m 37s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e6fafd4007f040d2975443fbcdba373c : SUCCESS in 30m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/71e8c598a45e4bbf90b31f2d0c95a9f1 : SUCCESS in 1h 08m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27ec673591214c0c803cc0b54f6c1486 : SUCCESS in 29m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/59971b1697a34611a33d79242e71ee5a : SUCCESS in 32m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1dea076581ce424790f8977bbd8e2f67 : SUCCESS in 26m 56s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f22eb18287ca4cdda476cc719ce3d76c : SUCCESS in 32m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49afeeba251f4709902c27739e8fa852 : SUCCESS in 5m 52s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/21b797cc9081488cb680a6b0a49d5ccd : SUCCESS in 32m 26s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/0719ba99e0284dddb7c83cc2379d539b : SUCCESS in 50m 41s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/b3722ccc52d245c1b6f9143abfcca441 : FAILURE in 43m 57s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a0a57ce5d8db434f817117a34250ce31 : SUCCESS in 1h 04m 13s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/7bfc2e681a3143e490590640d42b7604 : SUCCESS in 36m 05s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/72f99c900d394239a2b616b104944daf : SUCCESS in 1h 11m 34s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/82263b643ec24a52820a7ca0781f7536 : SUCCESS in 1h 45m 02s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/2169f1fff4754f5396d44e90b722e7dd : SUCCESS in 55m 56s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/180396234a5848288a3f543c013f23c8 : SUCCESS in 41m 33s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/27cf0c88428a4b2d822531dfae271d38 : SUCCESS in 1h 27m 54s","accounts_in_message":[],"_revision_number":3},{"id":"fdcc615f1db86da2b5a3b63a4055e75598b414ab","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-03-10 16:54:52.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d4d83fd2cbdd3c6600dc44dfdf3738e035d34d5b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-03-10 17:30:51.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"25dfd972d5b4da8e96f47e8f04153012b83e2fd6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-10 18:28:18.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9baf4a8350c241f582440f7e434b94bf\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/689524b31dd34702a08cc25a8821ab2e : SUCCESS in 27m 25s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2f44447999ca4075bcb93b1b459885ff : SUCCESS in 20m 26s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/3bb155b05cb9442396413dd8013c6cc7 : SUCCESS in 47m 31s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"8c4032952aea1a85c45d77893e897a154bd040ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-10 19:20:34.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6e6a5a6245844e0e868494d9214d5b17\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/01c2b2a4030e41e9af34d3d04df18a65 : SUCCESS in 1h 31m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/297bf11bf982490b8bfc5497c6eb5ab1 : SUCCESS in 1h 04m 51s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/d4b6b5b24c2b48a6a3980c44efd42f4e : SUCCESS in 1h 19m 25s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/6252a60e155645cca3723a05373110cb : SUCCESS in 1h 46m 04s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1546932a043643f0825a71bb5c057186 : SUCCESS in 48m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/46d86e8009b1456c8d6dd0bf0a073f34 : SUCCESS in 39m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37019ac2009e4bafa021c011ab70244e : SUCCESS in 24m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/531053a8a2f744acb09f716a99d3696b : SUCCESS in 21m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5f0cd454e934491dae36afb1ed4f2b8c : SUCCESS in 25m 22s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d2fd41e67b94417fbb3a8e0f179b6ca3 : SUCCESS in 30m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/052644cbce984ee78c14e361678035f8 : SUCCESS in 11m 43s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/b1a588c60194493eab7e5e0b31fe3e5a : SUCCESS in 57m 35s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/73064cca5d5b4023a740f82e2aa06cab : SUCCESS in 1h 04m 45s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/e07a513b84b44e358457b51976aa880d : FAILURE in 35m 21s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/47487c2fbf9e41d6a114fd3fdce3633a : SUCCESS in 1h 14m 23s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/a58cc9783e8c46cca28294be7939b81f : SUCCESS in 1h 02m 34s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/54635f345f9a4667a4219d013155bd82 : SUCCESS in 1h 44m 38s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/152ee73b4d304141ad90540b0ad6afac : SUCCESS in 1h 41m 42s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/37ae780446414acbb0715c74196639cc : SUCCESS in 1h 11m 04s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/444ee51e2da242499782806bfb546410 : SUCCESS in 1h 35m 34s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/74d9c77206f0412281c97d83de1d79ac : SUCCESS in 1h 08m 13s","accounts_in_message":[],"_revision_number":4},{"id":"6a8c027854c7c592a755880df072d5e616e747d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-03-11 09:34:31.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"5da238d32d61b3f70d818496fe7fec734bfb058a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-11 10:06:36.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e87aac1fed2495f947d3bec4de51162\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8caba75120ee40e39709668af3417fab : SUCCESS in 25m 32s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e3307c1de7b44cb3937e71cdb0b84ffc : SUCCESS in 30m 10s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/c9972746d36d4b3eb8555a3ddfa0289d : SUCCESS in 25m 35s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"1223b6fb4d56358973663298ba9c9fcc0954cf6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-11 11:43:13.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec4623ebe48046ecbb7059bd0af62ed5\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/02c0a67573024447a6c6c211462ac1e9 : SUCCESS in 1h 27m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4bb2bce0b15a43789f552d24dd965514 : SUCCESS in 34m 03s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/9deddf4aefcf405883f954d5b4fc1b0f : SUCCESS in 1h 23m 42s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/588829545f314de09879bc83c83d9980 : SUCCESS in 2h 06m 20s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3bd315ca2b27456dbe2de43f1ab0064e : SUCCESS in 34m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a41fd2ef93bf4059b16181e75842892c : SUCCESS in 38m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9e567be870b4bf98abefd8b0da0e41c : SUCCESS in 24m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7664c54a87bc461581fc620627f7cce0 : SUCCESS in 34m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1dfc5a6a4ab0445c91e4e0d16d824fef : SUCCESS in 24m 58s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/dce5e64097914a90bc7732ab89a06cb9 : SUCCESS in 33m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42c50c420db44533960b145e9f660292 : SUCCESS in 12m 01s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/17205b5b11a34432a833e34af4e97c55 : SUCCESS in 59m 23s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/29edf65d640845f4b5b4213386d33ad9 : SUCCESS in 1h 18m 09s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/65cf9ca82d9b4e2faaa05730a4e6b9de : SUCCESS in 1h 20m 19s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/3fadb49e35bf4ec8af613182d8c32e6f : SUCCESS in 1h 11m 30s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/d99ca5c35efa4f3b90797a1e09f42210 : SUCCESS in 1h 07m 19s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/729612fc7d86447b88faea23c5adf932 : SUCCESS in 1h 44m 31s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/cbd732ddc91740a6b6f909b8911c82a6 : SUCCESS in 1h 02m 27s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d5315adfc2e949b9906a85d1ef35578b : SUCCESS in 54m 18s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/f78e12c0060d447cbbbcc98080a3f441 : SUCCESS in 1h 07m 38s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/70864f914676469a87aa260019e66e05 : SUCCESS in 1h 34m 15s","accounts_in_message":[],"_revision_number":5},{"id":"c22b0608cfa5bd91517ca7370883e368dc4c1ff5","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-03-11 11:51:47.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"13406dbc01097e9e67d9187e40819faf71551311","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2026-03-11 13:28:01.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"aa9ad4b7a2dc4073e43cac3ab3d6b8232ba8fe6a","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-03-11 13:58:38.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"212fbf4e722dd48b8e1abbaa8f29fae1be9c93d3","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2026-03-11 14:06:33.000000000","message":"Patch Set 5: Code-Review+1\n\n(7 comments)","accounts_in_message":[],"_revision_number":5},{"id":"24e4067c496c9238719fb7aabff7e33cf067dfac","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2026-03-11 14:18:08.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"20a7b97a59884ccd319c85c89e0fe04139650d80","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-03-11 14:51:16.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"66a00f7562c40bad6c5b8157ae471007e171c3c3","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-03-11 14:52:43.000000000","message":"Patch Set 6:\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"2bedcd69c0ebdf9531c3eb926714d2c25d53a93e","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2026-03-11 15:22:40.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"2131bf3053831dd76e78a7cb31d0cdd2f7563f9a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-11 15:23:27.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2de4ab5ff4914366aeb8ecc9c29c00dc\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/39f40186fd7248b79a5f1657dc242c47 : SUCCESS in 29m 46s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/76bc8cda15d7439a9564661e2bf0c990 : SUCCESS in 27m 36s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/c3c87de2e3da4610b6cb83086c6c16d6 : SUCCESS in 29m 53s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"05b2a3102ec5986231ed6239869dcecf0e6dce0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-11 16:59:53.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76237264f3a14e038760bcb46c28c26e\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/644f1a6cd2fa4442b305d72c27f979eb : SUCCESS in 1h 38m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8cb1cc97f72d4e2fa175538856267efb : SUCCESS in 1h 03m 50s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f7b7dd977cd44183adbbe22664ef7e4c : SUCCESS in 1h 16m 23s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/462d4602d57a4f53ad672bef01f8d31e : SUCCESS in 1h 46m 03s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/326d288dfc8043338692de52cba4b3e9 : SUCCESS in 33m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bc319e2546664e1bb51d6dd8b6d862a3 : SUCCESS in 37m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/435de717a0b449a99e3ebf81dc0aa508 : SUCCESS in 25m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/508611333f764fab9c5d1b2c87d95cdf : SUCCESS in 27m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/aaee0c2300b34e4dba70a054c20e7cad : SUCCESS in 22m 41s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/764a4feff65d4bf7bd87bf0c4df5ba02 : SUCCESS in 34m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f43c0139a8bb45148168ea67b663b7ad : SUCCESS in 9m 46s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/2b9afbf32e2541278f3ab6db9930b9f9 : SUCCESS in 36m 25s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/55a387fbe6aa437bb3da68bd4b8db816 : SUCCESS in 1h 07m 27s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d3cc8e0c4eb9441c891407c7434f5a30 : SUCCESS in 1h 02m 02s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/196c91249eba482abaaad16b9f40d66d : SUCCESS in 47m 15s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/c334dfeb4e514f68823c1ad3ef25f20b : SUCCESS in 1h 14m 36s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/4a944a2cea314f56990db377ed24c14f : SUCCESS in 1h 59m 41s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1bbaf6caefd5484aa1d1d3503dd4ba28 : SUCCESS in 1h 51m 25s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/53d31f58f0d74c3fba08808268e2f7b9 : SUCCESS in 57m 33s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/9224c4f1421840aeb98075f4c585f6f4 : SUCCESS in 31m 34s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e48ef82995b346c6a30479a0bcb9dc6f : SUCCESS in 1h 15m 07s","accounts_in_message":[],"_revision_number":6},{"id":"9d0e3f4f96b38a9b39b8984f66804a3da0089329","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2026-03-11 17:08:13.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"57d3dfd5d3f49ada15094bc00341a7cb7e59d528","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-03-12 07:43:39.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"919e3142c278e562f9d5ff4176a93cdfeef2a457","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 07:44:04.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"519ac6568bff26de17f3e30b84172f30e589b064","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 10:03:11.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2481992f31ae4c4eb3a5aa73d5684981\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/bee7dd9dc244402caa14056d5db86109 : SUCCESS in 1h 05m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8d19ae7e2db04587bba14e8d230b66ce : SUCCESS in 44m 09s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/dcf37fed44ac49fa88f75ab4da275533 : SUCCESS in 1h 25m 00s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/7ff9da9f3dee48deaf7c4b0f576e4e1a : SUCCESS in 2h 06m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c4967fd2a3a4b9eb52c6ce4cde5ff0f : SUCCESS in 11m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5d48cfd4298e41f3b533d1033af59989 : SUCCESS in 34m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fbd9e1ed57d64975912b7c15ec560209 : SUCCESS in 19m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b97e16a9e7714374a1c08ffa1f00ab39 : SUCCESS in 6m 38s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d20130db4ff4442d826766a6298747ed : SUCCESS in 1h 17m 45s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/4b0899e4d50c42088fbc8b8ec90d41b0 : SUCCESS in 2h 12m 30s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/b5e420b38aa14427bf8ba221d583c369 : SUCCESS in 56m 04s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/eb7c189c1ffb4a7bb24a6a16ea088604 : SUCCESS in 42m 32s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/713a72ff9cf04dcd8c43ad42ec82ba86 : SUCCESS in 50m 17s","accounts_in_message":[],"_revision_number":6},{"id":"294286d14fdfb548af846b7aef1da7d90c0a051f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 10:03:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"d969b6aa0914eb81f9db71ef27cfe0643cfe9747","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 10:04:33.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f316193cbeda4e19ba998ecff8af4bdc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e37188b922e4faaa0adb367d6c59074 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"51e7110c7d1072082a9e7bd9b509dc2388401763","revisions":{"4683dcec5375cb702dd0c584adede86bf5478aae":{"kind":"REWORK","_number":1,"created":"2026-02-26 23:28:48.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/55/978155/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/55/978155/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/978155/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/978155/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/978155/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/55/978155/1"}}},"commit":{"parents":[{"commit":"ccd1e29c3d13f7d13a127c5b1cd61a489e96d98a","subject":"bgp: Connect BGP and Neutron worlds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ccd1e29c3d13f7d13a127c5b1cd61a489e96d98a"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-02-26 16:26:00.000000000","tz":0},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-02-26 23:28:33.000000000","tz":0},"subject":"bgp: Connect provider switch when created","message":"bgp: Connect provider switch when created\n\nImplement events that watch for provider network\nbeing created. If such a corresponding logical switch appears in OVN\nthen the event gets triggered and wires BGP and Neutron worlds together.\nAdditionally a DHCP options tables is monitored for any new GW IPs. This\nworks with IPv4 only.\n\nRelated-bug: #2111276\n\nChange-Id: I240c5bdebb78419ab62473f8e555df012b21aa56\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4683dcec5375cb702dd0c584adede86bf5478aae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4683dcec5375cb702dd0c584adede86bf5478aae"}]},"branch":"refs/heads/master"},"56c0c46213c1893f3629ecac80e9cf85fe7b3721":{"kind":"REWORK","_number":2,"created":"2026-02-27 15:33:25.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/55/978155/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/55/978155/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/978155/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/978155/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/978155/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/55/978155/2"}}},"commit":{"parents":[{"commit":"ccd1e29c3d13f7d13a127c5b1cd61a489e96d98a","subject":"bgp: Connect BGP and Neutron worlds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ccd1e29c3d13f7d13a127c5b1cd61a489e96d98a"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-02-26 16:26:00.000000000","tz":0},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-02-27 15:33:05.000000000","tz":0},"subject":"bgp: Connect provider switch when created","message":"bgp: Connect provider switch when created\n\nImplement events that watch for provider network\nbeing created. If such a corresponding logical switch appears in OVN\nthen the event gets triggered and wires BGP and Neutron worlds together.\nAdditionally a DHCP options tables is monitored for any new GW IPs. This\nworks with IPv4 only.\n\nRelated-bug: #2111276\n\nChange-Id: I240c5bdebb78419ab62473f8e555df012b21aa56\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/56c0c46213c1893f3629ecac80e9cf85fe7b3721"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/56c0c46213c1893f3629ecac80e9cf85fe7b3721"}]},"branch":"refs/heads/master"},"5a16bced04acdc4d78f8305e0338b50c79305495":{"kind":"REWORK","_number":3,"created":"2026-03-06 15:28:42.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/55/978155/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/55/978155/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/978155/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/978155/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/978155/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/55/978155/3"}}},"commit":{"parents":[{"commit":"0a1fb53097dca5811b95c426b5cb1755ad8278fe","subject":"bgp: Connect BGP and Neutron worlds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0a1fb53097dca5811b95c426b5cb1755ad8278fe"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-02-26 16:26:00.000000000","tz":0},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-03-06 15:25:24.000000000","tz":0},"subject":"bgp: Connect provider switch when created","message":"bgp: Connect provider switch when created\n\nImplement events that watch for provider network\nbeing created. If such a corresponding logical switch appears in OVN\nthen the event gets triggered and wires BGP and Neutron worlds together.\nAdditionally a DHCP options tables is monitored for any new GW IPs. This\nworks with IPv4 only.\n\nRelated-bug: #2111276\n\nChange-Id: I240c5bdebb78419ab62473f8e555df012b21aa56\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5a16bced04acdc4d78f8305e0338b50c79305495"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5a16bced04acdc4d78f8305e0338b50c79305495"}]},"branch":"refs/heads/master"},"3984da16d926272501ae0c7eabd44a9acc22a1ff":{"kind":"REWORK","_number":4,"created":"2026-03-10 17:30:51.000000000","uploader":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"ref":"refs/changes/55/978155/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/55/978155/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/978155/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/978155/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/978155/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/55/978155/4"}}},"commit":{"parents":[{"commit":"11de3dadce99ebb380967c084883f1adb38919c8","subject":"Merge \"[Docs] Remove FWaaS scenario document\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/11de3dadce99ebb380967c084883f1adb38919c8"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-02-26 16:26:00.000000000","tz":0},"committer":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2026-03-10 17:25:15.000000000","tz":60},"subject":"bgp: Connect provider switch when created","message":"bgp: Connect provider switch when created\n\nImplement events that watch for provider network\nbeing created. If such a corresponding logical switch appears in OVN\nthen the event gets triggered and wires BGP and Neutron worlds together.\nAdditionally a DHCP options tables is monitored for any new GW IPs. This\nworks with IPv4 only.\n\nRelated-bug: #2111276\n\nChange-Id: I240c5bdebb78419ab62473f8e555df012b21aa56\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3984da16d926272501ae0c7eabd44a9acc22a1ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3984da16d926272501ae0c7eabd44a9acc22a1ff"}]},"branch":"refs/heads/master"},"10d0eeaf706b3f61bed9e0cc8ebca126a8f6deed":{"kind":"REWORK","_number":5,"created":"2026-03-11 09:34:31.000000000","uploader":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"ref":"refs/changes/55/978155/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/55/978155/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/978155/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/978155/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/978155/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/55/978155/5"}}},"commit":{"parents":[{"commit":"11de3dadce99ebb380967c084883f1adb38919c8","subject":"Merge \"[Docs] Remove FWaaS scenario document\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/11de3dadce99ebb380967c084883f1adb38919c8"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-02-26 16:26:00.000000000","tz":0},"committer":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2026-03-11 09:32:32.000000000","tz":60},"subject":"bgp: Connect provider switch when created","message":"bgp: Connect provider switch when created\n\nImplement events that watch for provider network\nbeing created. If such a corresponding logical switch appears in OVN\nthen the event gets triggered and wires BGP and Neutron worlds together.\nAdditionally a DHCP options tables is monitored for any new GW IPs. This\nworks with IPv4 only.\n\nRelated-bug: #2111276\n\nChange-Id: I240c5bdebb78419ab62473f8e555df012b21aa56\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/10d0eeaf706b3f61bed9e0cc8ebca126a8f6deed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/10d0eeaf706b3f61bed9e0cc8ebca126a8f6deed"}]},"branch":"refs/heads/master"},"51e7110c7d1072082a9e7bd9b509dc2388401763":{"kind":"REWORK","_number":6,"created":"2026-03-11 14:51:16.000000000","uploader":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"ref":"refs/changes/55/978155/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/55/978155/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/978155/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/978155/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/978155/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/55/978155/6"}}},"commit":{"parents":[{"commit":"11de3dadce99ebb380967c084883f1adb38919c8","subject":"Merge \"[Docs] Remove FWaaS scenario document\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/11de3dadce99ebb380967c084883f1adb38919c8"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-02-26 16:26:00.000000000","tz":0},"committer":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2026-03-11 14:40:20.000000000","tz":60},"subject":"bgp: Connect provider switch when created","message":"bgp: Connect provider switch when created\n\nImplement events that watch for provider network\nbeing created. If such a corresponding logical switch appears in OVN\nthen the event gets triggered and wires BGP and Neutron worlds together.\nAdditionally a DHCP options tables is monitored for any new GW IPs. This\nworks with IPv4 only.\n\nRelated-bug: #2111276\n\nChange-Id: I240c5bdebb78419ab62473f8e555df012b21aa56\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/51e7110c7d1072082a9e7bd9b509dc2388401763"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/51e7110c7d1072082a9e7bd9b509dc2388401763"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
