)]}'
{"id":"openstack%2Fneutron~978164","triplet_id":"openstack%2Fneutron~master~I73e54f1ea47ba1319a8fd84905ebac6646b98378","project":"openstack/neutron","branch":"master","topic":"bug/1990229","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-03-03 11:08:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I73e54f1ea47ba1319a8fd84905ebac6646b98378","subject":"Since OVN 20.06, config is stored in \"Chassis.other_config\" (2)","status":"MERGED","created":"2026-02-27 07:23:14.000000000","updated":"2026-03-03 11:13:01.000000000","submitted":"2026-03-03 11:08:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"978164-bug/1990229","meta_rev_id":"9420459ff518abaf3dad1c92fa3a9ce460783a36","_number":978164,"virtual_id_number":978164,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-03 11:08:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":2,"date":"2026-03-02 11:31:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2026-03-03 08:50:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2026-03-02 12:17:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2026-03-03 08:50:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-27 08:04:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-02-27 09:55:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-28 01:37:46.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2026-03-02 10:52:58.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2026-03-02 10:52:58.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"},{"updated":"2026-03-02 11:31:16.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2026-03-02 12:17:07.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2026-03-03 08:50:06.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"c945c49be907dc06db76c89a1e959cf16332d276","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-02-27 07:23:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ab66a0386ca8697e2398e898c893d3242cdc3886","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 08:04:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad1609d7f85c402f98a9def9c0512125\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7c745a146f814f6eb04dfdd6c4b8e7d8 : SUCCESS in 27m 18s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/059b5f5e3ef14612b5ab0041dd008373 : RETRY_LIMIT in 4m 04s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/0ba4e58b10aa40d7b1f4461c5f16081c : SUCCESS in 26m 23s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6679779a4090458fa5a921a63c9df769 : SUCCESS in 39m 23s","accounts_in_message":[],"_revision_number":1},{"id":"4b32ec53955273186fb4a0084d00e7f7dd8d81e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 09:55:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c828886611244727a6fc5b75a545138c\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/df07dea6982c414b967e83394d926095 : SUCCESS in 1h 28m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9b31ffc70ca74978bf0bf281650d9e95 : SUCCESS in 1h 07m 49s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/4191b703ed5d473a929593c26326b908 : SUCCESS in 2h 31m 30s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/26d8226892a84e5fbe70dfca2965d542 : SUCCESS in 32m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/14ab4109197a4eea916e8537cc2c3fe5 : SUCCESS in 1h 06m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5975beb77844c2094b38f4a7612330d : FAILURE in 7m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e15b2af4449541619a48ea7ea5b8aae9 : SUCCESS in 28m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/be2aa035255c4996ad06ebd2f9197091 : SUCCESS in 17m 25s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/75fabee64f0e453e97cf25e59b20fce4 : SUCCESS in 33m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/636e01fe8ee54f1888b06fa583af9928 : SUCCESS in 13m 51s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/e30d041272114627a6f15c4eb3224512 : SUCCESS in 51m 14s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/1ceecf6d53c244238555ddd8a9eb9d1a : SUCCESS in 41m 59s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/b9e340ca8cab4912a6bdc465e04ca5aa : SUCCESS in 51m 52s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/f7547d207a2b4f72ba2c1fc2b01160c8 : SUCCESS in 1h 07m 38s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/56ebc295f9f1442eac9df59cedee6fae : SUCCESS in 44m 42s","accounts_in_message":[],"_revision_number":1},{"id":"e9336e9b5af2e079f2bb7486a97de06c2aa2cddd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-02-27 13:58:54.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"31b7d97ca69236f0297c4bdecd049a92f57265e2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 15:30:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9da33ded888a45f7840e727e0c932604\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/10243a9b10a84293a5e805a2f12cd82b : SUCCESS in 25m 41s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a4dd2e80a5924e6cb398f3bf98025986 : RETRY_LIMIT in 6m 44s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/e70714f4fc934f7fb126c71f0cb85aa8 : SUCCESS in 35m 18s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/708d7fd1075341e1aa7ad4957fd61e66 : SUCCESS in 21m 08s","accounts_in_message":[],"_revision_number":2},{"id":"b35f3628538f4cacc652343d18bae50ab877c96c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 17:54:46.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8f9b2727086641449f79c84ba8f41279\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/06710f6cd7df487d9edcc8850b3bc531 : SUCCESS in 1h 26m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/61ee01356b044cf1b6d8b52286b64850 : SUCCESS in 54m 07s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e73cad952cbd40cda67b9cfb25ad788e : SUCCESS in 2h 02m 49s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/12b5b022668849ec837272d3fd5b6e21 : SUCCESS in 34m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/08467225c9e24a3e893a4f8032724d6a : SUCCESS in 1h 16m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ca022d43e234042b24f8dde74196ca8 : TIMED_OUT in 30m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0eb9df054cdf452b9c22c80c2f9ea2ee : SUCCESS in 36m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9c533e06a5034a4aab6e0049b4918646 : SUCCESS in 16m 17s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/20a81e8946904eb2927375ebf6cdc9c1 : SUCCESS in 20m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee133c943ac74e87b9e0196887e5b12b : SUCCESS in 11m 42s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/43f14b95d169447f9577edb88bc9e98e : SUCCESS in 1h 11m 09s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/0fa4a8657e5b417fa46584f9c2393c28 : FAILURE in 35m 47s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/d3b45d3f6d6d412387a3e502c1d129a2 : SUCCESS in 1h 15m 38s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/38908daad0d444abbf998f9c4a9fc569 : SUCCESS in 1h 06m 52s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8210b639c302479fac34e3d4391bc077 : SUCCESS in 43m 21s","accounts_in_message":[],"_revision_number":2},{"id":"876f3483dd2a6055093285d9edad711294cd77b5","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-03-02 10:52:47.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"02422670043dcaba5fe0679950037a39f28d6cfd","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2026-03-02 11:31:16.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"744334428085abd880ce768c1bbcb58881f9b9e2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-02 11:52:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/607dcecb2872421cbfcd05b6dadfb84b\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d23e8c733fe94da6b533b143d4a7d1f2 : SUCCESS in 26m 58s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2378bcac1bd44dcd8c6d549dbac8d8fd : RETRY_LIMIT in 4m 27s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/0c685cc7607d45d6877d33c47a6c4da6 : SUCCESS in 35m 47s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f71dc17bd8844f1fb6264f357e29682e : SUCCESS in 43m 11s","accounts_in_message":[],"_revision_number":2},{"id":"c52d913ba3ca76f3f3e611f8cee0d26c393c4021","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-03-02 12:17:07.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"90327178f3fe24cff875f2e0dff85bcee897f985","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-03-02 12:17:42.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ca4e1a2eddb771129700f769b799228a3238c36d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-02 13:27:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c7b384efc54419abf4106a41a2bf9ff\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/00bb02306ff24e5684cf809b2ba55dc6 : SUCCESS in 1h 14m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d4f92a33c831412fafe981505f454e3f : SUCCESS in 1h 01m 55s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/88f093cbc63446d1a4ea69cf1d04afaf : SUCCESS in 2h 02m 02s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/928303fc8c214781845eac3efb27e73d : SUCCESS in 35m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e0db1c70dac247c2ab5032262d77b3fc : SUCCESS in 1h 21m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d284c00e7dfe4fc59863c28ddd73be86 : SUCCESS in 27m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7e62e4010d3941d2b926f535ac1c2cb5 : SUCCESS in 31m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ed2aebcf7f734dca901662a696ce1744 : SUCCESS in 20m 29s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a7df19a53f494f77a857a132ec51b4f4 : SUCCESS in 30m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ea5624d0d5b4ee8a4ef34669e47b490 : SUCCESS in 13m 00s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/6d0b9df6f8f94abeb2cead9ad241bc59 : SUCCESS in 1h 25m 05s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/57ae5d84a257446f9c6c793bb140f5c6 : SUCCESS in 1h 20m 31s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/e086ce8b629a44e4aacb86831ba46bb2 : SUCCESS in 40m 41s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d736b90f352446deaa90de539a7d67e0 : SUCCESS in 1h 12m 34s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/655e5d69abf84379a4eb2cf937190559 : SUCCESS in 46m 49s","accounts_in_message":[],"_revision_number":2},{"id":"1639127cbbae5463bfe24a9ea7424c8236e82db5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-02 19:46:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a2ea20d4a4a4ccbbccd64b276d62c17\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/62452627a5c340c6ba0fe29a47b0da42 : SUCCESS in 35m 31s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8c77313cb0624cffbe2f3dc94bc2649f : RETRY_LIMIT in 3m 34s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/2719f0d9763041d6a323fe32197141cb : SUCCESS in 53m 38s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c8aaa3b4467d4b559b137039226f8878 : SUCCESS in 37m 07s","accounts_in_message":[],"_revision_number":2},{"id":"66b864da67099b0d41549fce560125101ad079d6","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-03-03 08:50:06.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"984ca8390876708e20f4d9ac7675c76e8bbadf26","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 08:53:00.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d1021cdb90e89fbdeaf6c200689ba313dea51651","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 11:08:29.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d01e423ea074f3196d6ad9fb70acd8c\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/b7e553e56d454c26b38cc3176944a41b : SUCCESS in 1h 22m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8b152847666645a8b14772bd87d29f55 : SUCCESS in 1h 01m 17s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/4feb599fbe394c91b8611c3b78dc9d0b : SUCCESS in 2h 13m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7076c75daf58483895f2813382b85de2 : SUCCESS in 10m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a77d3278464a41d0b428958337849c96 : SUCCESS in 36m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c85e0a1a7cd14df0b8a717f761f559f2 : SUCCESS in 15m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74813396e0bf4caaa15bb5461051df71 : SUCCESS in 11m 51s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/239e51e7e17d426d8895900715f9ea9f : SUCCESS in 42m 16s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/e495521f843f40ba951b117d9026b6c3 : SUCCESS in 1h 11m 00s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/7d330a38aabb4d05904e0e01ba68bc08 : SUCCESS in 1h 07m 56s","accounts_in_message":[],"_revision_number":2},{"id":"b525f8eb5d54648d2f487544d2da1e35d01eed08","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 11:08:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"9420459ff518abaf3dad1c92fa3a9ce460783a36","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 11:13:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/beeafb1cad034b64a367e671493d9b33\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecbf0e6eec484bfcb34ab247e0da6292 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5c30152bfd9f119160da898d96b2a8ed9c625d01","revisions":{"88a96b6f2df3153ac7196300a021e26f4dfc779c":{"kind":"REWORK","_number":1,"created":"2026-02-27 07:23:14.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/64/978164/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/64/978164/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/978164/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/978164/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/978164/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/64/978164/1"}}},"commit":{"parents":[{"commit":"d09c1adfad922c5e5974037b6ecb78a41bba0513","subject":"Use ``Idl.cond_change`` to set a table condition","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d09c1adfad922c5e5974037b6ecb78a41bba0513"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-02-27 02:35:39.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-02-27 02:36:04.000000000","tz":60},"subject":"Since OVN 20.06, config is stored in \"Chassis.other_config\" (2)","message":"Since OVN 20.06, config is stored in \"Chassis.other_config\" (2)\n\nThis patch removes any reference to \"external_ids\" and auxiliary\nmethods. Since Dalmatinan (2024.2) [2], the minimum OVN version\nrequired is v24.03.\n\nRelated-Bug: #1990229\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I73e54f1ea47ba1319a8fd84905ebac6646b98378\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/88a96b6f2df3153ac7196300a021e26f4dfc779c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/88a96b6f2df3153ac7196300a021e26f4dfc779c"}]},"branch":"refs/heads/master"},"5c30152bfd9f119160da898d96b2a8ed9c625d01":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-02-27 13:58:54.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/64/978164/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/64/978164/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/978164/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/978164/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/978164/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/64/978164/2"}}},"commit":{"parents":[{"commit":"075eb2f22b506de7bf03dff50f66870283c63dad","subject":"Use ``Idl.cond_change`` to set a table condition","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/075eb2f22b506de7bf03dff50f66870283c63dad"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-02-27 02:35:39.000000000","tz":60},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2026-02-27 13:58:54.000000000","tz":0},"subject":"Since OVN 20.06, config is stored in \"Chassis.other_config\" (2)","message":"Since OVN 20.06, config is stored in \"Chassis.other_config\" (2)\n\nThis patch removes any reference to \"external_ids\" and auxiliary\nmethods. Since Dalmatinan (2024.2) [2], the minimum OVN version\nrequired is v24.03.\n\nRelated-Bug: #1990229\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I73e54f1ea47ba1319a8fd84905ebac6646b98378\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5c30152bfd9f119160da898d96b2a8ed9c625d01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5c30152bfd9f119160da898d96b2a8ed9c625d01"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
