)]}'
{"id":"openstack%2Fneutron~978463","triplet_id":"openstack%2Fneutron~master~I5f7ddf00a9ef4dfa375855972bbc975693bebe01","project":"openstack/neutron","branch":"master","topic":"bug/2142034","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-04-29 12:16:17.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5f7ddf00a9ef4dfa375855972bbc975693bebe01","subject":"Enable \"filter_validation\" extension by default","status":"MERGED","created":"2026-03-02 14:10:19.000000000","updated":"2026-04-29 12:17:33.000000000","submitted":"2026-04-29 12:16:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"978463-bug/2142034","meta_rev_id":"eeb12e948b9e8c040984de506efecf913a648707","_number":978463,"virtual_id_number":978463,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-29 12:16:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2026-04-29 09:54:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2026-04-29 03:16:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-04-29 09:54:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2026-04-29 06:57:16.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-02 18:46:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-03 02:50:26.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2026-04-07 08:24:02.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"},{"updated":"2026-04-29 03:16:30.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2026-04-29 06:57:24.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2026-04-29 09:54:30.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"c00ae6ba68f3a2ddb5ad62d24c12a468aae854e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-03-02 14:10:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2b64c0a58b96f5c45bfe74495f8fdc24f25e2127","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-03-02 15:39:10.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"dbafc1a422691f9424f6a6f5bdda2cd57d5f5d28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-02 18:46:19.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ef2e505df51047f087e9b74a2c6777cf\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/eef6c6ffcb2e48c5aab84fc7131ff883 : SUCCESS in 1h 37m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/96fb9c4de2314d91ba772f207fa0efe4 : SUCCESS in 33m 26s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c6bec987daa84bb48e9c5011e5cf1c0a : FAILURE in 1h 19m 24s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b5f4da908bd447f9bf7320409f84dbd6 : FAILURE in 2h 03m 22s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/18e680911ab54a0faad9bd5256e843ed : SUCCESS in 34m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a86c806502534c2d9dd359fe01622f47 : TIMED_OUT in 1h 21m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9a05a2061184358868a6ccf50ab5285 : SUCCESS in 26m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/189301abdeb34820812d60f564076994 : SUCCESS in 34m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5e5568b927dd42b78e7c2e699427a410 : SUCCESS in 15m 49s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9acf6d7acdb54570b6fdccdf4ccf5912 : SUCCESS in 27m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30ce26219f9b4a47bc03a5d57e735fb9 : SUCCESS in 13m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e9ebff64373e43418cd76be494fa6fb9 : SUCCESS in 22m 47s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/3e7c2235127048f9ac81fbdb1684fc41 : SUCCESS in 1h 07m 33s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/3467dfb0971b4f0e9186f69a541bdea3 : SUCCESS in 1h 14m 42s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d8fad5938d1245838581859cc31b22d4 : FAILURE in 43m 29s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/431d1d5b9f474eab89b03d8ac4ec5350 : SUCCESS in 56m 55s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/07992cef20ce4c96a368969f1a3abad8 : SUCCESS in 1h 14m 53s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/cfe67253e6234f9e816634501d5ab443 : SUCCESS in 1h 52m 53s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ece0fa1b7fca410f9b37538c8c41ce70 : SUCCESS in 1h 07m 56s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/43cd3774a3fc4d10bfd54ff4f217d312 : SUCCESS in 1h 09m 11s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/4059fe67195046c0bc19b1d1008636bd : SUCCESS in 39m 57s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/bda07d45cb8144d6bb3d5359d5486ad5 : SUCCESS in 1h 41m 28s","accounts_in_message":[],"_revision_number":2},{"id":"0276345bb18bcba09d86c410ddc2be3661b7ac2f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-02 21:17:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0cd9946ad6cd4b53a6a898fa2c740d0e\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6d0aa3b5be6b41cf9429381ab26dc348 : SUCCESS in 25m 49s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5726e253e2624453967e50eaa60a8b5f : RETRY_LIMIT in 3m 25s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/13f7c4cc72dd4a959795d60aefd9d45d : SUCCESS in 29m 00s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/dcd11f7f4e3f45ce92384a58f8de5c99 : SUCCESS in 41m 19s","accounts_in_message":[],"_revision_number":2},{"id":"5b7f0a100d7abc3b42fa204de467a8ef7f18f40d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-03-04 12:43:15.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"cd56e592ae4a794ae795f4aef04cee89442f528d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-04 14:10:13.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a78bd3e66d143ad9309fd541bb82fbf\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8c4ed9dae0524f12a189def88833ae48 : SUCCESS in 46m 12s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ebd38b8e83cd4c9e99411c1574e7debf : RETRY_LIMIT in 3m 51s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/b5d4d640460447b78eca98cc92b2ec5a : SUCCESS in 32m 33s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ff8c34df3d2c124a33a290bd3a8f598de8c7aecf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-04 14:41:30.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c3f238b1cfd246b38e25ccd500e0fb2a\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/7e00cc559680406abdbe01e4f8eec9ea : SUCCESS in 1h 02m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bedeb45bf3d9473f9249a9f919a64fc1 : SUCCESS in 1h 00m 41s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/6ba4324b3d0c48f8826f5547e30ecab8 : FAILURE in 1h 48m 37s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/1e5bd3009308407294bf56aa55ac4e83 : FAILURE in 1h 43m 24s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0948babda099401f9e2f65f4a05aed39 : SUCCESS in 33m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/735e661a430a4773b77c340b9c301ac3 : SUCCESS in 38m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db8e0cfeac3a49cdba829ddeb95d69c7 : SUCCESS in 10m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9adccbbb0add4b44a2b68796cf89661e : SUCCESS in 24m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/54278596b6a04d32a4fdb691eac91684 : SUCCESS in 38m 10s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6692e390ce4444ff9048a21115f9d9fc : SUCCESS in 49m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5a9c007f703466e838c9b3b2456c714 : SUCCESS in 9m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5fa49cecd91c41d588aa9fcbab9944b2 : SUCCESS in 18m 57s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/af9f238b0c884c3d8c0f8693bbf9b619 : SUCCESS in 1h 00m 53s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/4fcc3e13fcc4456a83d7adf9236044d0 : SUCCESS in 52m 27s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/7fa47c2898504bee8c84334c68b47e54 : SUCCESS in 48m 42s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/543b015ea16641dc91a2e928f028cf2a : SUCCESS in 45m 33s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/208194e157874ec4a98adbafbdf1d6b5 : SUCCESS in 44m 02s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/1d7088ded9654895ae46af55bfe8a503 : SUCCESS in 1h 52m 19s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/35aca26b74594dbab8cb293237c9e40a : SUCCESS in 1h 14m 09s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/fa7823c8952d4f20aa3d5d71e4e64608 : SUCCESS in 55m 19s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/5c7031ebc6b94295b51c2e2bd8de3806 : SUCCESS in 1h 02m 54s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8669fb7ce8504f9b8374a1669abed3ee : SUCCESS in 1h 28m 53s","accounts_in_message":[],"_revision_number":3},{"id":"f8535775957cc34ec5caf5c18981188c2086897d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-04-07 08:24:02.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":3},{"id":"aabe8dc1fab64188e7e69766724b59fba624e9b8","tag":"autogenerated:gerrit:restore","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-04-07 09:39:55.000000000","message":"Restored","accounts_in_message":[],"_revision_number":3},{"id":"8f325d6f6307b6b637d8a68c6482e0b352e37f9b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-07 10:11:00.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c7fdc4d2a6c4b31b20794de2ebff304\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/179da25a2cb34508adcbf9ba5f67e5d7 : SUCCESS in 24m 02s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5f5d1faf8bb34cc5abc54953a2c65788 : SUCCESS in 22m 34s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/5d95a3d0deda4bf3a5593d48c7b48388 : SUCCESS in 29m 25s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"366166da311a19d0eafaf59fd3b945eea566ce1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-07 12:04:48.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1620a9831bca4851a045add517844fcc\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/2c799a8f6fd94e53971b5df4ea541c0a : SUCCESS in 1h 24m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e77f9e5f04fc4afaabccd332637a4e38 : SUCCESS in 1h 00m 34s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/7907c966de6143329405daaca55e1b1a : FAILURE in 1h 23m 20s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/124283424d384e0c97889ed598fa3b06 : FAILURE in 2h 22m 06s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/51e911b5389440d198c0cd289858cbb8 : SUCCESS in 34m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/edb0c8aaed184d97a3a375824c6e4bd0 : SUCCESS in 44m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf05c8303d4b4056bc5e8cfcd086023f : SUCCESS in 20m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e62cfb5b03c746988a9f549cb4a899be : SUCCESS in 25m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ad4600ab012f40adb7333f39a7b3bac3 : SUCCESS in 24m 01s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b43e7d31e72b4072a8ee3520e2736b7e : SUCCESS in 23m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e85871a70ce4ba2b75397c50348546b : SUCCESS in 11m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/df90c99a2c4a4cc69217f83be6fab53d : SUCCESS in 20m 15s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/41ec2f75f4cc4bdea42246a1f397539b : SUCCESS in 1h 34m 15s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/bc0624db55ca4b3194af118bfb927538 : SUCCESS in 1h 08m 25s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/e3525e7f46e34a788c5d1bfc08d99aba : SUCCESS in 1h 18m 20s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/4083e87936974bcbb524315c497b2a15 : SUCCESS in 1h 51m 27s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ad9be288d4ba47a7b03f5a9ad961e67e : SUCCESS in 1h 43m 20s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/2b13a04126ef4f63a30107831a2eeb6a : SUCCESS in 53m 13s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/c203757b90ea46d482f8fefa72462170 : SUCCESS in 1h 19m 27s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a551808dd7f543338ecd01b94fe83bb7 : SUCCESS in 1h 18m 02s","accounts_in_message":[],"_revision_number":3},{"id":"8333a659ac33a66330a4e6622a1ae2abea896cd4","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-04-07 17:05:16.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"32df09ed9e1a5042eb764b1f641a55b339078459","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-07 18:30:53.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/17df3283733f4151b219735c9a96534d\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/20628841e46646819446c19bd00a339a : SUCCESS in 35m 43s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ce893a7839394a59a3deffebd4f7dd83 : SUCCESS in 19m 04s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/4c3e3625c94e4a448f29809759e4e8ab : SUCCESS in 43m 50s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0c3cdcd09621dec691f6c42e285486ecffde2296","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-07 19:33:08.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4bd196c7ab03465789218d6964a8f28a\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/523e930006204b86b3a7a3769a3c3996 : SUCCESS in 1h 15m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a2bcbf6049934f51a009766aae17970a : SUCCESS in 48m 08s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/15735c8c6e4d41a78d05ce10c3e0fce2 : FAILURE in 1h 20m 20s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f76a26eae9b74a68af09fe5ef171534f : FAILURE in 2h 22m 52s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2ea49bc554dd4df5ba0ecd3bfd6fcbfe : SUCCESS in 33m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5f9e896378524aadb5895333b3c5a75a : SUCCESS in 1h 07m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd561617851f48fd8ddc8788418187d0 : SUCCESS in 18m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2adcb19d574444cb99ccc3666dfd62fa : SUCCESS in 24m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8183568f93ce4cf49e377bcb6ec77480 : SUCCESS in 17m 47s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f839c3f6138d41418bb745b95cb77c7a : SUCCESS in 33m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9622bc22bab0443d96ec30f294170dae : SUCCESS in 12m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/629463efae0049e7bb616f4a0f1e4d46 : SUCCESS in 21m 25s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/df0251b3b68541b9a9031c8a443e0e04 : SUCCESS in 1h 25m 50s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e8043e2b847d4318ae8db45817019ff6 : SUCCESS in 56m 47s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/e610bb589cf74ce4ac1943dca76129ae : SUCCESS in 54m 22s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/edfc9debfc0a4e5eb8b84ff2084d70aa : SUCCESS in 1h 40m 22s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/db934462cd404a0099b0f97a75a428b9 : FAILURE in 2h 01m 50s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/c9fad737210841b182749500be3ab675 : SUCCESS in 1h 13m 09s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/fe86bf07b02a4d5cb956289c22c7d6a4 : SUCCESS in 1h 09m 53s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7a99da38567b47ee9b51020fd19e4bd5 : SUCCESS in 1h 09m 54s","accounts_in_message":[],"_revision_number":3},{"id":"b89d70e35df9c070f03e67a68a311774023dc768","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-04-28 10:06:52.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"ea877f33a147659a3802974b624a797e31522160","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-28 12:02:29.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c6d5af732f34b61b23a484d2c5e427b\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8ed732b743054fc8a505b48d6460c5d2 : TIMED_OUT in 1h 21m 10s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/61b83c615f8f4cc78439fd424d0d2f4a : SUCCESS in 19m 17s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/da25f121edab44b280a6232a618d177d : TIMED_OUT in 1h 21m 25s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f2b817f9170b6fe7fff55541f0e853f5f4e65fe4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-28 14:10:13.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c741e724f51c40a991190e82d90223c0\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/fbb9374adc0d4153ae158a46cca57f45 : SUCCESS in 1h 27m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d641841e1b474176b840b0dd5370be1a : SUCCESS in 35m 43s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f01f82ac2af8426ba46eae5dc72f7c28 : SUCCESS in 1h 27m 08s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/cc2b8c59e029496583b569d1b2abc904 : SUCCESS in 1h 46m 33s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5cce471758e043009604411f56dd1cc9 : SUCCESS in 34m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f5b0f8080eff4e7aa4a9de72f1b22a7f : SUCCESS in 39m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/358c6905181b43698c69a8a87c704f40 : SUCCESS in 24m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5723565e93354270b8b445c627455a53 : SUCCESS in 20m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ce21a1e8abf245ce9c84be25f3eddb51 : SUCCESS in 18m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/31da772def0540f9b3bf811bb8f8dbd5 : SUCCESS in 24m 30s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6c6bb96615e946b2b5a4fea36c2d562e : SUCCESS in 27m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ace86d6082e4c7dab828de1292ecd97 : SUCCESS in 11m 21s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a6883c2397014aa686475410660ae6e9 : SUCCESS in 2m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f1c2ec43dc334463ac1b40492d27ac6d : SUCCESS in 13m 11s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/6b118d80d77b4a2fb4428f6266d99ea6 : SUCCESS in 48m 51s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d41a36fda13746cbb6a27b15e72f7fae : SUCCESS in 39m 42s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/3f40cffb80464e02b4867ab3febb648a : SUCCESS in 1h 22m 44s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/8aed115c0cfc47eea48ea13977ab7986 : SUCCESS in 1h 07m 38s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/13e5203dc4c544f2bbb08076fb92a5f9 : SUCCESS in 1h 41m 17s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/fdaf4c862f264b5aafc580bdda389658 : SUCCESS in 1h 12m 07s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/c416b31e9ff84749885052893d6e74a1 : SUCCESS in 32m 48s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/99e194467cc24a838df49ad289d601b7 : SUCCESS in 1h 22m 51s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/8ffdb92622654bd58592ecb4a4638931 : SUCCESS in 3h 04m 57s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e3fdacd0bcd19b3e7b21b150a3a4b5facc8fb838","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-04-29 03:16:30.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"46b2b859f998e3d881eed3dd26884c9c56d1e055","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-04-29 06:57:16.000000000","message":"Patch Set 4: Review-Priority+1","accounts_in_message":[],"_revision_number":4},{"id":"aaef135b0479cd3efc601a072db45f5efffee00a","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-04-29 09:54:30.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"da20b115d2fe754c3e43c44dc1b7d5062a7ba423","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-29 09:56:08.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"5092461bee37140e104e7bd7943b969958f16548","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-29 12:16:17.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c28661854094ba6ba6f6780576f6438\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f54dfa90819a45899d63cd440b086c81 : SUCCESS in 1h 16m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d0aaf011f4f04e9bbf0c65cd7ecaff2e : SUCCESS in 33m 14s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/71085d97195248ccaaf8de7f6330b937 : SUCCESS in 1h 32m 34s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/31ed780bf1274672b89aa3d95bfbb891 : SUCCESS in 1h 57m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1bc0525d955b4a0abdf5ef0d1139f619 : SUCCESS in 5m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/24efd7839c2e44be91fea10496affe0d : SUCCESS in 42m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c07def093a3941aebd8ded0059ba0580 : SUCCESS in 23m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bd6096ed3d4b4e93a25590c0abe7845d : SUCCESS in 17m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c53e3a6864f7460ab9c4319c52e80fb1 : SUCCESS in 12m 56s\n- requirements-check https://zuul.opendev.org/t/openstack/build/9d94c9c6855f4c218007865e0eddb30c : SUCCESS in 2m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b7f58a8d52434182afce002c708113af : SUCCESS in 13m 07s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/8109af12c4da4a04a37ac7bc23647035 : SUCCESS in 1h 24m 05s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e25e7610573541fc9e8c6fae5a5e32ac : SUCCESS in 1h 54m 58s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/ff6eff4cdb7647cbad54c32285d859e8 : SUCCESS in 1h 14m 50s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/120a15e4016641caa1129a81d49379d2 : SUCCESS in 50m 33s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/01c1cbcbc6324bf39e5994f2c3e79005 : SUCCESS in 1h 05m 01s","accounts_in_message":[],"_revision_number":4},{"id":"a72621bafff5c16c9636d9b2a42c262d02afade6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-29 12:16:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"eeb12e948b9e8c040984de506efecf913a648707","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-29 12:17:33.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2087345a40e64be098187b9ee1327b4b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ff607a7553d45f890177daaf7eb07fd : SUCCESS in 1m 02s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1f10f47799ba46fc8a5a80232dc141c7 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"70eb118ac683e92473b1a3f1067db642eecf64df","revisions":{"a11166ec035f9ae82cff7e96bc9346267fec98c4":{"kind":"REWORK","_number":1,"created":"2026-03-02 14:10:19.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/63/978463/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/63/978463/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/63/978463/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/63/978463/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/63/978463/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/63/978463/1"}}},"commit":{"parents":[{"commit":"2f0f58f5a140fc87cbc5600ab753f29ce05d0999","subject":"Merge \"Add ``NetworkSegment.count_segments`` method\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2f0f58f5a140fc87cbc5600ab753f29ce05d0999"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-03-02 13:59:16.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-03-02 14:10:03.000000000","tz":60},"subject":"WIP \u003d\u003d Enable \"filter_validation\" extension by default","message":"WIP \u003d\u003d Enable \"filter_validation\" extension by default\n\nAll service plugins will have \"filter_validation\" extension enabled\nby default now.\n\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I5f7ddf00a9ef4dfa375855972bbc975693bebe01\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a11166ec035f9ae82cff7e96bc9346267fec98c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a11166ec035f9ae82cff7e96bc9346267fec98c4"}]},"branch":"refs/heads/master"},"afcbc25044af5889ea736b0fad885ca6dc20e508":{"kind":"REWORK","_number":2,"created":"2026-03-02 15:39:10.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/63/978463/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/63/978463/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/63/978463/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/63/978463/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/63/978463/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/63/978463/2"}}},"commit":{"parents":[{"commit":"2f0f58f5a140fc87cbc5600ab753f29ce05d0999","subject":"Merge \"Add ``NetworkSegment.count_segments`` method\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2f0f58f5a140fc87cbc5600ab753f29ce05d0999"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-03-02 13:59:16.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-03-02 15:38:57.000000000","tz":60},"subject":"Enable \"filter_validation\" extension by default","message":"Enable \"filter_validation\" extension by default\n\nAll service plugins have \"filter_validation\" extension enabled\nby default now. The service plugins inherit from\n``neutron_lib.services.base.ServicePluginBase`` that has now the\nattribute ``__filter_validation_support`` enabled by default.\n\nThis change also includes the ML2 plugins, inheriting from\n``neutron.db.db_base_plugin_v2.NeutronDbPluginV2``.\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-lib/+/978554\n\nCloses-Bug: #2142034\n\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I5f7ddf00a9ef4dfa375855972bbc975693bebe01\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/afcbc25044af5889ea736b0fad885ca6dc20e508"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/afcbc25044af5889ea736b0fad885ca6dc20e508"}]},"branch":"refs/heads/master"},"b3b2b99792b840bd07e6c915c0168065cb13831a":{"kind":"REWORK","_number":3,"created":"2026-03-04 12:43:15.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/63/978463/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/63/978463/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/63/978463/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/63/978463/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/63/978463/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/63/978463/3"}}},"commit":{"parents":[{"commit":"cf27b336bb0b15f63c1bd8d84158d6d6ccc9dff1","subject":"Merge \"Add tox-py314 overrides to job templates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cf27b336bb0b15f63c1bd8d84158d6d6ccc9dff1"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-03-02 13:59:16.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-03-04 12:42:39.000000000","tz":60},"subject":"Enable \"filter_validation\" extension by default","message":"Enable \"filter_validation\" extension by default\n\nAll service plugins have \"filter_validation\" extension enabled\nby default now. The service plugins inherit from\n``neutron_lib.services.base.ServicePluginBase`` that has now the\nproperty ``filter_validation_support`` enabled by default.\n\nThis change also includes the ML2 plugins, inheriting from\n``neutron.db.db_base_plugin_v2.NeutronDbPluginV2``.\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-lib/+/978554\n\nCloses-Bug: #2142034\n\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I5f7ddf00a9ef4dfa375855972bbc975693bebe01\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b3b2b99792b840bd07e6c915c0168065cb13831a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b3b2b99792b840bd07e6c915c0168065cb13831a"}]},"branch":"refs/heads/master"},"70eb118ac683e92473b1a3f1067db642eecf64df":{"kind":"REWORK","_number":4,"created":"2026-04-28 10:06:52.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/63/978463/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/63/978463/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/63/978463/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/63/978463/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/63/978463/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/63/978463/4"}}},"commit":{"parents":[{"commit":"3e9e57acc4161a9684d7af4fc75853f35dedb742","subject":"Merge \"Use HA_Chassis_Group in the OVN L3 scheduler\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3e9e57acc4161a9684d7af4fc75853f35dedb742"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-03-02 13:59:16.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-04-28 09:33:08.000000000","tz":120},"subject":"Enable \"filter_validation\" extension by default","message":"Enable \"filter_validation\" extension by default\n\nAll service plugins have \"filter_validation\" extension enabled\nby default now. The service plugins inherit from\n``neutron_lib.services.base.ServicePluginBase`` that has now the\nproperty ``filter_validation_support`` enabled by default.\n\nThis change also includes the ML2 plugins, inheriting from\n``neutron.db.db_base_plugin_v2.NeutronDbPluginV2``.\n\nBump neutron-lib to 3.25.0 that contains [1].\n\n[1]https://review.opendev.org/c/openstack/neutron-lib/+/978554\n\nCloses-Bug: #2142034\n\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I5f7ddf00a9ef4dfa375855972bbc975693bebe01\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/70eb118ac683e92473b1a3f1067db642eecf64df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/70eb118ac683e92473b1a3f1067db642eecf64df"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
