)]}'
{"id":"openstack%2Fneutron~978962","triplet_id":"openstack%2Fneutron~master~I2bd56f6d17b9cf3e4a07d4b9ea26474818cd5af8","project":"openstack/neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2026-04-23 14:16:10.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2bd56f6d17b9cf3e4a07d4b9ea26474818cd5af8","subject":"Migrate coverage configuration to pyproject","status":"MERGED","created":"2026-03-04 20:36:36.000000000","updated":"2026-04-23 14:17:43.000000000","submitted":"2026-04-23 14:16:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"978962","meta_rev_id":"3318be59ee80f4d1a4f91c19246511e4bc430a15","_number":978962,"virtual_id_number":978962,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-23 14:16:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-03-31 14:59:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2026-04-23 11:44:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-04-23 11:44:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2026-03-13 02:54:32.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-04 21:37:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-03-04 23:28:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-31 14:59:45.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2026-03-31 14:59:58.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"},{"updated":"2026-03-31 14:59:58.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2026-04-23 11:44:01.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"}],"messages":[{"id":"51163eeb374d5810eece0cf4ed9736e6bc278325","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-03-04 20:36:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0dd9661f464076d4bfba5e37848a9d5bd194d202","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-04 21:37:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f33bfbe33f484ddbbb4e4b719f774829\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/19f94e7d34334d4f9de8316a724779aa : SUCCESS in 58m 42s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/96951b6b7b33472b8f16a927acaea59d : RETRY_LIMIT in 2m 53s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/f926ff6e8b214b619b8277cea02c9adf : SUCCESS in 26m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"49e6d054bee314b18da9220164ae25dff88ab335","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-04 23:28:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/680ed1dea59f471b89b71956247c9c3e\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/23afa3d935a44620bdf4ce8e6ff3167f : SUCCESS in 2m 56s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d543f2f462ae47cbaa853e0ec0315f84 : SUCCESS in 1h 05m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/30a82e4043ba472abaa65fe885f56587 : SUCCESS in 32m 46s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/5b8ca846b69d49c18e09484cdc75eb5d : TIMED_OUT in 2h 49m 41s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/00c0e371a49441bd9b48c2e59f737b97 : SUCCESS in 2h 02m 46s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8e7f1505e8bf49a6a2178a388bb4af6a : SUCCESS in 33m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6150111e72f8442cb8b56064d3ef6d5f : SUCCESS in 49m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f320fd7ae0674304a90bfa612c58e925 : SUCCESS in 10m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/07afba38034a4270adae1107c53015b9 : SUCCESS in 20m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/990d366141ac4e4a83b934365e4a2386 : SUCCESS in 22m 34s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/fcab954c28414ac4914c9f695ecbfe66 : SUCCESS in 21m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5cef64fae094f9c902434d9882a8fc0 : SUCCESS in 10m 55s\n- requirements-check https://zuul.opendev.org/t/openstack/build/4bc9e61534f544f8b79fe5d5d7577f5f : SUCCESS in 2m 42s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/27b379a8216d47428b0c149c42d1caa3 : SUCCESS in 46m 58s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/1b8b7982666c46738c4b74c3aca1b88d : SUCCESS in 53m 47s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/cc57ab6bfb2d4689a053714cb7a637d1 : SUCCESS in 1h 14m 55s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/4da3e86121e443b292cf8fde26534f20 : SUCCESS in 43m 20s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/f2d25042b00845d98d0935d30f0624a4 : SUCCESS in 1h 13m 31s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/8c36fdba8c7f4faf80e2825be68de804 : SUCCESS in 1h 40m 38s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/721e2cec6b1a44efb193bc3b911313a9 : SUCCESS in 1h 44m 11s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/aa76e42cd54a43efa30d6c16606bc377 : SUCCESS in 58m 58s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/df1c7b7be8ac4ce8ae113cd39df89eb9 : SUCCESS in 45m 10s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3350b013f6384053859f68456957cb74 : SUCCESS in 1h 04m 43s","accounts_in_message":[],"_revision_number":1},{"id":"a93caf8a01f66d6026ae9760e813dbea8520e746","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-03-04 23:45:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ab61673fa931d1d359d4d1b5ce73cbf8eb04da00","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-05 00:24:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63313f59d3984fbf9b938f677ad6be19\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0d6ae37a2c8d41da8eb37534944c6790 : SUCCESS in 33m 33s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/077a665bce8c4701b4d9735ad5213214 : RETRY_LIMIT in 3m 14s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/ec4bfd7d455c468a9282ee287e2379e7 : SUCCESS in 36m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c7c2e35ca5bb49071c06e858ce13df1e6ce88599","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-05 02:34:14.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/07104c0886e343bab44e0ecdd7d8dcc4\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6aba5b81f02846249d707d26a07c6212 : SUCCESS in 2m 35s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/10f86b5059eb4e52be68e574e57816f8 : SUCCESS in 1h 41m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/470f0d1a0c354395af0573565fdab2c8 : SUCCESS in 35m 46s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/e025aa16edc34a03bcb789692a68f450 : FAILURE in 2h 46m 53s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b423ebb48d1a40b099ee3ab0263750ef : SUCCESS in 2h 33m 54s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3c1809d5c56d44cc8e421301e5d7d7fe : SUCCESS in 33m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/076c7f82ad9c4172872ba8445018d414 : SUCCESS in 35m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a75f65e46a7e4c63ba7c3b6800c42578 : SUCCESS in 16m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c6c9bc638e7248259bf74278aeac75c3 : SUCCESS in 20m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e3143fde764b48c6b8ec6b640f7dfbbf : SUCCESS in 25m 51s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/111aa3e85cd64d8387c257889ea278f2 : SUCCESS in 24m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0b65db3c30345e0ba8b33854aae5f2a : SUCCESS in 6m 18s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a19f9089c51f4f5ebeb436f4d6516718 : SUCCESS in 1m 43s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/f4548b7faeac478592cd9f5db1c38bcb : SUCCESS in 56m 51s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/40a7e0deabb545c5ac894c6e8963ed4d : SUCCESS in 42m 49s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a523a5ebab51498f89ad519dac5aaedd : SUCCESS in 1h 03m 08s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b08cd455997f445ead3f1048e48ad654 : SUCCESS in 1h 05m 44s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/a3919fc4ab6e420ea6fd05d07f7e2f33 : SUCCESS in 36m 34s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/0ac261659dfa4ac7b04b5b153ba74d8e : SUCCESS in 1h 43m 44s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/08983185c9954c25b9cf0da068873dbc : SUCCESS in 1h 20m 56s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/770087a7c1cc4b489d6eafc3438445e2 : SUCCESS in 46m 59s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/b6f6fbbbd68f4293b5532f1093c3cd2d : SUCCESS in 31m 19s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/fbb9873f1bdd4d9ca06a200aaaceefe8 : SUCCESS in 1h 37m 19s","accounts_in_message":[],"_revision_number":1},{"id":"8ef0e3b181b51cc7bd1c1b05fe8d699c753cc2bc","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-03-05 23:08:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c25b7c4d5a526bfb806de6cded9d404886bf135b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-05 23:54:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61562209182a421a83bf1152f044b08e\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/beb72511c58643c9baeb30d46c1b63ba : FAILURE in 8m 02s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/185c24d80ca846c5a870aef845cf6131 : RETRY_LIMIT in 3m 00s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/a9d84986026f49a09f42494a57a5d1e7 : FAILURE in 12m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fa0dcd48f32d9d0c39acf6950d22816592084ee1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-06 01:42:11.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3c1bd38207674fb3973e5be83f5ebd6a\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/04804b5fca4c44c1b17cac70d2d40b1b : SUCCESS in 2m 57s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/46b8ae798cc7447bb591e923ea9d4fd9 : SUCCESS in 1h 26m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/076c186081374427b9db04c529217ad4 : SUCCESS in 1h 08m 27s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/84665df08fdc426fbc90cd0c0691006a : SUCCESS in 1h 35m 03s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b2134b1361de40b8a93874c530bd1fa2 : SUCCESS in 2h 00m 43s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ca0906e438454616b6c4ca136bca4671 : SUCCESS in 30m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5f5af2303f9c4371b508c7f62e0666bc : FAILURE in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4762b0105234c8596e1accdf8e48652 : FAILURE in 5m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c6bc2a3780fd4afb879db96685fa32d9 : FAILURE in 3m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/53d270cec01643c7bed3d130d8ae3b7c : FAILURE in 5m 29s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f59746b3c5e543bc9ce623ae1bc41a2d : FAILURE in 5m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27abe44b435940bba9c885a053a7955a : FAILURE in 6m 39s\n- requirements-check https://zuul.opendev.org/t/openstack/build/07445600534c4742b9c4fe9fa87644a7 : SUCCESS in 4m 21s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/c04e9ad6f31d499f9bf48006c61b6257 : SUCCESS in 33m 22s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/e354168f56774f62948d2a4691a4bbd1 : SUCCESS in 1h 09m 43s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/7bea0b5b80354cd0b17942d1861620b3 : FAILURE in 16m 13s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/96965885481c4571b63bb47d6cf9e5e1 : SUCCESS in 41m 34s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/3c1b658270764cb4be578fcd965910b8 : SUCCESS in 1h 12m 31s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/89a973b0b66246e98a9c6be5570d2c33 : SUCCESS in 1h 54m 27s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d066fecedf844d2e8cd781b121d7db19 : SUCCESS in 1h 49m 43s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/58de60c110624d73b240f125a3b48a87 : SUCCESS in 46m 26s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/865b6e88ee684b96899655c25b0aa2f0 : SUCCESS in 55m 08s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/066f44e578ff4ca6a22bb6091663ab1a : SUCCESS in 1h 05m 13s","accounts_in_message":[],"_revision_number":1},{"id":"c7c09ebb278923a4360db2b59e697bbf00e7dc78","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-03-07 22:30:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b23747d3a7340aed5f0a5548b62869f4ccb3265e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-07 23:02:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/46ead33dc69947749f5da2d93163941d\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/08fbc2276ed2408f88abcdf72d389193 : SUCCESS in 26m 29s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/299b5c012f96425e99b48ff047ebcec0 : SUCCESS in 25m 10s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/174f1ae20b1a49eabeb9a05f3f82a823 : SUCCESS in 30m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"79aadc8b74b1c0a9fa30f05cc6d8b30f7e7f66d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-08 00:44:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb20af1353374b0486631dfdb860b7d8\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/dc3e4d3210fb4822ac65aac7495c61a9 : SUCCESS in 2m 35s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/20f3faeec3ab478595151182cb5dc86e : SUCCESS in 1h 28m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e4456cfc63844d75aab3b57a61717e7d : SUCCESS in 59m 18s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/71bb72b8d8354779855ce804c01fdeb0 : SUCCESS in 1h 34m 26s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/bf326f0920824c05a557ddbb2d6aa730 : SUCCESS in 2h 12m 37s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/47cda544151a42eba329e8a95f0b2980 : SUCCESS in 31m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/92618e61d79047b28dad6ecb7ccdb8e9 : SUCCESS in 1h 01m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b834cd3473424c10b24217f9b58eb566 : SUCCESS in 24m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b83afd4a6e89474c9f18075f50a83b19 : SUCCESS in 38m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/63946f6cb2ef423994326e1fef4366e1 : SUCCESS in 19m 44s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/aa824f41f6294319b44ed702515a7868 : SUCCESS in 32m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ce98349bccf442580d668bc9de40243 : SUCCESS in 11m 41s\n- requirements-check https://zuul.opendev.org/t/openstack/build/7273f3f168ef4a0bb5d331a5ca1837a9 : SUCCESS in 3m 41s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/f7276448fef24516b4a68e14006bd5c0 : SUCCESS in 1h 07m 39s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/c86bf563f9a84b7eb1e9a8a2d7a72e6c : SUCCESS in 1h 13m 13s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/01b6fd2e516f4dfd913a8cb09c0a3f15 : SUCCESS in 1h 01m 54s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/bf12c43a4389494dba4e2f826e78c871 : SUCCESS in 1h 01m 55s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/d2875e862fad48a98f1e52c195829af9 : SUCCESS in 49m 16s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/783262a6c36b44b884e3c1d8b4a2ba64 : SUCCESS in 1h 44m 51s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/839993e39a2c4c56bf8a9d8407e06b58 : SUCCESS in 1h 34m 24s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/e5354c129b084d808d6bf48a7489e6b9 : SUCCESS in 50m 00s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/f0a8697918aa4715b79494b93b7b9941 : SUCCESS in 1h 49m 43s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a58c200dcd31447cabd3bfabc43a672a : SUCCESS in 1h 30m 28s","accounts_in_message":[],"_revision_number":1},{"id":"46f4035aeeed05256e33ed90745665ef069a6182","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-03-13 02:54:32.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"7fec615de2029f539f6f465fbbcd630a09771110","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-03-31 14:59:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d2e818a6367de679224f5ec19aae852d84759d98","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2026-04-23 11:44:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a0433ef2c4dbb52649da99721d276567e6062222","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-23 11:44:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a8d91b375a5568aa5084d6df71c05d9cfac01726","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-23 14:16:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b799cf3aa34348c4acdd4d04825d4a93\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/dc6abce4388c4ca79b4ed2797ff2d75b : SUCCESS in 3m 14s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/737dfb9cb85a496ca69897d365aaa4ef : SUCCESS in 1h 13m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e9de953411854818ae963e77c873948c : SUCCESS in 47m 11s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/289987bbf0f64fa9a2a190675a4a92fc : SUCCESS in 1h 21m 27s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/9fc7a31b24624ec3b68a32d630154eab : SUCCESS in 2h 29m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad7a5879bbd34b62a71156c4a6ba66ee : SUCCESS in 23m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6be6f999329f4b9092efc8bb05844091 : SUCCESS in 36m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bc089d80c7be45d583d2994506a90a35 : SUCCESS in 20m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/772d6fffbfd4409ba3d3e198ad8dacb2 : SUCCESS in 23m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2688f866d4e4767871df8415a9d9dcb : SUCCESS in 11m 55s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f31e1adeb32447a48ab2d39cbb9eb5c3 : SUCCESS in 2m 45s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/35945eb4b88c4dd29646c66600a03728 : SUCCESS in 1h 07m 11s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/7e9acafee89641599fe057a27cb532da : SUCCESS in 1h 51m 02s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/a2eb370e16294d959103770280593e7c : SUCCESS in 1h 11m 58s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/8d387bc4d4f24e47b810016345bb6ec8 : SUCCESS in 58m 15s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/8eb49d0b18194c2282972b7bcc7a5e11 : SUCCESS in 48m 59s","accounts_in_message":[],"_revision_number":1},{"id":"de3c511a6d83ae4fa450d3dd1fba70319accaf0b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-23 14:16:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3318be59ee80f4d1a4f91c19246511e4bc430a15","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-23 14:17:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50f7f95b7443449e99ff3b9f75e21181\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99d0d53e317a4c1daf6941da703633a3 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ce3206c116571deda7a56d4ceca1b96ee8d1f931","revisions":{"ce3206c116571deda7a56d4ceca1b96ee8d1f931":{"kind":"REWORK","_number":1,"created":"2026-03-04 20:36:36.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/62/978962/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/62/978962/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/978962/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/978962/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/978962/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/62/978962/1"}}},"commit":{"parents":[{"commit":"cf27b336bb0b15f63c1bd8d84158d6d6ccc9dff1","subject":"Merge \"Add tox-py314 overrides to job templates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cf27b336bb0b15f63c1bd8d84158d6d6ccc9dff1"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-03-04 20:35:18.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-03-04 20:35:18.000000000","tz":-300},"subject":"Migrate coverage configuration to pyproject","message":"Migrate coverage configuration to pyproject\n\nCleaned-up files we should skip as some are no longer in\nthe tree after eventlet removal.\n\nChange-Id: I2bd56f6d17b9cf3e4a07d4b9ea26474818cd5af8\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ce3206c116571deda7a56d4ceca1b96ee8d1f931"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ce3206c116571deda7a56d4ceca1b96ee8d1f931"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
