)]}'
{"id":"openstack%2Fneutron~979005","triplet_id":"openstack%2Fneutron~master~Ib8fe38d5f22ef96bf92b61e34798d832ae8e7885","project":"openstack/neutron","branch":"master","hashtags":[],"change_id":"Ib8fe38d5f22ef96bf92b61e34798d832ae8e7885","subject":"[WIP] Dummy patch to Revert \"OVN ``NAT.gateway_port`` is supported since v22.06.0\" to unblock NetApp cinder CI","status":"ABANDONED","created":"2026-03-05 14:10:14.000000000","updated":"2026-06-24 05:25:06.000000000","total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"09efe8ff7da42382f7bc85edd5c16f51ecfcce8b","_number":979005,"virtual_id_number":979005,"owner":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-03-05 16:23:29.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2026-03-05 15:07:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-03-05 16:23:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"652c91467bd1d05fda561974917beb765d33c6da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"date":"2026-03-05 14:10:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a075fb1846c1247c9627a59b2e6a8f27edebfeb3","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-03-05 14:17:25.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"edba510f240a618cacf835246c2ee1e6b9a6b96d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-05 15:07:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/203482a680544059a5b30721e27a5477\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d7182626b5994218884750621c93ac01 : SUCCESS in 26m 56s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/bf6efbc5fffe486599acaabdfb94df99 : RETRY_LIMIT in 3m 11s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/24fd45b98a954569bf0a346707b7df8a : SUCCESS in 44m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"364de9b591bdeed59d5955885c1ce71edb58f13e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-05 16:23:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dabe4d98790d4be4831d87d1f1c95a4d\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d710ae8f4d8b493598d13331da268013 : SUCCESS in 1h 27m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e713378285ba4d3fb8d716ca60e440e6 : SUCCESS in 1h 02m 16s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/3f002b1a32314f42b44268561cdc1832 : SUCCESS in 2h 03m 48s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b646fe79305c49949340815c048b1986 : SUCCESS in 31m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1c81ec8f1a6940e8b11cd26420de8116 : SUCCESS in 39m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2051341ea02d4ca38bdf19da80fe2354 : SUCCESS in 24m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d47cc27cf5a940afa50d628eb9cd7ae6 : SUCCESS in 36m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/789ca4bb526a4bceb603115aa784153c : SUCCESS in 25m 47s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2594da5243754bd68d98499fdef2149d : SUCCESS in 21m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c941182716944066979c15f6a34d1d2a : SUCCESS in 8m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9bac90046c7b4322ad0a41133ebcbf97 : SUCCESS in 18m 59s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/c1d410b77e7e4adbaa84bd4e78891d8f : SUCCESS in 48m 06s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/1638f32436724df68ee40923922c9199 : SUCCESS in 1h 19m 51s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/6a6fc675a2b8414cb97a3353592520e4 : SUCCESS in 1h 20m 34s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/dcc4378387a543a79fef13a93339472f : SUCCESS in 1h 09m 55s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/905719541c224cadba3e38a663064054 : SUCCESS in 2h 06m 21s","accounts_in_message":[],"_revision_number":1},{"id":"09efe8ff7da42382f7bc85edd5c16f51ecfcce8b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"date":"2026-06-24 05:25:06.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"65d733411c6d63caa0cfa75085d1b45562985240","revisions":{"65d733411c6d63caa0cfa75085d1b45562985240":{"kind":"REWORK","_number":1,"created":"2026-03-05 14:10:14.000000000","uploader":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"ref":"refs/changes/05/979005/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/05/979005/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/05/979005/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/05/979005/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/05/979005/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/05/979005/1"}}},"commit":{"parents":[{"commit":"fd67650fad2430ca94685b7750e2508341c2dbba","subject":"Merge \"func: Remove ovsdb connection restarts\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fd67650fad2430ca94685b7750e2508341c2dbba"}]}],"author":{"name":"manideepn","email":"manideep.openstack@gmail.com","date":"2026-03-05 14:09:03.000000000","tz":-300},"committer":{"name":"manideepn","email":"manideep.openstack@gmail.com","date":"2026-03-05 14:09:13.000000000","tz":-300},"subject":"[WIP] Dummy patch to Revert \"OVN ``NAT.gateway_port`` is supported since v22.06.0\" to unblock NetApp cinder CI","message":"[WIP] Dummy patch to Revert \"OVN ``NAT.gateway_port`` is supported since v22.06.0\" to unblock NetApp cinder CI\n\nThis reverts commit 055e871db371388d5bf3674fcb32625df4f71eb1.\n\nChange-Id: Ib8fe38d5f22ef96bf92b61e34798d832ae8e7885\nSigned-off-by: manideepn \u003cmanideep.openstack@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/65d733411c6d63caa0cfa75085d1b45562985240"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/65d733411c6d63caa0cfa75085d1b45562985240"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
